)]}'
{"id":"openstack%2Fkayobe~773899","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I9c6b0d9d5e13e8b82ebb8e8a3a0f432efb865e28","project":"openstack/kayobe","branch":"stable/ussuri","topic":"story/2007297-stable/ussuri","hashtags":[],"change_id":"I9c6b0d9d5e13e8b82ebb8e8a3a0f432efb865e28","subject":"Fix copying Swift ring files","status":"MERGED","created":"2021-02-03 14:01:13.000000000","updated":"2021-03-09 14:48:38.000000000","submitted":"2021-03-09 14:47:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773899-story/2007297-stable/ussuri","cherry_pick_of_change":766059,"cherry_pick_of_patch_set":2,"meta_rev_id":"cfe4e9ca14e8b87774ee1702da8a5969221a34e5","_number":773899,"virtual_id_number":773899,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-09 14:46:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"all":[{"value":1,"date":"2021-02-04 10:39:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":1,"date":"2021-02-11 22:28:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":2,"date":"2021-03-09 12:28:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-08 12:19:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":1,"date":"2021-02-04 10:13:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":1,"date":"2021-03-09 12:28:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"}],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 15:56:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-04 07:40:05.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"CC"},{"updated":"2021-02-04 08:14:14.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"state":"REVIEWER"},{"updated":"2021-02-04 08:14:14.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":16198,"name":"Ilya Popov","email":"hebulrih@gmail.com","username":"IPO"},"state":"REVIEWER"},{"updated":"2021-02-04 08:14:14.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2021-02-04 10:13:27.000000000","updated_by":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"reviewer":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"state":"REVIEWER"},{"updated":"2021-03-08 12:19:47.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"c768d0b9d4c045bb7e9296e9267060c43d9d6e08","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 14:01:13.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e463e6d078a1a74af7145450ddccdb0a7a182daf","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 14:05:19.000000000","message":"Patch Set 1:\n\nTesting in https://review.opendev.org/c/openstack/kayobe/+/773902","accounts_in_message":[],"_revision_number":1},{"id":"18e078409d6511f8506f6a24c1c492273ab577ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 15:56:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5eced7f4b564cd596e7a0596cecd0ab : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50f5c64638fd448fb9ed14be6c8dec74 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/053febffcf0c4ec19de0e6b6c14df38d : SUCCESS in 4m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe68210bb6a143e4833112e232dc4791 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88dcb65c1bcb44ef86cc193b4776570b : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79dc3eeba3d64346beeaa8733efcbf5d : SUCCESS in 4m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/00df3534b25f4bb8b3b6cc78bce0d1e0 : SUCCESS in 5m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/90766474875a478da5b70514c354f5c5 : SUCCESS in 8m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0b3749c3f15741999dee7af305162377 : SUCCESS in 17m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ba48478526304ea698243ffff3be5ee2 : SUCCESS in 1h 09m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9a10285927df4fb4add5771aedb96d00 : SUCCESS in 15m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/eaf2df5d008c4d02ad3875888157ee07 : FAILURE in 1h 49m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7cb5db24747545d283ad5a7887ba760a : SUCCESS in 15m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bdc9cec18ffa45958d6dbc0b62780274 : SUCCESS in 23m 33s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0ae6b52f948c415aa6e29632aee804eb : SUCCESS in 9m 16s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"58c1ed3e622dbb90b28885958c079c4855e090d1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 16:12:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"baceb20020bd99cd3a0c01f384a789e2a2198bb0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 17:41:23.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e85397f6cc7e497f9838a647ec9881f8 : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce219c3ee884f51a9de6dab3f81d2f7 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f94f2c69ddeb42b89e4a57ba6da70d58 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1404b2304297496a96ef9d50b5260b09 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76e0bbd713564768a68574fe36f997db : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a28ef10f3931419f92529a42abdce66c : SUCCESS in 3m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f31e64b5fbec4e539dbe9d5aaa47186d : SUCCESS in 5m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6114e9b07cb745a3af0aaf86fa27a767 : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9a0a740ea3bf4098908040d869d9b8b0 : SUCCESS in 19m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a14e34ec49aa4761b413a243bbbfaf89 : SUCCESS in 1h 04m 52s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b4eaeb6050954e8e9c122ee88cc52c15 : SUCCESS in 14m 46s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e12836246ebb49f188a70c2ec09bcebf : FAILURE in 1h 04m 05s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/36f2c6906b134fc797e34db21225349c : SUCCESS in 16m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8d18806ab561497aa8a702eef36f636c : SUCCESS in 27m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/23ef028badc241ab823cebed98ff3048 : SUCCESS in 9m 19s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"6cba73cd8ac994a29f8e22b03ff94065054b520a","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-02-04 07:40:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dbc8824fa378990173527313bf10933794634354","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 09:30:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a54d2a4bad74c9ba32a5ddbf06bc8e3 : SUCCESS in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b50803c81b1f468f9396ea14ceb98b23 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c7e83512af44de298ccdadeb0289205 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/95abd9dc33054ce48f82b16496dfb997 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/004c3ec807f942c1a8057f059e4431ab : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52e7699da49f47c1b2d80b883f39f940 : SUCCESS in 4m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/744980df474a49e59cbffc20fb79dc57 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/abdf05e9cb7c457b82adcea4d6da07a6 : SUCCESS in 8m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e3352782706247e0befcb1cc822603c0 : SUCCESS in 16m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/aa3d5a3c005a4f7887dd904fdc248e7d : SUCCESS in 1h 00m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/77245034e6ee431ea376229f7820a0c7 : SUCCESS in 12m 07s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/07f52299f89f4baa9eda81a04e5c1ba4 : SUCCESS in 1h 46m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6b8ef4d9c32148fb843df8946b5a9cb7 : SUCCESS in 17m 19s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5ad66fb1232d45a0a6d6efbeb2d36843 : SUCCESS in 24m 27s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1f2e92e0978542d09c5e4b0eef54b996 : SUCCESS in 9m 25s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"ead2b5c80ceb525d63baaff8bbccef2b462ebd1e","author":{"_account_id":23871,"name":"MargaritaShakhova","email":"shakhova.margarita@gmail.com","username":"MargaritaShakhova"},"date":"2021-02-04 10:13:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5a209a3d29e768736ce8a2b824e210ec7caf2129","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2021-02-04 10:39:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"81e4c613a8fef49c8c8b4e623402d5da3e296adf","author":{"_account_id":30568,"name":"Sergei Raiskii","email":"connoreika@gmail.com","username":"connoreika"},"date":"2021-02-11 22:28:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b07b8c3d526cde4fdb12af15a3da23ba9438f8c5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-24 09:51:38.000000000","message":"Patch Set 1:\n\nTested in https://review.opendev.org/c/openstack/kayobe/+/773902","accounts_in_message":[],"_revision_number":1},{"id":"4065b6919fed2cfb2cc79286986b43eb81bbc944","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-08 12:19:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a2b10986fbeae12249e10e8850a0e05eeca75a4c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-09 12:28:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3012542067862f7a7d80377506eb68dd643af172","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 12:31:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"90e30e2c61be90ad03838c343ba9a9e183a18f14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 14:46:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ca8c29f44b1400abc0ad51c40d54fd7 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a71e5b9196c84a5ea72d905d7c676364 : SUCCESS in 6m 18s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fd44f809c3ba4e129681d6d131a732a4 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4475b7aadd514ff0832744080bec940b : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29fc8e9ab09344b99d08009b8fd95ad5 : SUCCESS in 4m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6e87a6ac15a74dc8a7d6903a76d746ba : SUCCESS in 5m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c014594a3ef40e5a59a52ad5fbd0b6d : SUCCESS in 11m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e6b9818804cd427c85956380a8366b35 : SUCCESS in 21m 07s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/105a945186614f4198e1c4e3b7483706 : SUCCESS in 1h 02m 00s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/01eb2639d4d545dda098eee58b39d81c : SUCCESS in 18m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/37e2db83b20c419394015dbc430793b5 : SUCCESS in 2h 10m 19s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dd1c4d03c7b44ed1a68b8cc85d864c34 : SUCCESS in 16m 55s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f86118243be145718fe7aa16ea674784 : SUCCESS in 27m 28s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e2fdca5034834106955c04a721046119 : SUCCESS in 10m 38s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"3f7013b2f1a9d74dc216c67f73af216a67936042","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 14:47:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cfe4e9ca14e8b87774ee1702da8a5969221a34e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 14:48:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17f7bacb63b943f6ae05774fe562a413 : SUCCESS in 1m 10s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dfedbea0f24480cb3a01ec20617c3fe : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9595873feea0e293e888c7ab0dabde2100156d8c","revisions":{"9595873feea0e293e888c7ab0dabde2100156d8c":{"kind":"REWORK","_number":1,"created":"2021-02-03 14:01:13.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/99/773899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/99/773899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/773899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/773899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/99/773899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/99/773899/1"}}},"commit":{"parents":[{"commit":"12f2bb18fc9c96d9df750fc8f9a1b1de95d5e7b0","subject":"Fix /tmp/swift-rings/backups/ files deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12f2bb18fc9c96d9df750fc8f9a1b1de95d5e7b0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-12-08 17:33:51.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-03 14:01:13.000000000","tz":0},"subject":"Fix copying Swift ring files","message":"Fix copying Swift ring files\n\nRing files are binary formatted and should not be templated. Attempting\nto template them causes an error such as the following:\n\n    Template source files must be utf-8 encoded\n\nThis change fixes the issue by implementing support for configuration\nfiles that are listed as \u0027untemplated\u0027.\n\nChange-Id: I9c6b0d9d5e13e8b82ebb8e8a3a0f432efb865e28\nStory: 2007297\nTask: 38774\n(cherry picked from commit c847ef91572afa87a6d76b265c88db2d9e50ccb8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9595873feea0e293e888c7ab0dabde2100156d8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9595873feea0e293e888c7ab0dabde2100156d8c"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
