)]}'
{"id":"openstack%2Fkayobe~773901","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Idab9e36fac0bbef250b50fb37637915388173324","project":"openstack/kayobe","branch":"stable/victoria","hashtags":[],"change_id":"Idab9e36fac0bbef250b50fb37637915388173324","subject":"DNM: Test Swift","status":"ABANDONED","created":"2021-02-03 14:03:42.000000000","updated":"2021-03-23 10:17:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"cherry_pick_of_change":764617,"cherry_pick_of_patch_set":15,"meta_rev_id":"327a77e15b5f0ab3621e3068bdfea2bc79e8cced","_number":773901,"virtual_id_number":773901,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-03 15:06:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2021-02-03 14:04:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2f7a698d9ca7f2c05549baaddc20f36dbc857039","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 14:03:42.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* zuul.d/project.yaml","accounts_in_message":[],"_revision_number":1},{"id":"b12911a21f05548571514adf2a3cf409911de0d4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 14:04:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fb10c6489a2b6e43c538c591af23f7e94cb96903","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 14:04:16.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kayobe on branch stable/victoria.  The error was:\n\n  while scanning a simple key\n  in \"openstack/kayobe/zuul.d/project.yaml@stable/victoria\", line 3, column 1\ncould not find expected \u0027:\u0027\n  in \"openstack/kayobe/zuul.d/project.yaml@stable/victoria\", line 4, column 14","accounts_in_message":[],"_revision_number":1},{"id":"aca50f5036220adccbef63a1573ffc8595d97a2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 15:06:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b95266612e7a44e9b47091622720f73d : SUCCESS in 57m 22s","accounts_in_message":[],"_revision_number":2},{"id":"327a77e15b5f0ab3621e3068bdfea2bc79e8cced","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-23 10:17:18.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"56341335ce2e41ee2c0b8c0c181c10165e8a087a","revisions":{"784b6e4319ff2814a81d32b8cceef064aad14cb3":{"kind":"REWORK","_number":1,"created":"2021-02-03 14:03:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/773901/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/773901/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/773901/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/773901/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/773901/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/773901/1"}}},"commit":{"parents":[{"commit":"49361ff2d11c193dab3ac75c997e09792f4b6e7f","subject":"Merge \"Fix /tmp/swift-rings/backups/ files deletion\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49361ff2d11c193dab3ac75c997e09792f4b6e7f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-30 09:40:49.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-03 14:03:42.000000000","tz":0},"subject":"DNM: Test Swift","message":"DNM: Test Swift\n\nDepends-On: https://review.opendev.org/773898\n\nChange-Id: Idab9e36fac0bbef250b50fb37637915388173324\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/784b6e4319ff2814a81d32b8cceef064aad14cb3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/784b6e4319ff2814a81d32b8cceef064aad14cb3"}]},"branch":"refs/heads/stable/victoria"},"56341335ce2e41ee2c0b8c0c181c10165e8a087a":{"kind":"REWORK","_number":2,"created":"2021-02-03 14:04:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/01/773901/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/773901/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/773901/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/773901/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/773901/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/773901/2"}}},"commit":{"parents":[{"commit":"49361ff2d11c193dab3ac75c997e09792f4b6e7f","subject":"Merge \"Fix /tmp/swift-rings/backups/ files deletion\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49361ff2d11c193dab3ac75c997e09792f4b6e7f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-11-30 09:40:49.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-03 14:04:01.000000000","tz":0},"subject":"DNM: Test Swift","message":"DNM: Test Swift\n\nDepends-On: https://review.opendev.org/773898\n\nChange-Id: Idab9e36fac0bbef250b50fb37637915388173324\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56341335ce2e41ee2c0b8c0c181c10165e8a087a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/56341335ce2e41ee2c0b8c0c181c10165e8a087a"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
