)]}'
{"id":"openstack%2Fkayobe~773944","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Iae321a30f135f85136e2c9f3667280c23c0693d9","project":"openstack/kayobe","branch":"stable/train","topic":"fix-centos7-ipa-train","hashtags":[],"change_id":"Iae321a30f135f85136e2c9f3667280c23c0693d9","subject":"Fix building CentOS 8 IPA images on stable/train","status":"MERGED","created":"2021-02-03 16:38:54.000000000","updated":"2021-03-10 13:36:55.000000000","submitted":"2021-03-10 13:34:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773944-fix-centos7-ipa-train","meta_rev_id":"5d0ab292b76f8eb4d9ac9084ced03a24bcf8415a","_number":773944,"virtual_id_number":773944,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 13:34:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-10 12:00:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-03-10 13:36:55.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-03-10 12:00:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-10 13:36:55.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-03 16:53:45.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-02-03 16:56:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-03 17:10:43.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"CC"},{"updated":"2021-02-03 17:11:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-02-03 17:11:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-02-03 17:11:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"d927284776c392e68cee9fa91628d0193bfe44ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-03 16:38:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7606e56e63913f12fb05bb5c405e620a68c498ac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-03 16:53:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b8580efb8c6274973768c9f4f5fb8aa97175e0a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 16:56:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5055214e69d4b59af379ecced27b12d : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e973a42d3b6546eaa516631fc64a9e29 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1284fdaeead9918c118ee5c57e3843830154b1c1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-03 17:10:26.000000000","message":"Patch Set 1: Review-Priority+1\n\nThis is useful for operators on Train given that CentOS 7 IPA images are broken.","accounts_in_message":[],"_revision_number":1},{"id":"229bef601acad2142381fc0930fec33a7b9eea0d","tag":"autogenerated:gerrit:setAssignee","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-03 17:10:43.000000000","message":"Assignee added: Doug Szumski \u003cdoug@stackhpc.com\u003e","accounts_in_message":[],"_revision_number":1},{"id":"36c505f3c738e9fc66f3c1b8642f4052c1848d6a","tag":"autogenerated:gerrit:deleteAssignee","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-03 17:10:47.000000000","message":"Assignee deleted: Doug Szumski \u003cdoug@stackhpc.com\u003e","accounts_in_message":[],"_revision_number":1},{"id":"d4269580c90a188613baf28b6908c7b0d089c72e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-12 11:19:16.000000000","message":"Patch Set 1: Workflow-1\n\nWe may choose to use a tag for IPA-builder instead.","accounts_in_message":[],"_revision_number":1},{"id":"c67fd6f267dba4b8324465c9136638f93a0c17ea","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-12 16:23:15.000000000","message":"Patch Set 1: -Workflow\n\nPinning to a tag is difficult due to breakage of old versions.","accounts_in_message":[],"_revision_number":1},{"id":"c07f201a333598a9a2f42bec940fd2e1f93946d3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 07:02:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6d8f437e5c8d33efcab947598f7c6a52cc9cfe6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 07:16:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc231c16f52d406e85a4d0f522fe9535 : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b55044b6740640328fe02bcd44553d0e : SUCCESS in 6m 28s","accounts_in_message":[],"_revision_number":2},{"id":"943b92a5202197550a07a1ace9fb648089cac14b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-17 10:12:40.000000000","message":"Patch Set 2: Code-Review+2\n\nSeems like we could be chasing these issues forever without stable branches for DIB and IPAB.","accounts_in_message":[],"_revision_number":2},{"id":"a984bb1a51a2129ce19f72b97d3f74d6a97571cc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-17 10:16:10.000000000","message":"Patch Set 2: -Code-Review\n\nProbably needs some changes.","accounts_in_message":[],"_revision_number":2},{"id":"621a34f2ca5054ccf967b4b749e0e9e8480c79bd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 10:23:38.000000000","message":"Patch Set 2: Workflow-1\n\nMore needed.","accounts_in_message":[],"_revision_number":2},{"id":"6253a721cfbac61c0d446d84e8a45d6fd3b1cee6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 11:19:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d87f86b02371e84589216d72e82e4079a103e85e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 12:18:34.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"df5b0d0a5505c131596425cc5594aa298925bcd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 14:07:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72ce24ea9c5d4c27b735db77941a2d2f : SUCCESS in 4m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5d566ac74ae54775a04dea9f6474a090 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a8f221353194aef9746d9e1fe36e672 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50aa902fa069476ca889a3407a361a0e : SUCCESS in 4m 36s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/31e4dc858f2842a0994100ae7d54b8d5 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ca95770a6864ce3b42247a70fd5e723 : SUCCESS in 6m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13f547b99c364e7f8608ef66b5009048 : SUCCESS in 4m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6b88f808538746e99c0404196620639b : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4a4e5d99212d4c549d7118f7566125b2 : SUCCESS in 9m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a29efd5282334709947bd5d9a7039a7a : SUCCESS in 17m 05s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/b60d1c156e324481a7a8f8afe2b6b769 : SUCCESS in 1h 05m 17s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/04990dd2b71d4d5cb903aaf9680f3592 : SUCCESS in 56m 39s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/5f281b9cb1c5445d86eee5fdaed26643 : SUCCESS in 14m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9c8f18361fa24bc08bc4750db38fd411 : SUCCESS in 15m 04s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/d55d73cd639446a7bbd1124bbc17100e : SUCCESS in 1h 27m 39s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/86a3a45e7abb465389e11a7ed9e69070 : SUCCESS in 18m 06s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e0cfb31e5332400bb2a4d21486c8ab10 : SUCCESS in 17m 21s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/14f1870b536346a1bb9edf5e201acace : SUCCESS in 36m 32s","accounts_in_message":[],"_revision_number":4},{"id":"1c887eef9012dcd3ab05f1c90ddaf52f07936704","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-22 10:17:58.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f0c1abbd1c4cd89b52d4f05aadf231a08efc2e5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-05 14:57:13.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"2e418949c4fb27e0bd4920fa8f99a505b8871fc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 16:47:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c77705a676864a19b030ac40bee0674a : SUCCESS in 9m 09s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/8826911f6bde434880ea8e965b7cfb1b : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08b61dd9c6e844f2a1dfcdc3cce40009 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7c609f33abe4aaab4658b20c567cadb : SUCCESS in 4m 31s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7e721c13441b4908864f3784da79584f : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d16bdc159c624828bda4a504f95e4383 : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2f6187f7fb9542beb72f41dbb30bc4ac : SUCCESS in 10m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d270e4288f654484b50e0f020c57febe : SUCCESS in 5m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/27cdef7b4d7b4e8eae976acace924f7d : SUCCESS in 11m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cee87dbfbd8b4e9c8d5a7c9a2e8f43ba : SUCCESS in 29m 23s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/3c743afd9c5f40d897577161229a4d84 : SUCCESS in 1h 18m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e7d0768a59da4ed7abfde01c89124dec : SUCCESS in 1h 08m 10s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/9843b0a5956b4bfc936b3b6050f415bb : SUCCESS in 20m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fa56be9dc8c346289ee9a635e44866f6 : SUCCESS in 12m 43s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/fddeb9be52344023bfbafd58b283c153 : SUCCESS in 1h 42m 41s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/d41d0dbc9faf42e181a30e0a2d7eb07b : SUCCESS in 23m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ffefe0dac5644e4f99c32909fe826ab8 : SUCCESS in 17m 16s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8036e292c8ea47aaa95543c98dd92b3a : SUCCESS in 29m 05s","accounts_in_message":[],"_revision_number":5},{"id":"237a863169723b60dfa501c7425d590f11a9b5d3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-10 12:00:47.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8e208ff663d8065ce7e823dc25671ae2abcc9fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 12:00:59.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"939e6de2d64dd4492021f8debcf7d16dfcc3e251","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 13:34:50.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ccdd6d37e71d4dd7a577439bc0bc81ff : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/279066672d834689a53b2d1bd65fec8c : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8bfcebbd48f472482d7bc94e55883b7 : SUCCESS in 6m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3ae0a83ea7a649919de3c0b1d8812066 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85a7157588754d309137184125b5a51d : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0e6eeffd3ab4738bdafae9ec7165782 : SUCCESS in 5m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/149d28fb50cd41c899525a716a13b96c : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/91037733a30b4ea5bf3fa0c24a4154fc : SUCCESS in 9m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/18eaa5f9b1af4c248e4c7a1885c08cdf : SUCCESS in 17m 24s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/1b3c2bd5aac2471193b65f9cb3c29c17 : SUCCESS in 1h 13m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cfac22353dd84fc297872a179dc9f9ef : SUCCESS in 1h 07m 33s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/b8803f30e1ad4707ad883465501e4652 : SUCCESS in 16m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/36ff5681570d40339f277bc7ba5e2b0c : SUCCESS in 17m 09s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/115eacc1d8ef4fe5838a814e5b83359b : SUCCESS in 1h 19m 28s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/046ddd964bab4868a3d7a01db3799cf8 : SUCCESS in 23m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1ba93d85b4bd4fdaab56df63a760800f : SUCCESS in 18m 05s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8d48a31086a44ac2a8c75b7acca549a5 : SUCCESS in 35m 14s","accounts_in_message":[],"_revision_number":5},{"id":"ec0b10aed20dad25f540a2ceb28e1a844e431d4a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 13:34:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"8b7ffb1967bf409d5507cdc5d95fddf548e6bcff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 13:36:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/969f811649a74eca9a95506cdf59e9a6 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e873571db46415793d8f4b126281ed1 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"41c4c68788a4b6d86daa9ce66b498f70c3a8cfa3","revisions":{"25551973cebb90a9dc2c51940f8e547fc56fee18":{"kind":"REWORK","_number":1,"created":"2021-02-03 16:38:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/773944/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/773944/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/773944/1"}}},"commit":{"parents":[{"commit":"e0e0db459d3c99fbc1cb266739fa7e6e17f0c00e","subject":"Fix Python setup when venv is not used","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0e0db459d3c99fbc1cb266739fa7e6e17f0c00e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-03 16:38:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-03 16:38:13.000000000","tz":60},"subject":"Fix building CentOS 8 IPA images on stable/train","message":"Fix building CentOS 8 IPA images on stable/train\n\nThis change fixes two issues with the documented configuration for\nbuilding CentOS 8 IPA images on stable/train:\n\n1) The ipa_build_dib_git_elements_default variable is only available in\n   Ussuri and later and thus has no effect in Train. Use the\n   ipa_build_dib_git_elements variable instead.\n\n2) The latest version of ironic-python-agent-builder uses a recent\n   feature of diskimage-builder which allows a list of parameters to be\n   used in package-installs (Idff7b067ad4255e6fc4138f7eff313a81b75c8ba).\n   This is only available from 2.37.2 while ussuri upper constraints\n   allow 2.35.0. Use victoria upper constraints instead (3.2.1).\n\nChange-Id: Iae321a30f135f85136e2c9f3667280c23c0693d9\nStory: 2008593\nTask: 41751\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25551973cebb90a9dc2c51940f8e547fc56fee18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25551973cebb90a9dc2c51940f8e547fc56fee18"}]},"branch":"refs/heads/stable/train"},"40d410907be8fe1af9b77bd2c5ad842f48c82f0c":{"kind":"REWORK","_number":2,"created":"2021-02-17 07:02:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/773944/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/773944/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/773944/2"}}},"commit":{"parents":[{"commit":"35e85e2399c630836b69a984d0d208cb93ddb560","subject":"Add ipa_build_upper_constraints_file variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35e85e2399c630836b69a984d0d208cb93ddb560"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-17 07:02:08.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-17 07:02:08.000000000","tz":60},"subject":"Fix building CentOS 8 IPA images on stable/train","message":"Fix building CentOS 8 IPA images on stable/train\n\nThis change fixes two issues with the documented configuration for\nbuilding CentOS 8 IPA images on stable/train:\n\n1) The ipa_build_dib_git_elements_default variable is only available in\n   Ussuri and later and thus has no effect in Train. Use the\n   ipa_build_dib_git_elements variable instead.\n\n2) The latest version of ironic-python-agent-builder uses a recent\n   feature of diskimage-builder which allows a list of parameters to be\n   used in package-installs (Idff7b067ad4255e6fc4138f7eff313a81b75c8ba).\n   This is only available from 2.37.2 while ussuri upper constraints\n   allow 2.35.0. Use victoria upper constraints instead (3.2.1).\n\nChange-Id: Iae321a30f135f85136e2c9f3667280c23c0693d9\nStory: 2008593\nTask: 41751\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40d410907be8fe1af9b77bd2c5ad842f48c82f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40d410907be8fe1af9b77bd2c5ad842f48c82f0c"}]},"branch":"refs/heads/stable/train"},"970721d2d53fc10d84ba2c0dc5c96a58ecf5d60e":{"kind":"REWORK","_number":3,"created":"2021-02-17 11:19:52.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/773944/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/773944/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/773944/3"}}},"commit":{"parents":[{"commit":"35e85e2399c630836b69a984d0d208cb93ddb560","subject":"Add ipa_build_upper_constraints_file variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35e85e2399c630836b69a984d0d208cb93ddb560"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-17 11:16:48.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-17 11:16:48.000000000","tz":60},"subject":"Fix building CentOS 8 IPA images on stable/train","message":"Fix building CentOS 8 IPA images on stable/train\n\nThis change fixes issues with the documented configuration for building\nCentOS 8 IPA images on stable/train:\n\n1) The ipa_build_dib_git_elements_default variable is only available in\n   Ussuri and later and thus has no effect in Train. Use the\n   ipa_build_dib_git_elements variable instead.\n\n2) The latest version of ironic-python-agent-builder uses a recent\n   feature of diskimage-builder which allows a list of parameters to be\n   used in package-installs (Idff7b067ad4255e6fc4138f7eff313a81b75c8ba).\n   This is only available from 2.37.2 while ussuri upper constraints\n   allow 2.35.0. Use victoria upper constraints instead (3.2.1). We need\n   to bump stackhpc.os-images to work with this newer DIB version.\n\n3) Kayobe defaults to using ipa.vmlinuz for compatiblity with the legacy\n   ironic-agent DIB element. However, ironic-python-agent-builder\n   creates a file called ipa.kernel. Update Kayobe ipa_images variable\n   to handle it. The file should be renamed to ipa.vmlinuz after being\n   copied: show the command to run on the seed host.\n\nChange-Id: Iae321a30f135f85136e2c9f3667280c23c0693d9\nStory: 2008593\nTask: 41751\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/970721d2d53fc10d84ba2c0dc5c96a58ecf5d60e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/970721d2d53fc10d84ba2c0dc5c96a58ecf5d60e"}]},"branch":"refs/heads/stable/train"},"739a4bac416192ad6454f1f247a8ea4afeb77462":{"kind":"REWORK","_number":4,"created":"2021-02-17 12:18:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/773944/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/773944/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/773944/4"}}},"commit":{"parents":[{"commit":"35e85e2399c630836b69a984d0d208cb93ddb560","subject":"Add ipa_build_upper_constraints_file variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/35e85e2399c630836b69a984d0d208cb93ddb560"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-17 12:17:45.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-17 12:17:45.000000000","tz":60},"subject":"Fix building CentOS 8 IPA images on stable/train","message":"Fix building CentOS 8 IPA images on stable/train\n\nThis change fixes issues with the documented configuration for building\nCentOS 8 IPA images on stable/train:\n\n1) The ipa_build_dib_git_elements_default variable is only available in\n   Ussuri and later and thus has no effect in Train. Use the\n   ipa_build_dib_git_elements variable instead.\n\n2) The latest version of ironic-python-agent-builder uses a recent\n   feature of diskimage-builder which allows a list of parameters to be\n   used in package-installs (Idff7b067ad4255e6fc4138f7eff313a81b75c8ba).\n   This is only available from 2.37.2 while ussuri upper constraints\n   allow 2.35.0. Use victoria upper constraints instead (3.2.1). We need\n   to bump stackhpc.os-images to work with this newer DIB version.\n\n3) Kayobe defaults to using ipa.vmlinuz for compatiblity with the legacy\n   ironic-agent DIB element. However, ironic-python-agent-builder\n   creates a file called ipa.kernel. Update Kayobe ipa_images variable\n   to handle it. The file should be renamed to ipa.vmlinuz after being\n   copied: show the command to run on the seed host.\n\nChange-Id: Iae321a30f135f85136e2c9f3667280c23c0693d9\nStory: 2008593\nTask: 41751\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/739a4bac416192ad6454f1f247a8ea4afeb77462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/739a4bac416192ad6454f1f247a8ea4afeb77462"}]},"branch":"refs/heads/stable/train"},"41c4c68788a4b6d86daa9ce66b498f70c3a8cfa3":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-03-05 14:57:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/44/773944/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/44/773944/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/44/773944/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/44/773944/5"}}},"commit":{"parents":[{"commit":"371ad145147f7f0be61714b38c96d7015265866a","subject":"Merge \"Add ipa_build_upper_constraints_file variable\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/371ad145147f7f0be61714b38c96d7015265866a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-17 12:17:45.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-05 14:56:48.000000000","tz":60},"subject":"Fix building CentOS 8 IPA images on stable/train","message":"Fix building CentOS 8 IPA images on stable/train\n\nThis change fixes issues with the documented configuration for building\nCentOS 8 IPA images on stable/train:\n\n1) The ipa_build_dib_git_elements_default variable is only available in\n   Ussuri and later and thus has no effect in Train. Use the\n   ipa_build_dib_git_elements variable instead.\n\n2) The latest version of ironic-python-agent-builder uses a recent\n   feature of diskimage-builder which allows a list of parameters to be\n   used in package-installs (Idff7b067ad4255e6fc4138f7eff313a81b75c8ba).\n   This is only available from 2.37.2 while ussuri upper constraints\n   allow 2.35.0. Use victoria upper constraints instead (3.2.1). We need\n   to bump stackhpc.os-images to work with this newer DIB version.\n\n3) Kayobe defaults to using ipa.vmlinuz for compatiblity with the legacy\n   ironic-agent DIB element. However, ironic-python-agent-builder\n   creates a file called ipa.kernel. Update Kayobe ipa_images variable\n   to handle it. The file should be renamed to ipa.vmlinuz after being\n   copied: show the command to run on the seed host.\n\nChange-Id: Iae321a30f135f85136e2c9f3667280c23c0693d9\nStory: 2008593\nTask: 41751\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c4c68788a4b6d86daa9ce66b498f70c3a8cfa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c4c68788a4b6d86daa9ce66b498f70c3a8cfa3"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
