)]}'
{"id":"openstack%2Fkayobe~774055","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I4719b4d397c01b35c78cb84c6d686dd27742d1c0","project":"openstack/kayobe","branch":"stable/ussuri","topic":"test-seed-ipa-build-stable/ussuri","hashtags":[],"change_id":"I4719b4d397c01b35c78cb84c6d686dd27742d1c0","subject":"Test building seed deployment images in the seed job","status":"MERGED","created":"2021-02-04 09:32:38.000000000","updated":"2021-03-10 15:21:10.000000000","submitted":"2021-03-10 15:19:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"774055-test-seed-ipa-build-stable/ussuri","meta_rev_id":"88a82cc79c34d5f6fd94055977af681d233a6d12","_number":774055,"virtual_id_number":774055,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 15:19:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-10 13:14:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-03-10 15:21:10.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-03-10 13:14:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-10 15:21:10.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-04 11:30:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-12 09:26:39.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-02-17 10:10:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-10 12:04:56.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-03-10 13:14:19.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"64cb98f5445f27a514c78286f3dcc3f9e21b0d0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-04 09:32:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99d1cc2f335ced53845441e5af4cced36e3b4b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 11:30:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb8bed7aae6d45aba76ee2e419f1cbd3 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea2755a1f6534bdb99dea3cda1861269 : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c974dffdd9384d549cc8bd123d013249 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/527f5a7ad4254d45bed02b11dd2e6d9e : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3122fbdf260f4499a00a0d2a097e51bf : SUCCESS in 8m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dcd3380884ed484697848d469f6e6dc4 : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ab8f93a1c19345b492fd3830ee0294bd : SUCCESS in 9m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8ceb5328754947baa25f37905e7da9ca : SUCCESS in 18m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ae2b50077a794ebdae9fb9a19dd85175 : SUCCESS in 1h 05m 23s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d45d57c81abf4a9e93b78af92e805819 : SUCCESS in 15m 44s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3efdbbe0c19543f19636c79d353352ea : SUCCESS in 1h 52m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/88a74f9010534869b6bdf155896a0b88 : FAILURE in 19m 27s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5fc1061053e54c66b694ed21f6a50db7 : SUCCESS in 25m 52s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/49d1067bd7354bb6873caf389702bdbb : SUCCESS in 10m 34s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"97667ad0bf4445654e860e5c0958cb1ce4c09cda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-04 13:45:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"587487fa32d8456560950eb8074ed45f582c9e3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 14:13:38.000000000","message":"Patch Set 2: Verified-1\n\n(17 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02937153367c42a2b741efc722ae08e6 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b259e7855fd04402b86ec38fdc3933da : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24258e6c2a834e7aa073d4d746c9e0e6 : SUCCESS in 4m 37s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8704b5628a74a6295b7d14ebaf74203 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9573cda20b854592b8152c6622a1135f : SUCCESS in 5m 57s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/899614ad4e3c4533974057a8168d2b40 : FAILURE in 20m 24s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"86c7f3d3050a2ac32af5813a2e7e9c07a0aaff1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-04 14:22:36.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ea0dc4c60048ac6cdaba7ea0a49834d64e6ecdff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 14:57:14.000000000","message":"Patch Set 3: Verified-1\n\n(20 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be2d941cc95a41a59ed220e4df7d8b00 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24e0254fac164173b20e49bbca56abd0 : FAILURE in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/667522b468ba402580ea85fa23eff7de : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d5b20eeb06be4d17913c2c1817eb976d : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bbb2e23a02d48e5b360dc98e2ecf42d : SUCCESS in 6m 26s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/63fd8e4054634a859dd39db0a05b98de : FAILURE in 18m 48s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"f8edf6e0776407af7917e23df8f9d4efe59cae3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-05 09:39:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"07528b9200a9505c2fb8486a219951fe2d3022ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 10:05:36.000000000","message":"Patch Set 4: Verified-1\n\n(20 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36152d7c70514846a22de3e0bbf84b9c : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b10a7f2a7b1e460fb401a63aec176c3c : FAILURE in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/09c239557b174e14802718f93d026b34 : SUCCESS in 4m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a4b8bba2af9541268ca1a7bdc4723fe0 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c1fb975460640fda4f52b17e1f95d55 : SUCCESS in 10m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e12672c4869842e5b2b5df867f87fe82 : SUCCESS in 25m 06s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":4},{"id":"a126c22fc8d44273da4eac3600ace5d5f6f1f347","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-12 08:49:55.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"edfd4dfac531dbc4f666f7162d1735b384b3b90e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-12 08:52:02.000000000","message":"Patch Set 5: Review-Priority+1\n\nThis is not just CI: it actually fixes IPA builds on Ussuri.","accounts_in_message":[],"_revision_number":5},{"id":"bb59695fe0d854fd8fc1132e3b57da40c9faa172","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-12 09:26:39.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5eb2e2d65c2b927bc69623314e323f16a3a7c78a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 10:35:42.000000000","message":"Patch Set 5: Verified+1\n\n(18 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5421361b49994382ae40b3733e3479bd : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeb3101d3fb34e9e827342f7eb1de7e9 : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/432d854e8fbd4516bdc82d8820d6eb6d : SUCCESS in 4m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eae4d97af30f4f7b874c7a4a14d0d8b1 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe3220332e904cc9afbb7c9ad392599c : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3628a9a0d5748b091783ed8d4c16725 : SUCCESS in 3m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/67ac295df9084aebb89b5a0524879912 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f6c632942def40ffb6800c733e771fad : SUCCESS in 9m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4364f90ca28f4afeb1362f05a12ed28d : SUCCESS in 17m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/38aa25d92ecc4a2b99c3b1fec1e428ac : SUCCESS in 1h 24m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/682e27e444524bc7aba2de9ac48c21a9 : SUCCESS in 15m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cb91b48440184991994818bd6d47f321 : SUCCESS in 1h 41m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7ade1adce6564b4dbea5545c103120d3 : SUCCESS in 24m 59s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a5e129b27fcb49b6af0f5c193391bcef : SUCCESS in 31m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/349c80986035404b9ec4b9dcd90422a7 : SUCCESS in 11m 33s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":5},{"id":"a60d3d2e80363435160e4def35836ab2c9c0dfa3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-12 12:09:57.000000000","message":"Patch Set 5: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"5ee8a93c64cafb06a73fee9377c56a50963bd694","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-12 13:55:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"d457a719a27bf1b04016d80533e6590d3a3a690f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-16 21:27:32.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"40ce4efe60cdbda618aafc3a17d24a674aadd624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-16 21:33:55.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"98b5c1462d61f462fe16d166927586fc933680c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 23:22:08.000000000","message":"Patch Set 7: Verified+1\n\n(18 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3f54dd39a264753ad2e2fa702be07e2 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e175624f1cc4c56849b84871bd88c46 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/943a5d2e25de464da6399bdd2a85c305 : SUCCESS in 5m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/14a5a323140c4a36a58c32b3fcab646e : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ede3ed349c6421987216d450fac9fe9 : SUCCESS in 7m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/becfda30aba847dd9081143536637279 : SUCCESS in 5m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ae2259b8e48e4c0e8237362b1af29880 : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ded6f3d027264132b9e763e022e67aaf : SUCCESS in 19m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b081c7665c7b40698cf63a9a179fd53e : SUCCESS in 1h 00m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/44f76b6ace914c709b6100aa1d4d9393 : SUCCESS in 14m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f1a1d398f3654b4eb6e32c43356186eb : SUCCESS in 1h 35m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4615ae99dfc347e786c12dddacf60cf2 : SUCCESS in 23m 27s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1990ab7932104daf997becee3e38c9b8 : SUCCESS in 27m 21s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4ef91812db5247209f42f7a6b8b7cf61 : SUCCESS in 9m 11s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":7},{"id":"ba1ccf060bbbfb5c44027e64cc6251264480524c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-17 10:10:14.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"ff53e3ed2f357c05c6a0b33df0a6df0b2ab45e6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-19 13:35:27.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"184c0d1a7aeedeea3a6fdb983541934a5a9b5002","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 15:34:53.000000000","message":"Patch Set 8: Verified+1\n\n(18 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ed3f3792a094d45afc10a921624187c : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be59f62efc8940d08e158576a4e0f77a : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca0d1999c99649139dbceeb39e79aa48 : SUCCESS in 7m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce7c28ff797143859a5ffa9bdb36e311 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1f6a192ed0c42c5abc843089e803257 : SUCCESS in 6m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d243a6a0241a471dbd7c46201529dc81 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/40052a2fa2d34c71a6a9fdb02295d01f : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/33fbac32db7d4218827b4d119dfb2dfd : SUCCESS in 16m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e86660e584c449678a3ac18aab0f7e1c : SUCCESS in 1h 21m 54s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ea37e01554f5469d92a9d6347edfe215 : SUCCESS in 21m 38s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f155a314400a4ea0a666ba68aa238d62 : SUCCESS in 1h 58m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/555e87d55d2e4ed8bf4fa5e465d6bc48 : SUCCESS in 25m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3d212d32031c43ee835a21fbcd617f61 : SUCCESS in 23m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5b68f2fbd2b647d28cdab7fb54a128c6 : SUCCESS in 9m 41s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":8},{"id":"57b3a7e38dc9741fce4d7f8fd7958a0d5bf31491","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 15:35:33.000000000","message":"Patch Set 8:\n\n(18 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ed3f3792a094d45afc10a921624187c : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be59f62efc8940d08e158576a4e0f77a : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca0d1999c99649139dbceeb39e79aa48 : SUCCESS in 7m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce7c28ff797143859a5ffa9bdb36e311 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1f6a192ed0c42c5abc843089e803257 : SUCCESS in 6m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d243a6a0241a471dbd7c46201529dc81 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/40052a2fa2d34c71a6a9fdb02295d01f : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/33fbac32db7d4218827b4d119dfb2dfd : SUCCESS in 16m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e86660e584c449678a3ac18aab0f7e1c : SUCCESS in 1h 21m 54s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ea37e01554f5469d92a9d6347edfe215 : SUCCESS in 21m 38s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f155a314400a4ea0a666ba68aa238d62 : SUCCESS in 1h 58m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/555e87d55d2e4ed8bf4fa5e465d6bc48 : SUCCESS in 25m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3d212d32031c43ee835a21fbcd617f61 : SUCCESS in 23m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5b68f2fbd2b647d28cdab7fb54a128c6 : SUCCESS in 9m 41s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":8},{"id":"204ac8362b14fda0014c7fdaa069391987137b05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 15:36:23.000000000","message":"Patch Set 8:\n\n(18 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ed3f3792a094d45afc10a921624187c : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be59f62efc8940d08e158576a4e0f77a : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ca0d1999c99649139dbceeb39e79aa48 : SUCCESS in 7m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ce7c28ff797143859a5ffa9bdb36e311 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1f6a192ed0c42c5abc843089e803257 : SUCCESS in 6m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d243a6a0241a471dbd7c46201529dc81 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/40052a2fa2d34c71a6a9fdb02295d01f : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/33fbac32db7d4218827b4d119dfb2dfd : SUCCESS in 16m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e86660e584c449678a3ac18aab0f7e1c : SUCCESS in 1h 21m 54s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ea37e01554f5469d92a9d6347edfe215 : SUCCESS in 21m 38s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f155a314400a4ea0a666ba68aa238d62 : SUCCESS in 1h 58m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/555e87d55d2e4ed8bf4fa5e465d6bc48 : SUCCESS in 25m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3d212d32031c43ee835a21fbcd617f61 : SUCCESS in 23m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5b68f2fbd2b647d28cdab7fb54a128c6 : SUCCESS in 9m 41s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":8},{"id":"37b72b83d2cc7399c16a9a5fdbf673821c7b3b52","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-04 20:27:44.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"6dbc65e4acb9395fc81e56aad8ae977f762f8758","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 22:17:49.000000000","message":"Patch Set 9: Verified+1\n\n(18 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/892e3c1b3a1441a88e5399fc6dacdf51 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/290675d1324046d1a5dd6969fa17e284 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b2a9ce20cab44461a591963d9f7fd11f : SUCCESS in 5m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33c77774093b444982841921c5fc19a3 : SUCCESS in 6m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70561fe38d604339acc60febbe1028ae : SUCCESS in 7m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/41643527b2b340f39e45d2e2283479a0 : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/10a2fd76352e4e1c899a2fc2fd32d741 : SUCCESS in 8m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f934eb32bc7f42ca97cda90d095f24c1 : SUCCESS in 17m 18s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5bb9304affc24f67b1669d1ca940ec83 : SUCCESS in 1h 07m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/18518f2001914ea5ba3834b431e2dc95 : SUCCESS in 14m 37s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/838e9b625cc44aeaa32963ebfee5ccfe : SUCCESS in 1h 48m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/156f5d241bcc4ea79fea2c5a63afa227 : SUCCESS in 30m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9a58d0df923a4d999dc3584e1dcb285a : SUCCESS in 27m 48s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2f386e56de6a4c7ba3a62eb012668627 : SUCCESS in 10m 38s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":9},{"id":"e357dd2c5b765101afdae0564f61c4ba326fdc99","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-10 12:04:56.000000000","message":"Patch Set 9:\n\nCan we get rid of these pep8 warnings? Zuul will always post those warnings after we merge this change...","accounts_in_message":[],"_revision_number":9},{"id":"3b3bc5efd61148a3c900bcf666ffd2b4f3c73b7f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-10 12:17:43.000000000","message":"Patch Set 9:\n\n\u003e Patch Set 9:\n\u003e \n\u003e Can we get rid of these pep8 warnings? Zuul will always post those warnings after we merge this change...\n\nMost of the warnings are unrelated to the change. It only posts warnings if you change the file.","accounts_in_message":[],"_revision_number":9},{"id":"311b5e4bc90ef0f061e3ec69f1a9ea58b1f317f9","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-10 13:14:19.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nWell then so be it","accounts_in_message":[],"_revision_number":9},{"id":"4e39524b4db197d6bc30a26864a7946407109b49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 13:14:36.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"a99ef8f44e0a916769b714d77316bfaa338b31c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 15:19:02.000000000","message":"Patch Set 9: Verified+2\n\n(18 comments)\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7438b6087f049f29e80746b410eef3c : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd811c998fdd4bedb35c521eec0a5fc0 : SUCCESS in 5m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/21152bad744c401985fe379f386a95c1 : SUCCESS in 9m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba8e556346824f5da39f38a99c2cc697 : SUCCESS in 7m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/41d6a9eb99df4153b66e96201b3d94a6 : SUCCESS in 7m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2939f9322edf417eaae4779b4f082798 : SUCCESS in 9m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f63162269aee411bb48c3a5d0b4f6881 : SUCCESS in 17m 57s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/467757e8202148d8a2ee55b758eb1fd5 : SUCCESS in 1h 04m 09s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/00463f4990d04cdd8e55ad0668544de6 : SUCCESS in 17m 16s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b039ecb8817a4586a03d06c3daeefa69 : SUCCESS in 1h 59m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5059b91493804ba680828575968f795f : SUCCESS in 28m 12s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ddf376a9f410442eb39caf565d854fdc : SUCCESS in 28m 29s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/abac2a36541a4fd38385d3c94c00d1ee : SUCCESS in 12m 34s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":9},{"id":"61438c3a87dd98918d81dc956a3c7e96dc75c8af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 15:19:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"0354a51da5e20dc9ed025b26366459725f6f9e60","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 15:21:10.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62974406180649b18d9e918c18bda229 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"fc53de47e1c7c0b5b8b69bb99f64376bb46152dd","revisions":{"cefa8b957479daf090b7dfc8578eb4f51e0655c3":{"kind":"REWORK","_number":1,"created":"2021-02-04 09:32:38.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/774055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/774055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/774055/1"}}},"commit":{"parents":[{"commit":"12f2bb18fc9c96d9df750fc8f9a1b1de95d5e7b0","subject":"Fix /tmp/swift-rings/backups/ files deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12f2bb18fc9c96d9df750fc8f9a1b1de95d5e7b0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-28 09:40:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-04 09:32:14.000000000","tz":60},"subject":"Test building seed deployment images in the seed job","message":"Test building seed deployment images in the seed job\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled. Update the role to v1.10.3 which is the\nlatest version available.\n\nIt also requires building seed deployment images only after deploying\nBifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit 1a012675cf73763ac42bed1cfc7117ff77071f35)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cefa8b957479daf090b7dfc8578eb4f51e0655c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cefa8b957479daf090b7dfc8578eb4f51e0655c3"}]},"branch":"refs/heads/stable/ussuri"},"a87d4af0e24d31293077f704b547bb9f18f159b5":{"kind":"REWORK","_number":2,"created":"2021-02-04 13:45:37.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/774055/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/774055/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/774055/2"}}},"commit":{"parents":[{"commit":"12f2bb18fc9c96d9df750fc8f9a1b1de95d5e7b0","subject":"Fix /tmp/swift-rings/backups/ files deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12f2bb18fc9c96d9df750fc8f9a1b1de95d5e7b0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-28 09:40:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-04 13:45:11.000000000","tz":60},"subject":"[DNM] Test building seed deployment images in the seed job","message":"[DNM] Test building seed deployment images in the seed job\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled. Update the role to v1.10.3 which is the\nlatest version available.\n\nIt also requires building seed deployment images only after deploying\nBifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit 1a012675cf73763ac42bed1cfc7117ff77071f35)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a87d4af0e24d31293077f704b547bb9f18f159b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a87d4af0e24d31293077f704b547bb9f18f159b5"}]},"branch":"refs/heads/stable/ussuri"},"42563a8d59264a2cf2be08f5ebfb4f21f79516bc":{"kind":"REWORK","_number":3,"created":"2021-02-04 14:22:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/774055/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/774055/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/774055/3"}}},"commit":{"parents":[{"commit":"12f2bb18fc9c96d9df750fc8f9a1b1de95d5e7b0","subject":"Fix /tmp/swift-rings/backups/ files deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12f2bb18fc9c96d9df750fc8f9a1b1de95d5e7b0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-28 09:40:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-04 14:22:26.000000000","tz":60},"subject":"[DNM] Test building seed deployment images in the seed job","message":"[DNM] Test building seed deployment images in the seed job\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled. Update the role to v1.10.3 which is the\nlatest version available.\n\nIt also requires building seed deployment images only after deploying\nBifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit 1a012675cf73763ac42bed1cfc7117ff77071f35)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42563a8d59264a2cf2be08f5ebfb4f21f79516bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/42563a8d59264a2cf2be08f5ebfb4f21f79516bc"}]},"branch":"refs/heads/stable/ussuri"},"c15d84f15517516865516c216dfb0633af2f050e":{"kind":"REWORK","_number":4,"created":"2021-02-05 09:39:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/774055/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/774055/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/774055/4"}}},"commit":{"parents":[{"commit":"12f2bb18fc9c96d9df750fc8f9a1b1de95d5e7b0","subject":"Fix /tmp/swift-rings/backups/ files deletion","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12f2bb18fc9c96d9df750fc8f9a1b1de95d5e7b0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-05 09:38:56.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-05 09:38:56.000000000","tz":60},"subject":"[DNM] Test building seed deployment images in the seed job","message":"[DNM] Test building seed deployment images in the seed job\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled. Update the role to v1.10.3 which is the\nlatest version available.\n\nIt also requires building seed deployment images only after deploying\nBifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit 1a012675cf73763ac42bed1cfc7117ff77071f35)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c15d84f15517516865516c216dfb0633af2f050e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c15d84f15517516865516c216dfb0633af2f050e"}]},"branch":"refs/heads/stable/ussuri"},"768f0229686dd1e76b2ef8ac1d62f95034e4d3c3":{"kind":"REWORK","_number":5,"created":"2021-02-12 08:49:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/774055/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/774055/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/774055/5"}}},"commit":{"parents":[{"commit":"1a6f11a896dae549d9d065f7b8a1039f89668459","subject":"Merge \"Fix --limit with commas\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a6f11a896dae549d9d065f7b8a1039f89668459"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-05 10:49:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-12 08:49:43.000000000","tz":60},"subject":"Test building seed deployment images in the seed job","message":"Test building seed deployment images in the seed job\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled. This Ansible role was updated to v1.10.2 in\nmaster and stable/victoria by I5efdbd52556721914fe69d7c6ba454b2c721b643,\nfor another reason. Bump requirement to the same version.\n\nOn Ussuri, we also need to use a newer diskimage-builder than what is\navailable via upper constraints, as ironic-python-agent-builder relies\non a recent DIB feature (see Idff7b067ad4255e6fc4138f7eff313a81b75c8ba).\nUse Victoria upper constraints instead (this applies only to the Python\nvirtual environment used to build IPA images).\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/768f0229686dd1e76b2ef8ac1d62f95034e4d3c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/768f0229686dd1e76b2ef8ac1d62f95034e4d3c3"}]},"branch":"refs/heads/stable/ussuri"},"e84d20e067617f03f98c21166b5b3f1e70f82bae":{"kind":"REWORK","_number":6,"created":"2021-02-16 21:27:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/774055/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/774055/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/774055/6"}}},"commit":{"parents":[{"commit":"be0153641b9300f19a327cc235d92cc61a4a557c","subject":"Fix IPA build on stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be0153641b9300f19a327cc235d92cc61a4a557c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-05 10:49:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-16 21:27:14.000000000","tz":60},"subject":"Test building seed deployment images in the seed job","message":"Test building seed deployment images in the seed job\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled.\n\nThis Ansible role was updated to v1.10.2 in master and stable/victoria\nby I5efdbd52556721914fe69d7c6ba454b2c721b643, for another reason.\nRemember to bump the requirement when backporting to earlier releases.\n\nIt also needs changes in the way we interact with Bifrost to avoid using\nthe env-vars file which has been removed. This is implemented by change\nI25078e69acdf41a4ef9957f99fe5047de54b778d.\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e84d20e067617f03f98c21166b5b3f1e70f82bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e84d20e067617f03f98c21166b5b3f1e70f82bae"}]},"branch":"refs/heads/stable/ussuri"},"bc57ef0a750da924ca6d12dababb7dc67e55a322":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2021-02-16 21:33:55.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/774055/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/774055/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/774055/7"}}},"commit":{"parents":[{"commit":"be0153641b9300f19a327cc235d92cc61a4a557c","subject":"Fix IPA build on stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/be0153641b9300f19a327cc235d92cc61a4a557c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-05 10:49:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-16 21:33:18.000000000","tz":60},"subject":"Test building seed deployment images in the seed job","message":"Test building seed deployment images in the seed job\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled.\n\nThis Ansible role was updated to v1.10.2 in master and stable/victoria\nby I5efdbd52556721914fe69d7c6ba454b2c721b643, for another reason.\nRemember to bump the requirement when backporting to earlier releases.\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc57ef0a750da924ca6d12dababb7dc67e55a322"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bc57ef0a750da924ca6d12dababb7dc67e55a322"}]},"branch":"refs/heads/stable/ussuri"},"1ff02f5acfbd95cd0148c021c4ac6061ed7d8ff7":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-02-19 13:35:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/774055/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/774055/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/774055/8"}}},"commit":{"parents":[{"commit":"e007a96939b9621481bf4c03588b7e7e85c4833f","subject":"Fix IPA build on stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e007a96939b9621481bf4c03588b7e7e85c4833f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-05 10:49:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-19 13:35:16.000000000","tz":60},"subject":"Test building seed deployment images in the seed job","message":"Test building seed deployment images in the seed job\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled.\n\nThis Ansible role was updated to v1.10.2 in master and stable/victoria\nby I5efdbd52556721914fe69d7c6ba454b2c721b643, for another reason.\nRemember to bump the requirement when backporting to earlier releases.\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ff02f5acfbd95cd0148c021c4ac6061ed7d8ff7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ff02f5acfbd95cd0148c021c4ac6061ed7d8ff7"}]},"branch":"refs/heads/stable/ussuri"},"fc53de47e1c7c0b5b8b69bb99f64376bb46152dd":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-03-04 20:27:44.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/55/774055/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/774055/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/774055/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/774055/9"}}},"commit":{"parents":[{"commit":"3cf2c992d957b6b5fbad5319c56c23a54ba2d398","subject":"Fix IPA build on stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3cf2c992d957b6b5fbad5319c56c23a54ba2d398"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-05 10:49:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-04 20:27:24.000000000","tz":60},"subject":"Test building seed deployment images in the seed job","message":"Test building seed deployment images in the seed job\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled.\n\nThis Ansible role was updated to v1.10.2 in master and stable/victoria\nby I5efdbd52556721914fe69d7c6ba454b2c721b643, for another reason.\nRemember to bump the requirement when backporting to earlier releases.\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc53de47e1c7c0b5b8b69bb99f64376bb46152dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc53de47e1c7c0b5b8b69bb99f64376bb46152dd"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
