)]}'
{"id":"openstack%2Fkayobe~775093","triplet_id":"openstack%2Fkayobe~master~Id5ed0b594d6270c4a03606caa776bbb291be2694","project":"openstack/kayobe","branch":"master","topic":"ipa-extra-hardware","hashtags":[],"change_id":"Id5ed0b594d6270c4a03606caa776bbb291be2694","subject":"Update IPA docs and test build with extra-hardware","status":"MERGED","created":"2021-02-11 10:24:01.000000000","updated":"2021-02-22 15:23:40.000000000","submitted":"2021-02-22 15:21:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775093-ipa-extra-hardware","meta_rev_id":"ec2ac94c0a40bed84cec43eea69470475788538e","_number":775093,"virtual_id_number":775093,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-22 15:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-02-19 14:51:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-02-11 14:07:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2021-02-19 14:51:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-02-12 08:25:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"reviewers":{"REVIEWER":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 11:03:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-02-11 13:20:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-12 08:25:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"state":"REVIEWER"},{"updated":"2021-02-19 14:51:54.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"4d849eb4be3d678f79d4d2f83c160267b79836c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-11 10:24:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f14fc1df4273f35cc838ea1c2651977ff3ff658","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-11 11:03:45.000000000","message":"Patch Set 1:\n\nThoughts? Maybe this should be in a periodic job instead?","accounts_in_message":[],"_revision_number":1},{"id":"0a8480bfb67298188513770e536dc6b3a1ef8162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 13:20:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/59f570c219b84daf8367dfd11a5cea73 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a9f1d41a13741d18626a8e0ae17da9c : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e2a5fbf9c7246ac861a50e0401741ed : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98d2e0e69d8348fb97d6b1860a5ae22e : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d8fdc94837646b3a25755849854fbab : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa96785f1cec4a219b90007e90cc66a0 : SUCCESS in 7m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5b7fa8432a9d49bf85f81b54cdb946dc : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e4b62065b0c041e1a52a583c85833f81 : SUCCESS in 9m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6073809430ed4c1394239097af785ef9 : SUCCESS in 19m 18s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f585fa11014a4fba9b9840c42b76dbd5 : SUCCESS in 1h 26m 57s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/42718b38307f467e893e0394379c183d : SUCCESS in 1h 05m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bb6825e0ff4c44e884c124a5c35bfb59 : SUCCESS in 16m 16s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d90302170a9b4ae4ba88a5538e4f3638 : SUCCESS in 2h 01m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d61554a1c6044559920c77161e6e4b80 : SUCCESS in 23m 18s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2f8dd905faa44c2380ad9903b55cf337 : SUCCESS in 35m 09s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/82f7fa9e17424db6b7aa0619bc48eb92 : SUCCESS in 9m 55s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"f905f5505c2bf274d5b16ee8ed2ff71028268160","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-11 14:07:31.000000000","message":"Patch Set 1: Code-Review+2\n\n\u003e Patch Set 1:\n\u003e \n\u003e Thoughts? Maybe this should be in a periodic job instead?\n\nIt seems reasonable to me, hopefully the overhead is not too high.","accounts_in_message":[],"_revision_number":1},{"id":"0a02d0f3ddc9d660895d0228dcd4e91fa63da5e6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-11 14:08:06.000000000","message":"Patch Set 1:\n\nOf course we have no way to verify whether it worked.","accounts_in_message":[],"_revision_number":1},{"id":"5f9d88416932d7823df5dbd800bc824db783cb00","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-11 15:49:03.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review+2\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Thoughts? Maybe this should be in a periodic job instead?\n\u003e \n\u003e It seems reasonable to me, hopefully the overhead is not too high.\n\nIPA build took 8 minutes, similar to without the extra-hardware element.","accounts_in_message":[],"_revision_number":1},{"id":"1e379cf933f0d57c16c88ba987a2b9756c9755a1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-11 15:49:31.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Of course we have no way to verify whether it worked.\n\nAt least we test the build works, it\u0027s better than nothing.","accounts_in_message":[],"_revision_number":1},{"id":"eafd40efabe46284fbfcaf7eb0adb2479feeb3cb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-12 08:25:27.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"e3776dd25c52b9066319dfca36a3388d0ec4b667","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-02-19 14:51:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"be57c6dd5e2cffdb7e4ba108b8c33bb4b08dafef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 14:52:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"30b35839bfde3c04c4329345b2938bf590462a67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 16:29:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0cdd0f146cd41e488b90d81ded1fce0 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bbe71de363794698bed002fc39d012b3 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf15f40a9f894367872823cec152b188 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cf4a345606645aea648b144273c82eb : SUCCESS in 7m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/377005736e0145848d69cc2a4c84ebfd : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/28491101adb540f38f8f51b63bb72036 : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/695ae93bb8a74323b312d592c87beb62 : SUCCESS in 19m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c0111387f7434a7490fc5aad9abb0a75 : FAILURE in 1h 10m 03s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/e0643f154b79465bad63506629640adf : SUCCESS in 58m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b05a9a42ade7421aac51d73b80e22101 : SUCCESS in 16m 52s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7f43dff78eaa4418adbb609bd55eccb8 : SUCCESS in 1h 31m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/36e1747af64044dfa22f46764213134f : SUCCESS in 23m 02s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1fa366d0bf4f4909bbfd2b0e5b1d699d : SUCCESS in 21m 08s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4b44bbb519b4426e92bfcdc36f7c1820 : SUCCESS in 9m 35s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"8a273c1f6569c1023b8bfda851919ac0c98d50b0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-19 16:59:45.000000000","message":"Patch Set 1:\n\nrecheck\n\nSSH issues to the test VM","accounts_in_message":[],"_revision_number":1},{"id":"44456cf649bb9cc5bf727e2e4e50d01bbbb1c7c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 19:15:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f9ecb50d9ca4d11b85ff93b886b6c51 : POST_FAILURE in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2efd80e186394ae5b770ddc2cc78f67a : POST_FAILURE in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b79d4e852c024eb9aef33d9670324748 : POST_FAILURE in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49ecda4aa936440d8ae8c670d42435b1 : POST_FAILURE in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63aad00f55b7423bb62baa78c5169105 : POST_FAILURE in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3301cd004ab943f691564e128bfb5972 : POST_FAILURE in 7m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/828c12bd090843338db2ba0f3dc59da5 : POST_FAILURE in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2d328a75150e4636a837109927586ad3 : POST_FAILURE in 9m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/04394bf9c36c420697cfb1b5346dd6b3 : POST_FAILURE in 19m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/76322d46e55743169336ced5e5746e48 : POST_FAILURE in 1h 03m 31s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/c1dd4683b9684de5a155d4379793cf04 : TIMED_OUT in 2h 02m 05s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2dcea110018649e190eda91a064736be : POST_FAILURE in 14m 21s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/643414d5f93a4c18a6b06cc59a2d9fb8 : POST_FAILURE in 1h 43m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cb33865dc38c47808f1396f2d0deca82 : POST_FAILURE in 33m 42s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d35f158995304a07b7907317baaf1e46 : POST_FAILURE in 20m 20s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b8d056dd5ebd43a89a99fe7347e12202 : POST_FAILURE in 9m 43s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"a502924fe2acde16c8d22cf9dafaa3021fcc764e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-22 08:16:47.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c78dec1980bc026b056dce1d7a24fa18abb4e86f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 10:22:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7af22d5aacdf4664982c8b85a8b1235d : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7898862f83d4614baffa347a4695f5e : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/169d75a27f0d4ed49e7cf23428f945ee : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/682e4b6a65c34e468687713244e9185d : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/523fdc061d1e4158bb6cfd57b801b768 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13b13063d2584b278c3f2525312cd99a : SUCCESS in 6m 47s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6ff6bf778d984db3b46f1ac0a0ed7e6c : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4ce590b447a24b5cb4a365083853f0a4 : SUCCESS in 8m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1c3c8c5912d84eadbedc3bbaddbfd25c : SUCCESS in 17m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/dc44d90260f44478866398736b87227f : SUCCESS in 55m 14s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/c1c8362fc81a4d958a2d15703efe9190 : TIMED_OUT in 2h 01m 54s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8ab0b5baa6b442b4b6c7b8c0b306caec : SUCCESS in 15m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a5007c6c1f284048a9e54946f34cef30 : SUCCESS in 1h 43m 05s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/10fc3d6c7ba943408c07f3e341f3d257 : SUCCESS in 29m 52s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f1872e63ce6c4adaa1acfd94dec080f3 : SUCCESS in 33m 06s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f1ceca6fcd7344a2a3ea3c888c8ce551 : SUCCESS in 10m 52s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"b8da17fc2546409533582a7a1d7c0276f4900ee4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-22 11:47:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4173e1ec5cdbae9f0bee57f2ae5f0c6afbc521d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 13:28:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67da9f8f669c471190ae15b291da6ea2 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f5d3cf9ff70496eaf691a8bcac47a06 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a08b598f74704f2d9f826645f0402f40 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d27477c5e584ef39bdc52889ebaedeb : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab1bda6d3401410295eba3c816322c63 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52236275f4974d4cbe0eaec57b997779 : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d006c98dba374a93a62055866bd7298d : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e0b7574cd59b47f4a90cb48aeea462cc : SUCCESS in 8m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/19717ebfe2054692adf693ed66e0c277 : SUCCESS in 17m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f94eca29990948c6ad7253f85c239c69 : SUCCESS in 58m 07s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/29675af499ac4eebb4464f23c4684531 : SUCCESS in 56m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/659a61b7882943ef927be340ba7ab2ee : SUCCESS in 15m 51s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a60938a6b79a4248a41d4dedb1f0e4d7 : SUCCESS in 1h 36m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/78c7cb26e5354406bd3ca5bc423d06d8 : SUCCESS in 24m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/06633b0a4b1a465381f420f3935f10f2 : SUCCESS in 22m 32s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f8f2d4ad60a14dfd87c717a6dd97860a : SUCCESS in 9m 40s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"c9e2764fa43dffc6af53d100a9012042e381cdab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 13:28:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"711e5ee5c13e6f0bd3a9ff2f05b2de98744751e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 15:21:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30eb8de2db5b4f699da9336ae6799f96 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd98c25ced3149888a9ceed0f4eb79ec : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e0196dc9c2a942fdbd6c9e7363794de1 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dee34471d3b4477891a722f07498d762 : SUCCESS in 6m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d2201034ad7c482589c5825583c72a8a : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4f9933b20e6f4ebf9129ddfa83ba5c51 : SUCCESS in 10m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dbb5bb73a7be44a4a8bbb6ac30c5face : SUCCESS in 21m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d5ca5388cbbe45c89f56982e8c787f12 : SUCCESS in 48m 00s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/4ed600e1fa314fb3afa691646a32369b : SUCCESS in 53m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3289b3535d154c0fa56c76c88a725895 : SUCCESS in 18m 54s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2e5ef65c859c4ba3bdc8c3db530103a5 : SUCCESS in 1h 52m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3fc9b5bbe35b465890ab98d7660222d4 : SUCCESS in 23m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4b6302b75d414c86b70007edf8c1a501 : SUCCESS in 20m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/49e5fa28cfaa449a8ea4113c5fc61966 : SUCCESS in 11m 25s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"4bb2280fa8bba60ccf84df4ead10d5c47e0f25a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 15:21:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ec2ac94c0a40bed84cec43eea69470475788538e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 15:23:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00da17c734cb4bfabd1819370a68628e : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d765fbc5ebb7c1a8f80d4c49f3846107e4f23189","revisions":{"d765fbc5ebb7c1a8f80d4c49f3846107e4f23189":{"kind":"REWORK","_number":1,"created":"2021-02-11 10:24:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/93/775093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/775093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/775093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/775093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/775093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/775093/1"}}},"commit":{"parents":[{"commit":"c84a9757dd449dbd8863bb1dcc55837abe26a716","subject":"Test building seed deployment images in the seed job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c84a9757dd449dbd8863bb1dcc55837abe26a716"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-11 09:50:43.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-11 09:50:43.000000000","tz":60},"subject":"Update IPA docs and test build with extra-hardware","message":"Update IPA docs and test build with extra-hardware\n\nChange-Id: Id5ed0b594d6270c4a03606caa776bbb291be2694\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d765fbc5ebb7c1a8f80d4c49f3846107e4f23189"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d765fbc5ebb7c1a8f80d4c49f3846107e4f23189"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
