)]}'
{"id":"openstack%2Fkayobe~775119","triplet_id":"openstack%2Fkayobe~master~Ic33d3e31096f7b0cec2e4e9284f5c1078a4ff04c","project":"openstack/kayobe","branch":"master","topic":"story/2004960","hashtags":[],"change_id":"Ic33d3e31096f7b0cec2e4e9284f5c1078a4ff04c","subject":"CI: add Ubuntu seed VM job","status":"MERGED","created":"2021-02-11 11:11:10.000000000","updated":"2021-03-04 18:53:24.000000000","submitted":"2021-03-04 18:45:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775119-story/2004960","meta_rev_id":"ee6fc342e10164faae888a0208f4ecebcac84a90","_number":775119,"virtual_id_number":775119,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 18:45:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-04 11:13:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-03-04 16:55:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-03-04 16:55:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 14:43:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-04 11:13:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-03-04 16:55:37.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"cd28a9457f86efb2f49f7cec17265d76331a10b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-11 11:11:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a2e35ea9b1b4f212455ee3536f47953cc250b384","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-11 14:16:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9fba8c4cf5e41f9fd92a7a8291a1502eae6d0219","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 14:43:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43a5132247a543ed8fa58f2714351c9b : SUCCESS in 9m 02s","accounts_in_message":[],"_revision_number":2},{"id":"d8c0dd43835eef775bad08bae01a9d8d989dd5ac","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-11 14:57:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"ff48ac91ede77eeb7e0308e675fbb789c0d35038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 15:16:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99e09e058c804549a89943789b0c9664 : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":2},{"id":"2942b693f0c14564d2a93cab3aaa46fe96a4baa0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-11 15:36:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6b3db720e546c6f2b767f550d82cfcbfd673a63d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 15:50:21.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0b71234400864bb0bf3ed2d53a323edc : SUCCESS in 9m 32s","accounts_in_message":[],"_revision_number":2},{"id":"e46efa0cb4dbb1eab17122d96a924c77ff1fcf19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-12 09:29:36.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"4da3633ed823f76c391e67243c025c547dfac56a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 10:56:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/48e468b917fc45e6b6652d6040bcd631 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe2ec073830142b9aaf1c8740df42cfb : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd1846c3add24e8d94b761a1d0f40097 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee149c320e44420484af82ff79441059 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d7345b39cfb4d37afa56ecd0b3100b5 : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e08fbd369e7c4fda88d45e06b3115e77 : SUCCESS in 7m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bd6c5fb50abb4fa89bfee73c462c0255 : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ed30dcb4623413cab3d394626b5ba3e : SUCCESS in 8m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/05febd3df1d141ffaf5de9ab00888e70 : SUCCESS in 18m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/80b82765da80411aa394f285da9e9195 : SUCCESS in 53m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4507751b43f6425fbf16bc8b73a763c4 : SUCCESS in 41m 34s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/bf8365302776442684acc321ff1932f3 : SUCCESS in 56m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4da1418088a5433fa1ca73137cfe5b5a : SUCCESS in 15m 24s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/418777944d5b470bb009d76d80ccaae3 : SUCCESS in 1h 25m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0cc80a1171224853aaf3ffcdb402c173 : SUCCESS in 14m 29s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/17cdda4e1ce84795b4a8861f97cc319f : SUCCESS in 24m 00s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/35e128c74dea488b81bf65da296833fd : SUCCESS in 10m 45s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9de6018292774f979cc7deb122d05095 : SUCCESS in 8m 08s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"9cb85c7bd308c2225b8ad727402cbdca6d920cbe","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-23 16:31:15.000000000","message":"Topic set to story/2004960","accounts_in_message":[],"_revision_number":3},{"id":"5a34eb7447e4581e1555b4ea1f6105b1b02f3863","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-01 17:59:31.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"448ad31d1ce257997491662f23d886939aa9ff47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 19:55:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/501c576ba4594f4ba4732b56d0d001c5 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/145a1906110047008e96efb421fb764d : SUCCESS in 4m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9a0a86fa787846aaa815b99ed98c2e2b : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c61082a25d24316a84378943479338a : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35192f9582ef4962840610b9aaa27dbc : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fed2fad72a34d09acb703812e37049e : SUCCESS in 10m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4cbf2bd953014554ad9f2780f0c64f86 : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c122564270c541f1bf21816c3a50dffb : SUCCESS in 9m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/34bdcf8875904d79925fba7c784bce21 : SUCCESS in 18m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/59f32d25b0f54cdfbf5a7d2fe929a68e : SUCCESS in 1h 03m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d901f4143e84ca8b831e4b6a96079fd : SUCCESS in 42m 47s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/44a134186b734917a87f54d4a5379826 : SUCCESS in 45m 52s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c0d4f9a27f854a75a19a301809e1d691 : SUCCESS in 11m 54s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7d4ef98ad2f34728bad10220d4f4c4f7 : SUCCESS in 1h 21m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ebe7e9012c8e442487e11dde27e80fc4 : SUCCESS in 26m 50s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ca243d2540d049248f8c7a78e5ff8096 : SUCCESS in 21m 37s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/d792297a748a4c85b5e419463eaf371f : SUCCESS in 10m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e8a895082044fdc836fbd347b4919f1 : SUCCESS in 7m 38s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":4},{"id":"fd2d062ce1330c5e0c402c84e2a80b57e955d943","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-04 11:13:12.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"5f4f3bb517bd97b59dde2ecef26993744890869e","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-03-04 16:55:37.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b3b9de148cbe82d563caebad80994937d5b6cf0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 16:56:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"87a36e7f58f726b1ada63a3f366239f72a700e59","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 18:45:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a70773b32ca42e396a351e44e6bbcde : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/255942a1980441c8b9f5edc89c2751a8 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efa8be03653a4a41aab365b675e259c3 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27b5730f1f1c4a458d76d472a9360c52 : SUCCESS in 7m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e30459e9eb14a3b82b0c422d4fe8836 : SUCCESS in 7m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6d5b516978c144f5a60514f9a6db5f6f : SUCCESS in 10m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dbc50feb87fe41a5ac4eeb0fb49814b3 : SUCCESS in 17m 44s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cf75faf8dfeb45b18ca1561ff7d05890 : SUCCESS in 56m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7361b02d640c4c7296ef868f8b9c510d : SUCCESS in 46m 33s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/30927de60f224bb4972506cdb573aba1 : SUCCESS in 1h 05m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c4aa72b9fa5a4a0e8a91ef0610e402f0 : SUCCESS in 23m 27s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fd6df6a7c776484692c1f041c0f3fcea : SUCCESS in 13m 51s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5b7775b438ea42f1b7d3e14692e528f2 : SUCCESS in 1h 43m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/16675593e18947e2bc443bda219c209e : SUCCESS in 26m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1477dacba9684d39ab4b723b59d9c425 : SUCCESS in 27m 51s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c44dc57a7d8841ef84bbac769597c3ec : SUCCESS in 10m 46s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76e9936181e246558a35716460de9476 : SUCCESS in 10m 21s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":4},{"id":"f33e99343f529037486b86668a6e81ed43b6f311","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 18:45:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ee6fc342e10164faae888a0208f4ecebcac84a90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 18:53:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7c2a6a6de8e4c9fad5b51be3564d4a7 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7a99b88ebcae4ae5d290fc33286ca91f5b9097f2","revisions":{"4d3fc532b7aaa9d9f2521f90e24b29ec6e963428":{"kind":"REWORK","_number":1,"created":"2021-02-11 11:11:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/775119/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/775119/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/775119/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/775119/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/775119/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/775119/1"}}},"commit":{"parents":[{"commit":"8762af4ae520f6af7cb559915526d469cf27528b","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8762af4ae520f6af7cb559915526d469cf27528b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-11 11:04:10.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-11 11:05:01.000000000","tz":0},"subject":"CI: add Ubuntu seed VM job","message":"CI: add Ubuntu seed VM job\n\n* Need to specify bash for \u0026\u003e syntax\n\nStory: 2004960\nTask: 41806\n\nChange-Id: Ic33d3e31096f7b0cec2e4e9284f5c1078a4ff04c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d3fc532b7aaa9d9f2521f90e24b29ec6e963428"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4d3fc532b7aaa9d9f2521f90e24b29ec6e963428"}]},"branch":"refs/heads/master"},"b14c572ff14fb1da045a8b9e03d16600c3e54f37":{"kind":"REWORK","_number":2,"created":"2021-02-11 14:16:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/775119/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/775119/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/775119/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/775119/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/775119/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/775119/2"}}},"commit":{"parents":[{"commit":"8762af4ae520f6af7cb559915526d469cf27528b","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8762af4ae520f6af7cb559915526d469cf27528b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-11 11:04:10.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-11 14:16:20.000000000","tz":0},"subject":"CI: add Ubuntu seed VM job","message":"CI: add Ubuntu seed VM job\n\n* Need to specify bash for \u0026\u003e syntax\n\nStory: 2004960\nTask: 41806\n\nChange-Id: Ic33d3e31096f7b0cec2e4e9284f5c1078a4ff04c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b14c572ff14fb1da045a8b9e03d16600c3e54f37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b14c572ff14fb1da045a8b9e03d16600c3e54f37"}]},"branch":"refs/heads/master"},"85ade6825c6fbef7aa230688247c8fa1b222ceac":{"kind":"REWORK","_number":3,"created":"2021-02-12 09:29:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/775119/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/775119/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/775119/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/775119/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/775119/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/775119/3"}}},"commit":{"parents":[{"commit":"8762af4ae520f6af7cb559915526d469cf27528b","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8762af4ae520f6af7cb559915526d469cf27528b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-11 11:04:10.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 09:29:31.000000000","tz":0},"subject":"CI: add Ubuntu seed VM job","message":"CI: add Ubuntu seed VM job\n\n* Need to specify bash for \u0026\u003e syntax\n\nStory: 2004960\nTask: 41806\n\nChange-Id: Ic33d3e31096f7b0cec2e4e9284f5c1078a4ff04c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/85ade6825c6fbef7aa230688247c8fa1b222ceac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/85ade6825c6fbef7aa230688247c8fa1b222ceac"}]},"branch":"refs/heads/master"},"7a99b88ebcae4ae5d290fc33286ca91f5b9097f2":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-03-01 17:59:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/775119/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/775119/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/775119/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/775119/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/775119/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/775119/4"}}},"commit":{"parents":[{"commit":"9f41cefc15e240602d2c6ef9076e36135e9138e4","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f41cefc15e240602d2c6ef9076e36135e9138e4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-11 11:04:10.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-01 17:59:31.000000000","tz":0},"subject":"CI: add Ubuntu seed VM job","message":"CI: add Ubuntu seed VM job\n\n* Need to specify bash for \u0026\u003e syntax\n\nStory: 2004960\nTask: 41806\n\nChange-Id: Ic33d3e31096f7b0cec2e4e9284f5c1078a4ff04c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a99b88ebcae4ae5d290fc33286ca91f5b9097f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a99b88ebcae4ae5d290fc33286ca91f5b9097f2"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
