)]}'
{"id":"openstack%2Fkayobe~775130","triplet_id":"openstack%2Fkayobe~master~I0ecf4f4ce9b7333d3e41c69c3f908bee83391781","project":"openstack/kayobe","branch":"master","topic":"story/2004960","hashtags":[],"change_id":"I0ecf4f4ce9b7333d3e41c69c3f908bee83391781","subject":"Ubuntu: update Apt cache before package installation","status":"MERGED","created":"2021-02-11 12:01:06.000000000","updated":"2021-03-04 18:24:12.000000000","submitted":"2021-03-04 18:21:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775130-story/2004960","meta_rev_id":"f9e0d22434e5d68f81774210800740e5a370a104","_number":775130,"virtual_id_number":775130,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-04 18:21:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-04 11:09:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-04 08:44:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-03-04 11:09:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 14:08:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-04 08:44:29.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-03-04 11:09:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"1d2dd51b7e91627de4ec4549c93cdf409b4212f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-11 12:01:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3719c814d8996d6c2cb9ed21f512fdec07e1908","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 14:08:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e11d839cd07d4a4580b4abe85e160227 : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8422866b5384b7d885bf071e9930bb6 : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ab54c2849d804e09bee12d4323d0f8ea : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44970a67e77f4fdba58b9396a831c1d7 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e8c46727c65479e835e10cd1cfefc64 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aedc44a31c1a485fa1f7add66758ffbb : SUCCESS in 6m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/883b36893b8143a5811499072b971ca5 : SUCCESS in 4m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/df8e2a0a85ea4eb0a74c8931a9bf22a9 : FAILURE in 4m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c551e76e533c411b80f3af370bce59dd : SUCCESS in 17m 45s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/77db7aaf3c89451d9b9595ec28107acb : SUCCESS in 53m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/28f2c37df482494c8aca7a51da3ad6bc : POST_FAILURE in 54m 43s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ed4e876c24434d79a8b62e5844afd576 : SUCCESS in 57m 09s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e49421aa66e24c918a53ce19b8ab7a34 : SUCCESS in 14m 17s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5fef383c915d4d0ea552f1fbfc6a6c6d : SUCCESS in 1h 29m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3335d324f43a49fdbe93f85420b8bbd8 : SUCCESS in 13m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/dcf8c785e6034ae98e625f005c3755ce : SUCCESS in 23m 15s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9bfcc1996c84431da2fd8c5549a98070 : SUCCESS in 11m 22s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"4fb727e99e8a588943c367719f61e31e9d12339b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-11 15:37:20.000000000","message":"Patch Set 1:\n\nLooks like the package module can be used to do this in one task (seen in an OSA role):\n\n  package:\n    update_cache: \"{{ (ansible_pkg_mgr \u003d\u003d \u0027apt\u0027) | ternary(\u0027yes\u0027, omit) }}\"\n    cache_valid_time: \"{{ (ansible_pkg_mgr \u003d\u003d \u0027apt\u0027) | ternary(600, omit) }}\"","accounts_in_message":[],"_revision_number":1},{"id":"b6a11a3486d01468025f0e23ba69522bd80702bd","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-11 15:37:31.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d2fa27307106d774b8a39b8e1faa9032506d1c59","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-23 16:31:02.000000000","message":"Topic set to story/2004960","accounts_in_message":[],"_revision_number":1},{"id":"4a9d2a36b1f76b99c0bbbe18a0bddde0b7cdb272","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-24 10:40:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c13cfb7adfbc73cd0a9a4dae9ce48573b1dc27da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 12:49:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d321d8d9e9e041299f287e69f5506a2d : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cf5177053704516b3e2253e901275c0 : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebf0b02342014224a68356dd2f5ba094 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62a7f52d1cb14504a6c21cb8ec658f53 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9341f3c3403643179300e1963181b07b : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f878cfac7164ad993397a690e651d0a : SUCCESS in 8m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/90772af9d89e4116a686ac2d58244317 : SUCCESS in 6m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3d30eac648974dec8cd4ba4d2995aa74 : FAILURE in 4m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4a00199349514fe2aad763d3e87b7518 : SUCCESS in 17m 51s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2a212bf66abc41649af339993f0a8d0f : SUCCESS in 51m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b9d07af61d714b0bb36afc13804a74b7 : SUCCESS in 43m 47s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/13721258c49f42b4a11b8283cf0e4cc1 : TIMED_OUT in 2h 02m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1e8fbb93a4724f0c9f776f6565c3b4da : SUCCESS in 16m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1541aa74a811499da469394bae84a2f9 : FAILURE in 1h 44m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/06eb44214cc74c10ace119f8ff764de2 : SUCCESS in 25m 34s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c4b7f2f341b246bb9f54226dc165b856 : SUCCESS in 26m 36s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b9ca4d42224b41fd94f77ec7ee1a2784 : SUCCESS in 11m 17s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"8b5247bbd18f4317e774b8a56372d9b93a94ee3b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-24 13:59:38.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"20d54631d469706d77469ce977156b922d193741","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 17:12:11.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26e36c9d65124c6f947b207b6556df06 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b8b481c6344489586972a8219f9e381 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c5d8e405a87241f29853b2d000fe7c3c : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c09a7a720444b6e98eb02edee39b86f : SUCCESS in 9m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6715c080b4f54ea3b1ecb06130e2db17 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28b5b3d840814c708cd1bfff40718d0d : SUCCESS in 7m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8604857058e444a5b366877bc9eec29d : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1fa298f9dde446c1b5ea221758b0e0c1 : FAILURE in 6m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/31f2d83c824a48c38755f39d5cdf79b2 : SUCCESS in 20m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7e74970dfaf84430ac2f7915518eed6f : SUCCESS in 54m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/15a572fa302b4b80992148e8b1b44eca : SUCCESS in 37m 35s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/1e1d4571d01e43f6a2e23c79a1f2171e : TIMED_OUT in 2h 01m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/32a3d4e6b25e40d6a9db1f194a1c1e3a : SUCCESS in 16m 06s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4bcdd9aee92f442f808104f7b1ed5fa3 : FAILURE in 55m 49s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b800ee26824e4d07b9939a464a820ce4 : SUCCESS in 24m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9002240c7b1d45388477f05748d4f63c : SUCCESS in 25m 24s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4384e4ae035b496c8aa42160e527b86b : SUCCESS in 11m 16s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"72dfd8083b818420c231e4983ef52e8c913f7622","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-24 17:33:40.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"49db1dd802809e46b2d48a036a11b2ac8520546d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 21:33:57.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae52d18da68a4d8cb40250bdb27b5df1 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d21d66754e245749ee07f160ef6d070 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/963fd8b0bb9b4c9e9430ffdd35a2fa26 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20ebf75f586c465ebfa056819d1017c6 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e76191d9275d453cb4aa97a047f66320 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b484e9ca1144e8c9e5bf7f8305d4f11 : SUCCESS in 6m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b33a1f2df6b3491f928e239696186e7b : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3ea88db85179440696ce466dcd2f7b28 : FAILURE in 8m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1f0e4741129b4ad3b6acc49536cce50a : SUCCESS in 20m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f215ac48d4a34be9b4135162fa301880 : SUCCESS in 58m 04s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/803faa0ae8174ee59ab4a8e084f7d4d5 : SUCCESS in 44m 18s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/658662cc91244ef187191858f4f086b6 : TIMED_OUT in 2h 02m 03s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c692c86f393d4a28a49e58b1c626e68d : SUCCESS in 15m 09s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ce79e75660f44d808c2d922c029f7734 : SUCCESS in 1h 55m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/81ffffd8164e4247a1797f68a8cf7c34 : SUCCESS in 20m 05s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5cf2dffb7e584884bb48982efeec2fe9 : SUCCESS in 22m 00s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ebdf8754574e43159bd137be77dd69c3 : SUCCESS in 10m 54s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"febc261c89c6e4e0d0e299f65a1a91e2818cce99","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-25 09:27:47.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"b7e406d84ff6b1eb36b35be38a175c2c662a3063","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 11:05:25.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8694ab77d434df7a5caf10599e6d618 : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd4996bf4e434f56b3651b8d09898250 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9543442329c74071a409a1b865f94beb : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24a7ec38f77f4ac2a63d14c21353e6dd : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/24b4edae26d94adfba81133dd25e8ac2 : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72226d2e7ad840bf86e3ba3d22d0510a : SUCCESS in 8m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/33a22252ab67451f926fb43ba6302f23 : SUCCESS in 7m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/70076c74d1ab4a54a43973123772df85 : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc5fea7dd31b46d2af6560bb83075492 : SUCCESS in 21m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/de1c00223fac42ed87d49fd6ea3db681 : FAILURE in 12m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3aba96e16bde4ceabd463a0528987771 : SUCCESS in 51m 22s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/83b7457b7794420cb60a42ae081769f8 : SUCCESS in 59m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/712d95cc258f48f089d1ad4eaeeeffcd : SUCCESS in 16m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ce2ef89196ee4ce48991e3e6a8b73b9c : SUCCESS in 1h 32m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8c0612f9652a45498cf43ae3cfa438c9 : SUCCESS in 33m 49s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b8465e5d0cd54e76bd5136dbec562bbc : SUCCESS in 26m 38s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/3f37080fa11e4f30a5f3abdf3ce3ca47 : SUCCESS in 11m 36s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":5},{"id":"973ed2a2c20530197def3ac08ee17b4f3434d8a2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-01 18:00:49.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"3278f515bc189659c4c064436b2b5e05ce880806","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 20:58:03.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/910a09f324b4401794cf43201a5b6063 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cd6f31297ad4230a261578060deac5c : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e7b5819d7e94efe97e6ea86d17649c1 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/267582ba422d43c7997255eb6faa71f4 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/537293ba9c0d4220a1f370cd5d89c556 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/482fad9166ce4fbc88c58e7b5b7a9c4a : SUCCESS in 6m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/390c13b5c3bd4ab9aa87e71d69a24723 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/11a4b1cc28e446c0a21db294724204ae : SUCCESS in 8m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c1d4d4607c95415a993bc066439f0e36 : SUCCESS in 17m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c9ebe137c6154bfe95d44846023565a6 : FAILURE in 11m 55s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/454098617144415b86f2eca5dd336cad : FAILURE in 11m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e98857bf47c948f1a821e41c64e37af7 : SUCCESS in 14m 28s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/812a2e1f76d54844982e494fa09d1f82 : SUCCESS in 2h 01m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/59ba0b9c25474a19af550482840cdb28 : SUCCESS in 30m 42s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/246df8051c9c4b49a3dead0b55ce2c60 : SUCCESS in 31m 46s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/82c403ed7fbd41ee86a4fa88c50216c8 : SUCCESS in 11m 51s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":6},{"id":"a362e49b5b795e16e96ac35dc3094a9134067ea6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-02 10:40:18.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"962259b5f474036320324e483a39c11e49f86a95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 12:35:27.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/323f9265b49d4554b14a269c451c57ee : SUCCESS in 5m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/311907c715374469ba3e9e0c38636562 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5715fd4351b241deae7cf61f835a1a3a : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edff590446704f7f98f7d171dd9e5f0f : SUCCESS in 8m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e779b405d8d40d6b1c70e19e18a03ce : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d604317118ba46589a404f6510325018 : SUCCESS in 9m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c3f828f2707046f4ba3ad70d7fa3c35a : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/869ac97a0a9747718dea3930821d7f4e : SUCCESS in 9m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6ffbf4afcf2c417eb6df830e95df8650 : SUCCESS in 20m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/00d79784b93247e086f4a88eb21183bc : SUCCESS in 55m 20s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/574f0e09282e407a881fb6876800a7b6 : SUCCESS in 1h 04m 14s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2261803e7d184e01a6920a9b82725c0d : SUCCESS in 14m 39s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f313815fcd384e0fab4ddf0191b4863f : SUCCESS in 1h 48m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/89bf40af5f384d0496176bba7d1df28f : SUCCESS in 27m 33s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5b3cb6a8a8b64726903bd892bff50928 : SUCCESS in 23m 03s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c91280180c194d98bf1757e41eb9f3eb : SUCCESS in 10m 52s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":6},{"id":"c2885ddb6274b3d652af53f9c7f2315b96c648f1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-04 08:44:29.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"4ce3d9bfa7c6e7beb7880add1c540143923bc48d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-04 11:09:29.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"2beb153030dea163d6b764c1390e836cc3acb00b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 11:09:47.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"6518a1778230e520fa3ef3079ac02fbabc22ef9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 14:22:16.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff33503b475d43ab94f73c08d462691c : SUCCESS in 9m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c869ec3f3d1b42eeb64ecc3757d2805f : SUCCESS in 11m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54cd97e8dd2e4797921e738db7496909 : SUCCESS in 9m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5218a6d98aca4988adb605b50093bf5f : SUCCESS in 9m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6661d1f118a6465aa92e099067237cf7 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f6e5b165ff8b4a7ab09f45f3c5c27fec : SUCCESS in 15m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/28f1ca5981f74bfcac42dfee34d48bf5 : SUCCESS in 20m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5d03acfef6094fbf82428f2ecb7cf891 : SUCCESS in 1h 08m 44s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/21e045eababb48f6a2ff3993106c032a : FAILURE in 49m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a921e63b9afe41f4ba91e37c9210b197 : SUCCESS in 23m 29s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a10f09470e51416da1d5611f87e7e4f9 : FAILURE in 1h 00m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/90bca6a6da3e4cc1b6ce1ee32587fcc6 : SUCCESS in 35m 32s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/69f399dac2054b24b69fe32ca244c016 : SUCCESS in 35m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/64e3c8e1cad04d78b8f84c72416793f1 : SUCCESS in 15m 16s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":6},{"id":"e9d0a129fab68a672219f7b65e0dc92055b07c09","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-04 14:41:12.000000000","message":"Patch Set 6:\n\nrecheck\n\nNetwork issues","accounts_in_message":[],"_revision_number":6},{"id":"162ea5e5d5a817e448cac1da35bb470cf81e0bcf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 16:33:46.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf852a4f3db84346a75a6e38d79ff523 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f487b6db69047c3a8297d80b3584357 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d80a51976df543f9a371155d7d70d925 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/440fd95eeabc4c6e9311b7bc047df023 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc311cf70e014410be366a81428dbac1 : SUCCESS in 5m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f7368f40df845a1aacd54ab468fd6a1 : SUCCESS in 8m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1f91e434c112488a85bc73ab87a8ab21 : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e756e4f204064392ad5f013be7f29f1d : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5c67b3625d204366ab1084bafcdc0ef2 : SUCCESS in 19m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ae7a5fe35d2a4b96828fdd7eb9644806 : SUCCESS in 56m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ab934e3124246c9a08a49688bd868dd : SUCCESS in 45m 58s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/691b2bbabad64c449840306b7ba79bed : SUCCESS in 1h 03m 46s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ff2796ee818b434d9d7afe51ac39d7df : SUCCESS in 18m 39s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8623e380d0984bc39e6474ed67f91032 : SUCCESS in 1h 44m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8bf17db8cebc4b3991cdf4821be73103 : SUCCESS in 26m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cb1908d1295644aea5dc7e611dac911d : SUCCESS in 22m 06s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f9a099cc0ec44a9dbc3ffc4f21028087 : SUCCESS in 16m 49s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":6},{"id":"b9476ad6bbc185ca4b88b9e06ba3c6ad733ee03d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 16:34:02.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c2b35731ef9ea54ab606a0b285eec99a92e8516c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 18:21:58.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4257281113cd4fddaaf5c8270d4ca7c8 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f577e8f8c11a4d1999efee7f2a6c1ba4 : SUCCESS in 9m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd19879528734a6a894f616aa9f344df : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/308fea5d2ce14b1688a87aa819d310c1 : SUCCESS in 10m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1ad59679a9a94417ad79e52ef4480e17 : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bf18a17288f04107bf666258bb99481c : SUCCESS in 10m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7fa50c84581f4d91aea2a9b0be1e6b83 : SUCCESS in 22m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/291f2d4166d4459ead7ccacb98b77da5 : SUCCESS in 1h 09m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ca8f50962d24218a8394fcd8581b72c : SUCCESS in 54m 17s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/c7362d62dd0e4e74bac9483a3ab6b00e : SUCCESS in 1h 01m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a88aeaf627f4445d9ba1d834d76c3376 : SUCCESS in 16m 09s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4277c9ae7bb8428bb56179cb03261604 : SUCCESS in 1h 44m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bff5633309fd4c5db33f0152f96d0380 : SUCCESS in 30m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6058bcb4fa2c47118779dd495fefe514 : SUCCESS in 22m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5f26cbcfb22e44169af74e6272cd39c9 : SUCCESS in 10m 17s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":6},{"id":"fd74f9bc9037488cbe9213c6fdc0c1136c4e452d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 18:21:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"f9e0d22434e5d68f81774210800740e5a370a104","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 18:24:12.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d35b044695b439eafc32b23984e2606 : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"0c309a18c8ab5de7cf5302ada713ba972381d352","revisions":{"08cb64287089a703528f61753490e4a09ae53bda":{"kind":"REWORK","_number":1,"created":"2021-02-11 12:01:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/775130/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/775130/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/775130/1"}}},"commit":{"parents":[{"commit":"eb1d2af6bae56cb0a0949ebe879b0c61f4cc3bff","subject":"Ubuntu: container image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb1d2af6bae56cb0a0949ebe879b0c61f4cc3bff"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-04 15:24:20.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-11 11:59:28.000000000","tz":0},"subject":"Ubuntu: update Apt cache before package installation","message":"Ubuntu: update Apt cache before package installation\n\nUpdate Apt cache prior to all package installation tasks.\n\nAdds apt_cache_valid_time, which defaults to 3600 seconds. This allows\nthe time for which the Apt cache is valid to be configured.\n\nChange-Id: I0ecf4f4ce9b7333d3e41c69c3f908bee83391781\nStory: 2004960\nTask: 41766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08cb64287089a703528f61753490e4a09ae53bda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/08cb64287089a703528f61753490e4a09ae53bda"}]},"branch":"refs/heads/master"},"52562f54cb8fc455a56604387dbe862b53f55014":{"kind":"REWORK","_number":2,"created":"2021-02-24 10:40:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/775130/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/775130/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/775130/2"}}},"commit":{"parents":[{"commit":"3ea234165bc47b14ee9a29e0478a83fa5d7a0f19","subject":"Ubuntu: container image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ea234165bc47b14ee9a29e0478a83fa5d7a0f19"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-04 15:24:20.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-24 10:39:34.000000000","tz":0},"subject":"Ubuntu: update Apt cache before package installation","message":"Ubuntu: update Apt cache before package installation\n\nUpdate Apt cache prior to all package installation tasks.\n\nAdds apt_cache_valid_time, which defaults to 3600 seconds. This allows\nthe time for which the Apt cache is valid to be configured.\n\nChange-Id: I0ecf4f4ce9b7333d3e41c69c3f908bee83391781\nStory: 2004960\nTask: 41766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52562f54cb8fc455a56604387dbe862b53f55014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/52562f54cb8fc455a56604387dbe862b53f55014"}]},"branch":"refs/heads/master"},"a27ca1e59a8a5882cf9380620534cd4f51b4d7bb":{"kind":"REWORK","_number":3,"created":"2021-02-24 13:59:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/775130/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/775130/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/775130/3"}}},"commit":{"parents":[{"commit":"3ea234165bc47b14ee9a29e0478a83fa5d7a0f19","subject":"Ubuntu: container image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ea234165bc47b14ee9a29e0478a83fa5d7a0f19"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-04 15:24:20.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-24 13:59:34.000000000","tz":0},"subject":"Ubuntu: update Apt cache before package installation","message":"Ubuntu: update Apt cache before package installation\n\nUpdate Apt cache prior to all package installation tasks.\n\nAdds apt_cache_valid_time, which defaults to 3600 seconds. This allows\nthe time for which the Apt cache is valid to be configured.\n\nChange-Id: I0ecf4f4ce9b7333d3e41c69c3f908bee83391781\nStory: 2004960\nTask: 41766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a27ca1e59a8a5882cf9380620534cd4f51b4d7bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a27ca1e59a8a5882cf9380620534cd4f51b4d7bb"}]},"branch":"refs/heads/master"},"de0b8764f8aa31c4eaf4521aa55903ecd0a945b5":{"kind":"REWORK","_number":4,"created":"2021-02-24 17:33:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/775130/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/775130/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/775130/4"}}},"commit":{"parents":[{"commit":"3ea234165bc47b14ee9a29e0478a83fa5d7a0f19","subject":"Ubuntu: container image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ea234165bc47b14ee9a29e0478a83fa5d7a0f19"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-04 15:24:20.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-24 17:32:46.000000000","tz":0},"subject":"Ubuntu: update Apt cache before package installation","message":"Ubuntu: update Apt cache before package installation\n\nUpdate Apt cache prior to all package installation tasks.\n\nAdds apt_cache_valid_time, which defaults to 3600 seconds. This allows\nthe time for which the Apt cache is valid to be configured.\n\nChange-Id: I0ecf4f4ce9b7333d3e41c69c3f908bee83391781\nStory: 2004960\nTask: 41766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de0b8764f8aa31c4eaf4521aa55903ecd0a945b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/de0b8764f8aa31c4eaf4521aa55903ecd0a945b5"}]},"branch":"refs/heads/master"},"193b99e2d2871f6139d392629285e704aab93f1d":{"kind":"REWORK","_number":5,"created":"2021-02-25 09:27:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/775130/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/775130/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/775130/5"}}},"commit":{"parents":[{"commit":"3ea234165bc47b14ee9a29e0478a83fa5d7a0f19","subject":"Ubuntu: container image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3ea234165bc47b14ee9a29e0478a83fa5d7a0f19"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-04 15:24:20.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-25 09:27:43.000000000","tz":0},"subject":"Ubuntu: update Apt cache before package installation","message":"Ubuntu: update Apt cache before package installation\n\nUpdate Apt cache prior to all package installation tasks.\n\nAdds apt_cache_valid_time, which defaults to 3600 seconds. This allows\nthe time for which the Apt cache is valid to be configured.\n\nChange-Id: I0ecf4f4ce9b7333d3e41c69c3f908bee83391781\nStory: 2004960\nTask: 41766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/193b99e2d2871f6139d392629285e704aab93f1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/193b99e2d2871f6139d392629285e704aab93f1d"}]},"branch":"refs/heads/master"},"0c309a18c8ab5de7cf5302ada713ba972381d352":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-03-01 18:00:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/30/775130/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/30/775130/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/30/775130/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/30/775130/6"}}},"commit":{"parents":[{"commit":"a9bf637cc56fe960ed7edd4af4624490bfc60b69","subject":"Merge \"Ubuntu: support non-persistent veth configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9bf637cc56fe960ed7edd4af4624490bfc60b69"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-04 15:24:20.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-01 18:00:49.000000000","tz":0},"subject":"Ubuntu: update Apt cache before package installation","message":"Ubuntu: update Apt cache before package installation\n\nUpdate Apt cache prior to all package installation tasks.\n\nAdds apt_cache_valid_time, which defaults to 3600 seconds. This allows\nthe time for which the Apt cache is valid to be configured.\n\nChange-Id: I0ecf4f4ce9b7333d3e41c69c3f908bee83391781\nStory: 2004960\nTask: 41766\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c309a18c8ab5de7cf5302ada713ba972381d352"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c309a18c8ab5de7cf5302ada713ba972381d352"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
