)]}'
{"id":"openstack%2Fkayobe~775267","triplet_id":"openstack%2Fkayobe~master~I839cd1b9d63c9bbf3e0c75def76b8bff68637c75","project":"openstack/kayobe","branch":"master","topic":"story/2004960","hashtags":[],"change_id":"I839cd1b9d63c9bbf3e0c75def76b8bff68637c75","subject":"CI: Add Ubuntu seed job","status":"MERGED","created":"2021-02-11 19:45:30.000000000","updated":"2021-03-10 16:16:03.000000000","submitted":"2021-03-10 16:07:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"775267-story/2004960","meta_rev_id":"0c470cb69e7b25af0796eb700ae82a3201182076","_number":775267,"virtual_id_number":775267,"owner":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-10 16:07:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-04 11:09:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-03-04 10:31:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-04 10:42:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-03-04 11:09:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-11 19:58:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-12 09:18:14.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-02-19 13:44:45.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2021-03-04 10:42:32.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-03-04 11:09:04.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"f06079fd1745cba5261274241225935e53d5b233","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-02-11 19:45:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7985fecb460a0104252f5822e721242fef073ea","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-02-11 19:51:57.000000000","message":"Topic change-767705-31 removed","accounts_in_message":[],"_revision_number":1},{"id":"681edb78d2fd3eb9e7e12f8da5e329731eba59f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 19:58:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5e6c79710a564766bfe583b7b86d013d : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/115c8a830b804f6b9c90cbe8b17152cf : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af020c23742d4a208bd71664b37d32e1 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fcac92dab95c47468cf8ca865659f4a8 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d624bde8194248fe8d1bb48bbff833f2 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acf9573b23d5420eb72cfa63176c1719 : SUCCESS in 7m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4641973604564d79ac1629e31f51cbda : FAILURE in 3m 28s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"9774a15329f519a0d5e9265fde3c9214053675e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-02-11 21:32:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"97ea27f6e206995563be27165015e3b8a1c57b16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 21:48:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae4c972950bd4d1785d18833d6ef6e04 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83dcffc5743347c7b7b47c10a4c5f066 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/142f6a6cc25a4fa4b405c4c172ddfce7 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/078f95c5ba3d40d79f060ea6e85e115f : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63670741ab194c40a233ecbc1f8801f4 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/936384f92257435c9362e5cde0a985c8 : SUCCESS in 7m 37s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1fb49d5475a140ed8dd8134833b087c3 : FAILURE in 7m 24s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"01bb22f68b4e7cae0f9825b419ce9fff552bc97c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-02-11 22:01:25.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"475c25da987c8c97601f35e065a7d281acd8519f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 22:23:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b29aeee6eed4e57a4c3e8cfca447675 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8abb237fa5f8440ebd31638dc1a4e45b : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4c96b86fde447b7b64d2279825d7a60 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d53c408b1d9b42dd8e63f35ca67cf5be : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f50abc75d4854ca1959838eb42855940 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd19d14781a442ff8c555825603be5df : SUCCESS in 8m 17s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09d17f1ed4da40bd83731acbf3a873cd : SUCCESS in 11m 37s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"27cd48b638b65f0ff2454214e488f434d2240dd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-02-12 07:42:27.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2c8b05a80a985bee1b748dd40b2984c590df2f0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 08:01:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/490f03cac44c49fd9fd7114629dbc0d0 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17038e10226e459baf370845ecbca40a : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f17e9107083543c5bc55f45d0965c51b : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58abb00be42f4a0c88d88ca720b3ad64 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8182fdcb6b5546a1a43fe11447a7763c : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec6c2c45482f40d5847bee93bc4478f2 : SUCCESS in 7m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1d8130360074c50a190a2ebcad58169 : SUCCESS in 13m 05s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"4740d4f3dd6d60b37c4c9f63350d60604f30d521","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-02-12 08:07:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"71671a475df94ff989a661be78498eba74aff53b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 08:22:17.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b457f2ea73cd42efad65366fab16113a : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bfe4bf7ba0343888c6cf4bb4035698e : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd1cf84fc93f4ddea64a25f8314acfa4 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/714e7017e59042b4bcdb65376dfc8382 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36a6a12b2e274b0392b0513cf357acac : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/029bcd4716bc40b6b4592d50dccbe64e : SUCCESS in 6m 21s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ff9c1be2a2f43bc8eddba1d307bb01f : SUCCESS in 12m 49s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"51adb083693372088a441d0961bebc4763df4763","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-02-12 08:40:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ac6dbfa9850f3ae0b27c18c937f2f0b57c9ed21f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-12 09:18:14.000000000","message":"Patch Set 6: Code-Review+2\n\nThanks!","accounts_in_message":[],"_revision_number":6},{"id":"5658738dd6c6163cf421a966762d03e47253053f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 10:43:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f17f8394a8044f48881d80e6d63f560a : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e34bba91c7a04e7aab3cfad7a1861b34 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ced428433cc4b05b49d037cd42e547c : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae536471d4b74bdb8fec97f0cf3db938 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4961160a2a484c6b8e90fe1e90d9c237 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b44d4d5e29ed4ec3880961e5360791fb : SUCCESS in 8m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7b75b62ac5fa4edcb2bc767797592d3c : SUCCESS in 5m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4618644fa44443eb819dccbaeb43b75c : SUCCESS in 7m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8b9fd2a5c37a41eb96ff08935b6f6542 : SUCCESS in 19m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7bb5183e156e44b2ad3c4b5984eca03c : SUCCESS in 59m 38s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35ee1f16d6394193a50f24540d225694 : SUCCESS in 46m 04s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/b9edffd230a442698aa46bcb2921528a : SUCCESS in 1h 11m 54s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4a607927b7694bf18a515d4f06dbf825 : SUCCESS in 16m 29s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/df0f542c123e4d90b93e2d4671e2db30 : SUCCESS in 1h 51m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e3d73cb58afc45f08b997546f08e97b7 : SUCCESS in 16m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d8a19d251b1491ea14293f2fcbbb0e8 : SUCCESS in 11m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8b270e18510b46769d9d513c3f22d57a : SUCCESS in 23m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/04393bf9f51d4ae5bc454386090ac0ac : SUCCESS in 9m 19s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":6},{"id":"cee435771170d29fbdcb076980cd63c28833a530","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-19 13:44:45.000000000","message":"Patch Set 6:\n\nrecheck\n\nI expect this to fail now due to IPA build tests.","accounts_in_message":[],"_revision_number":6},{"id":"95dbe3a4e303be9bf0b0ad14da7d322d8e03dd37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-19 15:39:39.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af9dd350a13a41c7a382f26e0b8ebf7d : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44a4fa996f674111ba2dd1ebd9db9687 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf2da3d0d39e4029b954f46df6c77739 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/224fa1a66b6d4d80a899f32ba03c916a : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82e37790c7844ed89e8537532079e606 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfb44cdf08564c2ab21a5f047ef5a157 : SUCCESS in 7m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/01da49b9a58645cbba863765323c7f03 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ea358ccbc0f4abebe8d9a1c7f99e264 : SUCCESS in 12m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b04206fd5d844dd3bacf537d90c17b78 : SUCCESS in 17m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/562a975e7aad4fe981f05da0edfade76 : SUCCESS in 55m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/623da3141cd54ac9bb23bef829377e07 : SUCCESS in 44m 14s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f68701f3a9d34fdfb79db4dbb4a07c5a : SUCCESS in 59m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d596791f1e9740a0ba8b4680fba8cc8c : SUCCESS in 14m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9a58057d1e52434e9e413fd685d87875 : SUCCESS in 1h 37m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/19ae01cf04d0474a8799d8929fe8c5be : SUCCESS in 27m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8483f9b1a64445d0ab4b06683e646b0a : FAILURE in 13m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/56da186993034e86b1491b62fc7b6b59 : SUCCESS in 24m 37s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0134d6301c2245e9a07b1767c0ea2b86 : SUCCESS in 9m 57s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":6},{"id":"4a3790b3b3e3e5a33c4d0e09fa41d00baab7721a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-19 17:14:20.000000000","message":"Patch Set 6:\n\nIt should work with Idaedcc58e00d06caf4e5e360011d1c5c5c7dbe44","accounts_in_message":[],"_revision_number":6},{"id":"f0c987a811144509cc6d29eea4067065c53896f9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-22 13:53:06.000000000","message":"Topic change-767705-31 removed","accounts_in_message":[],"_revision_number":6},{"id":"5e891103d75371f37553a5abb0c7298f1c7f1344","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-22 13:53:14.000000000","message":"Topic set to story/2004960","accounts_in_message":[],"_revision_number":6},{"id":"116345b564e26332ca235926be18a675571c4c10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-22 13:54:59.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"2e90939281f065a3fb56554336556f2f88113141","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-22 13:55:39.000000000","message":"Patch Set 8: Commit message was updated.","accounts_in_message":[],"_revision_number":8},{"id":"b9f3637c56327b8c7ff759c04395378cb3b2568a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 15:27:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a674e099b6624f6cbc493739317b3276 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/77a97343552c4ba88df1e5206d6603e3 : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9745f62d81e49e2aff408e4f442345b : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c91a8b42447f4af6b4c8a69b758c9144 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5939f2ebc9d042de84cbc3bf0d384c6d : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73e2097f334d4bc2b45457ec6c64b0e0 : SUCCESS in 6m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/73e5a7ca433649a1af09776db3676194 : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ee6f1713afea437189c41b25b5200f9e : SUCCESS in 8m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8edc6faf57054840a2d0623cdce006e3 : SUCCESS in 18m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fa9666923d814c31877f66df42f9e46d : SUCCESS in 51m 24s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee9712875e784b6c95ccd2a3ef9ab5ba : SUCCESS in 48m 01s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f598dc9703654b5aa82f581ccec2d083 : SUCCESS in 1h 09m 00s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8d56ff3f8f644e32ab8608eaa54c84bb : SUCCESS in 13m 30s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ea7f7cc8e3564784ad4a7c497d929ccd : SUCCESS in 1h 29m 00s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/02d13b27bfd9406ca7cde2186584253a : SUCCESS in 28m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66f6f797e1474aba8077e845c381901d : FAILURE in 13m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9eb7de06a84d4cbfb110fe3c1219766a : SUCCESS in 20m 12s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b392a018611a46bd8c5a5010564978f2 : SUCCESS in 10m 47s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":8},{"id":"244265962be5c52a97721cc8ed2e51a0879d5d73","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-23 09:03:21.000000000","message":"Patch Set 8:\n\nLooks like it\u0027s trying to build a CentOS DIB image: https://5c1bd3075fbf5f1be9e8-1e03f6a0971577a3603e6088d0a330da.ssl.cf5.rackcdn.com/775267/8/check/kayobe-seed-ubuntu-focal/66f6f79/primary/kayobe/ipa.stdout","accounts_in_message":[],"_revision_number":8},{"id":"cb5748e30a5159b6ce22b4bb14605b03c16e6ab6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-23 09:05:50.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"d4e926a6f17f0432584709c990fcddc9ebac442e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-23 10:10:39.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"ed195be73296f3e042d0b65296bd40c1a3c514c9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-23 10:16:48.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e80b53a4805f06056aa112d6ec49c40689601873","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-23 10:32:30.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e5b1b9c3887be83edd231060c166d51697fd30db","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-23 13:05:40.000000000","message":"Patch Set 8:\n\nrecheck\n\nThe Depends-On patch has been updated to ensure kpartx is installed.","accounts_in_message":[],"_revision_number":8},{"id":"24aca9807624ca2cfa76afa86ac187a730c4f47d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 15:15:03.000000000","message":"Patch Set 8:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ed375134c04464db9f33384bb913b05 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aa1e82f40474eb88a055d48616b15a6 : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a598a80351a64aa49b832a815fb65a67 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bba26f5f8a6e4411aea2d3ff5a663d5d : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0708d05ac3f44c191cbe6aee5f6ada1 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39af18f4bbbf4c94b5cca0a1d1c5eab4 : SUCCESS in 6m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/631962ccd93c42b581456efbe292d018 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0df301d3aa714547a89a79181d1a047a : SUCCESS in 8m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/67698009f36f49a5a368e4431db0d7e7 : SUCCESS in 20m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/61fe8ed5f1674bd596db01de8a6f7fb8 : SUCCESS in 53m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b389df1bde8e45d8a8c04eaada757a5e : SUCCESS in 41m 31s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/a9975254766240478f2f4a0890a56254 : TIMED_OUT in 2h 01m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b5b0bc7a5278450a89b058d324599c14 : SUCCESS in 14m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/762c3428687748998589d4eb636487ff : SUCCESS in 1h 33m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bd8e946798304e0a8f62488b2fe80a38 : SUCCESS in 26m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b38cbb0ee9ae40d0a7d6d840ef47454f : SUCCESS in 25m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1f13a15f4fa5411dbbfa7a23bab340c1 : SUCCESS in 23m 35s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/72b7318f5f6340739b62eefe26a1262d : SUCCESS in 10m 56s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":8},{"id":"b606755e6f01d9f9aa194fa1d05cc6deb7015dcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-01 17:59:37.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"f1a7e3516c5ba799fa8e6f7ef00045de81c195fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 20:28:18.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57653a97952547aa83e73d910cb359fa : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06f66653bbde418abee9ae7dc040789b : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c4e8f64588d4a3ab220ebd7460150c3 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b2b2af249204e3eb9cf6df602e5e9b3 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/11c67b8683dd4652b66ab3144d652756 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb5be4711c2d4157b6ee90c97a505367 : SUCCESS in 7m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/79f0dddc9e934077b52dbed6ffafc7d5 : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a6dd0109dadb4296938805a8e7f83345 : SUCCESS in 8m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/50702fec4a564e4e89bdcd2044f6fab2 : SUCCESS in 18m 47s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cbcdefde4b694d2891c7c5eb9e9e73cd : SUCCESS in 55m 13s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/57cbb6a540324db999577c03eb4c9143 : SUCCESS in 40m 28s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/420032fcccc94e1296e4524087ec171e : SUCCESS in 1h 02m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/67a01d0d9096433eb03301f291e06ac0 : FAILURE in 16m 26s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ca9f3fce7cd74a82b56e5cfacff20bcd : SUCCESS in 1h 50m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/818e5f6c23ec4fdaabcb3c3229b0a2b8 : SUCCESS in 24m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c51639be0f1462e9eedaa82adc1af1b : SUCCESS in 24m 22s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4207c55e3fba455380f1d6d141573a5b : SUCCESS in 24m 25s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4c1aa18d450d401290eba5f0da1fb13d : SUCCESS in 12m 31s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":9},{"id":"e49d754d00cbd15c87759f407c1a6ab727ab8966","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-02 10:44:16.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"56e09a644a42137759ea7f2ff42fc38ec76c8bf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 12:05:20.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c1afd7a95524901bf83ed76091a950d : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4765efd8dfe64b3e8d8b33635c6fd0af : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6c7538e5a2941a3886639a6b5c358e4 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d490b008ace4df8ba068f3d45ea3215 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dde7d387de08400ca407d0305d000ba2 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b15d50b0924840f18ad794f23e42375a : SUCCESS in 8m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/205dcd4ee9014b889d8ae112e3a6c03a : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6cb2a8c0a425418bba851b7678abe533 : SUCCESS in 9m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e8c563d699ed40a4a3737a308a1ae9c2 : SUCCESS in 17m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4fde23e8055b441e90191389bb9783aa : SUCCESS in 57m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7c735f005874a0b9f5515bd9216ce61 : SUCCESS in 41m 34s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/91f9d6bbfc7a48af98fb73a0f9dc6b22 : SUCCESS in 57m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fdccb1cfdfd94fb6a38ca0d558c8d661 : SUCCESS in 15m 03s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e248fc40e4e4498cb17abd76e84a4d3d : FAILURE in 1h 14m 19s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8b6f93c2b0dc4cce8b2b8e2a252f4bb6 : SUCCESS in 26m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cd7bbe120644411a42d9a5e601009f7 : SUCCESS in 25m 15s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fb9b6534f18e480fa370a449a5c98ce1 : SUCCESS in 25m 21s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1e249ddc12594a27a8ffb811f577d193 : SUCCESS in 12m 10s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":9},{"id":"0353514dbad4bce91d75612391926e0c3e3b3374","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-04 08:42:48.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"8c83d968d604fb773bff7d7214e6dc84b7511455","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 10:26:58.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9c22beba8b64c32ad80c6675e16ff4f : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c30bb38cd65492a80848bf57cf0c9cc : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56cd8566ca82488e8edc24d86d94ddac : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3266474c68d54484958b9e942339d6f1 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dba2ff5855844343b7198209f6bd50d2 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8182869dd436447ead0083427fa322fa : SUCCESS in 6m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b40aadc1cac438f8621f3984702dda4 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c5556334c644471f9b34f854cf78ed93 : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7fb77a932cd84c47a74d64757ed85731 : SUCCESS in 14m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cc95607353aa422c99c22cd932ad8441 : SUCCESS in 54m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caa3de9488ce475eb6048e4dc4c164bb : SUCCESS in 41m 33s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/8f85f40cdfdc45d6962ae4b27580d072 : SUCCESS in 59m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/168f3aa642a74893855716a0a0314689 : SUCCESS in 13m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f8c9679d216b4b2385bf60434c87f565 : SUCCESS in 1h 39m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c7cebb2081f64fc98c28d7745e88fef6 : SUCCESS in 24m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9aab7699f26342f791e7b54d0daf3b65 : SUCCESS in 22m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/28855114206e43ff9869731c055dcc64 : SUCCESS in 21m 23s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/35f3a4aee8bd4ad387ae3d97d186039f : SUCCESS in 10m 20s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":9},{"id":"fa5cf4db9086f624fc08f76bd7405e78db926927","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-04 10:31:23.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"a372e68734b3f81fc88e7471b1007620599a6f84","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-04 10:42:32.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"a6482a9ac3747876ed4d50ffa55efe59f6b19b37","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-04 11:09:04.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"d9642ca2a34bd03c82f14dd88639126b3e413684","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 11:13:58.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"e68968d3186d50585d61355941196de3d085acca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 16:20:25.000000000","message":"Patch Set 9: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":9},{"id":"716412ca750c32a05ebc401e0b5c2c24fa03d2f5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-10 12:07:55.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"b0b91ac6249083aba69887d748ab742ee751479e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 14:00:38.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4b19295471624699b6aa9ee0145633dd : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88ee8dd707834762846e18b1d06428f9 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/23df4e393232419995c8a84ab119b432 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/405ae17bcb944349921f9ff0d91462d5 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7196d8f631314c70b36478b863899c5e : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec6ddc5fe35648d088f38242270474d9 : SUCCESS in 7m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a227ade42acf4d9781b58dfbbb59162f : SUCCESS in 5m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f384f4b88cfc44049ee010dc5649b071 : SUCCESS in 10m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2509d28217ff449a96a0e1494a5f8cad : SUCCESS in 19m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/27509b1e9e32463cbd84957ab2596a99 : SUCCESS in 1h 02m 22s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43380043cf50489d8c4658e396638e74 : SUCCESS in 44m 24s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ce362fee7299470d98d98498901b35e3 : SUCCESS in 1h 01m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7b3a5bd92efd4475a9365e9bc14453b1 : SUCCESS in 17m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6fe459938661458bb5456c021e9f89be : SUCCESS in 12m 03s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/799f910f1faa46bdb89c4d44c1536939 : SUCCESS in 1h 42m 15s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ff031b091ef54a4f9c22b93a1cf0f944 : SUCCESS in 26m 43s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a02b29c6595f4182b65701bc9d09dcbf : SUCCESS in 25m 15s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fedbf7437e5042acbf17a85f9fbc3445 : SUCCESS in 20m 52s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e24e11a023bd4aa0b11bdf04eda8d865 : SUCCESS in 10m 21s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46191ff564bb4c94b5de81d4f0175a22 : SUCCESS in 8m 26s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":9},{"id":"2dd8055c67a12e58c4c0a5d47fa28e889faf81ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 14:00:57.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"5d776cd6070ecea65172b96e91212f6b6cf1b785","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 16:07:45.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96a152aafda9471a80b52db1b4be78ad : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3155f627cfa4630af61d8dda02ecf5e : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b443aa4e98dd4c3eb6d507e017984f76 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e13ff1b973a4967b0ae144c0d0db089 : SUCCESS in 9m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/16ec4a5e5d9448c5b084aed1e792c6de : SUCCESS in 7m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da42397a74e64b3a90123856bfce9961 : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eae862baf7be4dd4b6d2c0d47305d9de : SUCCESS in 20m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/64fab1d4b23747a08341d3aa69c51f45 : SUCCESS in 53m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/97d979d9fd42404498b1f822b6b18926 : SUCCESS in 48m 37s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/fc053e3e79c345a1a714bee934a13357 : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5e161c5b117747c7ab22e9133baf3cfa : SUCCESS in 20m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/654d2a7ca46a46578342b38e4eaceca1 : SUCCESS in 10m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bb56f139d4fb47f9960e8c2ae6fff627 : SUCCESS in 1h 21m 15s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/28c762879fad4f14a216f76c71af2759 : SUCCESS in 22m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/095f177a0f6c47fc8e143bcaabeaaa10 : SUCCESS in 26m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ff8df0c50221472a85262ff869f67df9 : SUCCESS in 21m 22s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7ca60e2856a64b58bc6c137ec178460a : SUCCESS in 12m 07s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/270d5fed9aef40d4be245a2d7f7c3ce9 : SUCCESS in 9m 24s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":9},{"id":"3e5ebe54ce4bb015d93084abcd040a72f8b14036","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 16:07:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"0c470cb69e7b25af0796eb700ae82a3201182076","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 16:16:03.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9949754258842dbb73e73e60a7387bc : SUCCESS in 55s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"534465797b4aabead7ddd5d20430101c0f64572d","revisions":{"ab4461a3bd5f82c6415218b773f66bee8d979c2a":{"kind":"REWORK","_number":1,"created":"2021-02-11 19:45:30.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/67/775267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/775267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/775267/1"}}},"commit":{"parents":[{"commit":"8762af4ae520f6af7cb559915526d469cf27528b","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8762af4ae520f6af7cb559915526d469cf27528b"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-11 19:39:24.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-11 19:39:24.000000000","tz":60},"subject":"[WIP] CI: Add Ubuntu seed job","message":"[WIP] CI: Add Ubuntu seed job\n\nThis commit adds Zuul job for deploying seed services against\nUbuntu 20.04 LTS (Focal Fossa).\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nStory: 2004960\nTask: 41807\nChange-Id: I839cd1b9d63c9bbf3e0c75def76b8bff68637c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab4461a3bd5f82c6415218b773f66bee8d979c2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab4461a3bd5f82c6415218b773f66bee8d979c2a"}]},"branch":"refs/heads/master"},"dfd24e3cb891fb661d71c62f3b0f96385a9a4ac0":{"kind":"REWORK","_number":2,"created":"2021-02-11 21:32:20.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/67/775267/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/775267/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/775267/2"}}},"commit":{"parents":[{"commit":"8762af4ae520f6af7cb559915526d469cf27528b","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8762af4ae520f6af7cb559915526d469cf27528b"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-11 19:39:24.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-11 21:32:07.000000000","tz":60},"subject":"[WIP] CI: Add Ubuntu seed job","message":"[WIP] CI: Add Ubuntu seed job\n\nThis commit adds Zuul job for deploying seed services against\nUbuntu 20.04 LTS (Focal Fossa).\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nStory: 2004960\nTask: 41807\nChange-Id: I839cd1b9d63c9bbf3e0c75def76b8bff68637c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dfd24e3cb891fb661d71c62f3b0f96385a9a4ac0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dfd24e3cb891fb661d71c62f3b0f96385a9a4ac0"}]},"branch":"refs/heads/master"},"c9b091be5e787a55e8d23839dfbee10942e8c7b0":{"kind":"REWORK","_number":3,"created":"2021-02-11 22:01:25.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/67/775267/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/775267/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/775267/3"}}},"commit":{"parents":[{"commit":"8762af4ae520f6af7cb559915526d469cf27528b","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8762af4ae520f6af7cb559915526d469cf27528b"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-11 19:39:24.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-11 22:01:15.000000000","tz":60},"subject":"[WIP] CI: Add Ubuntu seed job","message":"[WIP] CI: Add Ubuntu seed job\n\nThis commit adds Zuul job for deploying seed services against\nUbuntu 20.04 LTS (Focal Fossa).\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nStory: 2004960\nTask: 41807\nChange-Id: I839cd1b9d63c9bbf3e0c75def76b8bff68637c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9b091be5e787a55e8d23839dfbee10942e8c7b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c9b091be5e787a55e8d23839dfbee10942e8c7b0"}]},"branch":"refs/heads/master"},"821ff2b4baedfb468d6766213dc6843e78df1e6c":{"kind":"REWORK","_number":4,"created":"2021-02-12 07:42:27.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/67/775267/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/775267/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/775267/4"}}},"commit":{"parents":[{"commit":"8762af4ae520f6af7cb559915526d469cf27528b","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8762af4ae520f6af7cb559915526d469cf27528b"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-11 19:39:24.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-12 07:42:15.000000000","tz":60},"subject":"[WIP] CI: Add Ubuntu seed job","message":"[WIP] CI: Add Ubuntu seed job\n\nThis commit adds Zuul job for deploying seed services against\nUbuntu 20.04 LTS (Focal Fossa).\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nStory: 2004960\nTask: 41807\nChange-Id: I839cd1b9d63c9bbf3e0c75def76b8bff68637c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/821ff2b4baedfb468d6766213dc6843e78df1e6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/821ff2b4baedfb468d6766213dc6843e78df1e6c"}]},"branch":"refs/heads/master"},"5609161eb678ea86a6c4df94e3caaa3d0956f149":{"kind":"REWORK","_number":5,"created":"2021-02-12 08:07:16.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/67/775267/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/775267/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/775267/5"}}},"commit":{"parents":[{"commit":"8762af4ae520f6af7cb559915526d469cf27528b","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8762af4ae520f6af7cb559915526d469cf27528b"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-11 19:39:24.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-12 08:07:06.000000000","tz":60},"subject":"[WIP] CI: Add Ubuntu seed job","message":"[WIP] CI: Add Ubuntu seed job\n\nThis commit adds Zuul job for deploying seed services against\nUbuntu 20.04 LTS (Focal Fossa).\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nStory: 2004960\nTask: 41807\nChange-Id: I839cd1b9d63c9bbf3e0c75def76b8bff68637c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5609161eb678ea86a6c4df94e3caaa3d0956f149"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5609161eb678ea86a6c4df94e3caaa3d0956f149"}]},"branch":"refs/heads/master"},"3d78f70f34d87f9e02c89e070dfe6446afbdb3a9":{"kind":"REWORK","_number":6,"created":"2021-02-12 08:40:49.000000000","uploader":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"ref":"refs/changes/67/775267/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/775267/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/775267/6"}}},"commit":{"parents":[{"commit":"8762af4ae520f6af7cb559915526d469cf27528b","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8762af4ae520f6af7cb559915526d469cf27528b"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-11 19:39:24.000000000","tz":60},"committer":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-12 08:40:29.000000000","tz":60},"subject":"CI: Add Ubuntu seed job","message":"CI: Add Ubuntu seed job\n\nThis commit adds Zuul job for deploying seed services against\nUbuntu 20.04 LTS (Focal Fossa).\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nStory: 2004960\nTask: 41807\nChange-Id: I839cd1b9d63c9bbf3e0c75def76b8bff68637c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d78f70f34d87f9e02c89e070dfe6446afbdb3a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d78f70f34d87f9e02c89e070dfe6446afbdb3a9"}]},"branch":"refs/heads/master"},"4712c974b09d5fa1864de2b484101559b71b6aee":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-02-22 13:54:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/775267/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/775267/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/775267/7"}}},"commit":{"parents":[{"commit":"9b5fe8509fbe29f0ce67fef0035892e3f312fceb","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b5fe8509fbe29f0ce67fef0035892e3f312fceb"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-11 19:39:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-22 13:54:59.000000000","tz":0},"subject":"CI: Add Ubuntu seed job","message":"CI: Add Ubuntu seed job\n\nThis commit adds Zuul job for deploying seed services against\nUbuntu 20.04 LTS (Focal Fossa).\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nStory: 2004960\nTask: 41807\nChange-Id: I839cd1b9d63c9bbf3e0c75def76b8bff68637c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4712c974b09d5fa1864de2b484101559b71b6aee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4712c974b09d5fa1864de2b484101559b71b6aee"}]},"branch":"refs/heads/master","description":"Rebase"},"40a4a42fdac2ca2f667c5d6a1427b61b6faa6f8d":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2021-02-22 13:55:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/775267/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/775267/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/775267/8"}}},"commit":{"parents":[{"commit":"9b5fe8509fbe29f0ce67fef0035892e3f312fceb","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9b5fe8509fbe29f0ce67fef0035892e3f312fceb"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-11 19:39:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-22 13:55:26.000000000","tz":0},"subject":"CI: Add Ubuntu seed job","message":"CI: Add Ubuntu seed job\n\nThis commit adds Zuul job for deploying seed services against\nUbuntu 20.04 LTS (Focal Fossa).\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nStory: 2004960\nTask: 41807\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/772929\nChange-Id: I839cd1b9d63c9bbf3e0c75def76b8bff68637c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40a4a42fdac2ca2f667c5d6a1427b61b6faa6f8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40a4a42fdac2ca2f667c5d6a1427b61b6faa6f8d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"534465797b4aabead7ddd5d20430101c0f64572d":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-03-01 17:59:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/67/775267/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/67/775267/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/67/775267/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/67/775267/9"}}},"commit":{"parents":[{"commit":"9f41cefc15e240602d2c6ef9076e36135e9138e4","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f41cefc15e240602d2c6ef9076e36135e9138e4"}]}],"author":{"name":"Piotr Parczewski","email":"piotr@stackhpc.com","date":"2021-02-11 19:39:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-01 17:59:37.000000000","tz":0},"subject":"CI: Add Ubuntu seed job","message":"CI: Add Ubuntu seed job\n\nThis commit adds Zuul job for deploying seed services against\nUbuntu 20.04 LTS (Focal Fossa).\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nStory: 2004960\nTask: 41807\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/772929\nChange-Id: I839cd1b9d63c9bbf3e0c75def76b8bff68637c75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/534465797b4aabead7ddd5d20430101c0f64572d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/534465797b4aabead7ddd5d20430101c0f64572d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
