)]}'
{"id":"openstack%2Fkayobe~775327","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I4719b4d397c01b35c78cb84c6d686dd27742d1c0","project":"openstack/kayobe","branch":"stable/victoria","topic":"test-seed-ipa-build-stable/victoria","hashtags":[],"change_id":"I4719b4d397c01b35c78cb84c6d686dd27742d1c0","subject":"Test building seed deployment images in the seed job","status":"MERGED","created":"2021-02-12 08:29:35.000000000","updated":"2021-02-22 12:08:11.000000000","submitted":"2021-02-22 12:06:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"775327-test-seed-ipa-build-stable/victoria","meta_rev_id":"09c739e8c3019347228423d4736007577cd2f339","_number":775327,"virtual_id_number":775327,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-22 12:06:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2021-02-22 10:26:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2021-02-22 10:23:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2021-02-22 10:26:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-12 13:15:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-19 10:55:58.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2021-02-19 10:55:58.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"25e07c9a43cbb0aba6c2ae7a9a84e6653eac84b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-12 08:29:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75df30d1e640f5879b3cd3fd9cec5982c8a314a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-12 13:15:07.000000000","message":"Patch Set 1: Verified-1\n\n(18 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56273c7b8de84c548ac4932f66422722 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc1e7fc9f4174c9cac2a495e55c4ffb6 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6316fbd7a9bc47048bc04a6156c7179b : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/571b713c5f9c47b2a6843031b0d03d4c : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d6c44885d964da9b5ce4fd7eca6964e : SUCCESS in 6m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fd4818f2ceb04b0e8f534c3626c862b2 : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/13a955b28aa5457dbda4f325222dbf49 : SUCCESS in 7m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2ca7f20ed32a4dd8a4940735dbeeed15 : SUCCESS in 20m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/14f820ff5d3e4fcab5d654d18b5615ac : SUCCESS in 57m 50s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/eada374a7aca434b86cd48597d4e4063 : FAILURE in 10m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/35f7cb84df414cbf9a13fa07a1fc1cdc : SUCCESS in 16m 53s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a20cbd5dca44405d9f3646e6862d87fd : SUCCESS in 1h 47m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5df68020353c4050b20a8b1d10785244 : FAILURE in 9m 49s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5b0f986f942d4632a4d716ad08cacb12 : SUCCESS in 24m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ead5bb14e8724cc0a87694de60ec5844 : SUCCESS in 11m 19s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"b955059aea3721ebb5c345d4a0e8d12aa57bbe16","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-16 20:43:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"28a3bac9db9c269a9ba2edfd47f4a8fcd8ce06af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 21:10:05.000000000","message":"Patch Set 1:\n\n(18 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64fef5add9ec46abaaabb8d8b2070cf6 : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b552efb6c8040888a347b48f7039163 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7900ae13ee84474780dbf391a4362edc : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4fa56ffaf0c24d2e8c6929fd85f7e026 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e84de7a70544009f7cb96228c4b1c2 : SUCCESS in 6m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0fcb8ef72d8944cf91746d62f4343c5d : SUCCESS in 3m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8220f16aaa2a4d76b0edcc8138e7a20e : SUCCESS in 7m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/886d688063704758ba5ad701d149d94f : SUCCESS in 15m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/bf467fbf237644d089c3b6dba5566cf1 : FAILURE in 11m 08s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ab6674f77f764d0d9b3a7535ca1eed2b : FAILURE in 12m 03s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e5849eddf8644a259b839f400ddca40b : FAILURE in 13m 20s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/db8b33fb2be841888b7da63ee1eb2eab : FAILURE in 11m 42s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/16690eacce8140b7a7b042f56317f5b8 : FAILURE in 10m 23s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5fe6885b883c48d69ab04b83c1df9bd4 : FAILURE in 11m 08s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/3986683f8d944efbb8ea7e509da12295 : SUCCESS in 9m 17s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"d55ee3077d910a2c0e078a8f234370f804bdca6a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 06:53:10.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fd89bb624ecea37175e15c8236ecacf895390a11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 10:14:14.000000000","message":"Patch Set 1:\n\n(18 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b07e1650d8a4614979bb16b5638b11b : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b3920ca8a09487988e3bbd5c28f17fc : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b759c2ffcb1428b966b74f1351a67de : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30562054c0bd41c09ee633a6600ee10b : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13a5ab70935642d68f80800d71cf9f37 : SUCCESS in 6m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c74fd85b130a4f0695267276c267b514 : SUCCESS in 4m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e2e08bc2563b4189b36e32a7a9eb6291 : SUCCESS in 8m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5c826a88efc542a399d8489cc6145c65 : SUCCESS in 18m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d62d5871f5c040ef9df7396f3878df58 : SUCCESS in 56m 03s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/7b4e43b088cd42d5b2f4f85acd18b543 : SUCCESS in 1h 09m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/addedc9c14ed4090ad0aee9087bc84c0 : SUCCESS in 15m 19s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8722e28ed41746d2a96ecebc584539f5 : FAILURE in 1h 46m 24s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cb33841d9a5a487ca3459987d8d8f550 : SUCCESS in 27m 21s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9cc7c46675bc498ea1a0ce0e59656b81 : SUCCESS in 24m 40s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ef92d3f17ce7411d8c2e8f509ce81793 : SUCCESS in 10m 14s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"71a91b50c3c6127dc6d37b97ebe80dff0eb49832","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-17 11:26:20.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"373943a72b2dd8ab6d29b569cde8e16067f5591c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 13:14:35.000000000","message":"Patch Set 1: Verified+1\n\n(18 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/383dc3d4b57a48a49d51ae72ffdbe553 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3484dc56116437d8f7916647994f812 : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b14af84fe2604cc497d9522ad34bf12e : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6071f0915648431ea0d834259b2dba41 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3236e25c3e624078a50071080becbb41 : SUCCESS in 6m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b68e85ec6da46d4aa4a65adf9543a08 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ede43bbf13554b8f8072c8edf57c2bda : SUCCESS in 8m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c80a84577e744e7d95d0d9a3bbf95bab : SUCCESS in 19m 44s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4d6a94e8a9da4efe853365d72e1b7475 : SUCCESS in 59m 12s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/6159353d9f214a56b2c0cefb940600ae : SUCCESS in 58m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e063ac42096044a2b5374e11af698134 : SUCCESS in 16m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e5ecddfc0d154e5d813ec947108580e1 : SUCCESS in 1h 45m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c7389b68207f4bbfb66889b25200229f : SUCCESS in 28m 27s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f502d72a0e4a4c1bbe370c1b8c7a76e4 : SUCCESS in 23m 36s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2e0e74d084864610b2c2d66bb4b0480e : SUCCESS in 10m 37s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"9ba8baaa900504d2d9a6e1e678332071138c67aa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-22 10:23:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dbf60661c8d1af3452792c6cf01a95aa07ead1e4","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-02-22 10:26:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"65ad689fbdb6d2e57005afa3012a49101af2a9a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 10:27:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e87f7f9b74ce9c09fd9a44e4c83d8b9bd7d58ed9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 12:06:42.000000000","message":"Patch Set 1: Verified+2\n\n(18 comments)\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66c1996ef34b4e98b3c1aba16a7d574e : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e37ddabcd1664a14b13295cf7aaa8a5a : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe4ab8ee94fe452da82dae58099bfa7e : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef84f3605d914603a5658530c97c12bf : SUCCESS in 7m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a093dc70e67a4bdea9172dd885047e72 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3738cf00aa534dc78d34a65db76c9ae8 : SUCCESS in 8m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/63b2f71e3fe945e99a9e5a078718bcdd : SUCCESS in 20m 14s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/26648dde7f0c4cdc8ec0c1f6f512edde : SUCCESS in 1h 00m 09s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/d0263b8dd3964ec49101a8e9d96041fb : SUCCESS in 1h 18m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c1b5404fb2b6426d8375daf94d1696b9 : SUCCESS in 15m 00s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/81dba0a412f34fd1a47e28eb595c69b4 : SUCCESS in 1h 36m 48s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f0e3d8d97fd24e7e921fd424bfc8bf50 : SUCCESS in 23m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/be876bc61bb54d35a3a16e008b7c8a12 : SUCCESS in 24m 47s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/614c153c9bde4950a21e60393cec9db3 : SUCCESS in 9m 29s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"b331242d8712379530d8f2fd53d8e3aa235858cd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 12:06:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"09c739e8c3019347228423d4736007577cd2f339","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 12:08:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba7fd13156064ef9a8af7be5eabd3e6c : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3861f20daacc4c156e9e9efec3ef9fcbc6ef43f","revisions":{"d3861f20daacc4c156e9e9efec3ef9fcbc6ef43f":{"kind":"REWORK","_number":1,"created":"2021-02-12 08:29:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/27/775327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/775327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/775327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/775327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/775327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/775327/1"}}},"commit":{"parents":[{"commit":"1622b9dc5cf64b233a16a85a1c357e1173c37746","subject":"Merge \"Adapt to changes from the Bifrost Victoria release\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1622b9dc5cf64b233a16a85a1c357e1173c37746"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-05 10:49:13.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-12 08:29:17.000000000","tz":60},"subject":"Test building seed deployment images in the seed job","message":"Test building seed deployment images in the seed job\n\nThis requires stackhpc.os-images v1.10.0 or newer, for compatibility\nwith CentOS 8 when SELinux is enabled: we disable SELinux, but without\nrebooting it stays enabled.\n\nThis Ansible role was updated to v1.10.2 in master and stable/victoria\nby I5efdbd52556721914fe69d7c6ba454b2c721b643, for another reason.\nRemember to bump the requirement when backporting to earlier releases.\n\nIt also needs changes in the way we interact with Bifrost to avoid using\nthe env-vars file which has been removed. This is implemented by change\nI25078e69acdf41a4ef9957f99fe5047de54b778d.\n\nFinally, it requires building seed deployment images only after\ndeploying Bifrost, because the task copying images onto the seed expects\n/etc/kolla/bifrost to exist.\n\nWe also copy log files to identify issues when the job fails.\n\nChange-Id: I4719b4d397c01b35c78cb84c6d686dd27742d1c0\n(cherry picked from commit c84a9757dd449dbd8863bb1dcc55837abe26a716)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3861f20daacc4c156e9e9efec3ef9fcbc6ef43f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3861f20daacc4c156e9e9efec3ef9fcbc6ef43f"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
