)]}'
{"id":"openstack%2Fkayobe~776471","triplet_id":"openstack%2Fkayobe~master~I50f75f8f8ed200b1c70f834973463c2f7343ed1a","project":"openstack/kayobe","branch":"master","topic":"show-output","hashtags":[],"change_id":"I50f75f8f8ed200b1c70f834973463c2f7343ed1a","subject":"Add output printing to host-command-run","status":"MERGED","created":"2021-02-18 14:33:45.000000000","updated":"2021-04-02 15:34:54.000000000","submitted":"2021-04-02 15:33:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"776471-show-output","meta_rev_id":"7dc2139dc62e07dfd211a05c3b77b07bda79fa39","_number":776471,"virtual_id_number":776471,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-02 15:33:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2021-03-31 08:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-02 13:24:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-02 13:24:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}],"CC":[{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 17:04:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-23 10:19:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-02-25 09:47:37.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2021-03-01 10:06:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-26 13:39:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-03-29 07:12:26.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"}],"messages":[{"id":"0818a57a04ab3b339e8bcee53288cbae658e5955","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-02-18 14:33:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0de7698eeca16aa1e75b02204b52ba2fa64cfbf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 17:04:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efd759cc3adc4ac98d1b498cf742d51c : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4259b0eafc7d4746902b8a49492645ff : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8e693412ab5404b8ff84a2748765453 : SUCCESS in 5m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/583d0a169f5d47308c4165dee608cd39 : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4900b0d7d5254a209e42253aee0217e4 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c57ab47bf1a4e7dbd017ee895a9ce34 : SUCCESS in 6m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/90bc32d4c3704a82b638080707b6aded : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/583db3190e1840e1888500fbaba82c8a : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fb2f7454c50e4b7ab11670f9ace8f875 : SUCCESS in 18m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/39e091cd706d4bb18e27dd8adfbb3482 : SUCCESS in 55m 26s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/8c91fe45e98a4713bd532fad4c152480 : TIMED_OUT in 2h 02m 04s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b72a9311f09d4809bd6a7207457464ef : SUCCESS in 15m 56s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/edca8c789e7646b1afca25f3fa916e12 : SUCCESS in 1h 46m 51s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9bb703e1fde34601933e0d0d57643e37 : SUCCESS in 27m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1ea8e4be74264f508dc367dcffb6a420 : SUCCESS in 24m 35s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/33f78221a81a40a19ec4ee1c32753bc5 : SUCCESS in 11m 40s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"e070f268cd96079fee36998ecfb2a7bf60344fb0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-23 10:19:52.000000000","message":"Patch Set 1:\n\nOvercloud TLS job has insufficient memory in ironic. Should be fixed by https://review.opendev.org/c/openstack/kayobe-config-dev/+/777076.","accounts_in_message":[],"_revision_number":1},{"id":"342858d333d48fb417d23747cfcceb24df1a5d7d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-23 10:22:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bc5f70942ca4be51be258a6c638f0849f104fdf","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-02-23 17:09:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ac4f960df4562abe649513aefe32a39afa0cb4e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-02-23 17:14:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"54123f11cb66540f781bcfdbb653b473eae157f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 19:45:20.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79f1b9c283494553885526d7972c1a35 : FAILURE in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/852da11d7cbc4d11ae3cdc1a10bcaf2f : FAILURE in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9cd2b0198af4440894b53948d9293a70 : FAILURE in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/982dc34f0a0e4e8f8300f6c10322ef29 : FAILURE in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac2f0b4803b142c6add0607d7748c680 : FAILURE in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41d63ceb046b43edb167c851253e0d75 : SUCCESS in 8m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef68b098829246929d0df00fc28644ab : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ee77ceb7797f4bb0b5d6bd8afcadd6ee : SUCCESS in 9m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1fdb3a1d5c27487ebdecb3dba3f645a0 : SUCCESS in 18m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e67e29a8b1b4404fbd7ea0965ddae27a : SUCCESS in 57m 57s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/cfbb374cf86a480e929d32a317429605 : TIMED_OUT in 2h 01m 52s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f960132ed875430c99abc4771c7f017c : SUCCESS in 15m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cd12543eee9a4d4ab2c0108a69edd48c : SUCCESS in 1h 47m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7e443a8416cc41c285e5824001de09c2 : SUCCESS in 36m 39s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/caa941eb54f14a788abf83f36065e8ef : SUCCESS in 20m 49s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e7c97add2d0c4f99826f1f50f8c6689b : SUCCESS in 12m 16s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"86becbfd9b545bfcb5a692fbe8c9f569a6ee32b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-02-24 09:21:58.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"bae0c5e4e79959ae7fdfca55d2286f5aa436fdcd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 11:25:44.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/78a19570bf8b46198d64edde1183cc8e : FAILURE in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ac938919b334feda730d974a6695731 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d5775e6a73fe4703a1b58b6ab8eb7d3e : FAILURE in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83d442ba3e5b40fab569a0c80e37debc : FAILURE in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37144851cc7a4cc39925295ea0386d48 : FAILURE in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b3340fe30b34bf99bd6e4165a229f33 : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4fc2889a41fc42a9994df2c3226a197c : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/59fc24e473944ff4b0c79471c7f747b7 : SUCCESS in 8m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7eab53db55414c65847611cfa0c3f2c7 : SUCCESS in 18m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/afacbc266172480186bcff56dc3b7afe : SUCCESS in 1h 05m 20s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/17b4cae8867c4064a243acd7fc37e27b : SUCCESS in 1h 16m 05s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e420d5673401480ba700ff9bea355c7c : SUCCESS in 26m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8a9e21c7705b4087a355202c6a5da603 : SUCCESS in 1h 46m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/058c0e60a19946a1a831427840a807d9 : SUCCESS in 27m 13s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/676ef07302bd4abfa819f7b2a7010d55 : SUCCESS in 27m 10s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/d071d12e84c44485b8bf32f6a695dad7 : SUCCESS in 9m 51s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"de604797dbc4b7520337cf053888bf5247862830","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-02-24 16:53:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"dd9e90847d83a410d1a71df7dff6d47698a1de4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 20:31:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fce1ccc0ad1493fb80dc5de2804d733 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b04e1c512abd4a9bb87d5665b11df9ab : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2e587e4ac924261896b6a5d6526b2d9 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8daa2413cd4c4856aac5a70bbdefb180 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d09bbb21ace4cf3a9ec0e85b848f6c5 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/242a881f03eb4f9da9b4db0b1ca0381c : SUCCESS in 7m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d9ea451d30964163a28800225ebaa488 : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/902630bac5bc41ff9a4b6f5164e037ed : SUCCESS in 9m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/35c0283d70ba4476879467530293222e : SUCCESS in 21m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/efc206ed3fb546b38de8b900b0836089 : SUCCESS in 52m 12s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/cb5c90821264469baaf8d8e10c8e9460 : SUCCESS in 1h 23m 58s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c4ef9a0a7fc94262817af36de995870f : SUCCESS in 16m 24s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b36f71822d47477a8c570296fb00ac90 : FAILURE in 1h 46m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/694804ded2f84f3ca26bc0f5bef80cbb : SUCCESS in 31m 12s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/86a683f0616444d0a740fb416a3518e9 : SUCCESS in 26m 29s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5ad81a2c2faf4a8390fdbe0cdd772f9e : SUCCESS in 10m 47s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":4},{"id":"823da4b5eca9c86f8d784467d9bd813b5221e714","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-01 10:06:07.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"43bc30e2e0cf6c3dabe0ea39998c9e24a504e115","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-26 11:33:04.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8ef2a5b1c16983e160b483bcb867c6f2c6e04f18","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-26 13:39:27.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)\n\nCan you please add a reno?","accounts_in_message":[],"_revision_number":5},{"id":"90bbcbd2f64d66bec74b14c605a94c47fc4ea92b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 13:45:51.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce3826b1eff845ce806906c87a886fd2 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a991b63ee12b41af9d174e8c60922a1d : FAILURE in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/22320e5746694ddf978f4e9d0de6748f : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/712a948a2ab449d5ba7b893702d74816 : SUCCESS in 7m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc2d7b06f0644ebbb1dbd3b5e233cb89 : SUCCESS in 7m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ae7caa99e1a4440a318eeb258fec605 : SUCCESS in 10m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/867d31d9a16b467ea3f44bde3395cfa8 : SUCCESS in 7m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/56cbbc73cb0c44af9f6043b2263d1241 : SUCCESS in 11m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d99d4d2a102b4cca8b00dfbad06c9838 : SUCCESS in 16m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e4ab35aeba6f4a8eb19908dcd2b0840b : SUCCESS in 1h 02m 35s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/98cd28ba32364505809cf0222d07195b : SUCCESS in 40m 16s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/27c2b1a239ad4939ab7d1716496a210a : SUCCESS in 54m 51s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a621d8dd0e104f169bc1cce413351e9d : SUCCESS in 18m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/986176d0619d4af0b2cd9f70eff56b1e : SUCCESS in 12m 46s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a4a25cc516424ed5a09e5d044475903c : SUCCESS in 1h 42m 00s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3715465bb33c4dc5a132acb53fc2a450 : SUCCESS in 29m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e83ab3f2909499a9937554b29e74015 : SUCCESS in 25m 05s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0c7836531ff2445eb1095ff3634990a1 : SUCCESS in 25m 27s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/8d5ed24fc3d6478da7029269c551e6ef : SUCCESS in 13m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43aaf3bd74204f88809264a6d62e1056 : SUCCESS in 10m 24s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"096b3c32a72ce5958ba1cc22633a063cfe7f29ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-29 06:50:36.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"eae9c3b291250c03766716fe9c9b13cf06bc90a8","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-03-29 07:12:26.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"e93c9bed06da9138cafce350234398858ba3aeec","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-29 07:48:34.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"c63f2fc135d1e972ab24c88c7c03eb70a9f5183d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-29 07:50:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8f9664ad5142ac83a2dd86de0b8abca03095f3cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 08:20:11.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85beb6c3a6804e22a9b3a4a496067abf : FAILURE in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/218e1e62371c464d821362b993cca2b4 : FAILURE in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38c7392c406946a2a463237ffeca40bf : FAILURE in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26991acb44e94c72ad3c7ebb21de8548 : FAILURE in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03225279e7e04967948e678aa9b19711 : FAILURE in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54e2c913436d4efebc89a913b8e4d92d : SUCCESS in 7m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e74d84ebea5404d92b8103ea734731a : SUCCESS in 3m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2710d4f459b84783b9071d7f64026b2d : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/831e7a57844a4176af9c78a0b60bc3f0 : SUCCESS in 11m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b59aeff960d5447bb70466fde44502e8 : SUCCESS in 16m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9e84cd3572a54388a7d5b66a835bcafb : SUCCESS in 1h 01m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/05ae34c9793c46d0ba5ff38e129a8f03 : SUCCESS in 38m 48s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/9302f371c0954b8f8b1bce22e69ba9cd : SUCCESS in 1h 04m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b24c29d0cac4411aa399d62ab97a7e67 : SUCCESS in 14m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2155227a77d44d0f90c90a8e3d419688 : SUCCESS in 12m 56s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9292d8c39f4d4a27a0dcabaf635a06a7 : SUCCESS in 1h 27m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/80b94d8f9be6442fa3f2c325fb3c9264 : SUCCESS in 31m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7f9ecaf6096b4023b0d1447a7e505b55 : SUCCESS in 21m 57s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ab78dcefefab425dba00d3f8d47f4dbb : SUCCESS in 24m 51s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/dc8f60b2e11f4b44912a4807cb051d33 : SUCCESS in 13m 35s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b2cee54803545b4848a3da1986e4fd0 : SUCCESS in 9m 13s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":6},{"id":"895f597616436806b85816328dedde8be97c34ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-29 14:55:51.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8c2feeb4581de8c0d41aa848beb646d6835c5e10","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 16:44:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/805648f42d9d41fcaa318c38fd2a333f : FAILURE in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/027250fbcb5b4f8499a494ad7ef7d216 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8c84fd545736409696f85b8deb064882 : FAILURE in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9afab08208534c7fbd29cc11432872c5 : FAILURE in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21b7ca1378f84d07b4a8e77486f545a5 : FAILURE in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30b272e9423245e5a8e7170dd83a24ae : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f4ee0072ba04a22968ac2d60a8198bd : SUCCESS in 5m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/03ec1f940d7c4203b350951a67337ef7 : SUCCESS in 6m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9c5e4ce243394da890568b02ec6d5b8e : SUCCESS in 12m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cf6bc78e260a4370b1b7ff00bd7b8368 : SUCCESS in 20m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/41e2ac2bc6a846a4a0391df744037da9 : SUCCESS in 1h 04m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e492abb83f9641f9ae9f5bc1dd183b01 : SUCCESS in 48m 30s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/9b28490774b14086b727746cd6aab6b6 : SUCCESS in 55m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a5eda6a7ae0948b8ab90572bb2160b2d : SUCCESS in 17m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b6801f89ad4402d992097d0add02d5a : SUCCESS in 13m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/60210f1420ef406a945d433a4419fd15 : SUCCESS in 1h 47m 16s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9d95cb312f314ae5a6d75069e316328a : SUCCESS in 25m 23s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f9c0750c1ec748e5a784c06cc71a3c3a : SUCCESS in 24m 02s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/449d757c2d1d44868933e83d923b9393 : SUCCESS in 25m 17s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/91917a72e87a4157bb6b3a34382c3fad : SUCCESS in 11m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/740f7aa778ba4073b1e12f9300582ac4 : SUCCESS in 11m 58s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":7},{"id":"20ac0b899d20d1f22fd9f1f31a911e7b982188aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-29 16:53:40.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"f8b49cab2f4982cbdad1c506b052b192a158b2e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 18:33:11.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/632473d60c684849bb6322f77d590d0f : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dea66b008e8b4e0ca6f98cfc780dd4f2 : SUCCESS in 8m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a280bd6d8ec7448897dd586efc930b6c : SUCCESS in 9m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/566358f831494a1883b581f22ce1c9ca : SUCCESS in 8m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d933c82dd37147f696350c1dab1b393c : SUCCESS in 6m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0cd296a0b714466bfd519b28bf8466c : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8714a613f7d84e6da01c6504355d19b4 : SUCCESS in 4m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e719ae616131450c9df4c5bfb6a65375 : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e1260b8c116640bca2ec071f70e66183 : SUCCESS in 13m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bfe7ed09df9c4bde844dbc2683b97410 : SUCCESS in 18m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/52b8aada408646019ec4801e2240559a : SUCCESS in 47m 05s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/76c74b10a70f4ed290bbe54e630b28bd : SUCCESS in 53m 54s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/e2fb02d2069d44b5b71faf9a745ecd6b : SUCCESS in 1h 05m 37s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8c678ffcef8c4db39a41a9d0a0feff39 : SUCCESS in 27m 05s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/82053e8582b94115984ea37ff3b4142c : SUCCESS in 14m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3d0a7db91c424ec4a04235355ef95830 : SUCCESS in 1h 36m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7cee8efad92d4fabb953d3bd7959c37c : SUCCESS in 29m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e33350933a8044f1b307701d1033ef04 : SUCCESS in 22m 49s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9d129e877e524890b1ff55ce9496ef09 : SUCCESS in 44m 24s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/92c23ccd304b45f9972428c0d544c065 : SUCCESS in 11m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0371e42c06ab42bcb357d71291331449 : FAILURE in 5m 03s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":8},{"id":"781d41e720efde60f71e6ad33bbe56ff82edd9d7","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-30 05:28:15.000000000","message":"Patch Set 8:\n\nrecheck (galaxy error)","accounts_in_message":[],"_revision_number":8},{"id":"a4215d0d42f8f1a25ff0b5e661f73fe30895067e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 07:37:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f301852d820494dac840b35c96d84e9 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f3055154c864bd1a0b2214c1bee0b69 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1e4a69c75184898846b01f942778be8 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f2c594001c345308bd819f96c00183c : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c3f28e4c30f49b4b4755940b112ac40 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/852694291caf40c58e49dbbd0b509d5b : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b319320e0ba64cf69f3d7195fc9fa996 : SUCCESS in 3m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a16a08e281f438e93c72ba8952244f6 : SUCCESS in 6m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e0c0f65aac6c4e94b92ca2e852ced440 : SUCCESS in 11m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d87a206866cb4417b01f00034aa44f36 : SUCCESS in 21m 25s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e474aa2bbc464443b427ec7c28c50106 : SUCCESS in 1h 04m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d963b6ca293341a588eff0869ba348ee : SUCCESS in 37m 06s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ae5cfcd73bb948f9813c825271f9923d : SUCCESS in 1h 00m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9c94ba538bd2493bac0df8f6a5133c11 : SUCCESS in 18m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e10838f212ea42e3a58e55ca61e6e921 : SUCCESS in 10m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/45dab86957484e5a9bd0d825d5c46659 : SUCCESS in 2h 04m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4b63979da814425486ead15a495dcb35 : SUCCESS in 29m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/188c1a7b7f244594bfeacb405ab3f615 : SUCCESS in 23m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/41543f4602134d2a8449840517373fc5 : SUCCESS in 21m 44s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e7ebee69dcdb4610a98591da73a7a5bd : SUCCESS in 12m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c79ccd12101b423f84138a56414866c8 : SUCCESS in 10m 25s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":8},{"id":"4d24f1db9b0d356fd171a2960e4ef192af9590f8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-30 08:04:25.000000000","message":"Patch Set 8: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"c9c637422e327a8b43c18d891ad70b540007fc61","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-30 08:05:02.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"6bb8083d6c8ebb4e0aa49d84cb95be1f05aeb6b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-30 11:27:56.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"7dd1c5b045bfb11db9b251f8ee8287c4ba200ef0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 13:21:12.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f688f0adf2240ea81c35f4f7b42a770 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d13a9aa1f4904a2baf638dd76de82a84 : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef8fc9a5fe1947bc869fc60a57d54bdf : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31f510eff2a04309a3886f97e0dd41b0 : SUCCESS in 7m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/876fc639bd61414ab36a5f35182d0741 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec3904000b048dba7b3e6d818bba9e6 : SUCCESS in 9m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/64d0264c61694139920292f1054be161 : SUCCESS in 4m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6e05c21b6108471dbe1c150e8567036a : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/df6b98ef669445f0a3abde450ca33585 : SUCCESS in 10m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d1999a904c6043d0bd06528e572a557d : SUCCESS in 20m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/10e5bf42f7d5479eb11091961f02bd52 : SUCCESS in 1h 01m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2fa3d4a07cd54e09a1f290c1ec3fff63 : SUCCESS in 45m 15s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/394ec9fbf87640f9aef0bfd7c59b43be : SUCCESS in 1h 14m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e9b03b765db44617972fcabf90927065 : SUCCESS in 16m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab11375eda1a4c75b0d86a7103715041 : SUCCESS in 19m 18s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1812ebdb1e854954916ae957846e81cc : SUCCESS in 1h 48m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/841c48efbca3429bae3d4c663d4c886c : SUCCESS in 27m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1783c539feb4d098553c247b728ff68 : SUCCESS in 23m 59s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1a0215cbcc6b4ebb8fc4a90dc52d5257 : SUCCESS in 26m 25s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/01599691cb184226b0f30f5274a44028 : SUCCESS in 13m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a9eb951223b46688f71beeb66eb7716 : SUCCESS in 11m 01s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":9},{"id":"b82412b3f8082a038a95dd138d0f4f3ed37ce871","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-30 13:27:11.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"33af750520ad85ca602b14f00024564886de8245","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 18:59:49.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98f208116acc484891196a11cd58510f : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d9cfeb2318d46ecb15920146ebe1b3a : SUCCESS in 8m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad1d9d2c8a214e5d95de78255ab543ef : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7d64e99d6034d19ada2bdde4aa8fd3c : SUCCESS in 6m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6178f8370c1642c6b39c37a71a34d7ef : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/416264bf6d944f90af216342932e88d5 : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/321c486f56524593a9a6a2bd332c92f8 : SUCCESS in 4m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/de90a79a1a7542c29d5ed6cce4a77c97 : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6b6f9406b9f04efbb678b896131d5832 : SUCCESS in 11m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b664a35caa754bc799b151d29cbc341f : SUCCESS in 19m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/83a3c31a609e4f2a8fddcaf58c29ce40 : SUCCESS in 1h 01m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/11a9941e69e24a26900a0a47066185ab : SUCCESS in 41m 10s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/63f638f60a124e549f304beadcd98cce : SUCCESS in 1h 03m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/be04f92ff99a415aa574a0e61d8a876a : SUCCESS in 18m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/35ed2be683a94d6e956beae42462d107 : SUCCESS in 13m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/eb905ef77a034b08b54cd33c5e6dcb5a : RETRY_LIMIT in 18m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8af1e2344048496caa981ac8cde8b21b : SUCCESS in 31m 16s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9636e93c90554817af9f47212916019d : SUCCESS in 22m 07s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3a37baeb3eaf4dca84c163129a0c6ba5 : SUCCESS in 24m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/19bf8d854d734d1b8650e2c8250b8f06 : SUCCESS in 12m 02s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ab371e34c7eb46c7b161b4dbbdba64f9 : SUCCESS in 10m 01s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":10},{"id":"0dbc0b1c416b5813dfb784c3091121321e9470d6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-30 20:25:56.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"2b718013054d22295c83a579d7b0a186a5d32fb2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 00:04:28.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef2eb8044e8042ec9e5b4ce354f65afa : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/929e0839c6a54607a6c3098cfc9eb838 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b3316aec3ec04b7fb49de8e0518a1864 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6858e0ceba14862bbe0dc9fde32a0d8 : SUCCESS in 7m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abd0732cd55e4b17a061a78170913732 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1980c1b2ea4f4356bc0e80cf0231c9fd : SUCCESS in 8m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dfe92416b60c465995f88fdbeaf182b8 : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/78c4ce0806d94a55bdbf91efa7d8bbe7 : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/057d67131bc2446889e2e55fc2942504 : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f6e617de81c0448eae536e0678e9d056 : SUCCESS in 22m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c6d7da7e96f04061a3b6d63d0191a90e : SUCCESS in 1h 02m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6042aa896fdb43f6bcbf84455169f21d : SUCCESS in 44m 28s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/324116e4ec36468984bc1b53ade1deb1 : SUCCESS in 1h 01m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e390ab6d858242a98425987b0f921cd7 : SUCCESS in 16m 59s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e58a72f2a350479db88c0be261cec661 : SUCCESS in 11m 58s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cc2b2f0603514b3ea81d42c5e0e09099 : SUCCESS in 2h 03m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/07d33beda4b74b419dc77f389e2c79b9 : SUCCESS in 30m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34f640e95839498ab6616313c5919f3c : SUCCESS in 26m 47s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/51aa8f2f99db4821a477253faba72b6f : SUCCESS in 26m 07s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2800e4d55776470d81f92b3dae626bf5 : SUCCESS in 13m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/930c7cd23e3944fba2552028ea58cae4 : SUCCESS in 10m 43s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":10},{"id":"c687ac38691cc30f2fc5e61519591b420869f09a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-31 08:21:48.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"a68deb34f9f248c687359c1c67f96f2d2f6cdd5e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-02 13:24:57.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"f8c116f642c6ce282ba1cddb8e024361822c6a9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 13:25:17.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"a6b17d980a5dea742f741f76baca6e65b423184f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 15:33:33.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/818af349b33b4114ac382928a7ef974e : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b237f1166487411bbe8db22d3d1721b5 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4fe0240bb3040f494802c3f835822c3 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a2b6ecc843941bc8173c0bfc801adeb : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/868fdf3fe1314ec18b9795ec45b00fca : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/32ad482e16d04e7f81fe9fbdf91ce413 : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6eacd77b1c9340aeba04a7538cbec06a : SUCCESS in 10m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8656111d724249f4ace979ac0600328a : SUCCESS in 19m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e0b237bf75134f4da4c180318b67bfed : SUCCESS in 1h 00m 57s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c95484e8d6e94e84b3c88f65180be81c : SUCCESS in 49m 49s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/59856f3bd8bf42c3bb54c508c4f07dbf : SUCCESS in 1h 03m 09s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/053de39761c4497ebb84bedc59cc65d4 : SUCCESS in 17m 56s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5461b39f279b4e498ebe8a20b9eafe88 : SUCCESS in 12m 07s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3186045ae43542dca08f51eff1c5ca85 : SUCCESS in 2h 03m 12s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f7244cb2bf7346b5a1c35670487d93e1 : SUCCESS in 27m 38s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c74b524645a347688d61c2a5ff0c1ade : SUCCESS in 29m 30s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b507ecf1f4f443359d8b7668ebb5605e : SUCCESS in 26m 58s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/069953c4a52a44388c7e04aa5103c798 : SUCCESS in 14m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ebfd2a6af3d479eab42bf3b02d113ed : SUCCESS in 9m 56s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":10},{"id":"1c774231c4ea6a94736d559c542226dac88c046e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 15:33:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"7dc2139dc62e07dfd211a05c3b77b07bda79fa39","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 15:34:54.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e38cbdb090741a5888978fbe76c90c9 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/335209e884154ff983fa67966d9cf81d : SUCCESS in 45s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"847bb8666d4f08e076e789c8cd3d094e49ea03be","revisions":{"7910051682772b0db2ab853cc76a10eaed208684":{"kind":"REWORK","_number":1,"created":"2021-02-18 14:33:45.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/776471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/776471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/776471/1"}}},"commit":{"parents":[{"commit":"071ab367129960cf8bec249bac9f3dac7e435983","subject":"Merge \"Test building seed deployment images in the seed job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/071ab367129960cf8bec249bac9f3dac7e435983"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-02-18 14:32:57.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-02-18 14:33:39.000000000","tz":60},"subject":"Add output printing to host-command-run","message":"Add output printing to host-command-run\n\nChange-Id: I50f75f8f8ed200b1c70f834973463c2f7343ed1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7910051682772b0db2ab853cc76a10eaed208684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7910051682772b0db2ab853cc76a10eaed208684"}]},"branch":"refs/heads/master"},"335690213d07bc25c907147a5c01b908168c61f7":{"kind":"REWORK","_number":2,"created":"2021-02-23 17:14:00.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/776471/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/776471/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/776471/2"}}},"commit":{"parents":[{"commit":"071ab367129960cf8bec249bac9f3dac7e435983","subject":"Merge \"Test building seed deployment images in the seed job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/071ab367129960cf8bec249bac9f3dac7e435983"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-02-18 14:32:57.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-02-23 17:13:51.000000000","tz":60},"subject":"Add output printing to host-command-run","message":"Add output printing to host-command-run\n\nChange-Id: I50f75f8f8ed200b1c70f834973463c2f7343ed1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/335690213d07bc25c907147a5c01b908168c61f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/335690213d07bc25c907147a5c01b908168c61f7"}]},"branch":"refs/heads/master"},"d5d478b3131357b37f4308a7b9f0cf21f3f654f7":{"kind":"REWORK","_number":3,"created":"2021-02-24 09:21:58.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/776471/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/776471/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/776471/3"}}},"commit":{"parents":[{"commit":"071ab367129960cf8bec249bac9f3dac7e435983","subject":"Merge \"Test building seed deployment images in the seed job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/071ab367129960cf8bec249bac9f3dac7e435983"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-02-18 14:32:57.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-02-24 09:21:50.000000000","tz":60},"subject":"Add output printing to host-command-run","message":"Add output printing to host-command-run\n\nChange-Id: I50f75f8f8ed200b1c70f834973463c2f7343ed1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5d478b3131357b37f4308a7b9f0cf21f3f654f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5d478b3131357b37f4308a7b9f0cf21f3f654f7"}]},"branch":"refs/heads/master"},"45a8e1cb9f89983c6a377dab7f96b3e49809386f":{"kind":"REWORK","_number":4,"created":"2021-02-24 16:53:21.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/776471/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/776471/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/776471/4"}}},"commit":{"parents":[{"commit":"071ab367129960cf8bec249bac9f3dac7e435983","subject":"Merge \"Test building seed deployment images in the seed job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/071ab367129960cf8bec249bac9f3dac7e435983"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-02-18 14:32:57.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-02-24 16:53:13.000000000","tz":60},"subject":"Add output printing to host-command-run","message":"Add output printing to host-command-run\n\nChange-Id: I50f75f8f8ed200b1c70f834973463c2f7343ed1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45a8e1cb9f89983c6a377dab7f96b3e49809386f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45a8e1cb9f89983c6a377dab7f96b3e49809386f"}]},"branch":"refs/heads/master"},"2c141fa94c7bd52365aa93d0c53c96f7eb6991f1":{"kind":"REWORK","_number":5,"created":"2021-03-26 11:33:04.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/776471/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/776471/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/776471/5"}}},"commit":{"parents":[{"commit":"071ab367129960cf8bec249bac9f3dac7e435983","subject":"Merge \"Test building seed deployment images in the seed job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/071ab367129960cf8bec249bac9f3dac7e435983"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-02-18 14:32:57.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-26 11:32:59.000000000","tz":0},"subject":"Add output printing to host-command-run","message":"Add output printing to host-command-run\n\nChange-Id: I50f75f8f8ed200b1c70f834973463c2f7343ed1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c141fa94c7bd52365aa93d0c53c96f7eb6991f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2c141fa94c7bd52365aa93d0c53c96f7eb6991f1"}]},"branch":"refs/heads/master"},"22eae7a7b0ba3260909c697c8d373a8e69a2c3ff":{"kind":"REWORK","_number":6,"created":"2021-03-29 06:50:36.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/776471/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/776471/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/776471/6"}}},"commit":{"parents":[{"commit":"071ab367129960cf8bec249bac9f3dac7e435983","subject":"Merge \"Test building seed deployment images in the seed job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/071ab367129960cf8bec249bac9f3dac7e435983"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-02-18 14:32:57.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-29 06:50:32.000000000","tz":120},"subject":"Add output printing to host-command-run","message":"Add output printing to host-command-run\n\nChange-Id: I50f75f8f8ed200b1c70f834973463c2f7343ed1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22eae7a7b0ba3260909c697c8d373a8e69a2c3ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/22eae7a7b0ba3260909c697c8d373a8e69a2c3ff"}]},"branch":"refs/heads/master"},"f9f33fae2d2725ebdfd9d3a260c552be881b5cc0":{"kind":"REWORK","_number":7,"created":"2021-03-29 14:55:51.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/776471/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/776471/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/776471/7"}}},"commit":{"parents":[{"commit":"071ab367129960cf8bec249bac9f3dac7e435983","subject":"Merge \"Test building seed deployment images in the seed job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/071ab367129960cf8bec249bac9f3dac7e435983"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-02-18 14:32:57.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-29 14:55:38.000000000","tz":120},"subject":"Add output printing to host-command-run","message":"Add output printing to host-command-run\n\nChange-Id: I50f75f8f8ed200b1c70f834973463c2f7343ed1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9f33fae2d2725ebdfd9d3a260c552be881b5cc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9f33fae2d2725ebdfd9d3a260c552be881b5cc0"}]},"branch":"refs/heads/master"},"5629ae0ef266d87e2fe99b467907d5971aaa7db3":{"kind":"REWORK","_number":8,"created":"2021-03-29 16:53:40.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/776471/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/776471/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/776471/8"}}},"commit":{"parents":[{"commit":"071ab367129960cf8bec249bac9f3dac7e435983","subject":"Merge \"Test building seed deployment images in the seed job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/071ab367129960cf8bec249bac9f3dac7e435983"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-02-18 14:32:57.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-29 16:53:33.000000000","tz":0},"subject":"Add output printing to host-command-run","message":"Add output printing to host-command-run\n\nChange-Id: I50f75f8f8ed200b1c70f834973463c2f7343ed1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5629ae0ef266d87e2fe99b467907d5971aaa7db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5629ae0ef266d87e2fe99b467907d5971aaa7db3"}]},"branch":"refs/heads/master"},"ad2f9781ab1c320830cc865a5428790fcc127e60":{"kind":"REWORK","_number":9,"created":"2021-03-30 11:27:56.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/71/776471/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/776471/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/776471/9"}}},"commit":{"parents":[{"commit":"071ab367129960cf8bec249bac9f3dac7e435983","subject":"Merge \"Test building seed deployment images in the seed job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/071ab367129960cf8bec249bac9f3dac7e435983"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-02-18 14:32:57.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-30 11:27:52.000000000","tz":0},"subject":"Add output printing to host-command-run","message":"Add output printing to host-command-run\n\nChange-Id: I50f75f8f8ed200b1c70f834973463c2f7343ed1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad2f9781ab1c320830cc865a5428790fcc127e60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad2f9781ab1c320830cc865a5428790fcc127e60"}]},"branch":"refs/heads/master"},"847bb8666d4f08e076e789c8cd3d094e49ea03be":{"kind":"REWORK","_number":10,"created":"2021-03-30 13:27:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/71/776471/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/71/776471/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/71/776471/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/71/776471/10"}}},"commit":{"parents":[{"commit":"f8575af9c07aef00f0f80cf732f33c7fbcbb28ab","subject":"Merge \"Avoid fact gathering for group_by\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8575af9c07aef00f0f80cf732f33c7fbcbb28ab"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-02-18 14:32:57.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-30 13:26:56.000000000","tz":120},"subject":"Add output printing to host-command-run","message":"Add output printing to host-command-run\n\nChange-Id: I50f75f8f8ed200b1c70f834973463c2f7343ed1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/847bb8666d4f08e076e789c8cd3d094e49ea03be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/847bb8666d4f08e076e789c8cd3d094e49ea03be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
