)]}'
{"id":"openstack%2Fkayobe~776500","triplet_id":"openstack%2Fkayobe~master~I248b5bb9ce5a80a07a2a311cb3aca6daca920720","project":"openstack/kayobe","branch":"master","topic":"systemd-networkd","hashtags":[],"change_id":"I248b5bb9ce5a80a07a2a311cb3aca6daca920720","subject":"Ubuntu: support systemd-networkd","status":"MERGED","created":"2021-02-18 17:14:51.000000000","updated":"2021-04-23 11:51:54.000000000","submitted":"2021-04-23 11:50:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"776500-systemd-networkd","meta_rev_id":"62dbbc6c154aa0ba29eb1f073a0e541e3036e92b","_number":776500,"virtual_id_number":776500,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-23 11:50:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2021-04-21 09:38:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2021-04-23 09:40:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2021-04-23 09:40:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-18 18:03:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-08 09:14:10.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2021-04-21 09:38:11.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2021-04-23 09:40:32.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"34edac76ab25400c04083f4dda9068bb35c6e23f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-18 17:14:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d77b0432effe86d37460ac6a1074dd08e8a44ef1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 18:03:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a84d2ab6627489da76bcf419b0b08f3 : FAILURE in 46m 03s","accounts_in_message":[],"_revision_number":1},{"id":"986cff6b33ca3aed90c4f0aaca32cc72e195236d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-18 19:39:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"02d20f073445b7b783105c098353625643d798a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 20:22:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/edbfb4e0b72c4d9898f018d23475a8d4 : SUCCESS in 38m 33s","accounts_in_message":[],"_revision_number":2},{"id":"eeb974c12a483366c88ecfb2352e2fef59f545e0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-23 16:31:18.000000000","message":"Topic systemd-networkd removed","accounts_in_message":[],"_revision_number":2},{"id":"574dac0472c5d7ddc7a0f1577eb2d158469229ba","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-23 16:31:25.000000000","message":"Topic set to story/2004960","accounts_in_message":[],"_revision_number":2},{"id":"6f7109b16c8d20933ad3cf1e3f2c6b1821cdcfba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-23 17:42:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0e4c556ffb2664acc3050d68f83c8587c9c3f100","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 18:38:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f6593f5f66974f09ad89ecf5a7babbac : SUCCESS in 44m 35s","accounts_in_message":[],"_revision_number":3},{"id":"2a114626b4f2d13dfcce10b1419cd22ed17bf2aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-24 12:30:58.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"af69026817af032cae363a5faee1159eb6a85ccf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 12:45:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f41c2504ab0b4ca987d2f975bcfdfc3a : FAILURE in 5m 46s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43be40cc81114cbd829cba472999d9c3 : FAILURE in 6m 51s","accounts_in_message":[],"_revision_number":4},{"id":"3f0b76df8d6fa58ef3bbc0267c32b00606169d6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-24 13:39:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"446bb9960204bc20f1a5e353ed9c796c09aacfc2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 15:24:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/577e726db5a24469917a615ca8e2a3f1 : SUCCESS in 51m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/67c23505250d4150ba3582e02bb4e2eb : FAILURE in 6m 15s","accounts_in_message":[],"_revision_number":5},{"id":"831c445914a3a46094f1057cded4416d049ac999","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-24 16:10:52.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"11cce1f2d22b3e190fdcd8907b000418945bdf76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 19:24:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f194222628a34fc191124af7a47e8c36 : SUCCESS in 43m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a96baeddc5544907ac8126294cfaf5df : SUCCESS in 11m 34s","accounts_in_message":[],"_revision_number":6},{"id":"7e1571a10d85235d3695d733faa50987d6d0bb98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-04 17:22:08.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"feb8c7756a0f90f5ffc51eca9f7f9066e72493f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 18:44:40.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2fe82a403f9d4c0082032f9f65a3afab : SUCCESS in 1h 03m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c2738ef24e24d5685bde8d12b0e345f : SUCCESS in 50m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9097263bfbe74dcebf2de6c3d2b5de5c : FAILURE in 13m 54s","accounts_in_message":[],"_revision_number":7},{"id":"50973286df5098c87de34a298f4151cd5137ad22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-12 18:17:15.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1058b5fec8496bdba1d4529436dc323ecafc9c00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 18:30:28.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d9561e02c39a490e8e620cd135ca547d : FAILURE in 7m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/30313ee88db041feb56aa73af264b017 : FAILURE in 4m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0af8ca5fc8fb49fa9e2e34d8fac184dc : FAILURE in 5m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9fb76e3a67e447c9e4ecc99d0f5f442 : FAILURE in 5m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/49125db407644830ab9024c450d13d70 : FAILURE in 5m 27s","accounts_in_message":[],"_revision_number":8},{"id":"3989af10ddebfd4d25cb4201fedf9dd36e5ef6b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-12 18:30:46.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"16c178002cbf27c6f74ca9b91c865b295f258b00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 18:41:21.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cd59bd57d59b4775ab7dcc36902207d5 : FAILURE in 7m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/16b070d2316941cfba9b0bb89dd2e14c : FAILURE in 5m 46s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4cd60281442c4b18a61b44188b935330 : FAILURE in 6m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42e1494d2725423eacc66bb16e78a674 : FAILURE in 5m 23s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e07b8e0c624745e1ad8b84d86fdd8cf6 : FAILURE in 5m 27s","accounts_in_message":[],"_revision_number":9},{"id":"481da27690b032a2503025e06a152900550ff4a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-06 18:55:30.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"75012c7e813623068b0987891ab54a3d3d1d1dc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 20:12:03.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a94db1ae2dc747a2a70c999bb42af965 : SUCCESS in 58m 43s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cd6d7dd755714ffc8edf2f2423f94298 : SUCCESS in 40m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/78c913fb07e542cdbd11e24109c7b8c2 : FAILURE in 13m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a9577346796b4cd092b526b290a6e46f : FAILURE in 13m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea6a9ac6f8b94b149e53b561980d45c0 : SUCCESS in 10m 08s","accounts_in_message":[],"_revision_number":10},{"id":"563f6c4a4392a181131dae1ca561447f334b4d56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 09:24:00.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"79a46bb86f3c4c14fc1c0cca824a06168ba25782","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 10:37:26.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9b67502cf9f341738ebe356050ee20b7 : SUCCESS in 1h 07m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faec330f05f6443eb2f31d7ba9d599ec : SUCCESS in 48m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/602e8d85df2f4c899d261c5ea75e0dfc : SUCCESS in 13m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fe7915b926c473095baaa3cbb831f40 : FAILURE in 11m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/997614edaf4649699364bba65d6f7b32 : SUCCESS in 11m 26s","accounts_in_message":[],"_revision_number":11},{"id":"b47a823c02fa785a1fc80653067f9e82e24ed832","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 14:35:30.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b3a8fc8404e2a4e7503b2575ac25558f2ad7d3ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 14:39:16.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"994160f1afc7c742d88107083ce55cd6f3e32fa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 16:26:56.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9848ff1a50734ccea47daf434ab2afe3 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5c04a4586944c399d21c5e3174a00eb : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8dd001498f8948b481a1357ec052927d : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd459eca99874855935ad7775ca95a5e : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/310b837aca884652ae871477157ada16 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca46e162f92046a7bf83227fa8591761 : SUCCESS in 9m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a325a5b3fd5b4e53afa8abb4c50ade59 : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d296bcfef0434a05911ad794a30037aa : SUCCESS in 11m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d25ca2120e364ef1a8fef64c8d87ef90 : SUCCESS in 21m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ca852a883d094f94b852becbc1d0a5a1 : SUCCESS in 1h 04m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b8f05929cc4f434d9d14b743a4b9d46d : SUCCESS in 46m 21s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/c355fd3352184ae0876d773fd3c12918 : SUCCESS in 1h 08m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bdbc5231d9c34c0d9a7507b09bb3a60d : SUCCESS in 16m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8790adba0af14bc4ae2cd75a77b8b7c9 : SUCCESS in 13m 15s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/aaa62cedbe5c4335bcaf1d48df439f05 : SUCCESS in 1h 42m 15s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/995e16528c6e4c999e1b7719d3760733 : SUCCESS in 31m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d951baf955284567ab56733248878226 : SUCCESS in 30m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ab26f1aff25f44faa527eba9ea315951 : SUCCESS in 30m 22s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9026c95f0ff944f8a3dd4853aa2257d6 : SUCCESS in 13m 19s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89ecfb15c0d6421184844d25012b4a48 : SUCCESS in 10m 27s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":13},{"id":"c6525cbeefcf81222abf2a331f6aaa7c19387fe0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-08 09:14:10.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"ef3bfdc676804a43a279cabf199473c286467c90","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-08 09:25:13.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"b74adc85aab9237c234956d5b29c104ceafcf35f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-09 11:01:27.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"9637970c0b36f6df39bb6a66d0a19f929b0cfb36","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-21 08:10:58.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"f62a50b42600f2e0fe54f64a208afa361c26f7e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-21 08:16:24.000000000","message":"Patch Set 14: Published edit on patch set 13.","accounts_in_message":[],"_revision_number":14},{"id":"4d07c525eaf2e046a9d0129e6dd62a625792f9ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-21 08:23:49.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"876e380318ca9e6c67e3ab87ac85b712cbbdc2a6","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-04-21 09:38:11.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)\n\nA quality patch as always","accounts_in_message":[],"_revision_number":15},{"id":"8c542b20e0c37a922ce3465b29dc4559b18d1a5b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-21 10:01:00.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b47b57351a26c2123ac7318e9059ffe01aedefbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 10:06:11.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/831f3e59183645c7b5b4fb055547bcdc : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f0cbbc38421469684e88f8bee8783d1 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/830c9ebbde3b4b42966bc0d9521d8090 : SUCCESS in 7m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a587f86bb2674752b724af9a42ddfcb9 : SUCCESS in 6m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d534ea1be2f4d028eaf6264677faedd : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03fad544ee5942f39c9d47ccb68e6227 : SUCCESS in 8m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6f0e3070d8584d01b150d6109fcfb08d : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8da25a404e6f4634823708da28d9cf7f : SUCCESS in 10m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a37201b734c74b9f818be0c8cd87fe71 : SUCCESS in 21m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/599ff9ad694b45068680476ebc7db875 : SUCCESS in 1h 16m 33s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b02bec94d91840c5bc11b17cee9b9e53 : SUCCESS in 42m 03s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/086dc14dea594499b7a21d6e28f3f070 : SUCCESS in 1h 07m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9ad5a644ccb049529f0d4516bff2cabb : SUCCESS in 16m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcc3b1e42a494a029ce0e2041812f559 : SUCCESS in 14m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/59737401a7704380b11f9cc125b37314 : FAILURE in 1h 40m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0dc9dc4509244867b544531a5adf3f26 : SUCCESS in 33m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/589130318b4944c2a329d1c3d40f4aa8 : SUCCESS in 30m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4509dfe1918742fa870f6bc11c6fb12e : SUCCESS in 24m 39s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/249736974586474a84c34561e1a1b865 : SUCCESS in 12m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9a70d75d980246dea4664cd39d0df667 : SUCCESS in 9m 26s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":15},{"id":"9be0982ed1350ec35dc8565157e51d4ae36f8434","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-22 15:25:41.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"438fcacc8392c413799206cda6a14e98293b9943","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 17:20:20.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/338cdbc85e1c43199e302531ebbd31ad : SUCCESS in 8m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/586681dabd3b4323817b547d5bfe16fa : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e69dead3c3cb4918aa7034a21a9068cc : SUCCESS in 9m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4dc810dc485f4795b91d40386c84eef3 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df557d0d7942443488d66b837cdd704a : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd0529c20f9745d4b74ed5e009c430df : SUCCESS in 11m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1b842a48d07a4e98b9105ab23979c0d8 : SUCCESS in 7m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/32747ec3ba2e4a59b0b310dac0e5dcde : SUCCESS in 10m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/be2ebded55b642f090d3cd6d171ef115 : SUCCESS in 16m 43s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6f10d46925704430b8724bb01136b46a : SUCCESS in 52m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b289aaebb01c43bc865a36523af0b2d0 : SUCCESS in 40m 02s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/fcd65b4bf3dc4b8cb8cc662727579ad6 : SUCCESS in 56m 25s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2bb996b8357b4947bbcfc36ad6ac487b : SUCCESS in 15m 44s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44f8e5b78f5e410ea8af7a7fb3ec83b6 : SUCCESS in 17m 47s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/72c7705410754b419ad9bc57e6820cd4 : SUCCESS in 1h 47m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/5f46c36486f74a2e91326aa028eddc7b : SUCCESS in 34m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ff171c657c69426ba2e35ad8e62d83ea : SUCCESS in 19m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/91d7a8b4a8074d3ca176a8b4d8d29f35 : SUCCESS in 25m 31s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/927b2b155505431ba129e4cc81ef9b7f : SUCCESS in 14m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/360d3468fdf74592b7b46a383f7e2fad : SUCCESS in 10m 00s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":15},{"id":"d8643e9867e10aab0558206e2089ba2130842d32","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-23 09:40:32.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"24e82102e48912a77e8a3686587f32d4b1ddf3ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 09:41:42.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"ab5e56146a4f0a7378b32f4263066a593037fc77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 11:50:18.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd269bd9159a426e923da245755107e6 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a3710670c4e4907a4e05b313ce7b844 : SUCCESS in 7m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf2f01fadb114c0ca71ac15f3c57dbc3 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2c3f3d09e0a4d4c924da92c878b7cfb : SUCCESS in 8m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5fcb3ccb8b6b4f5dad22237f6e29948d : SUCCESS in 6m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6dff83baf1b74685955aa51301202a83 : SUCCESS in 10m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6ede2f0bd9434a54a7dde1b3f6f42e9d : SUCCESS in 20m 02s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6f2e457563dc406991c15ff99efacfac : SUCCESS in 58m 16s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a672c3e458354fe49373034c6b635eda : SUCCESS in 40m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/208e53ce89454057bc3e63db3246ceb0 : SUCCESS in 56m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/439e6e90e9ae4996acdfcb625cba2cb5 : SUCCESS in 26m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4265a294a4a84247987d30d87305e71d : SUCCESS in 14m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/14325f3736c94a85966cceb90fd40f3e : SUCCESS in 2h 00m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/1fa24472b3a84b5fbf2907fb5579c0a0 : SUCCESS in 28m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/291bbce33f7a4a12a8f819cfc3afc202 : SUCCESS in 24m 22s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/74f6fc7858804cfc90cabfd69e14f2ab : SUCCESS in 26m 23s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/944c38002a12489e8975c8f493b19324 : SUCCESS in 12m 28s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cdc004a9dab458ab1f0ae9d2a29899f : SUCCESS in 11m 05s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":15},{"id":"c8982197bf0e4cfc391bfbf228c8d2e7baf2c356","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 11:50:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":15},{"id":"62dbbc6c154aa0ba29eb1f073a0e541e3036e92b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 11:51:54.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0257c50d8dcb4ae89ea5b8abae34365b : SUCCESS in 58s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"3bbf736d8d4672460ab42576d1679b2465b06810","revisions":{"1a1706863644361374fd26e7ab0fc32142ba9fd3":{"kind":"REWORK","_number":1,"created":"2021-02-18 17:14:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/1"}}},"commit":{"parents":[{"commit":"8762af4ae520f6af7cb559915526d469cf27528b","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8762af4ae520f6af7cb559915526d469cf27528b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-18 17:11:39.000000000","tz":0},"subject":"WIP: Ubuntu: support systemd-networkd","message":"WIP: Ubuntu: support systemd-networkd\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a1706863644361374fd26e7ab0fc32142ba9fd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a1706863644361374fd26e7ab0fc32142ba9fd3"}]},"branch":"refs/heads/master"},"27a36ea93847e1dab2bb5bf82d91e5fa452ab710":{"kind":"REWORK","_number":2,"created":"2021-02-18 19:39:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/2"}}},"commit":{"parents":[{"commit":"8762af4ae520f6af7cb559915526d469cf27528b","subject":"CI: add Ubuntu overcloud deploy job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8762af4ae520f6af7cb559915526d469cf27528b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-18 19:39:41.000000000","tz":0},"subject":"WIP: Ubuntu: support systemd-networkd","message":"WIP: Ubuntu: support systemd-networkd\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/27a36ea93847e1dab2bb5bf82d91e5fa452ab710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/27a36ea93847e1dab2bb5bf82d91e5fa452ab710"}]},"branch":"refs/heads/master"},"1d97f1ac7af99c5b33c947090f5481e135762136":{"kind":"REWORK","_number":3,"created":"2021-02-23 17:42:14.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/3"}}},"commit":{"parents":[{"commit":"df1131481fd016e40f7b958b61f739cc65f1f98e","subject":"Refactor veth configuration role variable generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/df1131481fd016e40f7b958b61f739cc65f1f98e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-23 17:39:57.000000000","tz":0},"subject":"WIP: Ubuntu: support systemd-networkd","message":"WIP: Ubuntu: support systemd-networkd\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d97f1ac7af99c5b33c947090f5481e135762136"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d97f1ac7af99c5b33c947090f5481e135762136"}]},"branch":"refs/heads/master"},"6a636ec500150849752a5c5ba9b56ee53f65a313":{"kind":"REWORK","_number":4,"created":"2021-02-24 12:30:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/4"}}},"commit":{"parents":[{"commit":"50c2c14345751a310310cc4ffde8a662b7b4ba8f","subject":"Refactor veth configuration role variable generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50c2c14345751a310310cc4ffde8a662b7b4ba8f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-24 12:30:46.000000000","tz":0},"subject":"WIP: Ubuntu: support systemd-networkd","message":"WIP: Ubuntu: support systemd-networkd\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a636ec500150849752a5c5ba9b56ee53f65a313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a636ec500150849752a5c5ba9b56ee53f65a313"}]},"branch":"refs/heads/master"},"41e8915c375a955bd655ff4f7505002225bdc516":{"kind":"REWORK","_number":5,"created":"2021-02-24 13:39:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/5"}}},"commit":{"parents":[{"commit":"50c2c14345751a310310cc4ffde8a662b7b4ba8f","subject":"Refactor veth configuration role variable generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50c2c14345751a310310cc4ffde8a662b7b4ba8f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-24 13:39:16.000000000","tz":0},"subject":"WIP: Ubuntu: support systemd-networkd","message":"WIP: Ubuntu: support systemd-networkd\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41e8915c375a955bd655ff4f7505002225bdc516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41e8915c375a955bd655ff4f7505002225bdc516"}]},"branch":"refs/heads/master"},"610f62afa460d27a8ced7d607236c45f5d096c6d":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-02-24 16:10:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/6"}}},"commit":{"parents":[{"commit":"f82f13aefe6a4d9675489b9b346dce284b95cbe2","subject":"Refactor veth configuration role variable generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f82f13aefe6a4d9675489b9b346dce284b95cbe2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-24 15:41:33.000000000","tz":0},"subject":"WIP: Ubuntu: support systemd-networkd","message":"WIP: Ubuntu: support systemd-networkd\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/610f62afa460d27a8ced7d607236c45f5d096c6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/610f62afa460d27a8ced7d607236c45f5d096c6d"}]},"branch":"refs/heads/master"},"0d65af334d6b69c86b84c63a0ea6ddfd0cc3b4a3":{"kind":"REWORK","_number":7,"created":"2021-03-04 17:22:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/7"}}},"commit":{"parents":[{"commit":"f82f13aefe6a4d9675489b9b346dce284b95cbe2","subject":"Refactor veth configuration role variable generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f82f13aefe6a4d9675489b9b346dce284b95cbe2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-04 17:18:09.000000000","tz":0},"subject":"WIP: Ubuntu: support systemd-networkd","message":"WIP: Ubuntu: support systemd-networkd\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d65af334d6b69c86b84c63a0ea6ddfd0cc3b4a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d65af334d6b69c86b84c63a0ea6ddfd0cc3b4a3"}]},"branch":"refs/heads/master"},"2378601cbf8d0eb8b226ddaa51bfd9857878e150":{"kind":"REWORK","_number":8,"created":"2021-03-12 18:17:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/8"}}},"commit":{"parents":[{"commit":"97f2f3b5c7159c05f33268f01a9e10c10a68c82a","subject":"Refactor veth configuration role variable generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97f2f3b5c7159c05f33268f01a9e10c10a68c82a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-12 18:17:02.000000000","tz":0},"subject":"WIP: Ubuntu: support systemd-networkd","message":"WIP: Ubuntu: support systemd-networkd\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2378601cbf8d0eb8b226ddaa51bfd9857878e150"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2378601cbf8d0eb8b226ddaa51bfd9857878e150"}]},"branch":"refs/heads/master"},"f73cba831d31aea3338a2b544303b6fc72dd99ec":{"kind":"REWORK","_number":9,"created":"2021-03-12 18:30:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/9"}}},"commit":{"parents":[{"commit":"97f2f3b5c7159c05f33268f01a9e10c10a68c82a","subject":"Refactor veth configuration role variable generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97f2f3b5c7159c05f33268f01a9e10c10a68c82a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-12 18:30:31.000000000","tz":0},"subject":"WIP: Ubuntu: support systemd-networkd","message":"WIP: Ubuntu: support systemd-networkd\n\nTODO:\n\n* some attributes are currently not supported for systemd-networkd:\n  rules (string format), route options, ethtool_opts, zone, allowed\n  addresses\n* support dict format of rules for CentOS\n* test policy-based routing\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f73cba831d31aea3338a2b544303b6fc72dd99ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f73cba831d31aea3338a2b544303b6fc72dd99ec"}]},"branch":"refs/heads/master"},"fee81e34537e2023f1928a78e532f672b88b8fcb":{"kind":"REWORK","_number":10,"created":"2021-04-06 18:55:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/10"}}},"commit":{"parents":[{"commit":"e690ccfee1437e1c7c623cbc278ec88f62d2a5a4","subject":"Ubuntu: refactor network.yml playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e690ccfee1437e1c7c623cbc278ec88f62d2a5a4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 18:54:16.000000000","tz":60},"subject":"WIP: Ubuntu: support systemd-networkd","message":"WIP: Ubuntu: support systemd-networkd\n\nThis change adds support for network configuration via systemd-networkd\non Ubuntu systems.\n\nThis is implemented via an Ansible Galaxy role,\nstackhpc.systemd_networkd which was forked from\naruhier.systemd_networkd. Several improvements were made in\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/1,\nincluding:\n\n* Add support for removing unexpected config files\n* Use become where necessary\n* Refactor config generation into a single task to improve performance\n\nThe systemd_networkd role does not add much abstraction on top of the\nsystemd-networkd configuration file format, which provides a lot of\nflexibility at the expense of additional code in Kayobe. This code is\nimplemented as filter plugins, similarly to the existing\nMichaelRigart.interfaces role.\n\nThis patch includes support for:\n\n* Ethernet interfaces\n* bridges\n* bonds\n* VLANs\n* virtual Ethernet pairs (to connect Linux bridges and OVS bridges)\n* static IP addresses\n* static routes\n* MTU\n\nSome network attributes are currently not supported for\nsystemd-networkd:\n\n* rules\n* route options\n* ethtool_opts\n* zone\n* allowed addresses\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fee81e34537e2023f1928a78e532f672b88b8fcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fee81e34537e2023f1928a78e532f672b88b8fcb"}]},"branch":"refs/heads/master"},"6f6232676dd2650c929afdeaab865fdd65c41915":{"kind":"REWORK","_number":11,"created":"2021-04-07 09:24:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/11"}}},"commit":{"parents":[{"commit":"e690ccfee1437e1c7c623cbc278ec88f62d2a5a4","subject":"Ubuntu: refactor network.yml playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e690ccfee1437e1c7c623cbc278ec88f62d2a5a4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-07 09:19:12.000000000","tz":60},"subject":"WIP: Ubuntu: support systemd-networkd","message":"WIP: Ubuntu: support systemd-networkd\n\nThis change adds support for network configuration via systemd-networkd\non Ubuntu systems.\n\nThis is implemented via an Ansible Galaxy role,\nstackhpc.systemd_networkd which was forked from\naruhier.systemd_networkd. Several improvements were made in\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/1,\nincluding:\n\n* Add support for removing unexpected config files\n* Use become where necessary\n* Refactor config generation into a single task to improve performance\n\nThe systemd_networkd role does not add much abstraction on top of the\nsystemd-networkd configuration file format, which provides a lot of\nflexibility at the expense of additional code in Kayobe. This code is\nimplemented as filter plugins, similarly to the existing\nMichaelRigart.interfaces role.\n\nThis patch includes support for:\n\n* Ethernet interfaces\n* bridges\n* bonds\n* VLANs\n* virtual Ethernet pairs (to connect Linux bridges and OVS bridges)\n* static IP addresses\n* static routes\n* MTU\n\nSome network attributes are currently not supported for\nsystemd-networkd:\n\n* rules\n* route options\n* ethtool_opts\n* zone\n* allowed addresses\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f6232676dd2650c929afdeaab865fdd65c41915"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f6232676dd2650c929afdeaab865fdd65c41915"}]},"branch":"refs/heads/master"},"dd165b7e5c23f87e1f0206cdb52ca924d1714d69":{"kind":"REWORK","_number":12,"created":"2021-04-07 14:35:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/12"}}},"commit":{"parents":[{"commit":"ae2ed2215aeb6daee037d7376c883ae2830bce2e","subject":"Ubuntu: refactor network.yml playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae2ed2215aeb6daee037d7376c883ae2830bce2e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-07 13:48:53.000000000","tz":60},"subject":"WIP: Ubuntu: support systemd-networkd","message":"WIP: Ubuntu: support systemd-networkd\n\nThis change adds support for network configuration via systemd-networkd\non Ubuntu systems.\n\nThis is implemented via an Ansible Galaxy role,\nstackhpc.systemd_networkd which was forked from\naruhier.systemd_networkd. Several improvements were made in\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/1,\nincluding:\n\n* Add support for removing unexpected config files\n* Use become where necessary\n* Refactor config generation into a single task to improve performance\n\nThe systemd_networkd role does not add much abstraction on top of the\nsystemd-networkd configuration file format, which provides a lot of\nflexibility at the expense of additional code in Kayobe. This code is\nimplemented as filter plugins, similarly to the existing\nMichaelRigart.interfaces role.\n\nThis patch includes support for:\n\n* Ethernet interfaces\n* bridges\n* bonds\n* VLANs\n* virtual Ethernet pairs (to connect Linux bridges and OVS bridges)\n* static IP addresses\n* static routes\n* MTU\n\nSome network attributes are currently not supported for\nsystemd-networkd:\n\n* rules\n* route options\n* ethtool_opts\n* zone\n* allowed addresses\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd165b7e5c23f87e1f0206cdb52ca924d1714d69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd165b7e5c23f87e1f0206cdb52ca924d1714d69"}]},"branch":"refs/heads/master"},"f5f8468ac03d211f193807718f6e1ce75064fcc9":{"kind":"REWORK","_number":13,"created":"2021-04-07 14:39:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/13"}}},"commit":{"parents":[{"commit":"ae2ed2215aeb6daee037d7376c883ae2830bce2e","subject":"Ubuntu: refactor network.yml playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae2ed2215aeb6daee037d7376c883ae2830bce2e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-07 14:39:04.000000000","tz":60},"subject":"Ubuntu: support systemd-networkd","message":"Ubuntu: support systemd-networkd\n\nThis change adds support for network configuration via systemd-networkd\non Ubuntu systems.\n\nThis is implemented via an Ansible Galaxy role,\nstackhpc.systemd_networkd which was forked from\naruhier.systemd_networkd. Several improvements were made in\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/1,\nincluding:\n\n* Add support for removing unexpected config files\n* Use become where necessary\n* Refactor config generation into a single task to improve performance\n\nThe systemd_networkd role does not add much abstraction on top of the\nsystemd-networkd configuration file format, which provides a lot of\nflexibility at the expense of additional code in Kayobe. This code is\nimplemented as filter plugins, similarly to the existing\nMichaelRigart.interfaces role.\n\nThis patch includes support for:\n\n* Ethernet interfaces\n* bridges\n* bonds\n* VLANs\n* virtual Ethernet pairs (to connect Linux bridges and OVS bridges)\n* static IP addresses\n* static routes\n* MTU\n\nSome network attributes are currently not supported for\nsystemd-networkd:\n\n* rules\n* route options\n* ethtool_opts\n* zone\n* allowed addresses\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5f8468ac03d211f193807718f6e1ce75064fcc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5f8468ac03d211f193807718f6e1ce75064fcc9"}]},"branch":"refs/heads/master"},"33be25c92505f9cb906dc9c9e6d5a3d7c2b08d1e":{"kind":"REWORK","_number":14,"created":"2021-04-21 08:16:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/14"}}},"commit":{"parents":[{"commit":"ae2ed2215aeb6daee037d7376c883ae2830bce2e","subject":"Ubuntu: refactor network.yml playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae2ed2215aeb6daee037d7376c883ae2830bce2e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-21 08:16:18.000000000","tz":0},"subject":"Ubuntu: support systemd-networkd","message":"Ubuntu: support systemd-networkd\n\nThis change adds support for network configuration via systemd-networkd\non Ubuntu systems.\n\nThis is implemented via an Ansible Galaxy role,\nstackhpc.systemd_networkd which was forked from\naruhier.systemd_networkd. Several improvements were made in\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/1,\nincluding:\n\n* Add support for removing unexpected config files\n* Use become where necessary\n* Refactor config generation into a single task to improve performance\n\nThe systemd_networkd role does not add much abstraction on top of the\nsystemd-networkd configuration file format, which provides a lot of\nflexibility at the expense of additional code in Kayobe. This code is\nimplemented as filter plugins, similarly to the existing\nMichaelRigart.interfaces role.\n\nThis patch includes support for:\n\n* Ethernet interfaces\n* bridges\n* bonds\n* VLANs\n* virtual Ethernet pairs (to connect Linux bridges and OVS bridges)\n* static IP addresses\n* static routes\n* MTU\n\nSome network attributes are currently not supported for\nsystemd-networkd:\n\n* rules\n* route options\n* ethtool_opts\n* zone\n* allowed addresses\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33be25c92505f9cb906dc9c9e6d5a3d7c2b08d1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33be25c92505f9cb906dc9c9e6d5a3d7c2b08d1e"}]},"branch":"refs/heads/master"},"3bbf736d8d4672460ab42576d1679b2465b06810":{"kind":"REWORK","_number":15,"created":"2021-04-21 08:23:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/00/776500/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/00/776500/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/00/776500/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/00/776500/15"}}},"commit":{"parents":[{"commit":"ae2ed2215aeb6daee037d7376c883ae2830bce2e","subject":"Ubuntu: refactor network.yml playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae2ed2215aeb6daee037d7376c883ae2830bce2e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-12 16:42:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-21 08:23:44.000000000","tz":0},"subject":"Ubuntu: support systemd-networkd","message":"Ubuntu: support systemd-networkd\n\nThis change adds support for network configuration via systemd-networkd\non Ubuntu systems.\n\nThis is implemented via an Ansible Galaxy role,\nstackhpc.systemd_networkd which was forked from\naruhier.systemd_networkd. Several improvements were made in\nhttps://github.com/stackhpc/ansible-role-systemd-networkd/pull/1,\nincluding:\n\n* Add support for removing unexpected config files\n* Use become where necessary\n* Refactor config generation into a single task to improve performance\n\nThe systemd_networkd role does not add much abstraction on top of the\nsystemd-networkd configuration file format, which provides a lot of\nflexibility at the expense of additional code in Kayobe. This code is\nimplemented as filter plugins, similarly to the existing\nMichaelRigart.interfaces role.\n\nThis patch includes support for:\n\n* Ethernet interfaces\n* bridges\n* bonds\n* VLANs\n* virtual Ethernet pairs (to connect Linux bridges and OVS bridges)\n* static IP addresses\n* static routes\n* MTU\n\nSome network attributes are currently not supported for\nsystemd-networkd:\n\n* rules\n* route options\n* ethtool_opts\n* zone\n* allowed addresses\n\nStory: 2004960\nTask: 41881\n\nChange-Id: I248b5bb9ce5a80a07a2a311cb3aca6daca920720\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bbf736d8d4672460ab42576d1679b2465b06810"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bbf736d8d4672460ab42576d1679b2465b06810"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
