)]}'
{"id":"openstack%2Fkayobe~776947","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Id5ed0b594d6270c4a03606caa776bbb291be2694","project":"openstack/kayobe","branch":"stable/victoria","topic":"ipa-extra-hardware-stable/victoria","hashtags":[],"change_id":"Id5ed0b594d6270c4a03606caa776bbb291be2694","subject":"Update IPA docs and test build with extra-hardware","status":"MERGED","created":"2021-02-22 15:38:08.000000000","updated":"2021-03-05 21:52:58.000000000","submitted":"2021-03-05 21:50:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776947-ipa-extra-hardware-stable/victoria","cherry_pick_of_change":775093,"cherry_pick_of_patch_set":1,"meta_rev_id":"31a2558df62a22ab19d8495a681465c58cf64cf8","_number":776947,"virtual_id_number":776947,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-05 21:50:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-03-05 19:56:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-03-05 16:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2021-03-05 19:56:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-22 19:20:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-05 16:21:25.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-05 19:56:17.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"246dedd56ec0216bad42952f062c9c6fb2efad82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-22 15:38:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb7da9b54f5358cca12054d076058e64d1d081cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 19:20:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b82c94382878428fb02330c3ef992546 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a668e9013d14461a4e113e0764f7cd4 : SUCCESS in 6m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d02eaf0fa4e04e13ae8902be808d6f68 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce1635a077644891aea62872eb5613fd : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ed5520806054ff6a1bd9eda93cbd380 : SUCCESS in 9m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9d8bcf0a5d69468fb541a2137f8b9b2b : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/45193902da0847b6bc597f8dd2aaa36f : SUCCESS in 8m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/72f66dd945b64fb588c593d337aa8587 : SUCCESS in 18m 20s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cddbf148e87b4912b3b286faec00306d : FAILURE in 1h 46m 44s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/9443deab4fc947199b30c792db9a1c59 : SUCCESS in 1h 15m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/41745418a60445e9b7b41e291151b775 : SUCCESS in 17m 18s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b392082a491c408cb3cb004125ae5905 : FAILURE in 2h 09m 06s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cc50242b0b7d4ced9210394f84a19cbd : SUCCESS in 23m 57s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/159ffa255acd456abd45d5a64292a8ef : SUCCESS in 39m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/90983f35781246b4853656cd7eba677a : SUCCESS in 11m 08s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"cdc27b50c0012e709c1e119b27c314bf9e54bca5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-22 19:33:32.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3518ec2aff6d4af0c4f3615c3198641dc421f135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-22 21:26:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b994119f6b447218f96fb1111eb85ed : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a287b3cfd174020a999fd6ab5c8d92b : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd05275a8afc4e9d8ac8c9e6333a146f : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21bf69ce5f0944cca111200a3e805ccf : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/985ae6a7b8514ff5946259165bc348d3 : SUCCESS in 7m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cd8944ac11504db49daa004c7d6381fc : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6557ac723c234146863c47a780423308 : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f0f2c86b0753476f8492c636182dbece : SUCCESS in 19m 11s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/94ffaf5139c940e4a1a74ea917349d8b : SUCCESS in 1h 02m 37s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/6677b94f06b946feb0676671b4e81b48 : FAILURE in 1h 44m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9f69e9d09b1249acbee35652ae31a7f0 : SUCCESS in 12m 55s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0e8a5b6c7aa64da09096a071d025c9b9 : FAILURE in 1h 51m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/82487fb6c96d4b97be8feeebda5005f6 : SUCCESS in 20m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cd6301c2510c48dd9c3c63dd56884c20 : SUCCESS in 19m 37s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6f207c9e2a6546489e005aaf80fc74af : FAILURE in 15m 42s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"b25b9954135dbe383ec315cdd31ed75b1ccd4902","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-22 22:02:11.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bd73cb052810dbcd9d56ea097dcaa349ca55b6c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 00:36:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6c44924eacf4cbf95bed25e2220ec9a : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d63d7a5b25d45128e4e1ea5480ec014 : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f2137851a7e4f1ebef3dba90ea38d32 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7dfb1f8377b743b6afdb632085f3a1d1 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adcf2a93553845d9809f2700444cfb26 : SUCCESS in 7m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/014d9152154149af97c13e55424e0ae9 : SUCCESS in 4m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0195754b35ab4f979fa35607b5756979 : SUCCESS in 9m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9efc71d75d364e4d885aa82e423f3ad6 : SUCCESS in 20m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d0c531abfbf846558932c5a1469127af : SUCCESS in 53m 47s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/93200d6a8ef5422ca5ac4d45331766be : SUCCESS in 1h 06m 45s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ab3c6248ca6d44afaffb289a12299cc0 : SUCCESS in 27m 02s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b227845957d9463d84749e7a36a864b0 : TIMED_OUT in 2h 31m 44s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/46085eba4d4a4da9a6bca4c516dc4c16 : SUCCESS in 44m 13s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9608f6f253b842b58b6b0d465576df8d : SUCCESS in 18m 44s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c217d0d166d14588ad5dc7bf50b584a9 : SUCCESS in 8m 50s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"ab5fe8f84c40078189d7738a123d891539b5fef0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-23 08:43:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"78cbb2171b6cb8b7e25109f46cc6b2b7c017cb7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-23 08:49:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"33cec45a549c429e2848806bafdfbf8f4f04c112","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-23 12:43:02.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71ad6c6c060f43a586659de86164b5d1 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d31d72dac8745238f1e0be86fbb4a94 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41420080e0f145e5b5f2afd1f7fb407b : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07bb2d801805475b94f861d51c89a03a : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/392d10864b7e4732ad7b27919e270d87 : SUCCESS in 6m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/26de0ebae22443e284defc6c1bfb1b2b : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/698893acd9f24a599bd27cb2598549c4 : SUCCESS in 9m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fe9f396c6f564421be673b34f8cd1f72 : SUCCESS in 18m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cc0620de11b14725b61e94d0a3782a1c : SUCCESS in 52m 39s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/a899e63be0894df5b93cf683f187ce74 : SUCCESS in 1h 15m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d4c69e834bd8407997ad70cb27471993 : SUCCESS in 15m 36s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/436b25c7e9704c75ac959de35ef4b2f8 : FAILURE in 2h 07m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9b2e4ca02d3c47bf931d95a2a0cf4dbe : SUCCESS in 25m 05s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8e293de8f48f43b2bd1699786c9c1077 : SUCCESS in 23m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4259c57655dd4735b0e1b730e1914136 : SUCCESS in 10m 04s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"d52084f1c5b8c5fce768c34cf2f729f228d052b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-03 14:09:02.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"56f5d4041c8e34386e20de9ee919cafc744cbcb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 16:27:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7832494a5a646eea41c4bc5ed2a4161 : SUCCESS in 7m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/712d2076cbb34f50ba694ff058748f18 : SUCCESS in 7m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/698a9a6abbbc4835b999f4cc200d40c2 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d289e1f6bd7f4b6991158ece277893fa : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2288548c970946daaad7fc25a0c4e000 : SUCCESS in 7m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4c5f0c3de87d4857b8fa6788193f5414 : SUCCESS in 5m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d9c4a427b32d40ad86f7287ef967d352 : SUCCESS in 9m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc2b5adf79bf47bf8e7009a50e0f46eb : SUCCESS in 18m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/28629f9135be4e8b930e7384ffcb376b : SUCCESS in 1h 01m 50s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/32816a87608c4b69a7e7890fb82d0543 : SUCCESS in 1h 05m 03s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/972f89623648492b88038a8c66b926b3 : SUCCESS in 15m 21s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d4c2eec9e66346b8b038a6c1806c10a0 : SUCCESS in 2h 03m 16s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/13fa6e329fbb4b96b8303c17e48b11fb : SUCCESS in 27m 32s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fb9370b6091f4ef0a9d87bf80b3d657c : SUCCESS in 27m 38s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0644dc8ff40d468591ba83246a3e6e35 : SUCCESS in 12m 47s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":3},{"id":"dae0759fd00b97c86c1fa6238115d5e1d0be7c8b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-05 16:21:25.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9be036639ac35b9ad649e885db3fc2c6e0151dcc","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-03-05 19:56:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dee988da52141fbc07f6167d5b6c5862550b154d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 19:56:34.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dd7d3fe288208644d9e81619eea6f274efa6ef61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 21:50:18.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69bcfaec16ca4e07bd06e29724d461aa : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55238c56b7064a959586ac5712712c2b : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/017ada8ee5c94016b3c2aa4292391374 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c144e7c14744d1b901e5cf3b980025f : SUCCESS in 7m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ebd4c07200f44de2b2c3f89627d85f3a : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da812032a7b2406fb813ea2abcb7db89 : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8ba672f94a1446e7b8ea22de2a41c2a7 : SUCCESS in 18m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/78ab967f42f54adf925d97137d14686e : SUCCESS in 56m 09s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3d34fee458894250be3e0602abbe7f01 : SUCCESS in 58m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/61291ca5a11f409e89e33a69731810ce : SUCCESS in 14m 18s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2ec8de3e8e454e088c48efbfe9a70e56 : SUCCESS in 1h 52m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3119b4fcff8d4fcba2cabb2d342364b8 : SUCCESS in 25m 06s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/68f19d8590ea433681aaf94e8e0612fc : SUCCESS in 25m 31s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/2d652e4a856f43b393928cc8c781eefc : SUCCESS in 11m 09s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"20d047a37dec03b801020df642f23a76c0322346","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 21:50:19.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"31a2558df62a22ab19d8495a681465c58cf64cf8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-05 21:52:58.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faf9ff8fddcc4ea3a814dbe5bdf2ef2d : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"6df44cd4bfeff56c216b7c9441224b294093d56a","revisions":{"b9b0ddd1218c1540ab3535acecae5345c438565b":{"kind":"REWORK","_number":1,"created":"2021-02-22 15:38:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/47/776947/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/776947/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/776947/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/776947/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/776947/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/776947/1"}}},"commit":{"parents":[{"commit":"d3861f20daacc4c156e9e9efec3ef9fcbc6ef43f","subject":"Test building seed deployment images in the seed job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3861f20daacc4c156e9e9efec3ef9fcbc6ef43f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-11 09:50:43.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-22 15:37:47.000000000","tz":60},"subject":"Update IPA docs and test build with extra-hardware","message":"Update IPA docs and test build with extra-hardware\n\nChange-Id: Id5ed0b594d6270c4a03606caa776bbb291be2694\n(cherry picked from commit d765fbc5ebb7c1a8f80d4c49f3846107e4f23189)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9b0ddd1218c1540ab3535acecae5345c438565b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9b0ddd1218c1540ab3535acecae5345c438565b"}]},"branch":"refs/heads/stable/victoria"},"0f904e63f7b80b707fa5fe31ee6757b10ef3d092":{"kind":"NO_CHANGE","_number":2,"created":"2021-02-23 08:49:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/47/776947/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/776947/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/776947/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/776947/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/776947/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/776947/2"}}},"commit":{"parents":[{"commit":"d3861f20daacc4c156e9e9efec3ef9fcbc6ef43f","subject":"Test building seed deployment images in the seed job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d3861f20daacc4c156e9e9efec3ef9fcbc6ef43f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-11 09:50:43.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-23 08:49:40.000000000","tz":0},"subject":"Update IPA docs and test build with extra-hardware","message":"Update IPA docs and test build with extra-hardware\n\nChange-Id: Id5ed0b594d6270c4a03606caa776bbb291be2694\n(cherry picked from commit d765fbc5ebb7c1a8f80d4c49f3846107e4f23189)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f904e63f7b80b707fa5fe31ee6757b10ef3d092"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f904e63f7b80b707fa5fe31ee6757b10ef3d092"}]},"branch":"refs/heads/stable/victoria"},"6df44cd4bfeff56c216b7c9441224b294093d56a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-03 14:09:02.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/47/776947/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/776947/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/776947/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/776947/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/776947/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/776947/3"}}},"commit":{"parents":[{"commit":"cbb00dd88d0ed98f24d0450b1ddd0a0ac2429b69","subject":"Quote kolla limit when calling kolla-ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbb00dd88d0ed98f24d0450b1ddd0a0ac2429b69"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-11 09:50:43.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-03 14:09:02.000000000","tz":0},"subject":"Update IPA docs and test build with extra-hardware","message":"Update IPA docs and test build with extra-hardware\n\nChange-Id: Id5ed0b594d6270c4a03606caa776bbb291be2694\n(cherry picked from commit d765fbc5ebb7c1a8f80d4c49f3846107e4f23189)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6df44cd4bfeff56c216b7c9441224b294093d56a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6df44cd4bfeff56c216b7c9441224b294093d56a"}]},"branch":"refs/heads/stable/victoria","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
