)]}'
{"id":"openstack%2Fkayobe~777623","triplet_id":"openstack%2Fkayobe~master~I782be627c6e78370555bc7d155261fc15fc93348","project":"openstack/kayobe","branch":"master","topic":"docker-pull-rate-limit","hashtags":[],"change_id":"I782be627c6e78370555bc7d155261fc15fc93348","subject":"[DNM] Test Docker pull rate limits","status":"ABANDONED","created":"2021-02-25 16:53:33.000000000","updated":"2021-04-12 15:54:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d3edc14e15a6efeab88b60ab36f1f5d7e1fceb66","_number":777623,"virtual_id_number":777623,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-26 16:39:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-25 20:40:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9bff07927d7eb8a30abb771443606f07578506c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-25 16:53:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c532d30d0188a2848ee24d55323bec1fd96887a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-25 16:57:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"41a8ab42b40c680e87ffe148950bfc659dd44e01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-25 17:05:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"04e1b743db8ec126f9f7388f616bb6473dfc1a74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-25 17:11:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ad66593502f30c960e04fc5397509187f7bc8d9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-25 17:21:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a1f9c29cf7153debe66b3098657e4707acbd1a75","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-25 17:30:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"cfefba646b50139d7eb8ebba4bae9c8bba45a505","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-25 18:53:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0c295cc7914f6a80d8566ed8f3c7434f14115512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 20:40:50.000000000","message":"Patch Set 7: Verified-1\n\n(20 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3678d508b5a4d84b8160f390ebeea9e : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/603bb23d91574800895753c328d3d098 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8149d7451adf4ecea5ee28f81079126e : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e68a3b02ed241449ddd84e91c4cf425 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c76d473b35794b8980533a32782189fc : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43795ef2030b471797da515165742f1a : SUCCESS in 7m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9c018c9a10c74224b32aacd11e1d6e00 : SUCCESS in 6m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/63696726a85948b4831ebaf7033fd96a : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ee00338825364327b49b4ee7cbd5f159 : SUCCESS in 20m 40s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/62c5ba91ad194a9da78f3016219c54f7 : FAILURE in 1h 42m 09s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/9fc832eaf99144cd99d9ddbfbf33349d : SUCCESS in 1h 20m 06s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bc57a66992ae484cac74381da69917ce : SUCCESS in 15m 12s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/aa2c1ea53845401eb08dbb8bfed4abb1 : SUCCESS in 1h 33m 27s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7662449036174c149746ade6bdeb86b8 : SUCCESS in 25m 39s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/79b3d7d656074eb69edfa51d07fb864c : SUCCESS in 22m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/8009284a814a48ef9d43378fc84bea0c : SUCCESS in 11m 01s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":7},{"id":"d4bf77e391eeb90abaec34967c7d9bb26c9113bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-26 08:53:37.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"1b5288685ef1eba757111aa110ff8a519ec81212","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 10:28:45.000000000","message":"Patch Set 8: Verified+1\n\n(20 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2f00ed07b8240fc9e6161f335f360db : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac3e7d7bcd044103bd2fea917d6853ce : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c98e3e39b8d24533a67bf46647d0c6c8 : SUCCESS in 5m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1843dd427b7f46c8a0ad6a80cc12897a : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85e1a7d2433b4d6a944e8679490311f3 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30279827d47e48ad9223a626bee62edb : SUCCESS in 7m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/90a87eae813e43079b2d4a04247184a2 : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/df0908df2afe4504b4054738e10f163c : SUCCESS in 7m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a8f5e9c9c0a04955a0b6dda5677c10f7 : SUCCESS in 19m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3e52116826af4d1e931b676df37c75f7 : SUCCESS in 1h 05m 37s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/6e7501af4a174f39bc396c5a8496ef43 : SUCCESS in 59m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/95e52d67e4b147fd8e09ef598b22b9b5 : SUCCESS in 16m 12s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/148fece5087d46f180102b06739fdbfb : SUCCESS in 1h 31m 42s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ffcc04e9190f4c9dace7eb03c160291f : SUCCESS in 26m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4f8ce64583874dfa910439844062a847 : SUCCESS in 27m 48s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a1304507455d4b0badd518d2b0c767c1 : SUCCESS in 10m 57s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":8},{"id":"228b885a75f1df75a1c6ca82088675c808f04b56","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-26 13:45:12.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"21f98fe3ce8f25a234e62b1a959f57854aa0f2eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-26 13:51:17.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"15b74051403451dba0200dcf5eddf6063c6fba70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 14:29:13.000000000","message":"Patch Set 10: Verified-1\n\n(20 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65b479ff8caf45388f6d45ea777128a8 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89acba31c31b408bb5eea8ee2d46b027 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f1ccb49e87314dc4a8061353176a06b1 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24f6ab8735254f2a9ea647f6f2282e28 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c14e9dbccdcc4e51b3275cdad84135f6 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/389c068ab21a43958c2c8c5e8d48aa85 : SUCCESS in 7m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/907c039ebd6848239a19e70f191e87c9 : RETRY_LIMIT in 4m 32s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/bb132bcd83224f3090f7014506e1ae67 : RETRY_LIMIT in 3m 01s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bb7bd00f26614e7d89c2207d5824dc9e : SUCCESS in 16m 46s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":10},{"id":"55467adf6eaab4d13d499764439b0f1b42f6cc35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-26 14:36:33.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"8623b93b3dfc085d7b1d3773eebd658e3e21d457","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-26 14:54:51.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"5f6e030a25d11bf2ef97bd6b699731adf1f04e74","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-26 15:39:33.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"5c2228bf733aa1eee7e6a097678daa24595b9f87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-26 16:02:10.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"3a801165910e5eeb54032d818b4a2f937d327bb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 16:39:53.000000000","message":"Patch Set 14: Verified-1\n\n(20 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4379525729964b93ae0787e47be69392 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e170dbd2dc68411387ef088d94c81d68 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82123dda564b4727a039125731e642a8 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d77bebfe8c56451eaa3ef917e2f77985 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/17dde7bc3c8c4f43acdabcec37dcccef : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eeef71436a543aca8368107b2a4b829 : SUCCESS in 6m 14s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/55f64121a3724e079fd18a668c9246f9 : FAILURE in 10m 50s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/a49f22f35834463c9369fe3768f7e36c : RETRY_LIMIT in 3m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1e53d813996942b4ad7c30a4441d6ea7 : RETRY_LIMIT in 3m 12s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":14},{"id":"d3edc14e15a6efeab88b60ab36f1f5d7e1fceb66","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-12 15:54:19.000000000","message":"Abandoned\n\nThis is less of an issue at the moment.","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"d7b57811f5d500255b94c604eabc0df2112e9628","revisions":{"785cac928e74acb9d179d6da15ac060caed9f59b":{"kind":"REWORK","_number":1,"created":"2021-02-25 16:53:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777623/1"}}},"commit":{"parents":[{"commit":"b19222e6865ba7575d5c81e91cb7971178f68d37","subject":"Merge \"Update IPA docs and test build with extra-hardware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b19222e6865ba7575d5c81e91cb7971178f68d37"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 16:52:41.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 16:52:41.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/785cac928e74acb9d179d6da15ac060caed9f59b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/785cac928e74acb9d179d6da15ac060caed9f59b"}]},"branch":"refs/heads/master"},"60b98dc3169d8ff2b27da6ff9932ff626e12d92f":{"kind":"REWORK","_number":2,"created":"2021-02-25 16:57:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777623/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777623/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777623/2"}}},"commit":{"parents":[{"commit":"b19222e6865ba7575d5c81e91cb7971178f68d37","subject":"Merge \"Update IPA docs and test build with extra-hardware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b19222e6865ba7575d5c81e91cb7971178f68d37"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 16:57:04.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 16:57:04.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60b98dc3169d8ff2b27da6ff9932ff626e12d92f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/60b98dc3169d8ff2b27da6ff9932ff626e12d92f"}]},"branch":"refs/heads/master"},"d96843ab9a571d56f51d04aec58c9fe9628d91e1":{"kind":"REWORK","_number":3,"created":"2021-02-25 17:05:03.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777623/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777623/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777623/3"}}},"commit":{"parents":[{"commit":"b19222e6865ba7575d5c81e91cb7971178f68d37","subject":"Merge \"Update IPA docs and test build with extra-hardware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b19222e6865ba7575d5c81e91cb7971178f68d37"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 17:04:56.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 17:04:56.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d96843ab9a571d56f51d04aec58c9fe9628d91e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d96843ab9a571d56f51d04aec58c9fe9628d91e1"}]},"branch":"refs/heads/master"},"0fdde2cee34c45a8747ba4241863c682a9288688":{"kind":"REWORK","_number":4,"created":"2021-02-25 17:11:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777623/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777623/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777623/4"}}},"commit":{"parents":[{"commit":"b19222e6865ba7575d5c81e91cb7971178f68d37","subject":"Merge \"Update IPA docs and test build with extra-hardware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b19222e6865ba7575d5c81e91cb7971178f68d37"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 17:11:03.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 17:11:03.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fdde2cee34c45a8747ba4241863c682a9288688"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fdde2cee34c45a8747ba4241863c682a9288688"}]},"branch":"refs/heads/master"},"1df41558717736cebdc39cb033e44881e15064e8":{"kind":"REWORK","_number":5,"created":"2021-02-25 17:21:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777623/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777623/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777623/5"}}},"commit":{"parents":[{"commit":"b19222e6865ba7575d5c81e91cb7971178f68d37","subject":"Merge \"Update IPA docs and test build with extra-hardware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b19222e6865ba7575d5c81e91cb7971178f68d37"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 17:21:37.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 17:21:37.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1df41558717736cebdc39cb033e44881e15064e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1df41558717736cebdc39cb033e44881e15064e8"}]},"branch":"refs/heads/master"},"c2f6ca2a3673438eb60c1ce0eecbb0c147269e88":{"kind":"REWORK","_number":6,"created":"2021-02-25 17:30:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777623/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777623/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777623/6"}}},"commit":{"parents":[{"commit":"b19222e6865ba7575d5c81e91cb7971178f68d37","subject":"Merge \"Update IPA docs and test build with extra-hardware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b19222e6865ba7575d5c81e91cb7971178f68d37"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 17:30:09.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 17:30:09.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2f6ca2a3673438eb60c1ce0eecbb0c147269e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2f6ca2a3673438eb60c1ce0eecbb0c147269e88"}]},"branch":"refs/heads/master"},"175449c2baa0506e4ca385b8de58970b94a5895e":{"kind":"REWORK","_number":7,"created":"2021-02-25 18:53:28.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777623/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777623/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777623/7"}}},"commit":{"parents":[{"commit":"b19222e6865ba7575d5c81e91cb7971178f68d37","subject":"Merge \"Update IPA docs and test build with extra-hardware\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b19222e6865ba7575d5c81e91cb7971178f68d37"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 18:53:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 18:53:21.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/175449c2baa0506e4ca385b8de58970b94a5895e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/175449c2baa0506e4ca385b8de58970b94a5895e"}]},"branch":"refs/heads/master"},"558e6efc058b7b4e622e0c89483f10fbf03fb0be":{"kind":"REWORK","_number":8,"created":"2021-02-26 08:53:37.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777623/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777623/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777623/8"}}},"commit":{"parents":[{"commit":"47bbb96b29ab30764d6220cdd43e63a1d2072533","subject":"CI: retry overcloud jobs that land on vexxhost clouds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47bbb96b29ab30764d6220cdd43e63a1d2072533"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 18:53:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 08:53:19.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/558e6efc058b7b4e622e0c89483f10fbf03fb0be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/558e6efc058b7b4e622e0c89483f10fbf03fb0be"}]},"branch":"refs/heads/master"},"160163bbf2f161dcbacb02064af2033fcf6674f0":{"kind":"REWORK","_number":9,"created":"2021-02-26 13:45:12.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777623/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777623/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777623/9"}}},"commit":{"parents":[{"commit":"47bbb96b29ab30764d6220cdd43e63a1d2072533","subject":"CI: retry overcloud jobs that land on vexxhost clouds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47bbb96b29ab30764d6220cdd43e63a1d2072533"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 18:53:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 13:44:29.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/160163bbf2f161dcbacb02064af2033fcf6674f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/160163bbf2f161dcbacb02064af2033fcf6674f0"}]},"branch":"refs/heads/master"},"d0f6ff16f9b03956d29b14c41750649eb991137e":{"kind":"REWORK","_number":10,"created":"2021-02-26 13:51:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777623/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777623/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777623/10"}}},"commit":{"parents":[{"commit":"47bbb96b29ab30764d6220cdd43e63a1d2072533","subject":"CI: retry overcloud jobs that land on vexxhost clouds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47bbb96b29ab30764d6220cdd43e63a1d2072533"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 13:51:11.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 13:51:11.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0f6ff16f9b03956d29b14c41750649eb991137e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d0f6ff16f9b03956d29b14c41750649eb991137e"}]},"branch":"refs/heads/master"},"b668b1b8d487baae32907a589dd8d3793ab630c1":{"kind":"REWORK","_number":11,"created":"2021-02-26 14:36:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777623/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777623/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777623/11"}}},"commit":{"parents":[{"commit":"47bbb96b29ab30764d6220cdd43e63a1d2072533","subject":"CI: retry overcloud jobs that land on vexxhost clouds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47bbb96b29ab30764d6220cdd43e63a1d2072533"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 14:36:24.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 14:36:24.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b668b1b8d487baae32907a589dd8d3793ab630c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b668b1b8d487baae32907a589dd8d3793ab630c1"}]},"branch":"refs/heads/master"},"e2c8b7cf6f16146e548880a2196e95909a0661ad":{"kind":"REWORK","_number":12,"created":"2021-02-26 14:54:51.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777623/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777623/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777623/12"}}},"commit":{"parents":[{"commit":"47bbb96b29ab30764d6220cdd43e63a1d2072533","subject":"CI: retry overcloud jobs that land on vexxhost clouds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47bbb96b29ab30764d6220cdd43e63a1d2072533"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 14:54:44.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 14:54:44.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2c8b7cf6f16146e548880a2196e95909a0661ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e2c8b7cf6f16146e548880a2196e95909a0661ad"}]},"branch":"refs/heads/master"},"734d7e4518a3cb0b6074ce313768fa461076fa61":{"kind":"REWORK","_number":13,"created":"2021-02-26 15:39:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777623/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777623/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777623/13"}}},"commit":{"parents":[{"commit":"47bbb96b29ab30764d6220cdd43e63a1d2072533","subject":"CI: retry overcloud jobs that land on vexxhost clouds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47bbb96b29ab30764d6220cdd43e63a1d2072533"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 15:39:27.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 15:39:27.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/734d7e4518a3cb0b6074ce313768fa461076fa61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/734d7e4518a3cb0b6074ce313768fa461076fa61"}]},"branch":"refs/heads/master"},"d7b57811f5d500255b94c604eabc0df2112e9628":{"kind":"REWORK","_number":14,"created":"2021-02-26 16:02:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777623/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777623/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777623/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777623/14"}}},"commit":{"parents":[{"commit":"47bbb96b29ab30764d6220cdd43e63a1d2072533","subject":"CI: retry overcloud jobs that land on vexxhost clouds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47bbb96b29ab30764d6220cdd43e63a1d2072533"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 16:02:04.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 16:02:04.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7b57811f5d500255b94c604eabc0df2112e9628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d7b57811f5d500255b94c604eabc0df2112e9628"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
