)]}'
{"id":"openstack%2Fkayobe~777723","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Ie5ca04428855cbaf505adc57de76ea0b2d404080","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"Ie5ca04428855cbaf505adc57de76ea0b2d404080","subject":"CI: retry overcloud jobs that land on vexxhost clouds","status":"ABANDONED","created":"2021-02-26 07:19:15.000000000","updated":"2021-03-01 09:17:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ae8215838067174caebe8cf205ba2deb31e6335c","_number":777723,"virtual_id_number":777723,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-26 10:42:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-26 07:19:15.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-02-26 08:29:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c231f8680278c00ba41a46cf27875bc43c17476","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-26 07:19:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d24dd6f40c22408d53e1d1c50392daced58a2477","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 08:29:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8779c21b8a948aab81a86770088fda1 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c85dea86f2654648ac73113aeea26de5 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a26a596e4a3430f94423b585fe3384d : SUCCESS in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d4d6d87ff4994ce39153432d721700ca : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0153fa16f0a2404ea51adc1d5852e423 : SUCCESS in 7m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a7cd7afa51444c189b5d5f4833542758 : SUCCESS in 6m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cb8eb0e43deb4de78d890be43050766a : SUCCESS in 10m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/648c0d1cc3ba426cab1a86cd4155f36f : SUCCESS in 19m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/15bbe2d7f7834e2293b5dcd083ff6b74 : SUCCESS in 1h 04m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8679cd3382b84983bea02514aa4b1778 : SUCCESS in 17m 55s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b2d450562b6e4bc087992844b3f72d09 : FAILURE in 1h 09m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2f5e4d4c5db840878a786fce30733707 : SUCCESS in 20m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d5a02c377456430d991c87582c928bbe : SUCCESS in 23m 36s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1cd881a633224dacbd4909b9594fdffe : SUCCESS in 9m 08s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"d28bac68c02d208e4f3b8eb22291e613b0829cbd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-26 08:48:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4bafc82b18653749bf50a630cceaa9c045836a16","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 10:42:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2282e174634c43a88cc322f6734d595c : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16915fc9ec8447c0b087f94c7b7d1fed : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/08f12c06f9374b55b2f843c815e63e5c : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ec83099460e14ffe86dcc107fb8a9fe4 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9be73c271b644cc947ae9b6cd4017ca : SUCCESS in 6m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/46b8ba50c8ab4de0b325a148e239ae03 : SUCCESS in 4m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b7e750e0f6584a3c9001ec8a4d83935c : SUCCESS in 9m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b3b841b3103143b4bec1404f7635b7d9 : SUCCESS in 17m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1b8466ebb21b4bc6b2406b4c68ec558a : SUCCESS in 1h 10m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/303a9c46e817425abbb495e704a3b1cc : SUCCESS in 15m 58s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2b0c4826425d42b09674e0e53f1a523d : SUCCESS in 1h 51m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/200be31d576b4a64b7bb2e5b6780821a : SUCCESS in 16m 22s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/263da5871d28480591d325e8e815cc6d : SUCCESS in 19m 43s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ec67e4f8fe744f14a92bfbaafa3bb106 : SUCCESS in 9m 58s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"ae8215838067174caebe8cf205ba2deb31e6335c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-01 09:17:56.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a15a595aa0dabf5640340f98eb26371fad100d6","revisions":{"6a15a595aa0dabf5640340f98eb26371fad100d6":{"kind":"REWORK","_number":1,"created":"2021-02-26 07:19:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/23/777723/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/777723/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777723/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777723/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/777723/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/777723/1"}}},"commit":{"parents":[{"commit":"1a6f11a896dae549d9d065f7b8a1039f89668459","subject":"Merge \"Fix --limit with commas\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a6f11a896dae549d9d065f7b8a1039f89668459"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-25 09:23:06.000000000","tz":0},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 07:17:50.000000000","tz":60},"subject":"CI: retry overcloud jobs that land on vexxhost clouds","message":"CI: retry overcloud jobs that land on vexxhost clouds\n\nJobs that test baremetal deployment are currently failing when run on\nVexxhost infrastructure. The seed VM job also has trouble on Vexxhost,\nalthough does sometimes pass. It\u0027s not clear why yet, and we are still\ninvestigating. Until we have a solution, fail in the pre phase when we\nland on vexxhost clouds, in order to trigger the zuul retry mechanism.\n\nAlso use the normal number (5) of attempts for upgrade jobs, since these\nno longer deploy the previous release in the pre phase.\n\nNote: compared to the original commit, the custom configuration for\nironic is removed since it only applies to Wallaby.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2021-February/020685.html\n\nChange-Id: Ie5ca04428855cbaf505adc57de76ea0b2d404080\n(cherry picked from commit 47bbb96b29ab30764d6220cdd43e63a1d2072533)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a15a595aa0dabf5640340f98eb26371fad100d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6a15a595aa0dabf5640340f98eb26371fad100d6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
