)]}'
{"id":"openstack%2Fkayobe~777727","triplet_id":"openstack%2Fkayobe~master~I87b63966b55f04f80fd0255fbf2732d527732187","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I87b63966b55f04f80fd0255fbf2732d527732187","subject":"Revert ironic custom configuration","status":"MERGED","created":"2021-02-26 08:58:01.000000000","updated":"2021-03-01 15:24:04.000000000","submitted":"2021-03-01 15:22:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777727","meta_rev_id":"0258503f32db8daa6de122988e1b8d993c0f32aa","_number":777727,"virtual_id_number":777727,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-01 15:22:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-02-26 09:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-03-01 13:45:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-26 09:48:18.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-02-26 10:40:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dcd47e0633602ec9f66333c149431489a3ccadf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-26 08:58:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d2a055d95f1214c3f1bc5e97ee04a55131ce4c22","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-26 09:48:18.000000000","message":"Patch Set 1: Code-Review+2\n\nThanks","accounts_in_message":[],"_revision_number":1},{"id":"9d4676161f77e3bbac43e13b649f2d05f22c467b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 10:40:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa58f9d62abb4fc78dda5a6e57ce5ee9 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9032948898e044d4b79b02416e28fc43 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/70fd6c38668b4885be52535078a61d9a : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58bddbc642934ee3b52318a6281334ee : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da3b064b71dd4c4c9b19da7a1c5fe014 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/960f3ca101cc418994c60361b4c49510 : SUCCESS in 9m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2f98cd53547d4c1eaf589bc3430bb38f : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3b60189b26ba4e75bb3a1199f8c18bdf : SUCCESS in 10m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e77a5e65bd4148d59aff11cf78336640 : SUCCESS in 20m 35s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/17e3ea3b3cca4937aeca02e503795cfa : SUCCESS in 54m 17s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/e1cef35fc693412e9089cedb5cd226f8 : SUCCESS in 1h 05m 00s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/432f7fedda8d4e70b25415be7121213c : SUCCESS in 14m 36s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/502f2f290d67447a954b2deb1dbf5b5e : SUCCESS in 1h 40m 27s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ce139bd86ad14a368e6dd7a88c5378cf : SUCCESS in 37m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/767616725dbe4b6bada324eb39fc3bbc : SUCCESS in 27m 01s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/82997d3fcf684649a1d5c26a594adba9 : SUCCESS in 11m 59s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"6d88856b31ef1a997c7789b54938ead8312d8253","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-01 13:45:35.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"40dc2562d4a4db57ea88232bfba179ffabbdc798","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 13:45:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dcbe81f2414fcd41f3ca373801008eb096417b9e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 15:22:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0a7455f467744d09e9c05071618bf42 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2f716ae9c654eddbffc473675940d34 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35d4e2ac745e411989094336fc5ef1b2 : SUCCESS in 8m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63ad76f2b00944cc9fdf7182f8f2ad2e : SUCCESS in 6m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4f95a96d49a84257bc7f6b05c2c9005e : SUCCESS in 5m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4503001461a84655b8fdcb67fadf056e : SUCCESS in 9m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/23b98908d59c4ff0984743fb37865e8a : SUCCESS in 21m 18s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/10a1d416a177461b83c6fda5ec1f9822 : SUCCESS in 59m 31s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f8e661aeb1814eb29afb12c2835982a6 : SUCCESS in 56m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/79ec35fd02ae418daf8c9ae26bc36faa : SUCCESS in 28m 30s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7d2f0c8532a64f0e85bdb2e6220543e6 : SUCCESS in 1h 33m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/439a33bd30ab4e7690d0f450343b9dbe : SUCCESS in 23m 48s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c7bed6ee8dea4dcfa6b7eb9b1b8716bd : SUCCESS in 23m 19s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/04f5b68389a145a5bec4999f1d54c1a0 : SUCCESS in 10m 25s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"b24ab9a6e746bf29654cc11ffe0bbd58833c08c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 15:22:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0258503f32db8daa6de122988e1b8d993c0f32aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 15:24:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de04eebc86ff4819973d5d72db2d4db9 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5af7cac17ea72ff46da85d4c8bcdf373db91c768","revisions":{"5af7cac17ea72ff46da85d4c8bcdf373db91c768":{"kind":"REWORK","_number":1,"created":"2021-02-26 08:58:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/27/777727/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/27/777727/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/777727/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/777727/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/27/777727/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/27/777727/1"}}},"commit":{"parents":[{"commit":"47bbb96b29ab30764d6220cdd43e63a1d2072533","subject":"CI: retry overcloud jobs that land on vexxhost clouds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47bbb96b29ab30764d6220cdd43e63a1d2072533"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 08:56:18.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 08:56:18.000000000","tz":60},"subject":"Revert ironic custom configuration","message":"Revert ironic custom configuration\n\nThis is now merged into kayobe-config-dev [1].\n\n[1] https://review.opendev.org/c/openstack/kayobe-config-dev/+/777076\n\nChange-Id: I87b63966b55f04f80fd0255fbf2732d527732187\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5af7cac17ea72ff46da85d4c8bcdf373db91c768"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5af7cac17ea72ff46da85d4c8bcdf373db91c768"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
