)]}'
{"id":"openstack%2Fkayobe~777737","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I782be627c6e78370555bc7d155261fc15fc93348","project":"openstack/kayobe","branch":"stable/victoria","hashtags":[],"change_id":"I782be627c6e78370555bc7d155261fc15fc93348","subject":"[DNM] Test Docker pull rate limits","status":"ABANDONED","created":"2021-02-26 10:20:24.000000000","updated":"2021-03-05 21:08:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bf7cebaf7932a2fe6f84fb56669d3f0a42a01adf","_number":777737,"virtual_id_number":777737,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-02-26 12:22:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-26 12:22:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1df6b0a862c8f3430cf33c574ec7d0369f8d5483","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-02-26 10:20:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cee4599817991e1b9f55b78d7a33c26473a8ee32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 12:22:12.000000000","message":"Patch Set 1: Verified+1\n\n(20 comments)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08080e7b3d284d17a7a44339c8f97bbb : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9f154595e114cb3bc2ccc72c80d6d2d : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ca6e4aaf2644a08828f0aaf362a185c : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89adb7ad54c04369b0f34521d55ac322 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/971cee32e9a14a93b17489983e7952ae : SUCCESS in 7m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/08b13af83d1d44b99263decbc233017b : SUCCESS in 4m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d21b1f5c8c6e497e83609d9fcaf5625f : SUCCESS in 6m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/718cf3972aa6477da9c02e40463da801 : SUCCESS in 14m 00s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7ccc76fc94f54700af63ee66248e6a96 : SUCCESS in 1h 07m 52s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/4d9c5841b8cd467fa9ea7cc2d92fdf49 : SUCCESS in 59m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9371309bdd884ade91ece5ac6234e81c : SUCCESS in 17m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ca4d7a75fa0f422a893ce5b6c9723f4c : SUCCESS in 1h 55m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cefd0834df874ad68f740ba7b43b8740 : SUCCESS in 21m 40s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f0631d4b95434ca582ccbf0e7e57a86c : SUCCESS in 17m 49s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1415bd23caa44af2bc67468bf5e077c4 : SUCCESS in 11m 28s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"bf7cebaf7932a2fe6f84fb56669d3f0a42a01adf","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-05 21:08:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"57d06e1dfd0fe615c9e3a353269744fee6ab3040","revisions":{"57d06e1dfd0fe615c9e3a353269744fee6ab3040":{"kind":"REWORK","_number":1,"created":"2021-02-26 10:20:24.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/37/777737/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/777737/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/777737/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/777737/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/777737/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/777737/1"}}},"commit":{"parents":[{"commit":"7c0de4a579bc102566a996586198c5f5da847ade","subject":"CI: retry overcloud jobs that land on vexxhost clouds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7c0de4a579bc102566a996586198c5f5da847ade"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-25 18:53:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-26 10:20:10.000000000","tz":60},"subject":"[DNM] Test Docker pull rate limits","message":"[DNM] Test Docker pull rate limits\n\nChange-Id: I782be627c6e78370555bc7d155261fc15fc93348\n(cherry picked from commit 558e6efc058b7b4e622e0c89483f10fbf03fb0be)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57d06e1dfd0fe615c9e3a353269744fee6ab3040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/57d06e1dfd0fe615c9e3a353269744fee6ab3040"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
