)]}'
{"id":"openstack%2Fkayobe~777741","triplet_id":"openstack%2Fkayobe~master~I6bc8da2b8da903e09b97df8cd95c68a562c11db9","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I6bc8da2b8da903e09b97df8cd95c68a562c11db9","subject":"CI: Fix overcloud and seed VM jobs on vexxhost clouds","status":"MERGED","created":"2021-02-26 11:12:33.000000000","updated":"2021-02-27 00:48:38.000000000","submitted":"2021-02-27 00:47:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777741","meta_rev_id":"ff32e870f5a7cbebebe8bde4a133dbbd3cf475f2","_number":777741,"virtual_id_number":777741,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-27 00:47:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-02-26 21:07:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-02-26 21:07:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-26 11:41:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d8bce6e4fe30c35ce7948a2373f78ca26d623188","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-26 11:12:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4983e3f9eabd6bb6b2704ac4d9781c68b5a86c35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 11:41:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ec3c3ae73e342c99baacad18027992b : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9714eea9eb814b59b973ff3e6bbbe614 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/30af325fbd7f46c98df110f9b72b460d : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/56a167bf3c8d48aaaacd7432f723d7bc : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/908089a056ad486b996549fb7b0734e1 : SUCCESS in 4m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2127133153e543b4901d24f443181bb3 : SUCCESS in 6m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/89e4d39dc21b47aaa61b369ac97a7c5c : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f0ec5ae4854240d98aeebc193b4e9ae8 : SUCCESS in 8m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3bbcdb1cd37944bf80e988638d6f4ecf : SUCCESS in 16m 33s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/cdb1482328a14cadb2a8d5508a554224 : FAILURE in 13m 00s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/1348ce12d3ce4272bb893192178ae86a : FAILURE in 12m 58s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/59d4be00e28c4454828ce03536959bcb : FAILURE in 15m 48s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/aa40f28c75a543f984528fa660ad5a25 : FAILURE in 12m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d0f8f2df89d3441d8d8c684ebdfeb994 : SUCCESS in 27m 18s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0d57cdb2815f4b3db48a4c42393af876 : FAILURE in 17m 28s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7a703856497e4b1189a0a275ab3c770a : SUCCESS in 10m 59s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"e6b4d048fc13b2aad4e95ed19914377161255d4f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-26 12:19:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4fac2c0e1e08217860e4ddcb21372fba740612dc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-26 12:19:46.000000000","message":"Patch Set 1:\n\nLots of pull request limits in limestone","accounts_in_message":[],"_revision_number":1},{"id":"70259f69ac3197f2f35b68e154702dfee32f7437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 12:40:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d87ec9140f7d47ae959d727cbbd414b3 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eec828e27e1644a987e96e1c677d174b : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcf66e7c42284912ae8a05bea069ea68 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d0bf8b7cb6f4da588c203e2b4567425 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89813663e2144993bcba0ce6f629ec38 : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08068faad453476a9dabb0b2a158c60c : SUCCESS in 6m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/04ab8769d1544da0958542a998ed5c4f : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bc9cf8fa95f341b3a44072cdaa38d1bc : SUCCESS in 7m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/23fcbbef75f54f7baaacc48571e3c762 : SUCCESS in 15m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/297ecfc5f4e44221954ba2cb10ca2fc9 : FAILURE in 10m 30s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/b0f7c4cf9fd9483baa342c8707af6b36 : FAILURE in 4m 23s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f2c5d7bb599c49129a9f42aeb10f12f2 : FAILURE in 14m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/868a6305793543dd9df061b8d9ef355b : FAILURE in 10m 45s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0a0dd495dee04746acc6da8f312349fb : FAILURE in 11m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1c4e822dc84246acb7492b706bd99532 : SUCCESS in 17m 23s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/12dfba40144a4ec790a104ee0594412a : SUCCESS in 10m 10s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"65d971d5798ae06a820626c2b1951963bb1b2afb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-26 21:07:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2b978a7317819fd562fc1ac313e063cb8d173fc7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-02-26 21:07:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"809b6d215b0233d710479626729cf3c42b65a6c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 22:55:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ea611296e5e475eb285391759c17405 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/758294b176b0480eade7b203c0008625 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/944c78dd558f40daaf3a1ecaca8e9710 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e0431d2e3604824a9cf569a536aa8c2 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce08901d2641451bbadb3bf65712f0dd : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4870d5c00e04fc9b5210fde51ba5629 : SUCCESS in 7m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7cd805a34b60457394d35e62594726cc : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e14ded319df34d078e764baa83bbde76 : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/03e2c206f12c4130b2aca7da91b6259c : SUCCESS in 19m 07s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d3fc7fc46a0b4367b29652add40ef1f9 : SUCCESS in 1h 38m 34s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3d3df91273cf46de9f686c6686039c28 : SUCCESS in 1h 10m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c9ab63b1731c4b0ab299705f3155c355 : SUCCESS in 18m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cced77d99a8d4e319894847e8dfd3acb : SUCCESS in 1h 15m 27s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bda0d3566d254369aebeba7132b7aa7d : SUCCESS in 35m 25s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8dbd6e32cfe7417b9bdadb397fa1484d : SUCCESS in 28m 56s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/86ce49a03f034b9294f82fca29cdfa0d : SUCCESS in 12m 29s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"0e5e2a29e73e67b2bc9091e089644e79086415a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 22:55:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a9c65797864201fe6e282c85f9ec13b3bc89a851","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 00:47:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d811e65f7cb44d0a498da1a56fc0dbc : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8227d26bd42c4fc3979d4cbb60e00ca5 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e04ba6b527a94d60afdfe34414e9a816 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/394c0d5d4c91431f8541a5218aee1994 : SUCCESS in 7m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1d4c9631e50d47ce83a72187c8ef961d : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cfa3a9a32ffb41928f59a1971c4d26a2 : SUCCESS in 8m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9456f1b24f78462e84c7af68c9146712 : SUCCESS in 19m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6d510f3ac09748ed9da42d570cdb2a84 : SUCCESS in 1h 02m 11s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/405fa06963334d60b506a104e5e45d8a : SUCCESS in 1h 01m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0fbce627b8134e999f741b4293c477b4 : SUCCESS in 15m 48s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4f923d701319405f84fa20d37ed6bc83 : SUCCESS in 1h 47m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5c0fb4cd4c5e4f5a959cbe240618104f : SUCCESS in 38m 50s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4d556021a0874edcb6cdd23e871a3195 : SUCCESS in 25m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4b3c20e86ce04cfcb6ed5cfd0686b7ff : SUCCESS in 10m 17s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"ca9bb025a3c5c6366a561aee8e3257b63f87b2d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 00:47:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ff32e870f5a7cbebebe8bde4a133dbbd3cf475f2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-27 00:48:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57dde9de8df44b088f3a4a4f4be3c8e2 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9d9afcfba09ef73c6663a38fb60b703caeb56c7","revisions":{"f9d9afcfba09ef73c6663a38fb60b703caeb56c7":{"kind":"REWORK","_number":1,"created":"2021-02-26 11:12:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/41/777741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/777741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/777741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/777741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/777741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/777741/1"}}},"commit":{"parents":[{"commit":"47bbb96b29ab30764d6220cdd43e63a1d2072533","subject":"CI: retry overcloud jobs that land on vexxhost clouds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47bbb96b29ab30764d6220cdd43e63a1d2072533"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-26 11:02:47.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-26 11:11:18.000000000","tz":0},"subject":"CI: Fix overcloud and seed VM jobs on vexxhost clouds","message":"CI: Fix overcloud and seed VM jobs on vexxhost clouds\n\nThis partially reverts commit 47bbb96b29ab30764d6220cdd43e63a1d2072533\nwhich triggered a retry on vexxhost clouds.\n\nThe issue was introduced in Ie8fd965165e8d347d27528a2c16d0647e412ccdc,\nwhich applied some fixes for CentOS 8.3, and inadvertently removed\nthe Tenks variable that forces the use of qemu for \u0027bare metal\u0027 VMs.\nThis lead to autodetection of KVM, which does not work well when nested\nin all CI cloud providers.\n\nThis change fixes the issue by forcing the use of qemu for the overcloud\nonce more. It also adds a similar option for the seed VM job.\n\nChange-Id: I6bc8da2b8da903e09b97df8cd95c68a562c11db9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9d9afcfba09ef73c6663a38fb60b703caeb56c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9d9afcfba09ef73c6663a38fb60b703caeb56c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
