)]}'
{"id":"openstack%2Fkayobe~778107","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Ia885a226de30d72610dca2eb2f194cef10193095","project":"openstack/kayobe","branch":"stable/victoria","topic":"story/2008546-stable/victoria","hashtags":[],"change_id":"Ia885a226de30d72610dca2eb2f194cef10193095","subject":"Quote kolla limit when calling kolla-ansible","status":"MERGED","created":"2021-03-02 10:50:40.000000000","updated":"2021-03-02 18:46:44.000000000","submitted":"2021-03-02 18:44:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778107-story/2008546-stable/victoria","cherry_pick_of_change":771840,"cherry_pick_of_patch_set":2,"meta_rev_id":"cf47dd6abadee9b6c2d733b3a7d7e48ab3d9cc0b","_number":778107,"virtual_id_number":778107,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 18:44:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-03-02 15:46:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-02 16:46:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-02 16:46:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 10:50:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-03-02 12:32:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"70aa1a6a28e44226e729599cbc7a3d3c71fdee04","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-02 10:50:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cac6a252a0c5bbd5a8f74b61a97ed08293039393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 12:32:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b83ce06dc33d4800ba82f6d6b5edf739 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ae8b8dfa3b24348a4bdcee41c11901e : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d961de1194240bba7e2537a49a5f8f6 : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6f63199bd5549ccabb623ab97b56583 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4eeb565c3774f2593ea1257ae419583 : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af6b218440374474996f0f1230dddec7 : SUCCESS in 4m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7e0f99fee1124d9980eef3ca842594e7 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a7dd0942744d4090b37e4ffdb4b4f69c : SUCCESS in 8m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1428419a9a8a4baabc02f4a57160850d : SUCCESS in 18m 59s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a90d8b7da6964f739d2e2bf882d0432c : SUCCESS in 58m 18s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f02bb062a2414a67be9f15666218b5ce : SUCCESS in 1h 01m 20s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/49dd083b36e44a0885cb94184c28b445 : SUCCESS in 15m 43s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bac29dc70da34a5181f5c873c243d1b0 : FAILURE in 11m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/10f8081b54e94327aa44bfa13f10369a : SUCCESS in 33m 16s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/53ee83f18bf24613ad31760fca98b0c2 : SUCCESS in 27m 39s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/46b6be598c4e44eaa60f2fb2e8249694 : SUCCESS in 10m 17s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"f7d054358ab1592c20566ddf1e208934a7eca3bd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-02 13:20:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a51fe19dbe1082bb5eee8718405cb8d11ecc8296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 15:35:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e527eb19b0de457c8b87cf95369d15bb : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/626796142a2b4534a005ad4a119d616d : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/797b6dc8d12c4702b1d818bc05d94d37 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4c1bedb424944f0aa46977a7f7f0d06 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/199c24e86f584f43ba7680dd2e84bc58 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4a407f6176674bec9d03b219744b4c72 : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/92336bddaa7947de96f2c8831b58ccea : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e401054b81074334b509a9d652537789 : SUCCESS in 8m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d328433f5aed422b8d168364d4f20408 : SUCCESS in 20m 31s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a838c310b3394261873e07bb834fec41 : SUCCESS in 59m 25s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/9c6c5e73acba4a658491c2352c198bfb : SUCCESS in 1h 13m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0401e02a2d994eb4ad344789b1e98ff1 : SUCCESS in 16m 36s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d3c5ede8db5744e98d9261963606121f : SUCCESS in 2h 08m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3870f9d197944c289d2c64602786c865 : SUCCESS in 25m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/98d3cde46698420fa08c7bb1c63dce11 : SUCCESS in 22m 30s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/df21b8ad3deb4571b26b59fa70799de4 : SUCCESS in 8m 03s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"a4ed890cb81a9c702b151e61240c010142f8fa69","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-02 15:46:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"203cc7732568e799364c33e759c0b5d7df90ee3e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-02 16:46:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a31fb097c574000bfd750c99c18cdcf06555ed98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 16:47:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"133ecb2f147bc78fbc99c257b9579902fa9879de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 18:44:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d33eb2e9173a4143b7ff56e5caab0575 : SUCCESS in 7m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7d00a501f68417cacbcfe9623e98df0 : SUCCESS in 8m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb8a1ce4494346f7906e33d2598a7f5e : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e0b0dc5b45643ff945ee3001ca7ea9c : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1c671a81296646509e5d604f62671995 : SUCCESS in 5m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4104dae666f74f9f9196f1a34fe3934a : SUCCESS in 10m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/355a7fc2923e47899afa92c86a152bb4 : SUCCESS in 12m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/79ffa8bb45f4404db45f377a39e0f8ce : SUCCESS in 20m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e9202ee2b30641589129b843d2a716d0 : SUCCESS in 1h 03m 08s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/5076b9a5b62a4c8ca581bcfa1cb777dd : SUCCESS in 1h 06m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3996b2b2f9c1461a86bbe634712ffd00 : SUCCESS in 16m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d9e6b822820a40a69940e1f768066939 : SUCCESS in 1h 53m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dc4bda54b2954f52add09b5925e2262f : SUCCESS in 28m 02s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/66d97c5f159d493a8c4c7212a89d4a23 : SUCCESS in 30m 04s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ec6aa5d271dd4b23a6a4adedffe515f9 : SUCCESS in 12m 11s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"1bfd03e3955320d6f1092eedf02f6e2a2ba51443","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 18:44:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cf47dd6abadee9b6c2d733b3a7d7e48ab3d9cc0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 18:46:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f3d5e82c6e74752b1d98aa412b8e18e : SUCCESS in 1m 11s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b370c01929284bd6aaf844946cf6400c : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cbb00dd88d0ed98f24d0450b1ddd0a0ac2429b69","revisions":{"cbb00dd88d0ed98f24d0450b1ddd0a0ac2429b69":{"kind":"REWORK","_number":1,"created":"2021-03-02 10:50:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/07/778107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/778107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/778107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/778107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/778107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/778107/1"}}},"commit":{"parents":[{"commit":"94ffdeb5ce7152cf1f9c01b9e37db1d339609363","subject":"Merge \"Fix copying Swift ring files\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/94ffdeb5ce7152cf1f9c01b9e37db1d339609363"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-21 16:12:07.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-02 10:50:40.000000000","tz":0},"subject":"Quote kolla limit when calling kolla-ansible","message":"Quote kolla limit when calling kolla-ansible\n\nThis is necessary for correctly passing limits containing special\ncharacters such as ~ or \u0026.\n\nChange-Id: Ia885a226de30d72610dca2eb2f194cef10193095\nStory: 2008546\nTask: 41651\n(cherry picked from commit a1c9a8152508785d5babe4ed98eb6a743932a229)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbb00dd88d0ed98f24d0450b1ddd0a0ac2429b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbb00dd88d0ed98f24d0450b1ddd0a0ac2429b69"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
