)]}'
{"id":"openstack%2Fkayobe~778108","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Ia885a226de30d72610dca2eb2f194cef10193095","project":"openstack/kayobe","branch":"stable/ussuri","topic":"story/2008546-stable/ussuri","hashtags":[],"change_id":"Ia885a226de30d72610dca2eb2f194cef10193095","subject":"Quote kolla limit when calling kolla-ansible","status":"MERGED","created":"2021-03-02 10:50:50.000000000","updated":"2021-03-02 22:25:22.000000000","submitted":"2021-03-02 22:24:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778108-story/2008546-stable/ussuri","cherry_pick_of_change":771840,"cherry_pick_of_patch_set":2,"meta_rev_id":"a43b5efe03b23d18c7d0f0d1fe01d2027b37cd65","_number":778108,"virtual_id_number":778108,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 22:24:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2021-03-02 13:54:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2021-03-02 13:22:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2021-03-02 13:54:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 10:50:50.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-03-02 13:19:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-02 13:54:19.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"e8e29ee4e7e28c53d98b2f8d2f209ce2bb8232d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-02 10:50:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"eb38dc722a21c8c772b606a7bbd4b4fd7bd5eaff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 13:19:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9836c7b4a5d74d0c842750095ec11317 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07df87dc6b42446a8b4235b26de4063f : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1e1e1fece124f2dbcc71e86202c64e1 : SUCCESS in 8m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c61c20b9e87649bfa85613195f6c78b2 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfbe2c8933b9436b81a7679a110572d2 : SUCCESS in 10m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/44613ba96760461fa292c559ee6d8937 : SUCCESS in 4m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/58c292c6a25e43ff99b54310c27d97cc : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9a39da6e4c814f8eb8cf14972bb24dc6 : SUCCESS in 9m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6a5e5eb7353d44488ced9f84d2f75cf1 : SUCCESS in 18m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6e7bbb7efcc4434fb283643847f7532d : SUCCESS in 1h 00m 07s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4382eeef375c4afda65038d39bc9d4ea : SUCCESS in 14m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f268439584ab42c89d028219b72d6a5f : SUCCESS in 1h 45m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d888e8dbeb654b86928188e4f2f63f6b : SUCCESS in 16m 24s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/01961dcfd3834e6ba9e3f1105afa7a1c : SUCCESS in 23m 06s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/8df27f88b7da43c8962ea6ef1428c5f5 : SUCCESS in 11m 02s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"a1a9f890bd1b937e3675c456407e51abfa7a4911","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-02 13:22:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b0fed54792ca902deac61614a15e721008f625fc","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-03-02 13:54:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4b221123efb209b9c9f415b5868df8598e6221c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 13:54:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91f0f4bddb8ca03dbbae532c911252897833587f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 15:04:32.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1df1691a6474e0aa14a2503cff3dca5 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2debaa5369064683a4ddd7480f7a7f2d : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/92454b1e669744658d17a1b044418486 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c46a3a32fcc4f3ab5693efd9169d9cb : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf44eab5aeed46469383fd7d2cae6dde : SUCCESS in 3m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/40e6a969914c42c8bf4a4d6c21ec4d85 : SUCCESS in 6m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4b7852cae86c42c0bd6670bdca67b3a1 : SUCCESS in 10m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/69cfae0bd5184dc095a1a35bd40bf25a : SUCCESS in 19m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/36d34f8151994027bbce7145d43019db : SUCCESS in 1h 09m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3fbcdd7484054fb79d1a0de0574d4bd5 : SUCCESS in 14m 11s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/aa2dd1755fd84c00a877b443655b3a32 : FAILURE in 14m 15s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9266e96f76cc43babb984160c4b8396e : SUCCESS in 16m 12s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/541d571bec5349daa4bff58d87d3f1fd : SUCCESS in 25m 04s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/efa646eda6084f9c94385280058d5d83 : SUCCESS in 9m 35s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"9a88743feef8d40efeff9a3442b97875783b8730","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-02 18:50:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"71ddd466e84b304096cc2e40f19220b8ebd5855e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 20:31:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/47c96151e23a452f8da836e58199c6d0 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b2e80c3c00643dca4539341d83970a4 : SUCCESS in 6m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dcd581f50afe47dd8f433136a9419a51 : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/53213fbe72c44486a0b15fb96fa8b11f : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/850d265018d649e6af92cff4a3e839ec : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e524d2990f584b7c81ba905644206225 : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a513fd79bf0c45fc8eca0f4452c8adcf : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/82c6dc1c01f7417cbf0d592b4f89fec6 : SUCCESS in 8m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b01d26b3c3044292b20852c106dbd240 : SUCCESS in 17m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/797699be2687435e866e08d67ce2de24 : SUCCESS in 46m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/44eac3f02c9642c8bab55aa33867e9e1 : SUCCESS in 15m 15s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/22a714d6cb7c48fd84bea2821f97f4f4 : SUCCESS in 1h 39m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d495f61f8cf24ed690a7e29a6c3bef7b : SUCCESS in 15m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8aa7e1f387e24a85a9cff469f5ef243a : SUCCESS in 23m 12s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ee51b926c4dc401a91a697e1ee90a414 : SUCCESS in 10m 04s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"c8ad542a76e63c9b19b4f9fb526b47b2ebd409f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 20:31:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"480c227573fdac66cb7a0947a84f15b6c4c470d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 22:24:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c17bd0acd90473d8e12565988b7292b : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/643e2402e5484b34816d409750450603 : SUCCESS in 4m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c385c302ffc4477d9ca525b57dfa9493 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24819f913db24c6c99e6ae9d1dcfaf76 : SUCCESS in 7m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cd02ae855eb437eabf5d9c29a6316b3 : SUCCESS in 3m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0fffecf3f33f4f69869ddec7f84cc0ef : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/87eb038dc8954148b9da9fde263619eb : SUCCESS in 8m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/dc5a30447a064b23bf7cdf0d523fdba5 : SUCCESS in 18m 40s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e79c3535815e4cdd854b974a4b801108 : SUCCESS in 1h 04m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/aa87c6094aaa433791e078dce97153b6 : SUCCESS in 14m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4e61ce608a5e4f7b9c58c7ee7fed78a1 : SUCCESS in 1h 51m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c7d1511ea72b4e208ace37e2270eb861 : SUCCESS in 15m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/242bd7480c5a4abbae47a0c6ffe30b63 : SUCCESS in 24m 50s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/1e39ab0583e942ca9f3eae528b15da43 : SUCCESS in 10m 13s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"f64ec5bc9aff47041bb6eda06b9d9604f974ee69","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 22:24:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a43b5efe03b23d18c7d0f0d1fe01d2027b37cd65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 22:25:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e560669775b54e9bbf04a7ae09194f73 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a8deedca0e034b118cbc3f0687eb53c0 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d2c5f8e3f09a5027d80d689966e9d01fe9409774","revisions":{"d2c5f8e3f09a5027d80d689966e9d01fe9409774":{"kind":"REWORK","_number":1,"created":"2021-03-02 10:50:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/778108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/08/778108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/778108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/778108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/778108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/08/778108/1"}}},"commit":{"parents":[{"commit":"8eadb255a7edbe5cd1483f510b8108eaf9e5dff8","subject":"CI: Fix overcloud and seed VM jobs on vexxhost clouds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8eadb255a7edbe5cd1483f510b8108eaf9e5dff8"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-01-21 16:12:07.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-02 10:50:50.000000000","tz":0},"subject":"Quote kolla limit when calling kolla-ansible","message":"Quote kolla limit when calling kolla-ansible\n\nThis is necessary for correctly passing limits containing special\ncharacters such as ~ or \u0026.\n\nChange-Id: Ia885a226de30d72610dca2eb2f194cef10193095\nStory: 2008546\nTask: 41651\n(cherry picked from commit a1c9a8152508785d5babe4ed98eb6a743932a229)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2c5f8e3f09a5027d80d689966e9d01fe9409774"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d2c5f8e3f09a5027d80d689966e9d01fe9409774"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
