)]}'
{"id":"openstack%2Fkayobe~778720","triplet_id":"openstack%2Fkayobe~master~I18f67eda4e48058f3f402b9d0e692d6eeb02855f","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I18f67eda4e48058f3f402b9d0e692d6eeb02855f","subject":"Avoid fact gathering for group_by","status":"MERGED","created":"2021-03-04 14:58:58.000000000","updated":"2021-03-29 22:01:00.000000000","submitted":"2021-03-29 21:59:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"778720","meta_rev_id":"aa4eb88eb6b74d070a63477f8b531260b17255d2","_number":778720,"virtual_id_number":778720,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-29 21:59:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-29 14:59:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-29 20:11:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-29 20:11:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-04 16:41:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 14:59:45.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-03-29 20:11:46.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"c6cadde8c27c9274caa96493573adb8afe9213a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-04 14:58:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf8c6edc1fd64143faa103bba8a6c27be324a19a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-04 15:15:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c9bf73b00c1a3ae40c13185ad0c24318745f4342","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 16:41:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/916133b515f4484c9bcb86018a471edb : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4acc9a3843c4cba9c00eff83105b11f : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f56ec3218714c9b868b0d118695d4f1 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52bfcabb43df4ae691287e639d58bfed : SUCCESS in 8m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb57ef5354844e3b8fd41985473a26d1 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43f45e9251fe476b8a700565303734cf : SUCCESS in 8m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0cb3522e7eaf4ce79fa3c1ec67542a47 : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9b7387bbc33646c698d3698609310bf2 : SUCCESS in 9m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8704c34933684adf8bedf395b9433c54 : SUCCESS in 21m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/07a913464c684bc78474aa01f1c78efb : FAILURE in 44m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b0c7424a2bfd4922b25ddedfab919d34 : FAILURE in 37m 19s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/06d8a345e2204391befd83c2acc07f19 : FAILURE in 47m 38s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4b205952f6fd43acb1dca61e7dfd8879 : SUCCESS in 18m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b29edb36346943adb2a8e9054e8d02d6 : FAILURE in 1h 06m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/de2f1c37191c44388ed88e830a04b859 : SUCCESS in 27m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7f1ccd7e191f42c3a5b8746d2cb12b4d : SUCCESS in 24m 20s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f800baa667ae4639930b4602ecfa0a9c : SUCCESS in 11m 57s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"394fe316a18e35c21af396345fd40f362ee94c87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-04 17:32:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8b953902ed66682db927083abae8bd20e51ab065","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-04 19:27:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e732d8aeb1f94ef4b24c2897cd12fd4d : SUCCESS in 9m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85a38616c7494cac9a490753c9f79e3f : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8f49a040b8d94b7f80465000624c7fbc : SUCCESS in 8m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/984cc9bdb7c64d1ea6fafcb1b6dff550 : SUCCESS in 7m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/514738e663ca433a9c2a1b8447de6ea1 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee72c7f3116148d99f960419d82d907e : SUCCESS in 11m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5220ec8eec4a413a8441982fd25733da : SUCCESS in 9m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4a59c1ef72cd41a2bbe006c064f87b7b : SUCCESS in 11m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/162710488a3744f1a1f2f3ddfa2f092f : SUCCESS in 19m 59s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2a72c932573e4ddcb1b3e9998d024a3c : SUCCESS in 1h 00m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/393436ce49d945009be20ee82d6041a2 : SUCCESS in 41m 42s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/a0afdab1a80e49eab38a0a0841514b29 : SUCCESS in 1h 03m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bccf7cc94be249d897056faa40c0a7db : SUCCESS in 16m 42s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04e08ca4cbf0413db1db8d5db2a32b43 : FAILURE in 8m 11s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0bf17ddedaf5490caabd1cc986398f67 : SUCCESS in 1h 40m 24s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/34f9b9a60e124b17804cbe45ddd309f1 : SUCCESS in 30m 59s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2a589abd97fd4f19a8ebb36f3da243cc : SUCCESS in 27m 52s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/78f5db0f60d148fe804101c066b42c03 : SUCCESS in 14m 37s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2732de15f0b4e44b9a6883f8e5d4908 : RETRY_LIMIT in 5m 08s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"089404ac683414f8444e9f90b15718aee91548f5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 09:38:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"8a681b923c8a95a90be406021b8c474b2458a311","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 11:26:10.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd945d33061044db9a7960bec979c8f4 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1147acd55e7a4844a988a9229996479c : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e70a61e368d4af4ab9b799e739ac294 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd3070d2221d4ec9b787e4825f041d09 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b39a04c2ec3641eeb06f77acc41ec523 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e351872f0ff4628b5891d7332e05076 : SUCCESS in 7m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/219920e2286147d389ffa41db900476c : SUCCESS in 6m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1c669828a12c4a91a9a79b6517afca94 : SUCCESS in 10m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2ea6a6b672b640e2876dc601752b5628 : SUCCESS in 21m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/044704fb9e024cc79137d9e3e14c2a2a : SUCCESS in 1h 01m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fc2e7588a8945b292f6dc55ae5506af : SUCCESS in 44m 11s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/232a651fc5e64c458fb71f2b5beaa5db : SUCCESS in 1h 03m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9e4f4bbad2c84053a890bdaa8da8ffcf : SUCCESS in 15m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7cd309f19ec4b4ab770534cb50db503 : FAILURE in 4m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ab5e1a5141d74eb5a1a150226eee6b54 : SUCCESS in 1h 47m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/033d00c8a47347bb9abe6236c8199db0 : SUCCESS in 26m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/812499c7a736463ab9a58ddfc1623266 : FAILURE in 4m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3ce41c4a4908448fa7b319789a5fd863 : FAILURE in 5m 13s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/fdc2db6b847e485599f05b5c2f648711 : SUCCESS in 10m 55s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f3fadaeca23e486686d9ab924ecf0b8a : FAILURE in 4m 24s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"e6b1dba50bf7aecdc47f45f7c6a065bd12e84c85","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 12:01:23.000000000","message":"Patch Set 3:\n\nrecheck\n\ngalaxy issues","accounts_in_message":[],"_revision_number":3},{"id":"7d0f11ea62c843dfa62a8e557e1ea18aea2cd6fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 13:53:42.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e50961fb5fb849c08f49985fe1a04ddd : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/155ab2ba58244091a4f3291b92486a63 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c4d0dc121f24508b5e74a1953f7cb8e : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8947cd6f1a7c4339916a481a0fa2cd43 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bf8c23ffdf1488a814e3ff6eb6105e1 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f840c1210d82400d9fef7c2046c899b6 : SUCCESS in 8m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/af71269c749e4484954e33f9ea33a770 : SUCCESS in 6m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4d8650a1648c4e958454c90dc81f189f : SUCCESS in 9m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b6224cc587474f51a71d13c229ad040c : SUCCESS in 21m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b0be7d4e758346e6bc283004acce2fe3 : SUCCESS in 1h 00m 01s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/443abcdb301c429b9ad3cdb90b728855 : SUCCESS in 43m 59s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3ffca8dcc045435a8d7fab5aef99efab : SUCCESS in 1h 01m 15s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e1982b8775934803b28f419962132c93 : SUCCESS in 15m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e59b267038d74a50adaef47b34391608 : SUCCESS in 12m 46s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/726fcef150cb4e2d9112d592edd11c5d : SUCCESS in 1h 40m 00s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1274ac841faf4966858d876e074add7d : SUCCESS in 20m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9df6464d07ac4017ac592def652a3ec1 : SUCCESS in 23m 22s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f02c9c1b4b654c9cae035bab81315250 : SUCCESS in 23m 02s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/328758eebe1d4f2c876d9f9a17fdfef0 : FAILURE in 4m 40s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34e2914f3ee141cfa6a88346f0705aa6 : SUCCESS in 9m 00s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"b19598ad1a8eca818b838e08edceebe994ae0f6a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 14:00:21.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"eaedf99a04a383f6ec2608bff14f27038e788438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 15:57:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/954566b81cb344b29d14d50b1cae9732 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3123590720ab4a8e9f31e01592e21810 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d498e39de62a45d1aa91d60e69b95651 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da527e8b3ada4e539881854cbf564278 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d5402bbc79524ff2af72a66dd9f3706f : SUCCESS in 6m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cdaa6c0a9f54b67ad0599a142529134 : SUCCESS in 7m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/90d04f3f44234cd9a28a8c0a53463a6b : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/09b9debd23e34e8ca0a3fccd9c145378 : SUCCESS in 10m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/25842640b05b4af6b2bc4b4d006643de : SUCCESS in 20m 20s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b399ee55b8a14e069b4c9db2d40c19f2 : SUCCESS in 1h 11m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/483feacdfe984761a77b5c1b7ef5d2d0 : SUCCESS in 46m 03s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/b945c2530b394cf2b0ebdf95c3c1166a : SUCCESS in 1h 05m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6896a9a8b6774a669923f18ce282a7a5 : SUCCESS in 14m 34s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d67458bf7cac49718c34d2a0b036cf34 : SUCCESS in 12m 56s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/68e543a19f084018909c6d4301bdeeec : SUCCESS in 1h 45m 36s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b5e24fe77a1c4348b8a1ff5c8196a360 : SUCCESS in 26m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f0989b650eac498ab23b1e8531d77e62 : SUCCESS in 23m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/585a2b7668264bfc8c6abfd65003723f : SUCCESS in 22m 49s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4f773a03f3c741988ebcc8680832be6a : SUCCESS in 10m 38s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e360bbe6e0f048708a73d788937f59b0 : SUCCESS in 9m 12s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"1f1ed8b46a082b2bcafcaa8a7d4e5ad6c1d52611","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-29 14:59:45.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cfe48856af2bb1b3d4258d1e6a143ae495ae550c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-29 20:11:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c9aec77bceafb56a2b1dc82918335752765e79c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 20:12:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3f38855b01fce4d736ac0578e9b2fe4c7f630d33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 21:59:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f331c380ff34541885c722182f68b0c : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c49a1d9430fa4b5b88c2d00a74ea9a7c : SUCCESS in 8m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/efba95fa085343a1b9337aeb4a4aff94 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9241e0b46afd449fac6498952e2d8ac5 : SUCCESS in 8m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/023ca215bb22471ab241ac265d0e6610 : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a693de0e4a5f436195975815a49f9f9a : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/37cbe34de26f44c6a65b42035bcf91df : SUCCESS in 20m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fba222cc41454083a9df6bce1e6ef1fe : SUCCESS in 56m 37s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19f2a6f672d244d7946ba9416a3aa0de : SUCCESS in 45m 52s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/7550e8b1fdfd488ca269b801e76b379e : SUCCESS in 1h 04m 37s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b4705cc0d2da432fae0a7bbc031011bb : SUCCESS in 16m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0211432c38c40f8b0b2ae18e6862780 : SUCCESS in 13m 07s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/faac1bcb77364592a052668929ff839d : SUCCESS in 1h 46m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/373e5193728f448d91ab95bf4f416bb0 : SUCCESS in 26m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/36ec3214d5874ed1b7f1a840ac48a635 : SUCCESS in 23m 39s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/71966a4ec4b7430d9358ff8ce650cb3c : SUCCESS in 25m 25s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/bf9e8ce5992b48008af238f249b8cc35 : SUCCESS in 12m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/465c3da2e8544f6eac209ed4b1c92777 : SUCCESS in 10m 08s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"764cae6943a7b8a8d1582df85cb3fe0cfb2dfc90","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 21:59:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"aa4eb88eb6b74d070a63477f8b531260b17255d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 22:01:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15aeb065134e452ba42714c2b1d79ef5 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"388dfa369c9608f252dc946123c1801999931f6f","revisions":{"5793c0d31bce5668dafe0e6f3b2836333c91113d":{"kind":"REWORK","_number":1,"created":"2021-03-04 14:58:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/778720/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/778720/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/778720/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/778720/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/778720/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/778720/1"}}},"commit":{"parents":[{"commit":"a9bf637cc56fe960ed7edd4af4624490bfc60b69","subject":"Merge \"Ubuntu: support non-persistent veth configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9bf637cc56fe960ed7edd4af4624490bfc60b69"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-04 14:54:13.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-04 14:54:13.000000000","tz":0},"subject":"Avoid fact gathering for group_by","message":"Avoid fact gathering for group_by\n\nThere are a few places where we have a play with one task that uses the\ngroup_by module to generate groups that are used in a subsequent play.\nThese plays do not need to gather facts for their target hosts.\n\nIn particular, the kolla-openstack.yml playbook should really be able to\noperate without access to remote hosts.\n\nTrivialFix\n\nChange-Id: I18f67eda4e48058f3f402b9d0e692d6eeb02855f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5793c0d31bce5668dafe0e6f3b2836333c91113d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5793c0d31bce5668dafe0e6f3b2836333c91113d"}]},"branch":"refs/heads/master"},"7634d30868a6ee71c34608e6e4450ab127994dbc":{"kind":"REWORK","_number":2,"created":"2021-03-04 15:15:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/778720/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/778720/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/778720/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/778720/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/778720/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/778720/2"}}},"commit":{"parents":[{"commit":"a9bf637cc56fe960ed7edd4af4624490bfc60b69","subject":"Merge \"Ubuntu: support non-persistent veth configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9bf637cc56fe960ed7edd4af4624490bfc60b69"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-04 14:54:13.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-04 15:07:14.000000000","tz":0},"subject":"Avoid fact gathering for group_by","message":"Avoid fact gathering for group_by\n\nThere are a few places where we have a play with one task that uses the\ngroup_by module to generate groups that are used in a subsequent play.\nThese plays do not need to gather facts for their target hosts.\n\nIn particular, the kolla-openstack.yml playbook should really be able to\noperate without access to remote hosts.\n\nThis change also adds changed_when: false to all group_by tasks, since\nthey show as changed but haven\u0027t really made any changes outside of\nAnsible.\n\nTrivialFix\n\nChange-Id: I18f67eda4e48058f3f402b9d0e692d6eeb02855f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7634d30868a6ee71c34608e6e4450ab127994dbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7634d30868a6ee71c34608e6e4450ab127994dbc"}]},"branch":"refs/heads/master"},"388dfa369c9608f252dc946123c1801999931f6f":{"kind":"REWORK","_number":3,"created":"2021-03-04 17:32:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/20/778720/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/20/778720/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/778720/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/778720/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/20/778720/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/20/778720/3"}}},"commit":{"parents":[{"commit":"a9bf637cc56fe960ed7edd4af4624490bfc60b69","subject":"Merge \"Ubuntu: support non-persistent veth configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9bf637cc56fe960ed7edd4af4624490bfc60b69"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-04 14:54:13.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-04 17:32:15.000000000","tz":0},"subject":"Avoid fact gathering for group_by","message":"Avoid fact gathering for group_by\n\nThere are a few places where we have a play with one task that uses the\ngroup_by module to generate groups that are used in a subsequent play.\nThese plays do not need to gather facts for their target hosts.\n\nIn particular, the kolla-openstack.yml playbook should really be able to\noperate without access to remote hosts.\n\nThis change also adds changed_when: false to all group_by tasks, since\nthey show as changed but haven\u0027t really made any changes outside of\nAnsible.\n\nTrivialFix\n\nChange-Id: I18f67eda4e48058f3f402b9d0e692d6eeb02855f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/388dfa369c9608f252dc946123c1801999931f6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/388dfa369c9608f252dc946123c1801999931f6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
