)]}'
{"id":"openstack%2Fkayobe~779580","triplet_id":"openstack%2Fkayobe~master~I9d5166f17e0eefc4cb1660e9342204f0e3701892","project":"openstack/kayobe","branch":"master","topic":"update-cirros","hashtags":[],"change_id":"I9d5166f17e0eefc4cb1660e9342204f0e3701892","subject":"Use newly released CirrOS 0.5.2 image","status":"ABANDONED","created":"2021-03-09 20:53:54.000000000","updated":"2021-03-10 11:04:15.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"21338cb4eafb128e3b5f20db2acb884d14b211d8","_number":779580,"virtual_id_number":779580,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-09 22:16:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-03-10 09:55:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-09 22:16:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-10 09:55:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"1b498f91c8546ed98ab6c5a93ca23af48b9d0bcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-09 20:53:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e00aed018878aa343df4b2c7cc8b82ad701ca1b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-09 22:16:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f11d1e07417046469a28d4930a2bc2be : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee4e0a20ef254667b4fe78065ea831a6 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4eaf16f4cdc4deb9109da746594cfa7 : SUCCESS in 6m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ac1ecbe6e1e43d38ab8b6ae8f505058 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/df484e1bce9d4284be09cbb539c8a92f : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848fbd88b30a4af692c81662af0e63c1 : SUCCESS in 8m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4615f10c768a42d9a52b05ab661a5c20 : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c12541275a6f457c94cbe12884a2eb40 : SUCCESS in 10m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f7b350bc700d4a35aeb828be5c335569 : SUCCESS in 20m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/be2120ad0cb14771bfd1f819515b806e : SUCCESS in 1h 01m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/23bd0d2e4bc740cfab92a2ee651f97f9 : SUCCESS in 46m 42s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/778d538c0a17439b9fa04249077f7536 : SUCCESS in 58m 01s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7ffc8681b96e4a838b6a999576292ba8 : SUCCESS in 16m 15s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a4fa17f275724dd5bb7bca384a5121af : SUCCESS in 11m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2bb97376f31d423c90040808a4612816 : SUCCESS in 1h 17m 57s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8bfe2ec26ec04d2da6669a78c55bd04e : SUCCESS in 21m 48s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/718a1573bbb743da91e48036629cc816 : SUCCESS in 21m 52s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6804e33d501745a8af9df2c4e3616c1c : SUCCESS in 14m 39s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bd499c48b45645c1b0b7d3a7d57a0130 : SUCCESS in 9m 45s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"85de3fbb3b8537a8692e892c2be38c6a606e265e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-10 09:55:03.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"97d9b6b1ca50c71b45d019f8d3fa48e020d466c3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-10 09:55:39.000000000","message":"Patch Set 1:\n\nShould we merge this with https://review.opendev.org/c/openstack/kayobe/+/773163 ?","accounts_in_message":[],"_revision_number":1},{"id":"21338cb4eafb128e3b5f20db2acb884d14b211d8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-10 11:04:15.000000000","message":"Abandoned\n\nMerged into https://review.opendev.org/c/openstack/kayobe/+/773163","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7f6b53c8766d1e90eafb14e51ccaf1204d282945","revisions":{"7f6b53c8766d1e90eafb14e51ccaf1204d282945":{"kind":"REWORK","_number":1,"created":"2021-03-09 20:53:54.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/80/779580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/779580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/779580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/779580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/779580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/779580/1"}}},"commit":{"parents":[{"commit":"d42e3229c90200627a511c374a821676e398d812","subject":"Merge \"Ubuntu: fix deployment image build\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d42e3229c90200627a511c374a821676e398d812"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-09 20:52:30.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-09 20:52:30.000000000","tz":60},"subject":"Use newly released CirrOS 0.5.2 image","message":"Use newly released CirrOS 0.5.2 image\n\nChange-Id: I9d5166f17e0eefc4cb1660e9342204f0e3701892\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f6b53c8766d1e90eafb14e51ccaf1204d282945"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7f6b53c8766d1e90eafb14e51ccaf1204d282945"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
