)]}'
{"id":"openstack%2Fkayobe~779811","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Id5ed0b594d6270c4a03606caa776bbb291be2694","project":"openstack/kayobe","branch":"stable/ussuri","topic":"ipa-extra-hardware-stable/ussuri","hashtags":[],"change_id":"Id5ed0b594d6270c4a03606caa776bbb291be2694","subject":"Update IPA docs and test build with extra-hardware","status":"MERGED","created":"2021-03-10 15:25:23.000000000","updated":"2021-03-15 18:03:55.000000000","submitted":"2021-03-15 17:57:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779811-ipa-extra-hardware-stable/ussuri","meta_rev_id":"0dea5c5f4cc196fd2a327e704cda21ea5477e36c","_number":779811,"virtual_id_number":779811,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 17:57:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-15 09:03:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-03-11 09:15:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-03-15 09:03:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-10 17:33:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-11 09:15:09.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-15 09:03:24.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"efffd675a014e673598f1c4554af427680a3dff3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-10 15:25:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68689b8517be7630579ca927d1f0da77a8bf8a57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 17:33:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e44aeb981dd4f2b99d30db6dfd9f51d : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b31d83c8d1144c1caf66fd7c40afe408 : SUCCESS in 5m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3cf9b5f0305a45b1b05741dd2de1b9d9 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4b7d446c65d6467a961b683a5de63974 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1bbf176e48b49c788e24d4a8f3b2240 : SUCCESS in 8m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a89251bf51f543a0a8cb6ade05218985 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/585bff742092406d89c150928ff32003 : SUCCESS in 9m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7b96e23f95fa425299f064fa33c2485a : SUCCESS in 20m 31s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3043e1a9cd414fe48638c757c5cfca38 : SUCCESS in 1h 12m 49s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/806e776477054d7485ae19c27ad6e9d3 : SUCCESS in 14m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/318185142e3b4454a887c4bb241e8cc9 : SUCCESS in 1h 54m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/727854fc005b4a5a82a44c97cb52230b : SUCCESS in 32m 57s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/040f31f9ad564fa8acab7efe7b92fadf : SUCCESS in 25m 44s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/8a33726ddd2e494591a7114de8512ad0 : SUCCESS in 10m 23s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"3dfe9e5156f77e593e3a6ab69f529777b232ca8c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-11 09:15:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"92f656a2696954b5fa3184e6ff54cb7303681a77","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-15 09:03:24.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a602c96a619fa5540eabe8a926067aff8d75425f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 09:03:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c407c82767663394dfb83a7cb808f957ee075fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 17:57:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54b984bd3aac4101ba28ed02609c1fca : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80f4714bdce849c99d18cf9216f3f70c : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b1cb13422a384b5bab46f857d1b56194 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce6fec79d1af4137ae6c21a9b524d723 : SUCCESS in 9m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a8f9a9e59554419e95a1ee9ace4aa8cd : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/004c07c6ec6c449499454fcfb0917244 : SUCCESS in 8m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b3ad9290d2847d5a2fff59c7a0a299f : SUCCESS in 19m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6eeffdc96a504fc1b4f09eec20bbefb8 : SUCCESS in 1h 01m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4e5cff34726c425ea65a97fbf55e799e : SUCCESS in 15m 19s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a5c7ba4044c9425a8d62e218b2104a15 : SUCCESS in 1h 42m 25s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/04cc1a2adc2e4d31ac7d75d6372dbe34 : SUCCESS in 38m 29s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ffa6f675745749d1a510fa35ef95d7f8 : SUCCESS in 26m 00s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/12a1989262784038bd910aa814c411ea : SUCCESS in 11m 09s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"fdac27f6ab5b171fb2c276dc6b70265f127f2837","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 17:57:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0dea5c5f4cc196fd2a327e704cda21ea5477e36c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:03:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9da5ecdde6224f4ab931a248873b0175 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9d862eacd191644d68a4af3007b5286edb649b24","revisions":{"9d862eacd191644d68a4af3007b5286edb649b24":{"kind":"REWORK","_number":1,"created":"2021-03-10 15:25:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/11/779811/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/11/779811/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/779811/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/779811/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/11/779811/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/11/779811/1"}}},"commit":{"parents":[{"commit":"031411828e26a302dd3bcb3364cb7043bb536631","subject":"Merge \"Test building seed deployment images in the seed job\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/031411828e26a302dd3bcb3364cb7043bb536631"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-02-11 09:50:43.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-10 15:24:05.000000000","tz":60},"subject":"Update IPA docs and test build with extra-hardware","message":"Update IPA docs and test build with extra-hardware\n\nChange-Id: Id5ed0b594d6270c4a03606caa776bbb291be2694\n(cherry picked from commit d765fbc5ebb7c1a8f80d4c49f3846107e4f23189)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d862eacd191644d68a4af3007b5286edb649b24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d862eacd191644d68a4af3007b5286edb649b24"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
