)]}'
{"id":"openstack%2Fkayobe~779932","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I95fc2c3873c8cd17edf928f613a06d2154196786","project":"openstack/kayobe","branch":"stable/victoria","topic":"cirros-stable/victoria","hashtags":[],"change_id":"I95fc2c3873c8cd17edf928f613a06d2154196786","subject":"CI: Use cached cirros image for seed vm job","status":"MERGED","created":"2021-03-11 17:55:02.000000000","updated":"2021-03-16 13:13:28.000000000","submitted":"2021-03-16 13:12:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779932-cirros-stable/victoria","cherry_pick_of_change":773163,"cherry_pick_of_patch_set":4,"meta_rev_id":"28f6e35e915e7012a1dff063dd0462c0b133c7f8","_number":779932,"virtual_id_number":779932,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-16 13:12:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-15 09:02:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-12 08:40:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-03-15 09:02:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 21:16:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-12 08:40:10.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-03-15 09:02:29.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"0d37167ed65da7fe8c3c08581821f986080af876","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-11 17:55:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"db261aaf4340a19e9c649826b8a40115789bb602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 21:16:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/863472dda0ac42c58215b22110ff2e2c : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfa2a342c8e04525ba7cab2594b3b4ce : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5582763342424b19a106d9a49b385f87 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d6eee230b46400e898dd9949797425a : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f5aad41a37b4811829c669ea008f76c : SUCCESS in 8m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f22bae4bf48a43d88e66a9f8f3e6d44e : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/49e6c1912b36477293037ae80a3cf5df : SUCCESS in 11m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7312fe49353746d89474d01d7639b316 : SUCCESS in 15m 20s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e0b480caefc047c1854e7661a313edc1 : SUCCESS in 1h 00m 06s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/8270468150ae45008b2224e10d065448 : FAILURE in 14m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1bf935255e8d443b97f8ebd851df04ed : FAILURE in 14m 55s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4a97c62b8eb7494ca7be89a27a62ccb6 : SUCCESS in 1h 45m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/93b4001497e84c21bfef6624242c144c : FAILURE in 10m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8ca650203a284c14b882edfc603c6758 : SUCCESS in 25m 44s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a2601c13d81b4226abce1f8dcb33f6aa : SUCCESS in 13m 20s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"f672f2164e953d2b09711dfe56495939f72be59e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-12 08:40:10.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck\n\nDocker Hub pull rate limit","accounts_in_message":[],"_revision_number":1},{"id":"977fb9245c2cd99c4ad930613ddc946fdfaed50f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 12:44:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adabe02b17a54968a8c800f2f0ad759b : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b2e4ea79c20481da3928922e59fc627 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/caeb793a7f8b497bba8e59c9ccf09bb0 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e11e3bd9b7ce41be86a5d751d6cba19a : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a14304dab6d4e029d24e909881bb8cd : SUCCESS in 7m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0ed6c2c71ec64c6eab6f5398e152977f : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a394675df874417a8ad1832411fd1130 : SUCCESS in 8m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8cf07c080f7b49e9a90a9675d3a30fc0 : SUCCESS in 22m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8404e3e163a3436b9e8c4941f9fd1926 : SUCCESS in 58m 37s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/9bd0c9496da1454583ea52659d5e123b : SUCCESS in 1h 15m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5bcced901a774e759dd6ae7fd1a73538 : SUCCESS in 14m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e16599a45cf746e8abfe0ec55135526d : SUCCESS in 2h 25m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d65f284b8b7743a3b5182366df82d691 : SUCCESS in 26m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9810ae8424ca481fae8c47c03743338d : SUCCESS in 23m 05s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/dfd204c55ecd45f19ba050d04a8b59f7 : SUCCESS in 11m 21s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"8534905974bec9acaeab645a3a923db7b20b4a19","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-15 09:02:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6592176eeec314aaade429539f42b1aa66cd8602","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 09:02:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85f3ccc99b55c888ec66b1671237319d8e526869","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 14:31:21.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de5687d859704196b7c68cb38bedc896 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17bb80e5d18545b7b1b60f6c76e554d8 : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ded402d988ee43dd806aa6ef3acbf210 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ec70fc5190b4471b6685d125a04b0ad : SUCCESS in 6m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0c8fd2e6e82a48d4a6919ea8699f79d8 : SUCCESS in 5m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/674f6f135fe64849bb094be3c09bbbd9 : SUCCESS in 8m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/edda6ba9c7a74097b5a7866e838288ef : SUCCESS in 19m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/074475365ebd411eb3e186c6bbcadf76 : SUCCESS in 1h 07m 07s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/1dd877ec000140ec87c8f8b0a37ed6fa : SUCCESS in 1h 05m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/771bce9d338541c2a3c637c18196befa : SUCCESS in 15m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/967d644f95644ad59a06790f55726fc3 : FAILURE in 2h 13m 24s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/db019be89ceb42238805ab2373f6d4b3 : SUCCESS in 29m 36s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8cfeacccb73d47bf99aabbda0bed705f : SUCCESS in 23m 23s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/12bba45ef35943ca99289d5c60f2cc83 : SUCCESS in 10m 11s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"41f77e463d9f6c601c26a86429592fe21fe5876b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-15 20:50:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"934093d08c377183e103bdd9b7b23936d63550d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 22:45:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc55c7c5422e4f6e85c0d275000ff597 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb13c3141afb4753bb3552434903142d : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d29c90962bed4be19da3cb5798220d05 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5c190ca931dd42d4bc3cd5d3054bff8e : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c07a19355b4011acbcdc944092445c : SUCCESS in 6m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/64a7ac302d2846c9a6ba0a12d624588f : SUCCESS in 5m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/21ab1d205a1d4981b806246d58e1811d : SUCCESS in 9m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94ad83d1088e4af28bcddea16b79e2f6 : SUCCESS in 18m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ca1bd52ae8ba4fa0930147353e8a1a84 : SUCCESS in 58m 27s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/fce8fe599a064b4aaac8d051554d7f39 : SUCCESS in 1h 03m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c70cf25bec1746038ebf04663c5b62a6 : SUCCESS in 15m 22s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/02bb5e0bf4da4bc3936af11dd35fe070 : SUCCESS in 1h 52m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/58c8763ebb054dd982a426a2fa56f476 : SUCCESS in 24m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6e922f8d67f144fbb49a1130c4ffe910 : SUCCESS in 24m 12s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ae1b55864dc744e7a92b62662cf5ec57 : FAILURE in 4m 59s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"cb63ffaec2142ce44d3e06a49c85784a6a68fc7c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-16 08:11:26.000000000","message":"Patch Set 1:\n\nrecheck\n\nGalaxy issues","accounts_in_message":[],"_revision_number":1},{"id":"1571bac7b7b5e0a7cb493a60aa62e32e4bbcd315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 09:49:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35492ed90dd2410b9628ecb5bd619872 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7846117eea145dcad4bfee4e243b200 : SUCCESS in 5m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15671e1fa1d64aa5ba403d50faebe89e : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0c5c89fedbf4662a80499068ca2a9f4 : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d0e83887a03450a813e9ff8443aa7ac : SUCCESS in 7m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8f71a6860888434db668543a59373240 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0bfc7b9621884db3aa7344ac822f993c : SUCCESS in 8m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f077b3e8d4ab4ee6a032847fc07344b4 : SUCCESS in 19m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/24918bb6382c4ed684c80ff1070a27b8 : SUCCESS in 54m 27s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/3e4d6ca9ea7240c985a14337f8471012 : SUCCESS in 56m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/18da47a5a6484d8e9bda49fd87d5dfb6 : SUCCESS in 13m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3ed22ccf491643fc8681ace75922287f : SUCCESS in 1h 33m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a5d38701f87b4513891b0b5828c43370 : SUCCESS in 26m 14s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/012fc6d753fd4a1583bdd40351e41f6d : SUCCESS in 28m 58s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/36c72e41ea27496fa6d9be8e68285c95 : SUCCESS in 10m 42s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"4f34535dcd58a62ebef37edd8f69449df65a3106","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 09:49:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ec5d1897a2a4468aed7e0f1dde3fe85a750a0fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 13:12:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/297a411b129c4b2f86285ea653a943ff : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0b18da34d5240668af23e3fb0ce3111 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0cdb6a81f2994d87b270dd4238cc786f : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36229eeed4fc4e23a0d273ec45bd7543 : SUCCESS in 7m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1bb79067e3ec4779a715803de657543e : SUCCESS in 5m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ad6c258183b4eecb38a004eb7967cac : SUCCESS in 8m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/359c35ddade3412e88bf97cce487663d : SUCCESS in 22m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/de15a7f9026a429995ef6bfc95be5d43 : SUCCESS in 56m 52s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/649221d0809b4e7cb2d1ddbf63300d7d : SUCCESS in 1h 09m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2a26275c4e734fd08546005ccbec8be7 : SUCCESS in 14m 23s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b646bccf15e444dea5956af9b82b94e3 : SUCCESS in 1h 48m 48s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a282640e99374f73b68ef89c7a2deb28 : SUCCESS in 25m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/22f1fef272214e3e9270162277b9e4f2 : SUCCESS in 27m 35s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/debd66ea7b734cd38d3871f2af8ed2c8 : SUCCESS in 10m 10s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"5aed06d97a004ca349a6a90ffc3dde92000ece47","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 13:12:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"28f6e35e915e7012a1dff063dd0462c0b133c7f8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 13:13:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd33d8ee0ade4bd09b7d35c7bc66bceb : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff8d5e9d15410db59f08e90a5e068f0714559be7","revisions":{"ff8d5e9d15410db59f08e90a5e068f0714559be7":{"kind":"REWORK","_number":1,"created":"2021-03-11 17:55:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/779932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/779932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/779932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/779932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/779932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/779932/1"}}},"commit":{"parents":[{"commit":"9ee3aae6311852ee7bbd34298a03a036ca19baab","subject":"Merge \"Update IPA docs and test build with extra-hardware\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ee3aae6311852ee7bbd34298a03a036ca19baab"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-01-29 18:02:54.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-11 17:55:02.000000000","tz":0},"subject":"CI: Use cached cirros image for seed vm job","message":"CI: Use cached cirros image for seed vm job\n\nThe nodepool images include CirrOS images.\n\nAlso update to newly released 0.5.2 image.\n\nChange-Id: I95fc2c3873c8cd17edf928f613a06d2154196786\n(cherry picked from commit 39b3fe70c6cddc22e4b75c70379bcb0a0a717c90)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff8d5e9d15410db59f08e90a5e068f0714559be7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff8d5e9d15410db59f08e90a5e068f0714559be7"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
