)]}'
{"id":"openstack%2Fkayobe~779935","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I81cabf27f44af3c8135efe8e427db1ffee5f0091","project":"openstack/kayobe","branch":"stable/victoria","hashtags":[],"change_id":"I81cabf27f44af3c8135efe8e427db1ffee5f0091","subject":"CI: display VM console log on ping or SSH failure","status":"MERGED","created":"2021-03-11 19:33:20.000000000","updated":"2021-03-15 18:02:01.000000000","submitted":"2021-03-15 17:57:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"779935","cherry_pick_of_change":777372,"cherry_pick_of_patch_set":1,"meta_rev_id":"161d766f2a6b11851e28cfa4b15eac7c8f86ce5f","_number":779935,"virtual_id_number":779935,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 17:57:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-15 09:02:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-12 10:21:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-03-15 09:02:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-11 21:36:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-12 08:35:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2021-03-12 10:21:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-03-15 09:02:50.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"61e3c7f32861c590421042eb34ddf4c6b581b14c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-11 19:33:20.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"93e6e52c44678826f46d489a666c490574cee4fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 21:36:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f821e230cce745dd8e54ee696af00cc4 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49492c0776d84500b53bcc06545ef171 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48f60e7a22714b96a68c9bbbbbfbea3d : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6be2bf92aaec4b8aa6f73332a8570719 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa1a28b668f54ba59b606a65271d9e06 : SUCCESS in 7m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ea49f1bdaaff48ac8ca914bab07f7e6e : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ada2ee799ad4f5e88628f6b6b5312ab : FAILURE in 11m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d16f4945d06a4849a1a306a22ea0d610 : FAILURE in 7m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/98c9e8838bb440678c8365b04d4ffc6a : SUCCESS in 59m 07s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/0814fb6728ef4c9497d2ba3a1f84b9ef : FAILURE in 6m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/227a7ecdd2b24d6a9836c57ba42421a3 : SUCCESS in 15m 22s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/90db0455dab0443ab8986a3deabd8a40 : SUCCESS in 1h 59m 38s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2af95df99dee4dd58a0f2d4035a2dec7 : SUCCESS in 27m 48s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bd2fd309897a43ebbfeca7b26ee5abef : FAILURE in 6m 23s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/7888180540b240f38e7a07244465cadb : SUCCESS in 10m 44s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"82ccc7f05bc7daea0aa5e4bfc4d04584bdebbf12","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-12 08:35:08.000000000","message":"Patch Set 1:\n\nrecheck\n\nAnsible Galaxy flakiness","accounts_in_message":[],"_revision_number":1},{"id":"d7990c391054d217693b64ecf95b46afc4ad6409","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 10:09:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cb49fe06f8e4f9cadfea9b18d9ffd65 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/971e2ee2092844438fc1514023333677 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2fae276e04f042f6b610952fa19d39b1 : SUCCESS in 7m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3a0b822afb034ee684baa48817ba6000 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f61563c2fc35474686631849427dbb48 : SUCCESS in 8m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e73c66f206294b899f9fdb3cd91e1f32 : SUCCESS in 4m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/22469ad09ed542828dad5c7e6c571474 : SUCCESS in 10m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/98e8388189d545abac715ccd2db1597f : SUCCESS in 17m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f61e78cc6643419691aa753f18d8abc2 : SUCCESS in 59m 28s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/9fe81620430d44649cc6d8defe230105 : SUCCESS in 59m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cb20683179224323b438f6ecf123b2f1 : SUCCESS in 17m 19s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a00000ec3f1d45cb84ee948c4ac308b7 : SUCCESS in 1h 32m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6254814a46c74bedb57af691e0abed24 : SUCCESS in 27m 14s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/77e324d95d6c4d1388241c0863c3ec6a : SUCCESS in 27m 33s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/53bb536a88f543e78b18d49647123040 : SUCCESS in 12m 02s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"60c7370bf2d4e46504330ef4a47cb291d7145b74","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-12 10:21:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0c9440df5cfd6a8de4280399b74b95912d0c4d3f","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-15 09:02:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8dc07c06aa677b9230ed7c46848fcf791b78f006","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 09:03:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5de3e893f475ddeef4dfdf742fba680ad015a3fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 17:57:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb0ebbf740de416d89caa1c429dd39a8 : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c93287f0b874e26ae4fc3d73d602711 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f6bf0a533fd455eb3910163cea23274 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddc57bf9f36e4576afe33fd686feb063 : SUCCESS in 7m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/808137efc61a4bacadf229cdcd3508a3 : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7f09419de5ac49d9860020966cc10a5d : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3211e367f54149f1ab4435dbfe1dfd85 : SUCCESS in 19m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3baaa611bc5f47d19ccef7ee1d2b7304 : SUCCESS in 1h 01m 23s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/f06913b4bcb7409f9c2b4ed134d9048b : SUCCESS in 1h 07m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/097e21a2498c4aa58856af4ce48e37e0 : SUCCESS in 14m 01s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5586814a837c4979a8104c71e7134fd6 : SUCCESS in 1h 51m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b365a6efa2d145009b8c837bb84d6e79 : SUCCESS in 32m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/04d8ad2afdad4e68a5d27560b8fb4f5c : SUCCESS in 25m 06s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/77c7aae225bd484bb0f4bf9de808bb95 : SUCCESS in 15m 06s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"b1f87686957f0815c42e0e83a6788330553adec5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 17:57:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"161d766f2a6b11851e28cfa4b15eac7c8f86ce5f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 18:02:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ff50dda380e4bb0bac6f867c276f8eb : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9fb2c5d165cbdda1c542aeda21a2886bdb45f296","revisions":{"9fb2c5d165cbdda1c542aeda21a2886bdb45f296":{"kind":"REWORK","_number":1,"created":"2021-03-11 19:33:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/35/779935/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/35/779935/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/779935/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/779935/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/35/779935/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/35/779935/1"}}},"commit":{"parents":[{"commit":"9ee3aae6311852ee7bbd34298a03a036ca19baab","subject":"Merge \"Update IPA docs and test build with extra-hardware\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ee3aae6311852ee7bbd34298a03a036ca19baab"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-24 11:54:36.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-11 19:33:20.000000000","tz":0},"subject":"CI: display VM console log on ping or SSH failure","message":"CI: display VM console log on ping or SSH failure\n\nAlso increase attempts to 12, in line with Kolla Ansible CI.\n\nChange-Id: I81cabf27f44af3c8135efe8e427db1ffee5f0091\n(cherry picked from commit 25ae0be2f914cbaefcf144f0fe20c89c8adbe1fa)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fb2c5d165cbdda1c542aeda21a2886bdb45f296"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9fb2c5d165cbdda1c542aeda21a2886bdb45f296"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
