)]}'
{"id":"openstack%2Fkayobe~780080","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I7435bd06d58501e361312bacfa4e55394aba5b57","project":"openstack/kayobe","branch":"stable/victoria","hashtags":[],"change_id":"I7435bd06d58501e361312bacfa4e55394aba5b57","subject":"Wait for overcloud manage and provide state transitions","status":"MERGED","created":"2021-03-12 08:53:02.000000000","updated":"2021-03-15 11:30:47.000000000","submitted":"2021-03-15 11:29:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780080","cherry_pick_of_change":775135,"cherry_pick_of_patch_set":1,"meta_rev_id":"03f9e654a797b5c2dc68fcbd058b194568f06dd2","_number":780080,"virtual_id_number":780080,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-15 11:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-15 09:02:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-12 09:22:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-03-15 09:02:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-12 09:22:47.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-03-12 10:46:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 09:02:05.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e3f9fe7735d2314f73a95f7c67fe4814e5e1b77c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-12 08:53:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e0f7494e4a4c5e6d34a1f1987647aacfceb501dd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-12 09:22:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"50abda45d3bbd5ef437a3b8590a6ae3efe04a61b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 10:46:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5b23be73dd345369e1bda4bc9826647 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf68e5ac94b4851a2b628a01babe3df : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f66b4f25e3a74951b7cd472c20ba0871 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1b7d4fa51ce4157a1c6a729ac69f092 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b9967e67f6149b89a3587a01fab4624 : SUCCESS in 8m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7c750635d2cd47ad848c9215d9f91592 : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e29ecc5418694daebef6f40cad3af1b1 : SUCCESS in 7m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/930b06b2f1a14f219d5de15a7180cd15 : SUCCESS in 20m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/86a56593045b4117b93d52b7d940c91f : FAILURE in 4m 48s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/e51ce3d2f6214598b76c93eeb5c794c7 : SUCCESS in 58m 37s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1177552b401f47c0ba68a55c792fbde9 : SUCCESS in 12m 00s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0dc8e763580547febf9002e15b348d3f : SUCCESS in 1h 51m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f658ce0ce20e49a1a45ab1548f298433 : SUCCESS in 20m 50s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3e6c8a91dc514b1789ed3ba13d7794a5 : SUCCESS in 28m 19s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0154046d9b8a47cc92250bffd3177b48 : SUCCESS in 10m 09s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"f77d9c32eab5ae3c5df7efcedb10d5029d1fb92b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-12 12:07:50.000000000","message":"Patch Set 1:\n\nrecheck\n\nAnsible Galaxy issue: (HTTP Code: 520, Message: Origin Error)","accounts_in_message":[],"_revision_number":1},{"id":"1e025fd38ef661fd76c75c1b3b6240f4a83e6d54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 14:04:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12633160f7b74a17b7ddd527525541d8 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8da40891564344d88b367c9d9b96c91c : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47b9b8b943254b56aeac0b4734f195df : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/300cec23984b4c34b65049afd6653e91 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f588146825ab4464bbd45a236f0d5c13 : SUCCESS in 8m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bc224b0ee92c439c814aeaa7612b6fe7 : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/63ff52eda9ec45bca3d157bb5edf646e : SUCCESS in 9m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/85c5e75c16c44055ae1178b49e7d5d9a : SUCCESS in 19m 19s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2c472b5a02d24547a13a5a71ccd81a99 : SUCCESS in 53m 36s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/d158f2b8935542b99806b38173681109 : SUCCESS in 59m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/51a8cefb921c4e7fa50683426a52c56a : SUCCESS in 16m 43s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/625c3337b5d14dbd834006fdfaea215d : SUCCESS in 1h 50m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3c9398557ffd432cba97944dc1846a38 : SUCCESS in 27m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d8e90728ee9d41c595cb8613a5e032d9 : SUCCESS in 25m 57s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0e8abc2dd4a34acdbb1388c9f0a8eba8 : SUCCESS in 11m 41s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"f6ce27de9762ea37b8e1898eb36d1a103906d99e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-15 09:02:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b5af882dcc88c2a041b973dd3d16f9d30df9674a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 09:02:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36a2b57c7d933e6e08111ac21b268f377e732fb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 11:29:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdb05541368244f4b684d6b68976b5a7 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e9de76877e3469381082fac861962ee : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc37b709de1341f8bcad27f03e0e9f1c : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5915d968faf54fbfa6b0d71910b71b76 : SUCCESS in 7m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f957a2618380425d9e4cd090c185e40b : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1f101e67f6ba43a5a38527edbe65f940 : SUCCESS in 10m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/df94d1f8c48146e397b25be3e19c8828 : SUCCESS in 18m 31s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f5647a1b65d74d188980fc73b8c4917a : SUCCESS in 59m 41s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/47d54028cccc49a3b12afa0b7509aebe : SUCCESS in 58m 55s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/416baa2fe0494dd08f35ecc4b18603e3 : SUCCESS in 12m 24s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7f61f618a6204d8c89fafb9dd4b40541 : SUCCESS in 2h 13m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7d0d13cb011b47b39de1ee57a05eca02 : SUCCESS in 28m 18s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ce7e1b131ed04eafb3608a3933955e9f : SUCCESS in 24m 29s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ae44908bbfb44b02ae0a095fcea81897 : SUCCESS in 10m 06s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"2f73dd01493d0459d3b3bed53c207a1f6427cd0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 11:29:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"03f9e654a797b5c2dc68fcbd058b194568f06dd2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-15 11:30:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/160c08592597467cb7fe9d8fe052fc54 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5446355e6e9e8ad86b3df1a1dc4f740210502b3c","revisions":{"5446355e6e9e8ad86b3df1a1dc4f740210502b3c":{"kind":"REWORK","_number":1,"created":"2021-03-12 08:53:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/80/780080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/780080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/780080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/780080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/780080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/780080/1"}}},"commit":{"parents":[{"commit":"9ee3aae6311852ee7bbd34298a03a036ca19baab","subject":"Merge \"Update IPA docs and test build with extra-hardware\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ee3aae6311852ee7bbd34298a03a036ca19baab"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-02-04 15:44:40.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-12 08:53:02.000000000","tz":0},"subject":"Wait for overcloud manage and provide state transitions","message":"Wait for overcloud manage and provide state transitions\n\nDuring the \u0027kayobe overcloud provision\u0027 command, wait for the\nnodes to become manageable and available.\n\nAdds two variables, \u0027wait_manageable_timeout\u0027 and\n\u0027wait_available_timeout\u0027, to control the wait timeout.\n\nChange-Id: I7435bd06d58501e361312bacfa4e55394aba5b57\n(cherry picked from commit aaf9572d082aa679ad452df1b09f88da0fb18576)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5446355e6e9e8ad86b3df1a1dc4f740210502b3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5446355e6e9e8ad86b3df1a1dc4f740210502b3c"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
