)]}'
{"id":"openstack%2Fkayobe~780326","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I5aa1a32015bb0090689447f977d7840c9e5fefdd","project":"openstack/kayobe","branch":"stable/victoria","topic":"fix-extra-kernel-options","hashtags":[],"change_id":"I5aa1a32015bb0090689447f977d7840c9e5fefdd","subject":"Fix formatting of extra kernel options for Bifrost","status":"MERGED","created":"2021-03-12 20:41:18.000000000","updated":"2021-03-29 10:37:35.000000000","submitted":"2021-03-29 10:35:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"780326-fix-extra-kernel-options","meta_rev_id":"c845f8a3779eddd5af199b8732f6a9730cd9a8e8","_number":780326,"virtual_id_number":780326,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-29 10:35:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-15 09:01:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-03-29 08:07:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-29 08:53:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-29 08:53:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-12 23:12:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-15 09:01:51.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-03-29 08:07:31.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"15d448b21db61544d0c20d5656053ac2b6d7538f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-12 20:41:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ab738b020c85541b56dc7744cfdf71b4965740ef","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-12 20:42:27.000000000","message":"Topic set to fix-extra-kernel-options","accounts_in_message":[],"_revision_number":1},{"id":"3756cc3b85816beaf49e52b6911c614b57f529b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-12 20:42:28.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"464fb13934a0d72bcd04956a15f3e09de4595a89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-12 23:12:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41f447413b4740ff87353d029e1c8ea3 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5a5b0271a9d4bc18d79211499f0bac9 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e524e0383d24fa2abe2498418128fb9 : SUCCESS in 5m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bfb7d1531c04952bc1f4b42330d2c85 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f92d28d96ce493a914b2299ae9b970c : SUCCESS in 10m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bbfa9b30c084c66926d5305be54735a : SUCCESS in 6m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fd52fd15a57040e5a3d05755726a8d07 : SUCCESS in 5m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fade2dcfb99a4ba5a05f086705f0aafb : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8b0e205752dc449cac996f388a77f1f7 : SUCCESS in 24m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b0911f33c73941838dc75b4177458401 : SUCCESS in 47m 50s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/1b973162deca470a8640862e5930881c : SUCCESS in 1h 04m 45s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/38a862634c524cc7b339afd5d433ec22 : SUCCESS in 13m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8dbd56ba8ade471488533033c69d93ce : SUCCESS in 1h 58m 15s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/79ed2d129f374006be29bdbf37f94256 : SUCCESS in 22m 37s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e83dc17337b8458fafd8c2d80f26218c : SUCCESS in 25m 08s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/51a0a76b770a411baea10ba8fe62b5e2 : SUCCESS in 8m 53s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"b9c90b03d2138a30bba7455ac818ba6c4ec23489","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-15 09:01:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bf47466a78bc00c4360bfa567f3e18ac4015f75b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-29 08:07:31.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"849efdb98a14b796b3fd76b46e67ef82f9b76a12","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-29 08:53:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"adcd3c08191623cda40fd0a2ca15d89541c87be9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 08:53:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cce03d57a527b7358899121aeff18aa724d43654","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 10:35:52.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d28cf7c6075945f1848f251f26e264bb : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf3ac8a469734a2d821fab339eb92c75 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c9726b2e3964da88b939eb3388bbc6d : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43c4fa61da664d19b6d2e6cee7f9a445 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/842a7db1b3c742a7bb4bbeb3e425a813 : SUCCESS in 4m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a0446056cd534de3b0ae9c974c6a9cf8 : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ce1f69a257b41ba8000a9d2f4400a47 : SUCCESS in 11m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3ccf4b6ec54b46b0b8bc107e9021c20e : SUCCESS in 20m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1c2adc3407d649649938d140968016b7 : SUCCESS in 1h 07m 14s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/be73d9251461427c94c24c9579d0fd2b : SUCCESS in 1h 15m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8a87847cc42c49a380a6716aaa83ca2e : SUCCESS in 15m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d5aed6c4094840dcab987dd52bb45a8e : SUCCESS in 1h 36m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7e1619c9755143cb8b360d0f7767826f : SUCCESS in 26m 02s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/eafb3e4af44b42b1bf8508df11519417 : SUCCESS in 25m 36s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/52034c6a3e114a5b8a0c6b48b76b19f4 : SUCCESS in 10m 04s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"bfa2662ed65f9e65a6d897dacc93c295ece739e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 10:35:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c845f8a3779eddd5af199b8732f6a9730cd9a8e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 10:37:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e562c6072448ae86933d6e29c7587a : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec3091a3d2c6425cb4934f2d680f886b : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"46cd7e1337c8053659b5056d37cb64f23defb53a","revisions":{"c1f6f8d11d80eaa3183703ca554536c140e44c08":{"kind":"REWORK","_number":1,"created":"2021-03-12 20:41:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/26/780326/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/780326/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/780326/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/780326/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/780326/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/780326/1"}}},"commit":{"parents":[{"commit":"9ee3aae6311852ee7bbd34298a03a036ca19baab","subject":"Merge \"Update IPA docs and test build with extra-hardware\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ee3aae6311852ee7bbd34298a03a036ca19baab"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-12 20:39:00.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-12 20:41:11.000000000","tz":60},"subject":"Fixes formatting of extra kernel options for Bifrost","message":"Fixes formatting of extra kernel options for Bifrost\n\nChange-Id: I5aa1a32015bb0090689447f977d7840c9e5fefdd\nStory: 2008714\nTask: 42049\n(cherry picked from commit c693202c4293df40a19f5416897d4fd98219ace1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1f6f8d11d80eaa3183703ca554536c140e44c08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c1f6f8d11d80eaa3183703ca554536c140e44c08"}]},"branch":"refs/heads/stable/victoria"},"46cd7e1337c8053659b5056d37cb64f23defb53a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-03-12 20:42:28.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/26/780326/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/26/780326/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/780326/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/780326/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/26/780326/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/26/780326/2"}}},"commit":{"parents":[{"commit":"9ee3aae6311852ee7bbd34298a03a036ca19baab","subject":"Merge \"Update IPA docs and test build with extra-hardware\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ee3aae6311852ee7bbd34298a03a036ca19baab"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-12 20:41:49.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-12 20:42:15.000000000","tz":60},"subject":"Fix formatting of extra kernel options for Bifrost","message":"Fix formatting of extra kernel options for Bifrost\n\nChange-Id: I5aa1a32015bb0090689447f977d7840c9e5fefdd\nStory: 2008714\nTask: 42049\n(cherry picked from commit 99f6addaca960fa80efe0c14dc436bccbf5aa3a5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46cd7e1337c8053659b5056d37cb64f23defb53a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46cd7e1337c8053659b5056d37cb64f23defb53a"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
