)]}'
{"id":"openstack%2Fkayobe~780898","triplet_id":"openstack%2Fkayobe~master~I0299f5126ebd080f3912a76294b0013abe7747a0","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I0299f5126ebd080f3912a76294b0013abe7747a0","subject":"Bump max Ansible version to 2.10","status":"MERGED","created":"2021-03-16 15:25:42.000000000","updated":"2021-03-25 16:21:57.000000000","submitted":"2021-03-25 16:20:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"780898","meta_rev_id":"3022c56d97cd199a38dffabf613ea0c8f79a3f42","_number":780898,"virtual_id_number":780898,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-25 16:20:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-03-24 13:51:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-25 14:13:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-25 14:13:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-03-25 13:18:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}],"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-16 15:46:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-03-16 17:26:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-19 17:39:48.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-03-23 14:42:15.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"}],"messages":[{"id":"136c7b886ba43265632fd2429229b870908e8d3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-16 15:25:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f4105538417beb31b8d1f75b13fa9c97ff99ef88","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-16 15:46:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7da0802a922638b68888d5a50abc180f370b16fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 17:26:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1494324cb4f0423aa0bc11d6c96b8e37 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e24404135d140148baba82692703757 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02b306124a5e43f2b4ba5ef4b6541a1b : SUCCESS in 7m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30cfa161118249fba2b46af30363578f : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbcce866565348ac87735952bb262285 : SUCCESS in 7m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3d05d115e7346fe950cb25dc55a881b : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38458b231a9440b9839d20074ef5bc30 : SUCCESS in 4m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6090894c00ce4325944370ef1f295c48 : SUCCESS in 9m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d4e6b78e6cb14ad38e190650d665e237 : SUCCESS in 11m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b3b1a93f36344e8e80a39b521818a103 : FAILURE in 9m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7b64e26cc79c4e49b547b983966e81c2 : FAILURE in 48m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e25f3cb53901410eb3d69ac24bf7680d : FAILURE in 39m 43s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ee12c90b8cf04a15b21b19ae22f06741 : FAILURE in 50m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3683cc3ad01e4ac2ae15112f6ccd920d : SUCCESS in 18m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a58699be61fa43a1bd0e539c7ae9dba9 : SUCCESS in 12m 56s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e1db7e8771834caba3df3ca9f111c52b : SUCCESS in 1h 50m 27s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/aafcc4294ad942ffb463c5eb1005373e : SUCCESS in 30m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d984f300b07b44b5a65bd80a03aacbeb : SUCCESS in 43m 55s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/344ad9dcdc4f4282a37984e65a98043d : SUCCESS in 21m 31s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/987d1139c772400d8b02c7dc364a171d : SUCCESS in 12m 24s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e88ef10a65e34678b6924fa437e8003c : SUCCESS in 11m 20s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"d472aa46a7bf8bee43e189a8efaf75aeb2f66732","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-16 17:34:35.000000000","message":"Removed Code-Review+1 by Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"032160557ccb1c7c9d0475975f7694e7fa45204c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-16 17:35:05.000000000","message":"Patch Set 1:\n\nBreakage in multiple Ansible modules.","accounts_in_message":[],"_revision_number":1},{"id":"7e508458c62a3cf79fd7a53dd3f165e6547a5639","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-17 09:04:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a7455a4b44b28e67e3ac01563fa04102220d29fa","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-17 09:08:25.000000000","message":"Patch Set 2:\n\nMolecule fixed, let\u0027s take a look at the breakage in overcloud job.","accounts_in_message":[],"_revision_number":2},{"id":"63b6d0b831c00c907fc19219751ab30812834b40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-17 09:17:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"75bb4ed9cad9844c06bfd3db16d29f62fcd5bb44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-17 09:45:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"becae034fadd71ef0d6aeaef67e95372a5a014fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-17 10:47:49.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e0fc8a013ee2f237dfe2a2294c1238225720eaf0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-17 12:45:45.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b33d45ae00940389860b992ccb811fa : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08570f8c4c464794af8fe4221983208d : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b9c25ee9ed214adab93e6bb9acc13238 : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d728b96a94f7421da6b3e4fc292207c3 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fe8df8508ab44baa202b37d31efdfc8 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d77511c2acc4716b2cd019edb257b53 : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ab2f28cf20244bbb37cecf3eb8eb5d2 : SUCCESS in 3m 50s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3173d071e9534b91b71a457a2fa685fa : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e9fcfe0155654765b56a4ed6931d4dd4 : FAILURE in 7m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/035b871a50194bdb85859f8b8bc22869 : FAILURE in 12m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ad2b93c554b549e0aefb530b0fdb3c82 : SUCCESS in 59m 23s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/46de3ccc62eb473796a238baf5b396cb : SUCCESS in 54m 53s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/e72c5ffa83314d3bbdbef1dd7ca30fa5 : SUCCESS in 55m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3d8e72eb5f634aca94278ff72ffa835a : SUCCESS in 20m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59cb5017fb5f465b8af179a77b2ea533 : SUCCESS in 13m 38s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/866a1a17efc74228a385467c0c7e4ca5 : SUCCESS in 1h 53m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/af28ac1ce38042ec8e6e3121404274d4 : SUCCESS in 23m 05s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/55ca4a9b7e174c43af27e1ee48a6f843 : SUCCESS in 24m 10s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4b23ce4127294b2cbba074a72f46c352 : SUCCESS in 25m 06s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4b588b52aed8458baef5ee4d134ef9ae : SUCCESS in 15m 56s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/41e7678724794d2d8e5d849d6a333608 : SUCCESS in 10m 12s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"6603494072cbdf62e5e72af792272463e7686d8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-18 10:37:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c9ce6c1194d67494890db49bf53e097702bc8db5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 11:15:04.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e12c3df1dbb14c61b78da92ad01fd03e : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5e5f9b59c0514d20a5150069866d20c5 : SUCCESS in 11m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fb2284e9796e46a3b9279bef9a9e01d1 : FAILURE in 6m 38s","accounts_in_message":[],"_revision_number":6},{"id":"e9d7d67c94898c6081d7dd29d15a1d24406e8704","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-18 11:30:12.000000000","message":"Patch Set 7: Published edit on patch set 6.","accounts_in_message":[],"_revision_number":7},{"id":"d4d1b6a6c2242161d7812c7148aabb29dd2b13f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 14:00:39.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8de85602d2fa4ab58b88ded72ef58e0b : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b478c244c884d848136c3dfdd622a0b : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/267fa5231d9848c1bf71894e238ad7e0 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9932a58f93a3437b89b67d972fa2ff45 : SUCCESS in 8m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1fb0081a1f7a4191903b9a1fbda2d0d0 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bbb021ee2614dbe8a3717d55f48894b : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/101886c5d69343dfb717ade50bab7fbb : SUCCESS in 5m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/561d58368c1642d983d0c10e71c9fd84 : SUCCESS in 7m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/994d885a5c9143999112a34f196271af : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3fafb772c5d840d49db81a96720a6089 : FAILURE in 12m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/131de1ee231f4d9780a797a7fb7252a1 : SUCCESS in 1h 06m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7617492b81ca4ae081b866e5e4ab505b : SUCCESS in 52m 02s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/46cfcd4bd3f547c69cb36009302eaba6 : SUCCESS in 1h 07m 16s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/24f7e8ca7a2d4fbeb4a9ab32fd849653 : SUCCESS in 17m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f57753c57e245fa897b5e685d2fd065 : SUCCESS in 14m 11s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0f087be2c6b243a180bea126ea1ec4fd : SUCCESS in 1h 47m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/46a050a48b7243b7ae374124b5e87b73 : SUCCESS in 29m 16s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1350265d6a074b8497f3908808b41991 : SUCCESS in 23m 06s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ebd1d9f1c6684a7d82f2867a109f0bfd : SUCCESS in 24m 13s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/62982f0883d7470191ad9ce9f4aa0f3f : SUCCESS in 14m 11s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87818f61ae1c40a4aee9ed72d0afa112 : SUCCESS in 11m 00s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":7},{"id":"dc6fc5f9d74de8f4f74e271c399112cc58d876d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-18 14:14:02.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"829b6b9cf1828d6773cda7a259185a10f985793c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:26:08.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1671e8f682774195ab3f14ece3445460 : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64ae837f1c514bceaea13e016c13ddbf : SUCCESS in 8m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b48b047393654220894957bfa16b4ee7 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc5fda0d060c44719efec36124605d5a : SUCCESS in 8m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea483375a7b1426a9ab318970ac2bfa9 : SUCCESS in 8m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f66572167541a7981f0ed0895661a0 : SUCCESS in 10m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b945eaae3a4484dbcdbde6818f249fe : SUCCESS in 4m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7f310d0ed1bb4cf9912bd2e5ad53b0ea : SUCCESS in 8m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0d0305a9a7724c9ca08a88f7315771c7 : SUCCESS in 13m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/584f0941c2064b0bbd0dbb48dd5f00a3 : SUCCESS in 20m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d484b329caa645419b2c3f6d5e18f1fd : SUCCESS in 52m 50s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b5350a550aee424aa0d761f7547d9bb6 : SUCCESS in 37m 41s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/7a868600ad404580bca75a6df73413a2 : SUCCESS in 53m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4bcb62776ec04c21a34fde931326915f : SUCCESS in 15m 53s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d2ae4c6c1dea4617bfd0774d228be9c7 : SUCCESS in 12m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7a368b32f0cb4a13ada792cfcf944fc3 : SUCCESS in 1h 53m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bc0ea787b77a4448b008844695e46fb5 : SUCCESS in 39m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0a17418909274c65ac69857833e58c88 : SUCCESS in 23m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b3e8344b11f7410ba534d22dcc37b92d : SUCCESS in 24m 55s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/17d22e4411db41a8a5f8ee10a051bbce : SUCCESS in 12m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d49f147ba1d744b7a0b9c5ca3dae41b8 : SUCCESS in 10m 25s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":8},{"id":"26c623581ee01a3a37bde65ea35fd1e3e474c077","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-19 17:39:48.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"a113ce55a8f32585d5a3d4ad7bf7917512e2ace9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-22 10:04:32.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"b5bd0725fb265aca7cb1d445de7513a1428d046c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 12:19:18.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7a2955669c5485d80eed5083fea2e31 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c6c3baf5f7b46b79169ad2738f90679 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed6c2677c9c040bbb4a41a8153f14890 : SUCCESS in 7m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ac2f5f0dca1445ab45462abb8b4f4c7 : SUCCESS in 7m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c759a3b4421488b85eb49cbd38ba383 : SUCCESS in 7m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c6f82c595b445babdc606fac61d0b53 : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e2dd75ac39a6406782534c0cc8a30a67 : SUCCESS in 5m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b4fb8fee415b4bb7af6f294a501d1429 : SUCCESS in 9m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/20a73bb2eb924511b326a604f0dd5bac : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/037805b71ea547ea93430b371696130a : SUCCESS in 22m 25s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5d72ce6c12e247d48413b5fc548c2ebf : FAILURE in 52m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/abc1331e6788485ca2fc1bdc3facb104 : FAILURE in 30m 39s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/e6e05dd2c40146f89016ab78cbaf2ba1 : FAILURE in 52m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/44b89ebf032b4cb0b82fca4e731192b5 : SUCCESS in 18m 31s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/524eabb4fe81498ebaeadbb7b9ba521d : SUCCESS in 15m 21s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/21ed161368a94ce2831cb7a163939335 : FAILURE in 1h 54m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5ff689fb7138461395b99ecde6485604 : SUCCESS in 24m 14s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e389fab89d6e4efcb5c8bfaf78eeab45 : SUCCESS in 25m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ccfe5e4f015347fc93fd6a7b7afdc408 : SUCCESS in 27m 57s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/0d1080e01ada4e5e98195b708402e03d : SUCCESS in 12m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bc3fe3d070e4c5baa128d3738e99dbf : SUCCESS in 11m 11s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":9},{"id":"fef994e018eb9a071d3f20c64aee8ad9d7faab2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-23 08:40:13.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"cfb221667dfe0d46f5078a652a1ddda876b86084","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-23 09:20:23.000000000","message":"Patch Set 10: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":10},{"id":"785e506a28d867b080bb60adfd9f8effdf534160","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 10:33:53.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7b959316f274c7a892c5566df4997a3 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10d5a49b3dbf4b8a8f16eb2ca28f86f1 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cd48138fa0b49598b55a13d20b55111 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee5374405099441f9f030da4b63cd2b0 : SUCCESS in 6m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7c12880355e474986bc3072ef28dbcb : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97938e7cbe284233aad1e53873893047 : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a09245482ceb40948961fc5aaaabc219 : SUCCESS in 5m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5ac1c00494724c9ab4ccc8c3e6186a75 : SUCCESS in 7m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/753401ed606f43e5b6a8b4432be4fe08 : SUCCESS in 10m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/305a1c42b48f4a7fb8b88bbe499df1be : SUCCESS in 21m 19s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d67fe5729c60417a92c8235c6ec2a044 : FAILURE in 1h 06m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96ffc4979eb244be989141af2072d6e3 : FAILURE in 39m 21s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/c2a52e2e72074327b9e984a8f958b214 : FAILURE in 1h 15m 55s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fad34f5eba584a749899f48ec1391f3b : SUCCESS in 19m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ed61b05251b4660a6c5510b7c12f566 : SUCCESS in 11m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3ff154b9c86748cab2f2393b5c934fe8 : FAILURE in 1h 50m 21s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d42d4a7fab464724a59fd18a43171cd6 : SUCCESS in 22m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/818a2cd0ec2d4606b6f51c1499d542f4 : SUCCESS in 25m 37s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7cd6c4f4f1514008969fc5de14aaeac2 : SUCCESS in 25m 47s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/5de11021128c4958979788f8d7ebd3d3 : SUCCESS in 12m 14s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2a684b1da30a4697bba84f2f526ffa24 : FAILURE in 5m 21s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":10},{"id":"ec7ea245d9093c3956adc6ab450440dd8f2ee7ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-23 16:51:39.000000000","message":"Patch Set 11: Published edit on patch set 10.","accounts_in_message":[],"_revision_number":11},{"id":"15d34cb5af05f078b3f7b12a172cfac6077715bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 19:07:29.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/708f0d65575e418681a075c056a36872 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f527bcf9fb413eab44b705075de95a : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a4296d109264610b8418471ef470a4d : SUCCESS in 8m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/806ae57c88004f78a53dfe6c9a741242 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac7031407e95416488ea2040d64992bd : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03451a3b09364ba486600849afe057f2 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ed94bd16bf24c0283949ffa135ce75a : SUCCESS in 4m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aef54dfe6e9c4828bd297b09ae823525 : SUCCESS in 7m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/da3b1ce02fe040dbbc2111d6961cf856 : SUCCESS in 10m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4674eec33e2b4ca3bfb375a58affb002 : SUCCESS in 19m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/99b4a40113d84a7096b0e8d8f2667b8f : FAILURE in 56m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/96ec08f833c247fe9c80f47d94f227f3 : SUCCESS in 43m 06s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/588957e984d943f2a1563c27ba7501a7 : FAILURE in 1h 08m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/65f24edc0e424397ae3d5d41ca77df49 : SUCCESS in 17m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0f4ebbbd5c5449abe4b5e2a4cb5fa4b : SUCCESS in 14m 15s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/77e3ca3aa8bd40fca207665fd4c3d453 : FAILURE in 2h 08m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/34cf07b25b954ac1bff6eff872649107 : SUCCESS in 30m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/615361bf98774ce69c053694423f6865 : SUCCESS in 19m 34s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/40907f61d2954d83965bfeeb75439d7d : SUCCESS in 28m 17s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/00309a853a6146ea83a6e641913ab223 : SUCCESS in 12m 58s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6d537eea90214c4792332294fcd0657b : SUCCESS in 11m 56s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":11},{"id":"552dcf0cc54a6315c5264ab399fe972f65019ab6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-24 07:25:28.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"61210134fe9be2b9b88b66d3757607b2143646cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 09:07:20.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2bd838484f3f4f4cb276731633aa4cc2 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edc2d7763a3949c08ccbfcd969c2efef : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42bcc8514e3d4b1281f42aeb3224fe25 : SUCCESS in 6m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b70387d63122446dbe76eccc33a4d753 : SUCCESS in 6m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3229406e17524d6f9f9754ab9d8736c2 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59a814caa87d44a9a2348c7cf36f6fef : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/267977b498aa4c0ba23a0f1db4d85d44 : SUCCESS in 4m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c43301debf7c4d358e7c1a49d193d3cf : SUCCESS in 5m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/991014eee3b2477e9a45ef9759dec2ac : SUCCESS in 11m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a760fe82eb204c50b98c3cbf819a4731 : SUCCESS in 17m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/27aa1cfee6ae4c17aad29dc08fdae320 : SUCCESS in 1h 38m 58s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df3901711c7942ceb476bae104601811 : SUCCESS in 47m 28s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/14aabb048d344d08bc7e886bb467facb : SUCCESS in 53m 04s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ca8f281934ec424696ca8d6252907b2c : SUCCESS in 14m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13390d6e3c1c4d50848aa9d81b4c5f6b : SUCCESS in 15m 39s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8a6348d3c23e4fea939a9e3ee6ee8dac : SUCCESS in 1h 30m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8ff22b38cc4b4628aee113b608ce5906 : SUCCESS in 26m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b33cedf2f15541b7ae333b2be0149848 : SUCCESS in 18m 54s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4a74b51b460b4b97ad0b9d0e9cd180fe : SUCCESS in 31m 12s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/80305de7af4043608caabb83a2b3e7ee : SUCCESS in 11m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/db7c3c6fae484655897f5c2afa654165 : SUCCESS in 12m 12s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":12},{"id":"21c805f4a0975568a71db00bf058f9bc72d8b485","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-24 10:58:50.000000000","message":"Patch Set 12:\n\nTrying to test IPA here: https://review.opendev.org/c/openstack/kayobe/+/782689","accounts_in_message":[],"_revision_number":12},{"id":"72fe94ebba198d93d2ba32d9a396a2ee690259c7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-24 13:49:27.000000000","message":"Patch Set 12:\n\n\u003e Patch Set 12:\n\u003e \n\u003e Trying to test IPA here: https://review.opendev.org/c/openstack/kayobe/+/782689\n\nLooks good.","accounts_in_message":[],"_revision_number":12},{"id":"222c3395222636f148fc830766612f51a2ba583f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-24 13:51:16.000000000","message":"Patch Set 12: Code-Review+2\n\nNice","accounts_in_message":[],"_revision_number":12},{"id":"c35b3357b269656ff03d8ed69eb83a0ad48f5623","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 13:18:08.000000000","message":"Patch Set 12: Review-Priority+1","accounts_in_message":[],"_revision_number":12},{"id":"9bdf206133bb2f053aab12c3e946274c74f28c57","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-25 14:13:12.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\nThanks Michal!","accounts_in_message":[],"_revision_number":12},{"id":"b7d9f9116a944e5535c2c714c4b8532af8d5c147","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 14:13:31.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"10ccde76424b9140a58987af292d83f071c3dabd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 16:20:10.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75067416121e420099936f69ae35d6ba : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aab0c054fd3a4e0bacecbbf2100fbcdf : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf88474c6a4a4605bfafa5256c12ea17 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a354f716f01422692d271551cba3d58 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05094598cbfc40178afdbd0d751a9fc3 : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fdc6454c3738424ea888e355a1c1e0d0 : SUCCESS in 7m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/53689ce53efb41e5b70084fb480fb426 : SUCCESS in 10m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e46e90d39488473aa850775ab3991e72 : SUCCESS in 21m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4543b30ef50a4b4e88fcae04d74f76d0 : SUCCESS in 52m 29s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c2a9a2789b14d6ba4265ca0b6a30ccc : SUCCESS in 44m 40s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/8ff69b20bb2f432191b1c5922d10a5b7 : SUCCESS in 1h 07m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/31e5c3f3fb394c4089f0924ea3c1a7d5 : SUCCESS in 16m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/470e564130e248c1b778bfbde7f1612d : SUCCESS in 16m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a8e7b00cb75c42c89fc83322df2fd4b6 : SUCCESS in 2h 01m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c4c36942251c4d9da0068f63b49ad85d : SUCCESS in 29m 20s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6404087f10324f74bea8c8b6b30a0798 : SUCCESS in 26m 49s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3fe37bf084494211828702fc0abbd71e : SUCCESS in 29m 04s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/91d0596f71e3436fb4981e3e771a1f53 : SUCCESS in 12m 55s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/204af9e0d6414d7082f03fe87381f7b9 : SUCCESS in 10m 59s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":12},{"id":"b02cb3ed77a493bea6da6f6c5062b916e4af3571","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 16:20:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":12},{"id":"3022c56d97cd199a38dffabf613ea0c8f79a3f42","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-25 16:21:57.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71d0cabee2b9468cad96368afe3cad1e : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0e223f4caee4e1e8bd37ef5793b278a : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"16cce92a74315370061ec154764d03745435f60d","revisions":{"16385ff3bad9d391d9869d4b954adf92ea0c9a83":{"kind":"REWORK","_number":1,"created":"2021-03-16 15:25:42.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/780898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/780898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/780898/1"}}},"commit":{"parents":[{"commit":"f09070a41623c162e814de132dc056d72991696d","subject":"Merge \"Remove patch from kolla-ansible dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09070a41623c162e814de132dc056d72991696d"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-16 15:24:53.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-16 15:25:39.000000000","tz":60},"subject":"Bump max Ansible version to 2.10","message":"Bump max Ansible version to 2.10\n\nChange-Id: I0299f5126ebd080f3912a76294b0013abe7747a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16385ff3bad9d391d9869d4b954adf92ea0c9a83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16385ff3bad9d391d9869d4b954adf92ea0c9a83"}]},"branch":"refs/heads/master"},"bf01fc4997ed46e1554450285698eebf29f97423":{"kind":"REWORK","_number":2,"created":"2021-03-17 09:04:16.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/780898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/780898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/780898/2"}}},"commit":{"parents":[{"commit":"f09070a41623c162e814de132dc056d72991696d","subject":"Merge \"Remove patch from kolla-ansible dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09070a41623c162e814de132dc056d72991696d"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-16 15:24:53.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-17 09:01:59.000000000","tz":60},"subject":"Bump max Ansible version to 2.10","message":"Bump max Ansible version to 2.10\n\nMolecule also needs to be bumped to \u003c3.1 because of [1].\nlint: section changed in Molecule 3.x - adapted.\nmolecule-docker added do test-reqs because it code has moved to it from\nmolecule core\n\n[1]: https://github.com/ansible-community/molecule/issues/2872\n\nChange-Id: I0299f5126ebd080f3912a76294b0013abe7747a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf01fc4997ed46e1554450285698eebf29f97423"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bf01fc4997ed46e1554450285698eebf29f97423"}]},"branch":"refs/heads/master"},"144e01ecf8d53c0216da9faf53d0b87457111bfb":{"kind":"REWORK","_number":3,"created":"2021-03-17 09:17:37.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/780898/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/780898/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/780898/3"}}},"commit":{"parents":[{"commit":"f09070a41623c162e814de132dc056d72991696d","subject":"Merge \"Remove patch from kolla-ansible dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09070a41623c162e814de132dc056d72991696d"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-16 15:24:53.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-17 09:17:29.000000000","tz":60},"subject":"Bump max Ansible version to 2.10","message":"Bump max Ansible version to 2.10\n\nMolecule also needs to be bumped to \u003c3.1 because of [1].\nlint: section changed in Molecule 3.x - adapted.\nmolecule-docker added do test-reqs because it code has moved to it from\nmolecule core\n\n[1]: https://github.com/ansible-community/molecule/issues/2872\n\nChange-Id: I0299f5126ebd080f3912a76294b0013abe7747a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/144e01ecf8d53c0216da9faf53d0b87457111bfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/144e01ecf8d53c0216da9faf53d0b87457111bfb"}]},"branch":"refs/heads/master"},"c859b6023eaa8ec1c00b61bc8001c61f9b24b5f1":{"kind":"REWORK","_number":4,"created":"2021-03-17 09:45:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/780898/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/780898/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/780898/4"}}},"commit":{"parents":[{"commit":"f09070a41623c162e814de132dc056d72991696d","subject":"Merge \"Remove patch from kolla-ansible dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09070a41623c162e814de132dc056d72991696d"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-16 15:24:53.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-17 09:45:40.000000000","tz":60},"subject":"Bump max Ansible version to 2.10","message":"Bump max Ansible version to 2.10\n\nMolecule also needs to be bumped to \u003c3.1 because of [1].\nlint: section changed in Molecule 3.x - adapted.\nmolecule-docker added do test-reqs because it code has moved to it from\nmolecule core\n\n[1]: https://github.com/ansible-community/molecule/issues/2872\n\nChange-Id: I0299f5126ebd080f3912a76294b0013abe7747a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c859b6023eaa8ec1c00b61bc8001c61f9b24b5f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c859b6023eaa8ec1c00b61bc8001c61f9b24b5f1"}]},"branch":"refs/heads/master"},"6e26149bfd39166494ad5c574854fc23939e457c":{"kind":"REWORK","_number":5,"created":"2021-03-17 10:47:49.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/780898/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/780898/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/780898/5"}}},"commit":{"parents":[{"commit":"f09070a41623c162e814de132dc056d72991696d","subject":"Merge \"Remove patch from kolla-ansible dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09070a41623c162e814de132dc056d72991696d"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-16 15:24:53.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-17 10:47:40.000000000","tz":60},"subject":"Bump max Ansible version to 2.10","message":"Bump max Ansible version to 2.10\n\nMolecule also needs to be bumped to \u003c3.1 because of [1].\nlint: section changed in Molecule 3.x - adapted.\nmolecule-docker added do test-reqs because it code has moved to it from\nmolecule core\n\n[1]: https://github.com/ansible-community/molecule/issues/2872\n\nChange-Id: I0299f5126ebd080f3912a76294b0013abe7747a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e26149bfd39166494ad5c574854fc23939e457c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e26149bfd39166494ad5c574854fc23939e457c"}]},"branch":"refs/heads/master"},"ea4d923fd9513702098da1cb5de244458769c76f":{"kind":"REWORK","_number":6,"created":"2021-03-18 10:37:52.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/780898/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/780898/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/780898/6"}}},"commit":{"parents":[{"commit":"f09070a41623c162e814de132dc056d72991696d","subject":"Merge \"Remove patch from kolla-ansible dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09070a41623c162e814de132dc056d72991696d"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-16 15:24:53.000000000","tz":60},"committer":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-18 10:37:40.000000000","tz":60},"subject":"Bump max Ansible version to 2.10","message":"Bump max Ansible version to 2.10\n\nMolecule also needs to be bumped to \u003c3.1 because of [1].\nlint: section changed in Molecule 3.x - adapted.\nmolecule-docker added do test-reqs because it code has moved to it from\nmolecule core\n\n[1]: https://github.com/ansible-community/molecule/issues/2872\n\nChange-Id: I0299f5126ebd080f3912a76294b0013abe7747a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea4d923fd9513702098da1cb5de244458769c76f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea4d923fd9513702098da1cb5de244458769c76f"}]},"branch":"refs/heads/master"},"17e0ac51b77fa542efa68fb5361296ec56b80ec6":{"kind":"REWORK","_number":7,"created":"2021-03-18 11:30:12.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/780898/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/780898/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/780898/7"}}},"commit":{"parents":[{"commit":"f09070a41623c162e814de132dc056d72991696d","subject":"Merge \"Remove patch from kolla-ansible dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09070a41623c162e814de132dc056d72991696d"}]}],"author":{"name":"Michał Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-16 15:24:53.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-18 11:30:09.000000000","tz":0},"subject":"Bump max Ansible version to 2.10","message":"Bump max Ansible version to 2.10\n\nMolecule also needs to be bumped to \u003c3.1 because of [1].\nlint: section changed in Molecule 3.x - adapted.\nmolecule-docker added do test-reqs because it code has moved to it from\nmolecule core\n\n[1]: https://github.com/ansible-community/molecule/issues/2872\n\nChange-Id: I0299f5126ebd080f3912a76294b0013abe7747a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17e0ac51b77fa542efa68fb5361296ec56b80ec6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/17e0ac51b77fa542efa68fb5361296ec56b80ec6"}]},"branch":"refs/heads/master"},"670a99777a61ee8c20febb222affd8f7f03cc7c3":{"kind":"REWORK","_number":8,"created":"2021-03-18 14:14:02.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/780898/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/780898/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/780898/8"}}},"commit":{"parents":[{"commit":"f09070a41623c162e814de132dc056d72991696d","subject":"Merge \"Remove patch from kolla-ansible dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09070a41623c162e814de132dc056d72991696d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-18 14:13:24.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-18 14:13:59.000000000","tz":0},"subject":"Bump max Ansible version to 2.10","message":"Bump max Ansible version to 2.10\n\nMolecule also needs to be bumped to support Ansible 2.10\nlint: section changed in Molecule 3.x - adapted.\nmolecule-docker added do test-reqs because it code has moved to it from\nmolecule core\npytest-testinfra and pytest-molecule needed for testinfra tests to work\n\n[1]: https://github.com/ansible-community/molecule/issues/2872\n\nChange-Id: I0299f5126ebd080f3912a76294b0013abe7747a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/670a99777a61ee8c20febb222affd8f7f03cc7c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/670a99777a61ee8c20febb222affd8f7f03cc7c3"}]},"branch":"refs/heads/master"},"8b8e66ab0439d2cc3db3e3900d26d391a2d836d0":{"kind":"REWORK","_number":9,"created":"2021-03-22 10:04:32.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/780898/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/780898/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/780898/9"}}},"commit":{"parents":[{"commit":"f09070a41623c162e814de132dc056d72991696d","subject":"Merge \"Remove patch from kolla-ansible dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09070a41623c162e814de132dc056d72991696d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-18 14:13:24.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-22 10:04:25.000000000","tz":0},"subject":"Bump max Ansible version to 2.10","message":"Bump max Ansible version to 2.10\n\nMolecule also needs to be bumped to support Ansible 2.10\nlint: section changed in Molecule 3.x - adapted.\nmolecule-docker added do test-reqs because it code has moved to it from\nmolecule core\npytest-testinfra and pytest-molecule needed for testinfra tests to work\n\n[1]: https://github.com/ansible-community/molecule/issues/2872\n\nChange-Id: I0299f5126ebd080f3912a76294b0013abe7747a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b8e66ab0439d2cc3db3e3900d26d391a2d836d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8b8e66ab0439d2cc3db3e3900d26d391a2d836d0"}]},"branch":"refs/heads/master"},"18997ce3260c7ab4972f90a8b9c415104e8f0fa8":{"kind":"REWORK","_number":10,"created":"2021-03-23 08:40:13.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/780898/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/780898/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/780898/10"}}},"commit":{"parents":[{"commit":"f09070a41623c162e814de132dc056d72991696d","subject":"Merge \"Remove patch from kolla-ansible dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09070a41623c162e814de132dc056d72991696d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-18 14:13:24.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-23 08:40:04.000000000","tz":0},"subject":"Bump max Ansible version to 2.10","message":"Bump max Ansible version to 2.10\n\nMolecule also needs to be bumped to support Ansible 2.10\nlint: section changed in Molecule 3.x - adapted.\nmolecule-docker added do test-reqs because it code has moved to it from\nmolecule core\npytest-testinfra and pytest-molecule needed for testinfra tests to work\n\n[1]: https://github.com/ansible-community/molecule/issues/2872\n\nChange-Id: I0299f5126ebd080f3912a76294b0013abe7747a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18997ce3260c7ab4972f90a8b9c415104e8f0fa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/18997ce3260c7ab4972f90a8b9c415104e8f0fa8"}]},"branch":"refs/heads/master"},"6e4f2fd3969d0bf96876bf2614874dd550408bc4":{"kind":"REWORK","_number":11,"created":"2021-03-23 16:51:39.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/780898/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/780898/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/780898/11"}}},"commit":{"parents":[{"commit":"f09070a41623c162e814de132dc056d72991696d","subject":"Merge \"Remove patch from kolla-ansible dependencies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f09070a41623c162e814de132dc056d72991696d"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-18 14:13:24.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-23 16:51:26.000000000","tz":0},"subject":"Bump max Ansible version to 2.10","message":"Bump max Ansible version to 2.10\n\nMolecule also needs to be bumped to support Ansible 2.10\nlint: section changed in Molecule 3.x - adapted.\nmolecule-docker added do test-reqs because it code has moved to it from\nmolecule core\npytest-testinfra and pytest-molecule needed for testinfra tests to work\n\n[1]: https://github.com/ansible-community/molecule/issues/2872\n\nChange-Id: I0299f5126ebd080f3912a76294b0013abe7747a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e4f2fd3969d0bf96876bf2614874dd550408bc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e4f2fd3969d0bf96876bf2614874dd550408bc4"}]},"branch":"refs/heads/master"},"16cce92a74315370061ec154764d03745435f60d":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-03-24 07:25:28.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/98/780898/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/98/780898/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/98/780898/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/98/780898/12"}}},"commit":{"parents":[{"commit":"775620733bc2d848e5eeac3e18a2b0734e750a83","subject":"CI: Fix IP address detection in baremetal compute test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/775620733bc2d848e5eeac3e18a2b0734e750a83"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-18 14:13:24.000000000","tz":0},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2021-03-24 07:25:28.000000000","tz":0},"subject":"Bump max Ansible version to 2.10","message":"Bump max Ansible version to 2.10\n\nMolecule also needs to be bumped to support Ansible 2.10\nlint: section changed in Molecule 3.x - adapted.\nmolecule-docker added do test-reqs because it code has moved to it from\nmolecule core\npytest-testinfra and pytest-molecule needed for testinfra tests to work\n\n[1]: https://github.com/ansible-community/molecule/issues/2872\n\nChange-Id: I0299f5126ebd080f3912a76294b0013abe7747a0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16cce92a74315370061ec154764d03745435f60d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/16cce92a74315370061ec154764d03745435f60d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
