)]}'
{"id":"openstack%2Fkayobe~782272","triplet_id":"openstack%2Fkayobe~master~I20d8ee783cdf158bbb469e831518e36d3c8ed73f","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I20d8ee783cdf158bbb469e831518e36d3c8ed73f","subject":"Ubuntu: AppArmor config for seed VM provisioning","status":"MERGED","created":"2021-03-22 15:08:27.000000000","updated":"2021-03-26 19:42:21.000000000","submitted":"2021-03-26 19:40:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"782272","meta_rev_id":"8a444acbd9d0da64ded39e644bb7242ff3da0f7b","_number":782272,"virtual_id_number":782272,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-26 19:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-26 09:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-25 14:16:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-03-26 09:53:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-03-25 13:18:51.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 15:37:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-03-22 17:05:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-26 09:53:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"5991bb8804464c3654170487a91c609bc4f52c01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-22 15:08:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93358ecd2eec7f308157f12011eb767b4d3acd4d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-22 15:37:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9191feb20bb3b1222aa7c110deee63a586428035","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:05:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0328b0fd8cab43d0a19732bf0f94f2db : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f64428525d04cdabed978d19e0b6d1d : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77ad68cf737a43f59c350edb6d73721e : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce69715de40a42ca91d55c51c6dc7b05 : SUCCESS in 7m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fbb7ec64e7741eaa316710b186712c1 : SUCCESS in 6m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd7923adf7654101a510be761320924a : SUCCESS in 8m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5b557652da4e41bb89eb9e11c7a989a0 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6ee7b2091baa44ba8e2285c2e3266c64 : SUCCESS in 8m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f547bf974c814da6a796de9049a6b4a7 : SUCCESS in 20m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/bfc44129dc49475982f71fa7ea6878c6 : FAILURE in 54m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e060e71e816f43cfb9b18031d994b134 : FAILURE in 34m 03s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/1ae40c06c78c4bacb17267c7e6302cdd : FAILURE in 1h 00m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/564a7652748b4bd7818b0fdd3e99d7f6 : SUCCESS in 14m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f737ea6548cf4263805fd0a00c0eea35 : SUCCESS in 12m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/968392003dea45fca28bb742e94f5035 : FAILURE in 1h 47m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0c48a99c03a547ceb83f7fbfd766c3ed : SUCCESS in 38m 29s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7aaabb53953416d8142863374aec428 : SUCCESS in 24m 04s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a184a041467c436a955e08d513cffde7 : SUCCESS in 22m 28s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/21448f70b21b4923ba074aa283942393 : SUCCESS in 13m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/13113f2673f848abbc99a404fb087b6f : FAILURE in 8m 19s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"ca93d7f73fa07c713e4dd3ef0fcf21d1741213d1","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-23 10:58:54.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d04664a29913b6b6698655a3606a3cd1d1d78c98","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-23 11:07:03.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7b83c15972305dd2781d31299845a0a0c81c4244","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 11:21:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef84920b9f904208b00c51884bbe0222 : FAILURE in 6m 41s","accounts_in_message":[],"_revision_number":2},{"id":"dfb4521688452f1c7448e423e8c0b5d3e85747f5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-23 11:27:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e12cf66f37090821f028a3d8c5b47af68a61d805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 11:41:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebc3743e8dc140ce98963a6467fc2782 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":2},{"id":"9f14e78cc66c7df83a1f55c9da13301439b18fc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-23 15:28:45.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"104c39e12bec4a5bfc19a4b278576e3a8456a533","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-23 15:32:18.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"01227d1dc232a3cb69843b8fe6113da8eb63cc4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 15:42:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a95502d6c746431e9908797ec0d9de6f : SUCCESS in 8m 27s","accounts_in_message":[],"_revision_number":3},{"id":"9710dff123d9beebc39df3843f36f5c44fb66083","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-23 15:43:07.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"9ba9a426cb0d2e93b9b64dbddcfd3c3408c39557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 15:58:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e4d896abeb74dde94738c98882de46f : SUCCESS in 8m 52s","accounts_in_message":[],"_revision_number":4},{"id":"ea06c0faf1c38c5fcdd4f64e39c5e10e96e61a34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-24 10:41:15.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"b1faf1c019bddbbe112ca8c789a7f262e88e443b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 12:25:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf3000c5e826426e9b7373a387067174 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/771ec68b300744a0a5299fb188a7eed1 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c3683d48957478db8b7c6eff4ef997a : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45a23f6b749548eabefa8eaa438ad55f : SUCCESS in 15m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e7c237afc97453ab41e28ea09c4b2f4 : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa2eeae97334429782a8973dfed53574 : SUCCESS in 7m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/31e99b9a75144edcb0e886fbeb0ccb42 : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/910b22e1229e40aa92c1f067228f786b : SUCCESS in 9m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f9a9d172e89c453badc8446faca93888 : SUCCESS in 19m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8e499804483d4e7d8082a6ccae23ee18 : SUCCESS in 59m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2ee0015086b541a0ae41be32ea64a3bc : SUCCESS in 33m 56s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/471043b6af67415c81fde19ea0675105 : SUCCESS in 1h 13m 00s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7a5feb73a3bd4976a814311968ad96ad : SUCCESS in 14m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c78d5f5c2f84cc9937f3fc0507bbad9 : SUCCESS in 9m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8327849d01624f6b818e6a7ee8a464db : SUCCESS in 1h 41m 28s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1ab559bbfdd342fe8baac0c91fb491ac : SUCCESS in 27m 45s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4db57aa1fb204d7381c7e693dc8dc67b : SUCCESS in 16m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5b1499f9fa564c13bbb447af72d190b0 : SUCCESS in 17m 36s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/53bae60cdbf2417189665be082ef9266 : SUCCESS in 12m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/22ff0d48b6334e5dafb8586641bb7b14 : SUCCESS in 9m 46s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":5},{"id":"460d9d8dd4e404782a1e051cbe3d132c1ab040b0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-25 13:18:51.000000000","message":"Patch Set 5: Review-Priority+1","accounts_in_message":[],"_revision_number":5},{"id":"5b3428763d27b49eca641e7ab70e41f400e577d4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-25 14:16:03.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"25d829676c191c16565907fe112e7a51879aeaf2","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-26 09:53:12.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"95927cf94493f3e7e20a623715961ca4a210ae12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 09:55:19.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3bf813edf2c7e4af006f8072c6fc9a61cb951665","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 11:52:23.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b828e72b5e341df85bb0795e88e28ec : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cab8873577a54a7da87e68762a673e54 : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/17e05b1c57b94ef192383a70ab10f378 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae026c63c1484d5b92aeda760e30ccdf : SUCCESS in 8m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c888ee9b431d45e89a9cc16f347a49d4 : SUCCESS in 8m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/80f0ed790d134b37bbbde973cfb5051d : SUCCESS in 9m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a242a9208e95452db294e5cdced34752 : SUCCESS in 19m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/61612d93b4ba468ab9d311c9e6d4149b : SUCCESS in 1h 05m 32s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ecc9f1b7ae246818753f97685180f2c : SUCCESS in 42m 41s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/8ae5e064671844a0a6710abce2d2f8ac : SUCCESS in 1h 04m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/eb6938fd145b4fd4accf6a83f4ed32c6 : SUCCESS in 17m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bab782fcf881488197851a8bde0ebb82 : FAILURE in 12m 52s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b81986cd44aa4c969ee443e6c01771a7 : SUCCESS in 1h 50m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5478cc7cfb4f4ee88714c46845795ed8 : SUCCESS in 27m 12s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/222b9a6bd7bb4c4998c113e2450150a5 : SUCCESS in 23m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f5d3bea3a907442abe26c6b27884a519 : SUCCESS in 25m 34s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/9cc53177b7264c688bda19306a32201e : SUCCESS in 13m 57s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ec52b72a9659470e9e95be2f764f3a0c : SUCCESS in 10m 51s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"44dbb9c53f2239ca364c8ebd0144686a5f57a4c6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-26 12:27:37.000000000","message":"Patch Set 5:\n\nrecheck\n\nApt mirror issues","accounts_in_message":[],"_revision_number":5},{"id":"bce532901aa471a00ecac88939627fb11622e792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 13:37:37.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc29c22ce12f4335b22f18811e2a4ca5 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/633d4ba9ca3f4ebd8fe668a2af5cc831 : SUCCESS in 7m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58cc73e457784fb19eb5fd89fd4fd806 : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afe7c1dce1de4a118f4a8d2d08f33a1d : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c7e9a4c8e6941eb9f56b3256c6d0817 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f26e0ed0d94acfab8444ea98abe6c2 : SUCCESS in 11m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b0a15a4200424825b1a2d47fe9e1b012 : SUCCESS in 9m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ed6db62912e3490bb2e84e5d8d2f34f8 : SUCCESS in 8m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c858619893b745c8898796df242f77ae : SUCCESS in 17m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/637a1c731a4444ad9bc648e0314338b2 : SUCCESS in 1h 02m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19038897b597494c9f4a878e36407595 : SUCCESS in 49m 10s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/30a1ab638bcf44899378d01309e49fca : SUCCESS in 1h 06m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/85ebec4095214c86a8b57d8bf9946d22 : SUCCESS in 16m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e30bc1c22b8849df82226693ee1a52dd : SUCCESS in 12m 24s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fb9f3b699b834eb79c913cb39bc5cbfc : FAILURE in 1h 00m 26s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f63494ce8ea446a69890f7f2bd83d425 : SUCCESS in 25m 03s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14b5313b46c8486492e02faf0e5970b8 : SUCCESS in 21m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b07db85646ae43e19ccb6f9530b2d3f7 : SUCCESS in 25m 01s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4342553b3c7e448b8b8cae16f9da6121 : SUCCESS in 11m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f79b49cbda594da882d627ace04ecf25 : SUCCESS in 10m 10s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"2095b3646ccaf33f39ec4216e3741738306b972e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-26 15:48:25.000000000","message":"Patch Set 5:\n\nrecheck\n\nGalaxy issues","accounts_in_message":[],"_revision_number":5},{"id":"345774c96c82a1fa10cf8187e9644d4a2a216d38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 18:04:08.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a67cc7cb1724938b76478c06d0044a9 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9dc0fda3a4e142268ecbdc5797463edb : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61d345257b454606a593a5ebc8bdfb66 : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54e711981f6940f3997a01f07f976173 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/866d302e13074da5949e0d0045007e93 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/def025f0b54741f6b8676fa124292b0a : SUCCESS in 9m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4fe661847d6649b5a75ae58f2be439aa : SUCCESS in 7m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e17d2b8417454bcc82920241f9fc493a : SUCCESS in 11m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/080ba5174919444e91757bc99b3f941d : SUCCESS in 21m 45s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/28d6c5ecd09947ae95a68c118f906300 : SUCCESS in 49m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/701c33ef5534434c8c819d385174fb2d : SUCCESS in 45m 52s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/94a77506960245f1954395833c94329b : SUCCESS in 1h 00m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a597b5a1dc7540cb868c97d4636b58a7 : SUCCESS in 17m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3da296c5ba3249d1829d1b83a9847335 : SUCCESS in 13m 57s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f77718dc609548749ac0915ad6ba9a49 : SUCCESS in 2h 06m 35s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b97cb222901a4f50b6dd1b1e2c216f91 : SUCCESS in 29m 13s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/37c1fc7aae9341e8a26cd78fef609485 : SUCCESS in 37m 13s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7f46290ee7ca4be5ade7c7c334b2420e : SUCCESS in 24m 03s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/28f1018e66b745e7b5d3afdcb0dac54d : SUCCESS in 11m 16s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/413ce575be3749a0a6649c92e9261700 : SUCCESS in 11m 05s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"dac81b9630d955fa6a9e8846068c40d7bc69286b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 18:04:33.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"3bcb4e738a745021d0af0d896ce7761bac21fefc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 19:40:43.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4baaec30af964efd998868337484e7c8 : SUCCESS in 8m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/861a0ecc6f00498eb49d35440eea41b3 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e45a7dfc810845fb8f07a367f0ade765 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7712fdc890c348a79361cc2dfd1be875 : SUCCESS in 9m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/45b65b31c28a4a2e8e2f8a30cdc355d0 : SUCCESS in 5m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a1bf735655ec4248aedd68948ef8d7ef : SUCCESS in 9m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/703419a278904f38a6544cfb6b38a7c3 : SUCCESS in 16m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/38819b0f933046e68056ab570f5ba73a : SUCCESS in 50m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24cbdb1cff5946f0a1737179ffa85e87 : SUCCESS in 38m 27s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/23d0289e966b443eb5bb702c2500508b : SUCCESS in 56m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1ac357264ddb4d43b711a2bd5a3a33c7 : SUCCESS in 16m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5addf0810d5c425f8226d70cce2464ca : SUCCESS in 12m 21s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2965806f6c3444c3b5f9dfe57e575dab : SUCCESS in 1h 33m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dd5936247bc74852a45d006687c52651 : SUCCESS in 29m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b99f6ea58cb46a89243857f31890180 : SUCCESS in 23m 13s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0266454242c04c8cbecd2964d1de2430 : SUCCESS in 26m 44s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4895e8bf1df147b1abbdde9eb95c0ebc : SUCCESS in 13m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c99b29b07195442e9d58d5d3c40c406e : SUCCESS in 9m 52s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"ba368be1df69daceb038c97b0a001fd6ff95da3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 19:40:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"8a444acbd9d0da64ded39e644bb7242ff3da0f7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 19:42:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/527c1ea4aeaa4bbd8e7e62a11188b690 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"eaabd7ef30a84f9210bbe8be392660f83ac9ccaf","revisions":{"981b74ebc2abd47674ddcf17ecc3dbbd18c9c0f3":{"kind":"REWORK","_number":1,"created":"2021-03-22 15:08:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/782272/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/782272/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/782272/1"}}},"commit":{"parents":[{"commit":"ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9","subject":"Merge \"Change docker_registry network_mode to host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:00:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:00:55.000000000","tz":0},"subject":"Ubuntu: AppArmor config for seed VM provisioning","message":"Ubuntu: AppArmor config for seed VM provisioning\n\nThe stackhpc.libvirt-host 1.8.1 release includes support for modifying\nthe libvirt QEMU AppArmor template to include the path of directory\nbacked pools. This allows the \u0027kayobe seed vm provision\u0027 command to\nsucceed with AppArmor enabled.\n\nStory: 2004960\nTask: 42091\n\nChange-Id: I20d8ee783cdf158bbb469e831518e36d3c8ed73f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/981b74ebc2abd47674ddcf17ecc3dbbd18c9c0f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/981b74ebc2abd47674ddcf17ecc3dbbd18c9c0f3"}]},"branch":"refs/heads/master"},"23b7b52670cfe58515ce12be75fe9628bd4d530a":{"kind":"REWORK","_number":2,"created":"2021-03-23 11:07:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/782272/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/782272/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/782272/2"}}},"commit":{"parents":[{"commit":"ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9","subject":"Merge \"Change docker_registry network_mode to host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:00:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-23 11:06:53.000000000","tz":0},"subject":"Ubuntu: AppArmor config for seed VM provisioning","message":"Ubuntu: AppArmor config for seed VM provisioning\n\nThe stackhpc.libvirt-host 1.8.1 release includes support for modifying\nthe libvirt QEMU AppArmor template to include the path of directory\nbacked pools. This allows the \u0027kayobe seed vm provision\u0027 command to\nsucceed with AppArmor enabled.\n\nStory: 2004960\nTask: 42091\n\nChange-Id: I20d8ee783cdf158bbb469e831518e36d3c8ed73f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23b7b52670cfe58515ce12be75fe9628bd4d530a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23b7b52670cfe58515ce12be75fe9628bd4d530a"}]},"branch":"refs/heads/master"},"7ef27dc2cbacd41cde34356c35a074e90a0fc029":{"kind":"REWORK","_number":3,"created":"2021-03-23 15:28:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/782272/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/782272/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/782272/3"}}},"commit":{"parents":[{"commit":"ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9","subject":"Merge \"Change docker_registry network_mode to host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:00:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-23 15:28:41.000000000","tz":0},"subject":"Ubuntu: AppArmor config for seed VM provisioning","message":"Ubuntu: AppArmor config for seed VM provisioning\n\nThe stackhpc.libvirt-host 1.8.1 release includes support for modifying\nthe libvirt QEMU AppArmor template to include the path of directory\nbacked pools. This allows the \u0027kayobe seed vm provision\u0027 command to\nsucceed with AppArmor enabled.\n\nStory: 2004960\nTask: 42091\n\nChange-Id: I20d8ee783cdf158bbb469e831518e36d3c8ed73f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ef27dc2cbacd41cde34356c35a074e90a0fc029"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ef27dc2cbacd41cde34356c35a074e90a0fc029"}]},"branch":"refs/heads/master"},"25d7fa11773efd14402a03597ddaca5a525e1390":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-03-23 15:43:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/782272/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/782272/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/782272/4"}}},"commit":{"parents":[{"commit":"ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9","subject":"Merge \"Change docker_registry network_mode to host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:00:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-23 15:42:50.000000000","tz":0},"subject":"Ubuntu: AppArmor config for seed VM provisioning","message":"Ubuntu: AppArmor config for seed VM provisioning\n\nThe stackhpc.libvirt-host 1.8.2 release includes support for modifying\nthe libvirt QEMU AppArmor template to include the path of directory\nbacked pools. This allows the \u0027kayobe seed vm provision\u0027 command to\nsucceed with AppArmor enabled.\n\nStory: 2004960\nTask: 42091\n\nChange-Id: I20d8ee783cdf158bbb469e831518e36d3c8ed73f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25d7fa11773efd14402a03597ddaca5a525e1390"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/25d7fa11773efd14402a03597ddaca5a525e1390"}]},"branch":"refs/heads/master","description":"Edit commit message"},"eaabd7ef30a84f9210bbe8be392660f83ac9ccaf":{"kind":"REWORK","_number":5,"created":"2021-03-24 10:41:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/72/782272/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/72/782272/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/72/782272/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/72/782272/5"}}},"commit":{"parents":[{"commit":"ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9","subject":"Merge \"Change docker_registry network_mode to host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:00:53.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-24 10:41:12.000000000","tz":0},"subject":"Ubuntu: AppArmor config for seed VM provisioning","message":"Ubuntu: AppArmor config for seed VM provisioning\n\nThe stackhpc.libvirt-host 1.8.2 release includes support for modifying\nthe libvirt QEMU AppArmor template to include the path of directory\nbacked pools. This allows the \u0027kayobe seed vm provision\u0027 command to\nsucceed with AppArmor enabled.\n\nStory: 2004960\nTask: 42091\n\nChange-Id: I20d8ee783cdf158bbb469e831518e36d3c8ed73f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eaabd7ef30a84f9210bbe8be392660f83ac9ccaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eaabd7ef30a84f9210bbe8be392660f83ac9ccaf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
