)]}'
{"id":"openstack%2Fkayobe~782273","triplet_id":"openstack%2Fkayobe~master~I96827fc32c1594ca9a0535e259929c49d3f0e704","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2022-02-14 17:46:53.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2022-02-14 15:39:23.000000000","reason":"Pierre Riteau replied on the change"}},"hashtags":[],"change_id":"I96827fc32c1594ca9a0535e259929c49d3f0e704","subject":"CI: Enable bare metal testing for Ubuntu","status":"MERGED","created":"2021-03-22 15:10:40.000000000","updated":"2022-02-14 17:50:12.000000000","submitted":"2022-02-14 17:46:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782273","meta_rev_id":"008b1a45bd991b6685994dd9ed44a0ba06ea88d9","_number":782273,"virtual_id_number":782273,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-02-14 17:46:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2022-02-14 09:59:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-02-14 15:39:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-02-14 15:39:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 16:08:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-11 10:52:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2022-02-14 09:59:32.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"022fe762ffa1667808cd67d83b04c2a88141bf7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-22 15:10:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd77fe0be71c92d004c509bbde84cb0269fbb9d6","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-22 15:11:44.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"54dcbbce1632380938551fcf45fb0b4bfaaca956","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 16:08:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2805dd1bbc344065b19272486e2460eb : FAILURE in 36m 21s","accounts_in_message":[],"_revision_number":1},{"id":"cf024cdb00b33ba0c17b64de26dd879315d3a110","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-22 16:31:41.000000000","message":"Patch Set 1:\n\nTenks broken by recent libvirt-host role change. Fix: https://github.com/stackhpc/ansible-role-libvirt-host/pull/47","accounts_in_message":[],"_revision_number":1},{"id":"e6cb5b70d4592fbf27b2a3ffb44c7a163cbd00b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-22 16:34:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"97aed0923c7f3646f3da4875220bb13319f18a6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 17:19:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c5b277d89a94c64862b202227e8561f : FAILURE in 38m 56s","accounts_in_message":[],"_revision_number":2},{"id":"9bbc3e9630de6fbb0d90993a24fdc8db1baab636","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-23 09:25:50.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f65d0622b8654eda0bc67e0e8d85c85f98f89f63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 10:21:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1981e65959a94c7fa03976ba9f145b40 : FAILURE in 53m 45s","accounts_in_message":[],"_revision_number":3},{"id":"ee10201a29bbf0ed0492076c6997d02868496975","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-24 10:42:20.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"72d8388f256e59029cd2e164a752ba0de2465d8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 11:26:56.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/328511d8548a483aa34a5d305f95c1d0 : FAILURE in 38m 32s","accounts_in_message":[],"_revision_number":4},{"id":"1fc93b1d79bcb5e1036c30c878a956d96ce73c3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-24 11:33:42.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"c4a7573b24a93c0eddcedab9d0c466adee06fbef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 12:24:43.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3a3a700f968540dca1f72cd9b49dba82 : FAILURE in 50m 10s","accounts_in_message":[],"_revision_number":5},{"id":"1fcb8e2535073f41aae6dd4c1c014660cecef1ef","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-24 12:28:44.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":5},{"id":"f02d60291d030b23a19cfaadc6d79cb6d0f12b40","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-24 12:28:55.000000000","message":"Patch Set 5:\n\nFails as before, Error setting Chassis Boot Parameter 5\\nError setting Chassis Boot Parameter 0","accounts_in_message":[],"_revision_number":5},{"id":"fd1049e045630eefee974882b44f79bf6a7d32b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-10 10:04:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"1665c1b1eca454a4d6b950f5ed4e5782c058c84d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 11:00:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/efe33b9548c54248a7b6175ccba367f0 : FAILURE in 52m 20s","accounts_in_message":[],"_revision_number":6},{"id":"9c8c5147ed1a31efa58425f90b1b2ba12a9b138b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-10 11:32:39.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"4665c18d868e88bcc3927c276c3ce3eed5fb91bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 12:33:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70509f2e1bef4168b6b9750f1b8d4e01 : SUCCESS in 55m 06s","accounts_in_message":[],"_revision_number":7},{"id":"38d833052760eb563e18c6c984f4c01e61768288","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-10 12:40:39.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b39f1d48192636c34417108495dca75c543ed7d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 13:41:23.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d48c57f6ebab4d0888de754fdd17faef : SUCCESS in 55m 13s","accounts_in_message":[],"_revision_number":7},{"id":"f55fe7a246b658b521a9ee98e4920b46ce645a16","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-10 14:24:57.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b3a42cb4b918b489f27a0eb96af08d7ad82cb1ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 15:23:06.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ba79ee7b9fc34d9390c660314a9bfd12 : SUCCESS in 52m 40s","accounts_in_message":[],"_revision_number":7},{"id":"0c4b9535405024a044597c1a027bbee9055b0e45","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-10 15:37:37.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"25565c09b9d6ae364edcaa13c4b8f91fb4df93a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 16:29:58.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/269c143d2bea4b44b186fa2c6557458e : SUCCESS in 48m 24s","accounts_in_message":[],"_revision_number":7},{"id":"f69e758ec0ae3dea1ce00710382e16342dc74109","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-10 16:48:06.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"9b9c321e124572b1b2763ae501b12e843ccd23e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 17:43:14.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3d792f47f4004ed6ae73ad60b24abecd : SUCCESS in 50m 56s","accounts_in_message":[],"_revision_number":8},{"id":"a615a57cfdaac006ac0090d1520c857198a13f17","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-10 17:45:15.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b70e07dd1d840d473963b55d61d3d8aae09b3225","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 18:44:24.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90226cbc4f7d431fa2d62ff652d9a3ba : SUCCESS in 52m 00s","accounts_in_message":[],"_revision_number":8},{"id":"44152ce9a3e80bf22a72f26bf91e78dd06a46ed0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-10 20:21:50.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"68aa4803b96a2be703d3b780cb371b66cfa8b1df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-10 21:18:32.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2df3cbe91dd74887a3c4b920acb49abb : SUCCESS in 55m 16s","accounts_in_message":[],"_revision_number":9},{"id":"04ba67bc9e63b5023f2a70e130fa9b3b241df0dc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 09:00:54.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"dcd8f4287d651b04d7ff3635db0b38965242920a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 10:00:50.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/626689c9f0b24f6d9c78ec42f8a70c70 : SUCCESS in 55m 05s","accounts_in_message":[],"_revision_number":9},{"id":"9fc33d0b7bd83e56226bf157f50c291a4d99d27e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 10:46:51.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"77afe465a34ce2625357b0fce3541753ac1086ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 10:47:52.000000000","message":"Patch Set 11: Commit message was updated.","accounts_in_message":[],"_revision_number":11},{"id":"393a4ac11563fc379476b424eaf03f72e8b3e1cb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-11 10:52:09.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"6c362ff568096c7bfb5f8d4a7386e44ccc3c2897","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 12:40:21.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aada0b9a55e344c28d20242975e61f1c : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33e148938cb8414799ac56421e1e42ec : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/163599a19e564d3ebd6dad1dc35e789e : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29e1c44d792348138f31fa70de6e75c1 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf93396a63f24d5b9298bd7529172d23 : SUCCESS in 8m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4ab7ba3e8ca349a99a171770243067f2 : SUCCESS in 5m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/93c42fc683b84da1abcaa9d94ed6deeb : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a9ab7155f4be48cc917f8f4a109a14c2 : SUCCESS in 35m 00s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d51b552cf0bc4782a47119800a7e6d64 : SUCCESS in 1h 03m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08d3177cab8f4739a4201463a4c7cdcd : FAILURE in 58m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6ddbe162522042a1922721aabd305e94 : SUCCESS in 1h 15m 52s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/93d49f092c5e4a7eb27c503b11a5a854 : SUCCESS in 20m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43f61f67f9fe4ea68c969426565cf0bd : SUCCESS in 13m 29s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c27c5285e2cf4eebae77c2a655e0cbee : SUCCESS in 1h 49m 52s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6e2626296f6f4ee19df5369d43e106b8 : SUCCESS in 1h 10m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/315255c33cc941fe807ff91ac1be4877 : SUCCESS in 46m 53s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84d7944fe79b4a7db8685fb48def696f : SUCCESS in 35m 16s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5d78bcc09f624a3e876b49f8ae0af7ad : SUCCESS in 26m 40s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e402f82098b4914833ba51f857089bb : SUCCESS in 24m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/558366cf937c4f0f93a1ad9489898362 : SUCCESS in 18m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00f78419f5be457d96973aaea9cef9d3 : SUCCESS in 11m 44s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/3154e3e48e27439a843df06e3bdb75b1 : SUCCESS in 20m 15s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4d840624a9754d0fa2ebf1d20a201744 : SUCCESS in 15m 57s","accounts_in_message":[],"_revision_number":11},{"id":"8d02e9332354aea983efa4290b1ddd369e43c51a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-11 13:21:56.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"98f9cd6a3f6218451948e83a071aa8e621538572","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 14:09:41.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"197ac13a3cdd61b409ed877c46345ca18eb0dd83","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 14:13:13.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":11},{"id":"33f8545c4dd9e595e35d5bda720d47cb345cce37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 15:13:08.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4024e7c572d14e6185eebce7b5d8a2b2 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cbd9d318cfc4f0d9739bada3fc54c1b : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/abb056d6445d4d8d8b6b9a82d4247b3f : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/669480480da94ed782f06413eaa38f0e : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a21c35459b074ef39c2f1a80456c4069 : SUCCESS in 9m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6ee7b4dc4cbf4301a85ad1136478d96e : SUCCESS in 6m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/44ffed086bab4bacab451d53b1206d44 : SUCCESS in 13m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1242d8081270479098486b065df3f4cc : SUCCESS in 33m 52s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3fb485f7e72a4fdd8901fd65aebcf20c : SUCCESS in 1h 08m 02s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d42ec49dad334568a7ad134737066ec0 : SUCCESS in 55m 03s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e147e51f6da34d12a26072045c36dd6f : SUCCESS in 1h 12m 00s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/ab7d4d2cc896422ca48d4ade41735d9f : SUCCESS in 23m 28s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6cbddf73b8c1461e8df30668bb07590d : SUCCESS in 15m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/34b66746e62545e9b9e8d935353bc180 : SUCCESS in 1h 46m 16s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e26287045039450a969cea68e68ad3d7 : SUCCESS in 1h 18m 59s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c4255a9768184c6b89bbb188d948b517 : SUCCESS in 51m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a59e0dfcf5b4712903a483553e1cbee : SUCCESS in 39m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9a95f8dd106e4a00a131490445304298 : SUCCESS in 26m 33s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4a8d3e5aaa264a22b3c9f194a6d611c8 : SUCCESS in 20m 34s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/40e53097c5d24f2dadc275c6073dcfaa : SUCCESS in 13m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2b5f32205a8647a4bf506157199d1a94 : SUCCESS in 12m 35s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/97a522106bc640d7a1ec4e00bf106368 : SUCCESS in 16m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bcd8e18d13b64646ada26a9705d97f1a : SUCCESS in 16m 02s","accounts_in_message":[],"_revision_number":11},{"id":"eeffe2a55b275a0fb5a954cf1a6d758017ce87ec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 16:09:22.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"26feed7c30f3a093b55193577b906adac94ad052","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 16:10:01.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kayobe on branch master.  The error was:\n\n  Job kayobe-overcloud-ubuntu-focal1 not defined\n\nThe error appears in the following project stanza:\n\n  project:\n      check:\n        jobs:\n          - kayobe-overcloud-ubuntu-focal1\n          - kayobe-overcloud-ubuntu-focal2\n          - kayobe-overcloud-ubuntu-focal3\n          - kayobe-overcloud-ubuntu-focal4\n  \n      gate:\n        queue: kayobe\n        jobs:\n          - kayobe-tox-ansible-syntax\n          - kayobe-tox-ansible\n          - kayobe-tox-molecule\n          - kayobe-overcloud-centos8s\n          - kayobe-overcloud-ubuntu-focal\n          - kayobe-overcloud-tls-centos8s\n          - kayobe-overcloud-host-configure-centos8s\n          - kayobe-overcloud-host-configure-ubuntu-focal\n          - kayobe-overcloud-upgrade-centos8s\n          - kayobe-overcloud-upgrade-ubuntu-focal\n          - kayobe-seed-centos8s\n          - kayobe-seed-ubuntu-focal\n          - kayobe-seed-upgrade-centos8s\n          - kayobe-seed-upgrade-ubuntu-focal\n          - kayobe-seed-vm-centos8s\n          - kayobe-seed-vm-ubuntu-focal\n          - kayobe-infra-vm-centos8s\n          - kayobe-infra-vm-ubuntu-focal\n\n  in \"openstack/kayobe/zuul.d/project.yaml@master\", line 2, column 3","accounts_in_message":[],"_revision_number":12},{"id":"087f764ad11c97f76d12485d1b61fc976e2820a4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 16:16:16.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"90493b6864cfc0c70a4dc754fedb035a3c8da588","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 17:16:18.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal1 https://zuul.opendev.org/t/openstack/build/c7d38cc78fa447969770607f997335fa : SUCCESS in 53m 19s\n- kayobe-overcloud-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/7499276091af456095c590cafadbd6bd : SUCCESS in 49m 37s\n- kayobe-overcloud-ubuntu-focal3 https://zuul.opendev.org/t/openstack/build/866157506b3b4b3d902d7779fd62abe1 : SUCCESS in 55m 40s\n- kayobe-overcloud-ubuntu-focal4 https://zuul.opendev.org/t/openstack/build/daa5c12f64144ada861d2c4acbc9f898 : SUCCESS in 57m 41s","accounts_in_message":[],"_revision_number":13},{"id":"33a7b76fc9d038215e2852d6cdd8d59c3c4721b9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 18:47:46.000000000","message":"Patch Set 13:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":13},{"id":"4ffd198f9edfdba5113c73e986cd5812cd5f6490","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 18:48:44.000000000","message":"Patch Set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"64e746d6b6df1e283d2c364ed213d9b6e68edce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 19:43:41.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu-focal1 https://zuul.opendev.org/t/openstack/build/07339e36f9d740d6827e62986a6fea9d : FAILURE in 47m 29s\n- kayobe-overcloud-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/c2d52e46e1914a89b495d72048aecbda : SUCCESS in 53m 35s\n- kayobe-overcloud-ubuntu-focal3 https://zuul.opendev.org/t/openstack/build/055a1ab75e9c4136bdfb49ebf231f662 : FAILURE in 51m 43s\n- kayobe-overcloud-ubuntu-focal4 https://zuul.opendev.org/t/openstack/build/b9d14d6946e242018b8f90220c60d480 : SUCCESS in 52m 28s","accounts_in_message":[],"_revision_number":14},{"id":"f48ab276f47ac033912d3bbb1e0e0a8c950e2a7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-11 20:11:40.000000000","message":"Patch Set 15: Published edit on patch set 14.","accounts_in_message":[],"_revision_number":15},{"id":"698d0a7c7b508e384e94a74b5802d1499f20140d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-11 20:58:06.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal1 https://zuul.opendev.org/t/openstack/build/15070edf0b3c44f9b6ad9d401597ecd6 : SUCCESS in 45m 01s\n- kayobe-overcloud-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/2ab549e71c044791b86349d4434b3343 : SUCCESS in 43m 47s\n- kayobe-overcloud-ubuntu-focal3 https://zuul.opendev.org/t/openstack/build/fcf3174a6d04455f810c5ea1e0aa2e05 : SUCCESS in 44m 36s\n- kayobe-overcloud-ubuntu-focal4 https://zuul.opendev.org/t/openstack/build/faa2a51f8bfe42db832c7d561af33b8f : SUCCESS in 44m 21s","accounts_in_message":[],"_revision_number":15},{"id":"25134ebde2badc0a872c36d2993100c1fb15b9d2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-12 08:36:39.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"6e7374a6450873e51703d4ef7d9c26acf4b9e779","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-12 09:32:36.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-ubuntu-focal1 https://zuul.opendev.org/t/openstack/build/4c5c4482826a49f0b238339c115153d6 : SUCCESS in 54m 45s\n- kayobe-overcloud-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/48579f9e56dc4910bffbbf12ed106d2d : FAILURE in 40m 19s\n- kayobe-overcloud-ubuntu-focal3 https://zuul.opendev.org/t/openstack/build/3d911b0968a846e89db50474b4cbe3fb : SUCCESS in 54m 27s\n- kayobe-overcloud-ubuntu-focal4 https://zuul.opendev.org/t/openstack/build/7cd55389173943119ee644a1adf00613 : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":15},{"id":"ff6689ad6504f512f2e17692bda1a0cea6fffb99","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-12 14:02:57.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"9c9c8df0a20c3900a0f273587b50c9d02fa9d022","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-12 15:19:21.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal1 https://zuul.opendev.org/t/openstack/build/822a1fa5d9544872bdd7dbf9bb58df86 : SUCCESS in 1h 10m 16s\n- kayobe-overcloud-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/64b8ac2003db4436beb8914abb4759fe : SUCCESS in 56m 55s\n- kayobe-overcloud-ubuntu-focal3 https://zuul.opendev.org/t/openstack/build/4302931e3ec7413c84ff17f0e4360d14 : SUCCESS in 43m 48s\n- kayobe-overcloud-ubuntu-focal4 https://zuul.opendev.org/t/openstack/build/d540fb9aebb34ba6a65c2089b5b89c6e : SUCCESS in 52m 18s","accounts_in_message":[],"_revision_number":15},{"id":"d4471f542d7402f2095ae99ee03cd910147c2c1a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-12 17:34:29.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d3033ffc4adcf5645731f74f3f146f1775769867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-12 18:30:55.000000000","message":"Patch Set 15:\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-ubuntu-focal1 https://zuul.opendev.org/t/openstack/build/f70a3706afd04a3ab096e23d3078f189 : SUCCESS in 46m 27s\n- kayobe-overcloud-ubuntu-focal2 https://zuul.opendev.org/t/openstack/build/f643e8e603374bfcbc7cfe4547924464 : SUCCESS in 53m 14s\n- kayobe-overcloud-ubuntu-focal3 https://zuul.opendev.org/t/openstack/build/1d34ec2d5eb44ad484587baf24dfac96 : SUCCESS in 54m 56s\n- kayobe-overcloud-ubuntu-focal4 https://zuul.opendev.org/t/openstack/build/a44f400298c74988b76ec43755991633 : SUCCESS in 52m 26s","accounts_in_message":[],"_revision_number":15},{"id":"41e0f3e7809f0b2bc7a872ba1ad58dcead3f1105","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-14 09:18:43.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"ee6cbf231982f17ab25fa1bfa35dd12564f715d0","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2022-02-14 09:59:32.000000000","message":"Patch Set 16: Code-Review+2","accounts_in_message":[],"_revision_number":16},{"id":"b66113b280b68b1b548eee3e2bdfdc4a85d4bc53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 11:21:35.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a72d936fb66428bb8d3b1783500077b : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49373f2cf8674398bcb4c1123089f4ea : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/56f9521dc34c41d5bd7e939363f41828 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4de6d08275e549be817736a29fb33955 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ea212140b684ff182b18077d21f0e13 : SUCCESS in 9m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fc9088a149304aaa8d5754a21b1a4db5 : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/aaab4c0de5344ddca2a87083ae93fff7 : SUCCESS in 10m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/df145a2a92ea4a31b433a9f73a8ef23b : SUCCESS in 33m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/16a9fd7924934b798317e58ce8b99cd9 : SUCCESS in 1h 07m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/522e377ab9c8467199493e26de922cb2 : SUCCESS in 50m 30s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1b220aafcddf41ff8b7e7eac92bedb66 : FAILURE in 1h 14m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/29bbd00a36df494e9111a4be7d2a9c78 : SUCCESS in 22m 18s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/052a3ec347cd4fc48de507813a1cfc5c : SUCCESS in 15m 31s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/dda575bf02074501b0eda41680f99dcd : SUCCESS in 1h 58m 06s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c57e86d40a148b1b57691182d7e77f8 : SUCCESS in 1h 06m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/56798392bc9e4e18850feaeaf83f36d9 : SUCCESS in 46m 09s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2e53d98fada34094b1f6cd6222487eaf : SUCCESS in 37m 38s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bfdb47606ff54190969c85f0fc3f9a17 : SUCCESS in 27m 07s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caa0a0274fce4e42ac2150078a878840 : SUCCESS in 24m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c9f50ac0bfe14bcfa0991ab05750f042 : SUCCESS in 19m 43s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a7b37101e365489f94bca30a32c3f567 : SUCCESS in 13m 12s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/f90b78d05b654bad98d14ce29e2d316a : SUCCESS in 17m 44s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bef98a46efa742a9b58e8e5538b0d94e : SUCCESS in 13m 40s","accounts_in_message":[],"_revision_number":16},{"id":"0e6977e00844f60c08f132f181f34e1ed3c0902f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-02-14 11:40:25.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"1aa11d609cce7871c3c36a06cbb23bab4c5278ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 13:39:18.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbbdf731fd8340fd8789b82964d63739 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22db7b5b4cdc4de9a9542dfcb43a0ad1 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be5f28ff73ad47a5adb1752ca7ee1ff5 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acfa22b2c9c244019c3f99d79d9011f8 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd36fe899dc454aa5815fe8131bba9a : SUCCESS in 8m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/df8cbcc968cd47cfa85c82e4bf43fd47 : SUCCESS in 7m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a42ad869190a45a89d58c641334ac1ae : SUCCESS in 12m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fdcd67bab89d4bf39a9c3d975056aa9f : SUCCESS in 30m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/0fba57ec8aef4462b42bba456aed5184 : SUCCESS in 1h 01m 20s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c2274948a31e4cd3b064c05785e7dd54 : SUCCESS in 49m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/ccbf881a8edf48229845b716fe22f0e3 : SUCCESS in 1h 16m 32s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4f47d16262464e799f24a90790109b59 : SUCCESS in 26m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b428cd77e014c4f92160df02f02f701 : SUCCESS in 13m 30s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/192ca1a43613492cbc1d03d9aa28d4dd : SUCCESS in 1h 52m 38s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/324f7acab4f6405e9f82687e45ec02ca : SUCCESS in 1h 04m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3b571703b4bf4f43914b54c79b85da0a : SUCCESS in 51m 07s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b034fe1d0cf7427d95e4be66567941c3 : SUCCESS in 35m 55s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6f23e4e7279b4acc9443aa3fb08474b1 : SUCCESS in 28m 51s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9676e899882b48fb9d01226b3082e085 : SUCCESS in 24m 02s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/04c70504cd7d4b26b014055ab7746990 : SUCCESS in 17m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5a071018dea04eecb7d749de5130331d : SUCCESS in 12m 28s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/a44bf7947620488996140477508b0f22 : SUCCESS in 20m 24s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4ddbc1aaeb5a4eafbfcd4d7a35bca58e : SUCCESS in 13m 46s","accounts_in_message":[],"_revision_number":16},{"id":"5e69ff464db2fc52ee9fcb62c19dd1f64859146b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2022-02-14 15:39:23.000000000","message":"Patch Set 16: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":16},{"id":"bc44e45c0b9b721908377a249dde0de49771c6c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 15:39:58.000000000","message":"Patch Set 16: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":16},{"id":"e4219bc0b13cb0a66e1b88ea41fa424a413b35f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 17:46:52.000000000","message":"Patch Set 16: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76e0a419be194337ba976ced72bd802f : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f2df9003a4c4c29a6f50a23762c5f4f : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f31a045dff84e72bb6b7d864df6c014 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0596b896d64f4b3b8c2ed4b43d388abc : SUCCESS in 9m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4a650c1e9f8f4e64a1689e19a6403b67 : SUCCESS in 7m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eccdf89aa54e429294ae3656696c3ff3 : SUCCESS in 12m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/84c8088aad754da0a937cac5e9dbd73f : SUCCESS in 33m 24s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d2f3a2719ff14c338113d78ca0de9744 : SUCCESS in 1h 13m 26s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2d78cfde6cbe4a4081606ab1f257df2e : SUCCESS in 1h 07m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/72d7c3b3a19147cab68b7c37d3395cb5 : SUCCESS in 1h 25m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b40000ae362947ae84f7d363b45c6092 : SUCCESS in 22m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a22c52f8bdf14d3a8b269bdb3dc57c42 : SUCCESS in 21m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/a0ceb3727cc34b43b9b6e1356f3c9df3 : SUCCESS in 2h 06m 03s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f5a168488a7482998ea5d96a23d88d4 : SUCCESS in 1h 12m 07s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8aa793fe31ce4b01bfa5814611d0ae3f : SUCCESS in 48m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ffadefadf20140fc9de5d5e39484eb43 : SUCCESS in 36m 19s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6d42c5bc767a4eef90c8cdfd8aec9f04 : SUCCESS in 27m 56s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b51728cea8bd44bdb1066c92d08a9108 : SUCCESS in 22m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ffd33260f1af4be5a5319bfaa6937e4c : SUCCESS in 17m 26s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/501256798de64985bf014067592facb9 : SUCCESS in 12m 23s\n- kayobe-infra-vm-centos8s https://zuul.opendev.org/t/openstack/build/2999600ca66a4e70b5be6a1b29746ff2 : SUCCESS in 19m 10s\n- kayobe-infra-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51a7a9266e8d468a9dc9ccd1258b64e6 : SUCCESS in 14m 12s","accounts_in_message":[],"_revision_number":16},{"id":"7614934fa6d2383a4ab2d284d5decb7fe9042e14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 17:46:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":16},{"id":"008b1a45bd991b6685994dd9ed44a0ba06ea88d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-14 17:50:12.000000000","message":"Patch Set 16:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b95d4a0711d64fa9ba939fd9d216b9c8 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"883027afb0656a46369d3f4965a0c65b4a3e8dd8","revisions":{"c7bdba7cb16118ee4b55abee10b0fbde04f76d22":{"kind":"REWORK","_number":1,"created":"2021-03-22 15:10:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/1"}}},"commit":{"parents":[{"commit":"ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9","subject":"Merge \"Change docker_registry network_mode to host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c7bdba7cb16118ee4b55abee10b0fbde04f76d22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c7bdba7cb16118ee4b55abee10b0fbde04f76d22"}]},"branch":"refs/heads/master"},"209242f3aa58cf9eb3a005d1c33b4c6fce5a8c5b":{"kind":"REWORK","_number":2,"created":"2021-03-22 16:34:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/2"}}},"commit":{"parents":[{"commit":"ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9","subject":"Merge \"Change docker_registry network_mode to host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 16:34:41.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/209242f3aa58cf9eb3a005d1c33b4c6fce5a8c5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/209242f3aa58cf9eb3a005d1c33b4c6fce5a8c5b"}]},"branch":"refs/heads/master"},"3e1b910c831f95a6db7e7cc20a72bdef81fdf69e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-03-23 09:25:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/3"}}},"commit":{"parents":[{"commit":"ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9","subject":"Merge \"Change docker_registry network_mode to host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-23 09:25:46.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/782383\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e1b910c831f95a6db7e7cc20a72bdef81fdf69e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3e1b910c831f95a6db7e7cc20a72bdef81fdf69e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"93075cf2b195491f091651dc03ccd64dab2d3d61":{"kind":"REWORK","_number":4,"created":"2021-03-24 10:42:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/4"}}},"commit":{"parents":[{"commit":"ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9","subject":"Merge \"Change docker_registry network_mode to host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-24 10:42:18.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/782383\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93075cf2b195491f091651dc03ccd64dab2d3d61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93075cf2b195491f091651dc03ccd64dab2d3d61"}]},"branch":"refs/heads/master"},"5167f1badd91e1f568d19698f44f58cd1a464dc8":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2021-03-24 11:33:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/5"}}},"commit":{"parents":[{"commit":"ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9","subject":"Merge \"Change docker_registry network_mode to host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-24 11:33:31.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5167f1badd91e1f568d19698f44f58cd1a464dc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5167f1badd91e1f568d19698f44f58cd1a464dc8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"7a814c8b18f6183d59a76bb2ac6a127810e1816e":{"kind":"REWORK","_number":6,"created":"2022-02-10 10:04:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/6"}}},"commit":{"parents":[{"commit":"e6da84a094c04f97a37ead1dd193b49fc7dfb489","subject":"Merge \"Support building multiple disk images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6da84a094c04f97a37ead1dd193b49fc7dfb489"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-10 10:04:21.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a814c8b18f6183d59a76bb2ac6a127810e1816e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a814c8b18f6183d59a76bb2ac6a127810e1816e"}]},"branch":"refs/heads/master"},"45f5a76f97763f06e61af26757376b3218378b29":{"kind":"REWORK","_number":7,"created":"2022-02-10 11:32:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/7"}}},"commit":{"parents":[{"commit":"e6da84a094c04f97a37ead1dd193b49fc7dfb489","subject":"Merge \"Support building multiple disk images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6da84a094c04f97a37ead1dd193b49fc7dfb489"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-10 11:32:30.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45f5a76f97763f06e61af26757376b3218378b29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45f5a76f97763f06e61af26757376b3218378b29"}]},"branch":"refs/heads/master"},"8f0ef2e84a25f460444b6fc7205aa54e518410ab":{"kind":"REWORK","_number":8,"created":"2022-02-10 16:48:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/8"}}},"commit":{"parents":[{"commit":"e6da84a094c04f97a37ead1dd193b49fc7dfb489","subject":"Merge \"Support building multiple disk images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6da84a094c04f97a37ead1dd193b49fc7dfb489"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-10 16:48:06.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f0ef2e84a25f460444b6fc7205aa54e518410ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8f0ef2e84a25f460444b6fc7205aa54e518410ab"}]},"branch":"refs/heads/master"},"3266bc49fb196ee3315951ecd4bd8cdecde8c3f1":{"kind":"REWORK","_number":9,"created":"2022-02-10 20:21:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/9"}}},"commit":{"parents":[{"commit":"e6da84a094c04f97a37ead1dd193b49fc7dfb489","subject":"Merge \"Support building multiple disk images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6da84a094c04f97a37ead1dd193b49fc7dfb489"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-10 20:21:50.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3266bc49fb196ee3315951ecd4bd8cdecde8c3f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3266bc49fb196ee3315951ecd4bd8cdecde8c3f1"}]},"branch":"refs/heads/master"},"a4c5be1e3848fe5524d8c36cc97f11f444bd39b0":{"kind":"REWORK","_number":10,"created":"2022-02-11 10:46:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/10"}}},"commit":{"parents":[{"commit":"e6da84a094c04f97a37ead1dd193b49fc7dfb489","subject":"Merge \"Support building multiple disk images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6da84a094c04f97a37ead1dd193b49fc7dfb489"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-11 10:46:48.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4c5be1e3848fe5524d8c36cc97f11f444bd39b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a4c5be1e3848fe5524d8c36cc97f11f444bd39b0"}]},"branch":"refs/heads/master"},"f11b2e3ab6b5305af0f84dba865554a81a6019ff":{"kind":"NO_CODE_CHANGE","_number":11,"created":"2022-02-11 10:47:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/11"}}},"commit":{"parents":[{"commit":"e6da84a094c04f97a37ead1dd193b49fc7dfb489","subject":"Merge \"Support building multiple disk images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6da84a094c04f97a37ead1dd193b49fc7dfb489"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-11 10:47:52.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nPreviously we were seeing issues with ipmitool and virtualbmc on Ubuntu,\nbut this no longer seems to be the case. Example error:\n\n  Error setting Chassis Boot Parameter 5\\nError setting Chassis Boot Parameter 0\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f11b2e3ab6b5305af0f84dba865554a81a6019ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f11b2e3ab6b5305af0f84dba865554a81a6019ff"}]},"branch":"refs/heads/master","description":"Edit commit message"},"c7e5ed08b8394a31e560f1842cf0a451b1e8723d":{"kind":"REWORK","_number":12,"created":"2022-02-11 16:09:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/12"}}},"commit":{"parents":[{"commit":"e6da84a094c04f97a37ead1dd193b49fc7dfb489","subject":"Merge \"Support building multiple disk images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6da84a094c04f97a37ead1dd193b49fc7dfb489"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-11 16:09:15.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nPreviously we were seeing issues with ipmitool and virtualbmc on Ubuntu,\nbut this no longer seems to be the case. Example error:\n\n  Error setting Chassis Boot Parameter 5\\nError setting Chassis Boot Parameter 0\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c7e5ed08b8394a31e560f1842cf0a451b1e8723d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c7e5ed08b8394a31e560f1842cf0a451b1e8723d"}]},"branch":"refs/heads/master"},"bdfd7e8a2cfbd5a23c97ae5b94d26e4ebaad3add":{"kind":"REWORK","_number":13,"created":"2022-02-11 16:16:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/13"}}},"commit":{"parents":[{"commit":"e6da84a094c04f97a37ead1dd193b49fc7dfb489","subject":"Merge \"Support building multiple disk images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6da84a094c04f97a37ead1dd193b49fc7dfb489"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-11 16:16:10.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nPreviously we were seeing issues with ipmitool and virtualbmc on Ubuntu,\nbut this no longer seems to be the case. Example error:\n\n  Error setting Chassis Boot Parameter 5\\nError setting Chassis Boot Parameter 0\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bdfd7e8a2cfbd5a23c97ae5b94d26e4ebaad3add"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bdfd7e8a2cfbd5a23c97ae5b94d26e4ebaad3add"}]},"branch":"refs/heads/master"},"88352a07c6af0fff67b20f72a975ee3f217fba62":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2022-02-11 18:48:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/14"}}},"commit":{"parents":[{"commit":"e6da84a094c04f97a37ead1dd193b49fc7dfb489","subject":"Merge \"Support building multiple disk images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6da84a094c04f97a37ead1dd193b49fc7dfb489"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-11 18:48:44.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nPreviously we were seeing issues with ipmitool and virtualbmc on Ubuntu,\nbut this no longer seems to be the case. Example error:\n\n  Error setting Chassis Boot Parameter 5\\nError setting Chassis Boot Parameter 0\n\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/828901\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88352a07c6af0fff67b20f72a975ee3f217fba62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/88352a07c6af0fff67b20f72a975ee3f217fba62"}]},"branch":"refs/heads/master","description":"Edit commit message"},"01e36d3be2194b45481647461129754f4670d6c5":{"kind":"REWORK","_number":15,"created":"2022-02-11 20:11:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/15"}}},"commit":{"parents":[{"commit":"e6da84a094c04f97a37ead1dd193b49fc7dfb489","subject":"Merge \"Support building multiple disk images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6da84a094c04f97a37ead1dd193b49fc7dfb489"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-11 20:11:40.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nPreviously we were seeing issues with ipmitool and virtualbmc on Ubuntu,\nbut this no longer seems to be the case. Example error:\n\n  Error setting Chassis Boot Parameter 5\\nError setting Chassis Boot Parameter 0\n\nDepends-On: https://review.opendev.org/c/openstack/tenks/+/828901\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01e36d3be2194b45481647461129754f4670d6c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/01e36d3be2194b45481647461129754f4670d6c5"}]},"branch":"refs/heads/master"},"883027afb0656a46369d3f4965a0c65b4a3e8dd8":{"kind":"REWORK","_number":16,"created":"2022-02-14 09:18:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/782273/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/782273/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/782273/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/782273/16"}}},"commit":{"parents":[{"commit":"e6da84a094c04f97a37ead1dd193b49fc7dfb489","subject":"Merge \"Support building multiple disk images\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e6da84a094c04f97a37ead1dd193b49fc7dfb489"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:10:31.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2022-02-14 09:16:59.000000000","tz":0},"subject":"CI: Enable bare metal testing for Ubuntu","message":"CI: Enable bare metal testing for Ubuntu\n\nPreviously we were seeing issues with ipmitool and virtualbmc on Ubuntu:\n\n  Error setting Chassis Boot Parameter 5\\nError setting Chassis Boot Parameter 0\n\nThe dependent change fixes these issues, and this change enables bare\nmetal testing in Ubuntu CI.\n\nDepends-On: https://review.opendev.org/c/openstack/kayobe-config-dev/+/829006\n\nChange-Id: I96827fc32c1594ca9a0535e259929c49d3f0e704\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/883027afb0656a46369d3f4965a0c65b4a3e8dd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/883027afb0656a46369d3f4965a0c65b4a3e8dd8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
