)]}'
{"id":"openstack%2Fkayobe~782276","triplet_id":"openstack%2Fkayobe~master~I8462bf190e74eea9e87f676e856922c360abbb6c","project":"openstack/kayobe","branch":"master","attention_set":{},"removed_from_attention_set":{"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2024-08-20 13:20:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I8462bf190e74eea9e87f676e856922c360abbb6c","subject":"CI: refactor zuul config to use templates","status":"ABANDONED","created":"2021-03-22 15:17:09.000000000","updated":"2024-08-20 13:20:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d9ca54af4c967c04e923fa9dac565ab42933cb22","_number":782276,"virtual_id_number":782276,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-24 12:27:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-22 15:17:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 15:10:41.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-08-27 00:16:36.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"}],"messages":[{"id":"e9e7d4c257ddaa0e6e91841a4d05924ee3b290ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-22 15:17:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"79a24dbff2d3c88295ca223114043a20e253bdfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 15:17:27.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/kayobe on branch master.  The error was:\n\n  Configuration item not recognized.  Each zuul.yaml\n  configuration file must be a list of dictionaries, for\n  example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  The dictionary keys must match one of the configuration item\n  types recognized by zuul (for example, \u0027job\u0027 or \u0027project\u0027).\n  \n  The incorrect values are around:\n  \n  template:\n    name: kayobe-tox-jobs\n    check:\n      jobs:\n  ...","accounts_in_message":[],"_revision_number":1},{"id":"92acbdb594c68dcfa42d0f9f84acc1f42293e587","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-22 16:30:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3aa5d30fa46964772661ee77ca01254554bd96dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-22 18:38:36.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f075386a94274535aa74e0b1c0ee25f7 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bab489c1daf493bb67bad675e4e5502 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/926dba21efde45b2b8c73ac2aed0772f : SUCCESS in 6m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71cdec449b2b4fb5b1f38de712ebf150 : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5039956a8c8e4b5a862489184c379112 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eba3291b68d4131b920a27adb049716 : SUCCESS in 11m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b18634d380847399ef3f78040ab1936 : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f230f1b2edfb422693da5faada5091b4 : SUCCESS in 9m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a75635f31a5f49b8a03e87e78706f367 : SUCCESS in 21m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d3b21787401f43c783c1849bdb85c330 : FAILURE in 1h 01m 28s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecbaa980c6b24dbc9a63f27da8ef9a84 : FAILURE in 40m 08s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/176277b17b9e4e329271a292e78202f3 : FAILURE in 1h 01m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a61761d00cf6487e993fb156de49bd7c : SUCCESS in 13m 32s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bde8880cf46491799c7dea57f3a20a7 : SUCCESS in 14m 00s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b62dcb804fac43fda3c4b7675abfc0ef : FAILURE in 1h 55m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5f8534fee5434493849cf15bc2e784f2 : SUCCESS in 29m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08c85e872eb44b2aa7f57259a584b8b4 : SUCCESS in 26m 42s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6615fa9df6b44337b619979871fb44d3 : SUCCESS in 23m 44s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/cdf11902864648df81d386faacaa5ef7 : SUCCESS in 10m 49s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa6ec7f302a5434aafc14f392669d15f : SUCCESS in 10m 45s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"500607873e32c0ea092582caa0e8a910b6522b1d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-24 10:43:18.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"444960afea77ea58e2ab92aa90b09a2eb9610675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 12:27:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73dd8a5f834e4021b2f159f8e9f8b109 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e39665e43564a61a38ad89635b36a0e : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97f87f952e324f749b09f535bfff3b1e : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/681d2d8fd5334596a3b5765054f488ca : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60e760bd52944ddf8ec090609d4a7462 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3caa2dbe7cd74bbeba0a6b0806c0e144 : SUCCESS in 7m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/17b3262c9c774ecdbe3206295ebebbd4 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5520f7fd90a7437a9fc0ab5def3906aa : SUCCESS in 11m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/03b0eaaf6f1b470388eeb328a4bc0430 : SUCCESS in 20m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/564ac5834ab0408494ea0050191fce19 : SUCCESS in 58m 42s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4890937590cc409591e49f1fb037f221 : SUCCESS in 45m 03s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/5a39cbb5878040bdb25a7bcf0ad838a4 : SUCCESS in 56m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/eee244135ad74f5a856715b753bfd224 : SUCCESS in 16m 10s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00081c03a7494a709a2b80a91a8e7ee1 : SUCCESS in 12m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/05946555ebd84531a16c886c3aae46ee : SUCCESS in 1h 40m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f0a038c0077d492f9d5e30212f2c331e : SUCCESS in 27m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0972f86cbbc84aeb9dfc2a5ac215caa8 : SUCCESS in 23m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8b16fcb8a8454598bd8849d22d8466d2 : SUCCESS in 24m 22s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/8aa30202552a48dcb9da529f7bae5a7a : SUCCESS in 11m 13s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ba5a9d939094951ac9faa41a644582b : SUCCESS in 9m 05s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":2},{"id":"b1591b3fcc4360edc80c348808f8fa19cd06dee0","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-29 15:10:41.000000000","message":"Patch Set 2:\n\nLooks nice, but perhaps it would be better to split templates.yml into each template in a separate file? It would nicely fit in kolla/kolla-ansible, where eyes bleed from just looking in the zuul config ;-)","accounts_in_message":[],"_revision_number":2},{"id":"d9ca54af4c967c04e923fa9dac565ab42933cb22","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2024-08-20 13:20:59.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7e380d26eef46725628648b57994bf4c8a4462f7","revisions":{"23dec8f1ed841fd1fbafe4b3498e354b7d9938c6":{"kind":"REWORK","_number":1,"created":"2021-03-22 15:17:09.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/76/782276/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/782276/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/782276/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/782276/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/782276/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/782276/1"}}},"commit":{"parents":[{"commit":"ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9","subject":"Merge \"Change docker_registry network_mode to host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:16:52.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:16:52.000000000","tz":0},"subject":"CI: refactor zuul config to use templates","message":"CI: refactor zuul config to use templates\n\nChange-Id: I8462bf190e74eea9e87f676e856922c360abbb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23dec8f1ed841fd1fbafe4b3498e354b7d9938c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23dec8f1ed841fd1fbafe4b3498e354b7d9938c6"}]},"branch":"refs/heads/master"},"7e380d26eef46725628648b57994bf4c8a4462f7":{"kind":"REWORK","_number":2,"created":"2021-03-22 16:30:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/76/782276/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/76/782276/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/782276/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/782276/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/76/782276/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/76/782276/2"}}},"commit":{"parents":[{"commit":"ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9","subject":"Merge \"Change docker_registry network_mode to host\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad0800c2cbe6534232786e05fdc4cfea8e2ea5a9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 15:16:52.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-22 16:30:33.000000000","tz":0},"subject":"CI: refactor zuul config to use templates","message":"CI: refactor zuul config to use templates\n\nChange-Id: I8462bf190e74eea9e87f676e856922c360abbb6c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e380d26eef46725628648b57994bf4c8a4462f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7e380d26eef46725628648b57994bf4c8a4462f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
