)]}'
{"id":"openstack%2Fkayobe~782749","triplet_id":"openstack%2Fkayobe~master~Id4588f4529a4522e68e22ce58711cb927fa68a9d","project":"openstack/kayobe","branch":"master","topic":"merge-configs","attention_set":{},"removed_from_attention_set":{"28048":{"account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"last_update":"2023-09-01 16:02:23.000000000","reason":"\u003cGERRIT_ACCOUNT_28048\u003e replied on the change","reason_account":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},"22629":{"account":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"last_update":"2023-09-01 18:12:17.000000000","reason":"Change was submitted"},"14200":{"account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"last_update":"2023-08-27 00:02:48.000000000","reason":"\u003cGERRIT_ACCOUNT_14200\u003e replied on the change","reason_account":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},"14826":{"account":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"last_update":"2023-09-01 18:12:17.000000000","reason":"Change was submitted"},"15197":{"account":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"last_update":"2023-09-01 18:12:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id4588f4529a4522e68e22ce58711cb927fa68a9d","subject":"Use merge_configs and merge_yaml to generate Kolla custom config","status":"MERGED","created":"2021-03-24 14:03:07.000000000","updated":"2023-09-01 18:13:19.000000000","submitted":"2023-09-01 18:12:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":75,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"782749-merge-configs","meta_rev_id":"682cdb8729822ce30157b05d970b77f58f16f8d4","_number":782749,"virtual_id_number":782749,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-01 18:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2023-09-01 16:20:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2023-08-27 00:02:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":2,"date":"2023-05-30 19:17:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2023-09-01 16:20:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-24 15:12:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-24 09:34:46.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-07-12 08:59:11.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2022-12-12 16:19:13.000000000","updated_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"reviewer":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"state":"REVIEWER"},{"updated":"2023-04-19 14:33:55.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2023-05-30 15:45:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"fe29676764195402e2fa686a41c0ee04e618acdd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-24 14:03:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df6506a4640d90b2c0d0ab1e7078bffdcdb38144","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 15:12:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/562e2705eec341848c08b59df760e7df : SUCCESS in 6m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d56ed43d1cc24e67b5e7d0ef3137627b : SUCCESS in 9m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fb9c649cc92d4e78a007d6b697dea363 : FAILURE in 8m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d3bc7ec6933046acb99c83bc159129e6 : SUCCESS in 1h 01m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8da7ecf2430a49428fcf289bf7cd045e : SUCCESS in 26m 54s","accounts_in_message":[],"_revision_number":1},{"id":"82843e40edbc59bab617c91db51800b2b2d3ea96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-24 09:34:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cd9af2b8164d1a62df21e9a62730d0286c7bfd93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 09:47:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/aa72c048c63a40d38f6b917d4aeec4e5 : SUCCESS in 6m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/211fb95e343741c1843dab41c9a62644 : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/50d3d705d64943319033fbb5cd007649 : FAILURE in 7m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9e601636e9654ae49ac65913430d6eb0 : FAILURE in 11m 20s","accounts_in_message":[],"_revision_number":2},{"id":"5e5479293cd9013c9a3f2d017144da8feddc0196","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-24 15:12:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"273495d3480d705e0d65fc0a5fbf3c15e964bb1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 15:34:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0327ffdd50ab4f46aac1d59bbc191c2f : SUCCESS in 6m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/37c613e965ed406f84ee58205f36ee53 : SUCCESS in 9m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e45623fc15734f60849057c29c677abb : SUCCESS in 18m 45s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/faff7e574f0343379f7295c6964d885d : FAILURE in 12m 11s","accounts_in_message":[],"_revision_number":3},{"id":"0a7b2f0a5b44df99df467e72e110468b1c5c7441","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-25 15:33:48.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"aca4d5c6496c1becaad9f10563958349044cbbcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 16:12:20.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8b951eecc73941df8db52aed4243af3e : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d0b9eef2ee9a4670a90a22c3eb45864d : SUCCESS in 11m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8145a0b2027f471999c16c856cbc34b4 : SUCCESS in 32m 23s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/17011385779144408f3d7531f7f773df : FAILURE in 12m 09s","accounts_in_message":[],"_revision_number":4},{"id":"569ce87f86a993bfd4a21c655c3be149f49523d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-05-26 08:56:31.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"81ab9333f0abd75f111c24479d387e57fb091c96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 09:28:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4ad784a9ab7c45a0a719a72a3bfdc6a8 : SUCCESS in 6m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/61e2d334db4f4bc985617329724fa5ff : SUCCESS in 11m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0f95f46a9ba34086b2980b3af996ad7b : SUCCESS in 29m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/cb80857972a148b78d7ad06697cf819d : FAILURE in 16m 07s","accounts_in_message":[],"_revision_number":5},{"id":"5ec4693ac3f4abc4cc7961caa53e3fc9c4a2a946","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-07-08 16:56:06.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"3b36639c03d398ad8e094114233745266030de15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-08 17:53:52.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cb4619f4ea134149aca5758c6e604038 : SUCCESS in 6m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c20f5ea88d4d4b26933f41e17d63b148 : SUCCESS in 12m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6b47cdf906a14fa5b47bdba8df946e0c : SUCCESS in 21m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3c307bde2fc9481b8fc5af3fc8af6d4f : FAILURE in 52m 33s","accounts_in_message":[],"_revision_number":6},{"id":"090f9a38b7ad457f62eeb07036c64ed448018f49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-07-28 18:42:55.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"7f7007a8e2d1c73a990c345f717d1234d0c60a6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-28 22:04:42.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb7060581e88413aadb97053b30c8ba5 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab10af0c1c67440681d70e767919b1f5 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98fafac87903486c87d0f2977be41997 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/34c5f2a28e7e40f892d050a41a50d732 : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ff9633179584c53bd137a9ad298469b : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c482e85cbb534a3bb613dc17d6e44c39 : SUCCESS in 9m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ff2d301b29e74a36991bd4f877e0cb17 : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/149c2519a3ab4de4b52e57a5eca4a3b0 : SUCCESS in 13m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/abe5090a6f744daaa84fb773a580d8ff : FAILURE in 13m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/baeb763385f341fdb37ab82236b6f6c5 : SUCCESS in 1h 06m 23s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/476e31c688734c3e856e948de31a191f : SUCCESS in 48m 35s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/4d64a8b9e2894dc184b66e4008e2e26a : SUCCESS in 1h 11m 53s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/908ad232e5024fb5b30eac1db571e83d : SUCCESS in 20m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9f598f5d215f4ad4982c5e27477e5148 : SUCCESS in 14m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f7fe2c0e6ac04c33b57a6f5ac2859394 : SUCCESS in 1h 57m 56s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ab0effd9bd8d4de9ab506d3520a06788 : SUCCESS in 36m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/169768c250b24fe0a0786bef4a58fa59 : SUCCESS in 33m 05s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0c542c103ebd437bb1d761e05e30bd53 : SUCCESS in 27m 57s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/450926711c6e47cd88f0193585e61781 : SUCCESS in 14m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebdd765da97d487dbdd697cdb9c64125 : SUCCESS in 11m 14s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":7},{"id":"c1a6c386e62f4422749f699deab89039429060bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-07-29 08:43:50.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9bf30a68f50e1323d6ed187311358637283e5449","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-29 10:39:40.000000000","message":"Patch Set 8: Verified+1\n\n(1 comment)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cbb3663b072435e81e67b103d9b3115 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70224cebbec74d9383e875f3d4aff8c9 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/721568ff55fd4e4f971e84afe927b152 : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af36c391a24c484090c09671d5c4c6e2 : SUCCESS in 7m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bac6f5cdbf0400bbc2de702a58cda82 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/daf412c097a64fe3adfb7db5c8eeb90d : SUCCESS in 9m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/78fc833acc9f40cb82fa767fc910a2e8 : SUCCESS in 6m 38s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1e50ba215d574d2c9b2cba3e7a078d12 : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/388fbd82d02c4737b45365637cd9bab6 : SUCCESS in 13m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/56adf17e772a438b862f121e8008f132 : SUCCESS in 1h 02m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4eb4e17b470d421c96455df73d3f6927 : SUCCESS in 48m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/5df427b8807446d4ac66b57ab7d5a9e1 : SUCCESS in 1h 07m 01s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/c41aa0164b1b4d33830f222770d02964 : SUCCESS in 19m 20s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/267ab23425674a8a99ae39e253d97f45 : SUCCESS in 13m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/ec607fcdec2b4e8090209dd05c696a1b : SUCCESS in 1h 52m 35s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c0397424ec954f579ccb5ded40692e2f : SUCCESS in 36m 52s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/31da8cae50084405950215e41866a1bd : SUCCESS in 29m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9af8057cf5144c0d919b229d97172de2 : SUCCESS in 27m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/048a17bdc02b4136a4428ce38767c189 : SUCCESS in 13m 55s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b633a7b3a754c48b5b1540255a74746 : SUCCESS in 9m 24s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":8},{"id":"93b723ee72b51535830a690d26202e0ce6db2ccd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-10 11:28:47.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"264f63150ab46ff315f87e708fb86e376e7b5b82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-10 13:29:19.000000000","message":"Patch Set 9: Verified+1\n\n(1 comment)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf49adbb80164051aeb1a021f7d4bc26 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0a0b22138bf47d5a616ddee0402dedf : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8a6caced44b045a4972d6cec47b49ce7 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bdebdbfe5e544832a89f7993619e42b1 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18279a0e1a1d4e78bd40160c0d23f176 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eea71efa528e442d95f7afbdcfcce3ec : SUCCESS in 8m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/972efc19a5a94719b0ac7e95d685d492 : SUCCESS in 7m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/410fca9d32eb4659a76c482f766263e3 : SUCCESS in 14m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1cedf5c2054d4ddf9d98e584e52f20fa : SUCCESS in 13m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/aa590a74ce0e4570bb75b17a37047a6c : SUCCESS in 1h 07m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ce549b3866a4c9dafade52479e65dc8 : SUCCESS in 45m 24s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/21ae6396c84c465b8f8acdbdaf521564 : SUCCESS in 1h 01m 14s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/849b9844382c49d8a37ea1ed5b99aeb0 : SUCCESS in 17m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e2a18014e814f62898bf02aab2e67c5 : SUCCESS in 13m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/91860d94362a4326b6f75f1edb8266ce : SUCCESS in 1h 56m 10s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ffc6111b0570466c8d6ebf93c6a638a8 : SUCCESS in 31m 16s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/764b593914624778a37f6079cedb07d7 : SUCCESS in 31m 07s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c819d7d0c0bf43f590217de6db66339b : SUCCESS in 28m 10s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/fcc8dec41721438aae17b02bde5c4b55 : SUCCESS in 12m 09s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a656f16898304bbfa6059214d117a7f0 : SUCCESS in 11m 24s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":9},{"id":"c6fbeade78a9beb71889da20426fd6c122193328","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-08-10 14:45:41.000000000","message":"Patch Set 9: Code-Review-1\n\n(19 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a6799e968e4ce2deedc54976d485c74ef0630c60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-09-15 08:41:50.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"264c7ff15519fc993362085357bbf1fcc510ad2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 08:42:15.000000000","message":"Patch Set 10: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kayobe for 782749,10","accounts_in_message":[],"_revision_number":10},{"id":"53fc2723ed7db3a6b0cf53036c6bdbfecf21df3e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-09-15 08:42:46.000000000","message":"Patch Set 10:\n\n(16 comments)","accounts_in_message":[],"_revision_number":10},{"id":"16655575e5f670e58a7daca1a478547b1c8f81a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-09-15 09:20:36.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"790ea0369c7ceb6889560c30607e206327e92aca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 09:20:54.000000000","message":"Patch Set 11: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/kayobe for 782749,11","accounts_in_message":[],"_revision_number":11},{"id":"e1583aee3861f3eb00cf5ae0e3751988f5997e62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-09-15 09:22:58.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"80d7ef53ff4f267ae3e6723317c3f126a0516fe0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-09-15 09:33:55.000000000","message":"Patch Set 12:\n\n(7 comments)","accounts_in_message":[],"_revision_number":12},{"id":"78b4b2d6a2e8180c3757a2161798662ea9964285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-15 10:37:56.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41e6cb5c95ea49c5aa8ae4c1bd9ee614\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a1a9827c6694ce680fe961c6b57c82b : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90b4ea51f3854020b57426da51997b6e : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04716d9db07c44288e83aa471042cebb : SUCCESS in 7m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74202fb16bd5481aab195fc050ac6426 : SUCCESS in 7m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78a47deab1b94d68a5a358d9ff3e7c35 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b23912e0f3b046f1a63d1ddb7ca675aa : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef82e706408f4efb941b800bdc8667cc : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/444da9b3c5424f0f858edeeeba021788 : SUCCESS in 6m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a91c79b22fa49ef8747c382a2bc653e : SUCCESS in 14m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f91c845bf32741c18d50513edc5e8300 : SUCCESS in 13m 51s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61f22adea7564a2e9891ab95f18c1a46 : SUCCESS in 1h 01m 33s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27d3827ed16847ad8ece059dc8fb3af6 : SUCCESS in 20m 46s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c31d2dd1d254fabba9744ec8218bbc7 : SUCCESS in 1h 09m 02s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de2714f93dbf4008955b74c15a8fb6d0 : SUCCESS in 16m 52s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0391e28e93854337a6b409948ffb9822 : SUCCESS in 28m 41s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/873bbcfa0fd544fd873ce683adfd3e60 : SUCCESS in 12m 21s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d91a0483b64f4352b605a9635f571dea : SUCCESS in 13m 42s","accounts_in_message":[],"_revision_number":12},{"id":"7401f2f41177d0061b4a0b51a831d23ec0037d65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2022-12-12 17:05:01.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"99c707c92a0c46a93c51f9b5584632820c945f83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 18:11:52.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d40c1bc04fdb49b0b0a111ade67bb388\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/488a42af688a492fa3433c1c157b1eb2 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0f04fd11ebf4b62978a1c6a6bfc524b : SUCCESS in 6m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/edbcddbb33534fb19aec9cafcc2270de : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b04961c3ced9496ab23aa79763673e12 : SUCCESS in 6m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/22ab2a80d1c14519a799259ef6c2b8c2 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/664dcf1e6ccb45a1ab81c8f4d16eb946 : SUCCESS in 10m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cbd4ccb261f34fb99294984b7ba3fbf7 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/410bf8ee6f064d259f2d4c6b521b19a0 : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fbf1a6ae05a5421fb941d8680cdb6105 : SUCCESS in 16m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9a87df52afb04cfcb32edcfd8c8694f5 : SUCCESS in 14m 00s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5d179488903c4907b54f71b996b5abd9 : SUCCESS in 48m 37s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9d2f57ef1b574bd1b7a0cdb4d42542dc : SUCCESS in 17m 56s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ecf9d72e4744443883bbcec38e442249 : SUCCESS in 1h 04m 56s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1a342cd0caa947d79dfb32254110a6d7 : SUCCESS in 17m 41s\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/deb6ad0751e54f51a36acc72f4d084a3 : SUCCESS in 29m 22s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3e194bfce41841a8905460fa6684dc0d : SUCCESS in 12m 16s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/355571c5cabe405a898e1766c21802cd : SUCCESS in 14m 46s","accounts_in_message":[],"_revision_number":13},{"id":"59c37f61a056f13561a4771c8b29a87de350ddba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-01-11 13:56:26.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"009cc4bbd3f58835c5d7937ed4c64c5353cd3b31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 15:14:15.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f443bf305094967bc77eb594782c8e1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18ba7abb4e1f4dff8cbe6006aaee447b : FAILURE in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b34314176e448c28ab65dfee510ab8a : FAILURE in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ffa86714f874f899887593513534074 : FAILURE in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02536ddf56b6402280b4fbb84cd5ef54 : FAILURE in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8231181e030417688d8240c92a611ac : FAILURE in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/28cda344f59946e7ad88f95a1e3b5ae4 : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2be079115f0b437f8057ea1dcaebc318 : FAILURE in 3m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a2c9c1081b24ab7bfdd1f67e7b78947 : FAILURE in 3m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/50244f0c1ca44206a7a90939eef52a2d : FAILURE in 3m 51s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/d5a02f0ba9364fba930950f572838112 : SUCCESS in 57m 41s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/53d2bdaa82a54e8695c08c839e77eaf8 : SUCCESS in 57m 28s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3523f42782214c349659f4b444b8c893 : SUCCESS in 49m 36s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/4586e0d9725f4213a851d51696955651 : FAILURE in 19m 24s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/5c17bd106ec94ea99e60540caa777188 : SUCCESS in 15m 17s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/1e558a338bb64eceab9f358f504d66e3 : SUCCESS in 19m 58s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc756359188949319d66ba5338e6bea5 : SUCCESS in 12m 44s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5affe19e9da2455f990a6af3636b132c : SUCCESS in 1h 06m 07s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/37d5996b5abb4157b48f35fbd12f08be : SUCCESS in 17m 30s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/26f704ecc1b0428689bfb1882a2edc5b : SUCCESS in 13m 59s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/f7daa09cf77848d2a562cb67b9efa82d : FAILURE in 24m 14s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/87d3f4f5bbd444f4a0705ee5f0e35dcf : SUCCESS in 37m 17s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/8412e528db7e47cabdd32469d95d7de1 : SUCCESS in 11m 54s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da5f50893d574823a759b662dee0a13e : SUCCESS in 9m 50s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/487d9661469d4189ab40fd3ce160ccf4 : SUCCESS in 17m 10s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7c85042815ce4c19ba5b7cc33d2e0fad : SUCCESS in 13m 03s","accounts_in_message":[],"_revision_number":14},{"id":"90651ef2eee8dbfc22e35b352dfbce8f46898ee1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-01-11 17:27:51.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"8db506985c25f47566c750c3d95be54fcd49bdc6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-01-11 17:35:19.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"aba4683a8b463cd4319b3e99d5b3dc9b6e465ce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-11 18:36:57.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/286e599202884a92b1cc7da4087e4444\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bafeaff4909f45599ed175abc7e62406 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa270be556914f4b9b6d6adb77306377 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98803d88c395438eba2f593fbb756520 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e281409dffea4b44beffc3961f10e986 : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5310d6688a504f3abc0e415f2b36cec3 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c4ea51c78d145d787e1ac5c86e0829e : SUCCESS in 3m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d28f053601ef4b7696ddadf352c51023 : SUCCESS in 3m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4273c74840444d9c99168ba2772fe1ee : SUCCESS in 8m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b7b726089ca14e818caa1e750e290404 : SUCCESS in 9m 41s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/64f2923d49674a22a6e166be9436492c : SUCCESS in 43m 25s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/a9d744bcb40340869c243ae792f5cd77 : SUCCESS in 50m 11s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/72deb285177f49f682dbca406b7a90c4 : SUCCESS in 45m 02s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/026ddcd3520247c99155e47130bcc188 : SUCCESS in 52m 47s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/c39c46bd956246f1b0f2307f10b60394 : SUCCESS in 14m 08s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/7d1e95b77fe8458eadd025a15963ec93 : SUCCESS in 15m 07s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/540f496683b14852bca858c7f9cd7d55 : SUCCESS in 12m 20s\n- kayobe-overcloud-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9749db486bc4c438647ca69d0cb6d1a : SUCCESS in 59m 52s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/30abbd120d0842a3968f8d40bbe928f3 : SUCCESS in 13m 12s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b22334f9f0b4c4082a524e9ebceecef : SUCCESS in 12m 24s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/071a94b8425d468688c0c9b73703fb9f : FAILURE in 21m 10s (non-voting)\n- kayobe-seed-upgrade-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fa3851e4b47e43aa9e1c952006bcc5b4 : SUCCESS in 22m 57s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/eda7284f07034bb7aab9adc1e1f07409 : SUCCESS in 10m 35s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc99ddba9618430988a7f56ed7aad2e5 : SUCCESS in 10m 48s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c39ae8d9d688446691dd92f9be490085 : SUCCESS in 13m 39s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3acf951a05ff404aaa12bf0ff89c6f7e : SUCCESS in 12m 54s","accounts_in_message":[],"_revision_number":16},{"id":"c37b2471a430ce1def21eee2ece02263fe3011fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-03-07 10:22:48.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"99c2bdc59ef448772e97686307621777f349465f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-03-07 10:23:37.000000000","message":"Patch Set 17:\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"e7ad75e005f54993269fd69cba94bfee25adc213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-03-07 10:29:27.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"b424a902586a01e6fb59b701d1855fd83514a71f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-07 12:10:07.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e3d3f11d5be47569c30e6207bcb15af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38acab8765574be29df71d2f91925c95 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a842326ece3b4587b4f59f20900c2501 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac206df1085e407bbd2ed7180690f468 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20576db0700c4db1993fd8034d3e3939 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce7482ffa9f246f28132340e1d67ed16 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/652558b8345a49e5b56721044844966d : SUCCESS in 3m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6da749e20d354680aee530a9281a5b8e : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/527d1258cc854b1b9a7683027c28e58d : SUCCESS in 11m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d19305aaf9394956a1fbf0ba7b0018b1 : SUCCESS in 11m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/351dfa21af2b43d49431e682cdc25ee6 : SUCCESS in 55m 10s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/57c914c460ee49f1bb7239b189b757c3 : SUCCESS in 59m 51s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a22e036648845a1bdb20962263adef0 : SUCCESS in 49m 08s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/fb09963319264f44ac7c653551f54155 : SUCCESS in 57m 20s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/f2525a34b96449b18c25ad522e75b5ba : SUCCESS in 17m 05s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/851b13f4abcd4409a487a544fb11a918 : SUCCESS in 17m 24s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4ff911c151fe4f50aafb808c0e7ff133 : SUCCESS in 14m 33s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/49542c063b8f43ce839de506bc95ac0d : SUCCESS in 1h 37m 24s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6e21511f3db24e659cf072eb9efbeba5 : SUCCESS in 1h 03m 23s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/e5fd34a8e91a425b968e390f63e55fee : SUCCESS in 17m 52s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d055e7b2e2054e748e697a56d911b124 : SUCCESS in 13m 45s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/8bfb86f2e823427d8e873fa09fc86427 : SUCCESS in 30m 03s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/24a6a0dc713846e1b6adc74d62ebc775 : SUCCESS in 24m 48s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a48b6ecbbf7b42d187fcc2f6ab8f4d41 : SUCCESS in 20m 56s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1b152c8be69f48d08aac787e10c6c40f : SUCCESS in 11m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fac75789a44d42bf86239916ed80975c : SUCCESS in 10m 41s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/a867f81c9bdb40dba9cc1efc3e1d2985 : SUCCESS in 14m 34s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/13e6f652c0e84a7797870e6fb7ac0f29 : SUCCESS in 12m 41s","accounts_in_message":[],"_revision_number":18},{"id":"736f17d7e0919b463ddc166d83f989cb072deb19","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-19 14:07:43.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"5b71aac0adff3614ef70c5db5c016d215f4e9b06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-19 14:09:33.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"8740e89dc9f506f7f8639d90e0bc420a26c71fa1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-19 14:10:26.000000000","message":"Patch Set 20: Code-Review-1\n\n(10 comments)","accounts_in_message":[],"_revision_number":20},{"id":"554e600a7fe5c4b51d252d1bc41dd4d7b4627850","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-19 14:11:14.000000000","message":"Patch Set 20: Code-Review+2\n\n(8 comments)","accounts_in_message":[],"_revision_number":20},{"id":"4d7d528fb4d1cb2c07628620ce985046f23f06fd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-19 14:14:39.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"629ffe84565090944c52549f88d63fee1a443e82","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2023-04-19 14:33:55.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"f11b6f0518ca1954766e5eb1c252e5c0bcfa33bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-19 15:40:15.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":21},{"id":"d594b6f14b9e59c2a064b261e4d770fef37fe5af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-19 17:15:23.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4804f3c36ab419e81264cf10375c2fe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8109c61e38854d67a6d20ddab7c9b6fb : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/becdfed06e06491abb9e1046f4d3a0d8 : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/030997c9ce094486ac206547becf7662 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/214107bb0619463b8aaafbf5c1cc98f6 : SUCCESS in 3m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5fb4b03dfe94914ad460ab62ece05eb : SUCCESS in 6m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eda3c980d1354d73b18aba772eaebdb1 : SUCCESS in 3m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2c7caeef1ea64181bd367e1494e7437b : SUCCESS in 3m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e6c19b34e5a941759ba90b13a8e0f68b : SUCCESS in 10m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7fe1d7f3f1f14e54a4cff713034c14fb : SUCCESS in 11m 35s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/2a70732c51634f47b3c9f2df6fa808de : SUCCESS in 1h 02m 39s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/6a16a538350f49d0b23ee6ade9c7f2fa : SUCCESS in 1h 04m 28s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09077bdd6ca8410a9b2ce4d0682529a2 : SUCCESS in 49m 05s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e73f15eeda0f4134bb8b955151b21210 : SUCCESS in 1h 03m 54s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/707b7549617d46f99cb3c79bd9bfd77e : SUCCESS in 18m 45s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/656311d99d0b41088dedf37d1e309707 : SUCCESS in 15m 56s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c615b716a8154602a9e493c9770c6a11 : SUCCESS in 13m 49s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/b67589eacd0a4b03965dc41f6598de64 : SUCCESS in 1h 32m 40s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14fbe150b28b4401bc1886774c88a9be : SUCCESS in 1h 07m 45s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/48b0430753344c0f950e25cd01fcb9a0 : SUCCESS in 17m 38s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/54f936fda1a947808081c8bb6eb6f263 : SUCCESS in 13m 10s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/7f02ac6197eb4939bbec88517ee90b2d : SUCCESS in 30m 43s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/55a61fce24914a8ba2c98a17e9e6e909 : SUCCESS in 25m 46s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dc8146a7fb4e4b6ab9ea461e8ed5eba4 : SUCCESS in 22m 31s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/75413a47621d4442904fd96a61b1dfba : SUCCESS in 13m 31s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a32d86006d294c83a511281fc0e6799c : SUCCESS in 10m 05s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/d5bf812bd22a4070b9c64d16dc560785 : SUCCESS in 18m 15s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e3dd50d386584fbaa9930734538d0c88 : SUCCESS in 14m 07s","accounts_in_message":[],"_revision_number":21},{"id":"852eda9f041e3ff4efcb59f23cd91415f6df3b23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-20 09:19:43.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"33990bc0e24569ced5f05f0bbb102074c1acc581","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-04-20 09:19:55.000000000","message":"Patch Set 22: Code-Review+2","accounts_in_message":[],"_revision_number":22},{"id":"eb7406383d08ab0d15cab8acf3a3cd9ef10ea449","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-04-20 09:28:59.000000000","message":"Patch Set 22: Code-Review+1","accounts_in_message":[],"_revision_number":22},{"id":"6bfd7d2075581995cf1ae161cad2a8a7b39aa1bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-20 11:18:06.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad4e9562d3a4444292f8bcbdec5801d0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd6ece23d2d14a148c695606be226f6a : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4aaf23c5599142de89c2428700e2e78c : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ae46d304aff4a58bd0559dfdd4cb3e3 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0cf887d94732439e85e97ac918f1458b : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67f6c3dfa86f4cb6a6e2456d4524834e : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc829f7cd92b4f10bdf5d934cb796346 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f09f9f8a289449739b06eecd3c466556 : SUCCESS in 5m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e09495a13c134086bc44d0b380699cc3 : SUCCESS in 11m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1311f63d192c4957b48640aa6fb4483b : SUCCESS in 10m 20s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/df86a6bd2c564b02bd0e062916d289b0 : SUCCESS in 57m 15s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/1f92018de3604bf2981607e0195d0f7d : SUCCESS in 54m 21s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bac73261abbb48b088e62beeb7e0e48b : SUCCESS in 52m 18s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/d7ddbbae208b45ae8d120655e4472e4c : SUCCESS in 56m 55s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/1a426069e40649d29a1853431ea5bbae : SUCCESS in 19m 09s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/0b9324841c1b45db80dc61459220c51c : SUCCESS in 16m 27s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9b61e793b78b43708962879d11f60e22 : SUCCESS in 13m 51s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/ee3ea19866a7482f81446098a6d58f12 : SUCCESS in 1h 47m 44s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6c47b2482694bb1a6bad02d02abe5d1 : SUCCESS in 1h 08m 44s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/341ade7ac41240ee9bbbbc52d94b1976 : SUCCESS in 15m 08s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c428eae5787d4707894622e87c2bb576 : SUCCESS in 13m 22s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/0d0012f336b74d3fb47987f45173a13b : FAILURE in 25m 53s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/cd7c5edd85d74c17bcf0d539b9a18096 : SUCCESS in 24m 06s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/64cab257d1c84903b762aea56b2b6461 : SUCCESS in 20m 04s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/c933cc5bd6b3455aa4ebe3f32d3652fd : SUCCESS in 13m 08s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a682c44482244a858ff61e81c1858601 : SUCCESS in 19m 30s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/9b472ca9955348639f8eb95ecd5d44e4 : SUCCESS in 17m 06s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a5b500679f8444c09d97f175eb89baa8 : SUCCESS in 16m 50s","accounts_in_message":[],"_revision_number":22},{"id":"945f57e609ed711ff484470f352aaec81612304e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-05-30 15:45:06.000000000","message":"Patch Set 22:\n\n(4 comments)","accounts_in_message":[],"_revision_number":22},{"id":"13a7b5009dd61a520ba5bacbc3d4827b81e802fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-05-30 15:52:47.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"001b5a5d8fcdb1e52cb7bc0bb74e88475a04830a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-05-30 15:54:17.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"ddbc86dab7c3d8ec7da7b55c909e9e88fd9058d9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-05-30 15:54:20.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"aaac276281f9f2b3f336e4904eb6343829a4215d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2023-05-30 16:47:14.000000000","message":"Patch Set 25: Patch Set 24 was rebased","accounts_in_message":[],"_revision_number":25},{"id":"5a9c583804adcae1f5c1226f7998c17fc56e62ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-30 18:36:00.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/736c25df297a4bfb9af7435a40de3c0a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/776a5c78c17e43e1bd5a1d068efc6b7f : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d67ff4958e0e4c419df1d93812066952 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2a76b8c09d3d47318e0ae88be956fd40 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46e8311dfc844f00bc06fdab9f593c5b : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a9724fe6a9a4755b5e23063608e3165 : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ed3e829951346fabfe444e9fc9f3d1f : SUCCESS in 7m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/91d034017099404999787526a789dc08 : SUCCESS in 3m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3d872450b0a1492da553565a4d9af58e : SUCCESS in 4m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ee8600122ae64c52814e494ade63c18d : SUCCESS in 11m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/383f4d114c7644f1922a6bb9af8ef837 : SUCCESS in 16m 50s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/ddb1b20f5e9e44d484fb39ae4481f2e7 : SUCCESS in 55m 09s\n- kayobe-overcloud-centos9s https://zuul.opendev.org/t/openstack/build/46f77ce0d4494a588608f7de4d9b7ed9 : SUCCESS in 53m 17s (non-voting)\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c975dc0209d1442cb4bdc937ae0cf54c : SUCCESS in 1h 00m 16s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/7953970410a84eaab0f0c959ee741fee : SUCCESS in 59m 37s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/edb7ef57c8e24ec88d69ee714e6bce61 : SUCCESS in 16m 22s\n- kayobe-overcloud-host-configure-centos9s https://zuul.opendev.org/t/openstack/build/ee4f29b91e4248cc95900afcfbeef9b0 : SUCCESS in 17m 26s (non-voting)\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bd3c6dca1da5412d84cea38640ed6d51 : SUCCESS in 12m 53s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/da9c8aff1ef9421099e03ea22c242cea : SUCCESS in 1h 40m 16s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92ea10fd96724b01b1903ef722159336 : SUCCESS in 1h 31m 06s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/13b3e3beb233497b90be72aebf4fc318 : SUCCESS in 16m 22s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/56fc5d2dc08c4d31b97d8ce80fddc91a : SUCCESS in 16m 55s\n- kayobe-seed-images-rocky9 https://zuul.opendev.org/t/openstack/build/d50c5fba2c8845ae87c2dab91a54ea0f : FAILURE in 38m 37s (non-voting)\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/66b912a9a35c4c098886b431fb3b4817 : SUCCESS in 28m 47s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8e2fd6cd7c2542108ebaf84152207701 : SUCCESS in 28m 47s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/e5348f041a1342649c15cbd4b0593ccb : SUCCESS in 14m 19s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/731cef0c00164e3d9e61c103fc4e1b14 : SUCCESS in 12m 29s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/4907220943054e87a345beabd5dbcba9 : SUCCESS in 33m 29s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/27e3cf1142ff445fb4b3ad6ce32f51e6 : SUCCESS in 34m 08s","accounts_in_message":[],"_revision_number":25},{"id":"e108e6ae16e213884f53ee95156fdf2b8c17c0de","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2023-05-30 19:17:57.000000000","message":"Patch Set 25: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"2982f32a4b0243932eac5af3ad92d6ba80c7a8dc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2023-08-27 00:02:48.000000000","message":"Patch Set 25: Code-Review+1","accounts_in_message":[],"_revision_number":25},{"id":"f6fc5d1e6cbd581403b53b31f92fcbc8a567d2b8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-09-01 16:02:23.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"5e22858185cd0f2ace363bfb289dcec58d0ba200","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2023-09-01 16:20:44.000000000","message":"Patch Set 25: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":25},{"id":"779a6321334b41d24ca0f46a947a7abb526800cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 16:21:08.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"46448c091f04fe014dc49973827aefa29cfceef1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 18:12:17.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f679bdf40b074701b8acf40d8b15eba9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52eb0621f8d54517800c5debc4da27b1 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ae69ca38e0c43c597e3d13989d72bc5 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd0b8da1356d4c979d5af1516f92fef2 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/43a76fcd801242dfb3238046c073b455 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c6dba61f08f48d1918fd2c3030bc174 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15b7816d48b443c78c1f6c7ae7043539 : SUCCESS in 3m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/069a9f5375c245398b452f8015a7d8c0 : SUCCESS in 4m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bce38c91b7e849d0bae9af534ee744ee : SUCCESS in 11m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e63f3438331d4e96a95a540c9356c410 : SUCCESS in 10m 05s\n- kayobe-overcloud-rocky9 https://zuul.opendev.org/t/openstack/build/020067ba53af46a1b53a9fb3507b66cb : SUCCESS in 1h 04m 59s\n- kayobe-overcloud-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/37d73596377b4fc490434ed6d6fcef68 : SUCCESS in 44m 17s\n- kayobe-overcloud-tls-rocky9 https://zuul.opendev.org/t/openstack/build/e150d9dad4dd49e9ac64d3a151cdbca1 : SUCCESS in 1h 00m 56s\n- kayobe-overcloud-host-configure-rocky9 https://zuul.opendev.org/t/openstack/build/dd940b20fce2421a9861bccd53da9cb5 : SUCCESS in 18m 37s\n- kayobe-overcloud-host-configure-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2cc7526de0f54925973214e0e2531470 : SUCCESS in 15m 19s\n- kayobe-overcloud-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/f3e2fecf8cf44601acf6cc3b52cedac9 : SUCCESS in 1h 43m 06s\n- kayobe-overcloud-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d5874ed67a2d43b3a61bfba2275ad8ce : SUCCESS in 1h 45m 09s\n- kayobe-seed-rocky9 https://zuul.opendev.org/t/openstack/build/b9cc532f4711445487b3465dd875c670 : SUCCESS in 17m 37s\n- kayobe-seed-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4aec5f6b1f764a21bac8eb3a7b77d4c7 : SUCCESS in 15m 01s\n- kayobe-seed-upgrade-rocky9 https://zuul.opendev.org/t/openstack/build/3f7e94b5c2af4080908f26d66c05ed88 : SUCCESS in 26m 47s\n- kayobe-seed-upgrade-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9077375375144dc991fbddeeae92586 : SUCCESS in 22m 38s\n- kayobe-seed-vm-rocky9 https://zuul.opendev.org/t/openstack/build/0e82d490424942c28cb7cdb9d8477b24 : SUCCESS in 15m 00s\n- kayobe-seed-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4b8a504ec601493c8fbfaf648816409e : SUCCESS in 11m 13s\n- kayobe-infra-vm-rocky9 https://zuul.opendev.org/t/openstack/build/1c323ad5c32d4d39b9d89cdea7eed580 : SUCCESS in 27m 24s\n- kayobe-infra-vm-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/698bb185be3641a8b72151980d4f8b4c : SUCCESS in 21m 35s","accounts_in_message":[],"_revision_number":25},{"id":"8080f8dee2f3db786cb25dc0a4dc207d81865786","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 18:12:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":25},{"id":"682cdb8729822ce30157b05d970b77f58f16f8d4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-01 18:13:19.000000000","message":"Patch Set 25:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bd54298921947c28854d566ef55ba8f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3b3bf16b10b414fab2a55705ce9a354 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13efa92e21a54e198504ca25ef2fc9f8 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"428ef10fa4d4a281f0c690876fc3cbfff41c2bbb","revisions":{"1ce804e9f1e09fd9780ff9566a2590e5525faf64":{"kind":"REWORK","_number":1,"created":"2021-03-24 14:03:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/49/782749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/1"}}},"commit":{"parents":[{"commit":"99b795d6816c268f45c0772b822e6fc894ac7aa4","subject":"Import merge_configs and merge_yaml from Kolla Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/99b795d6816c268f45c0772b822e6fc894ac7aa4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-24 13:56:46.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-24 14:02:54.000000000","tz":0},"subject":"WIP: Use merge_configs and merge_yaml to generate custom config","message":"WIP: Use merge_configs and merge_yaml to generate custom config\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ce804e9f1e09fd9780ff9566a2590e5525faf64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ce804e9f1e09fd9780ff9566a2590e5525faf64"}]},"branch":"refs/heads/master"},"a2f1f941918a682382bb4f74b4c80140a0125db7":{"kind":"REWORK","_number":2,"created":"2021-05-24 09:34:46.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/2"}}},"commit":{"parents":[{"commit":"4ba5123aea50ce364f44af526d3e70b701dd6711","subject":"Import merge_configs and merge_yaml from Kolla Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ba5123aea50ce364f44af526d3e70b701dd6711"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-24 09:12:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-24 09:12:55.000000000","tz":60},"subject":"WIP: Use merge_configs and merge_yaml to generate custom config","message":"WIP: Use merge_configs and merge_yaml to generate custom config\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2f1f941918a682382bb4f74b4c80140a0125db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a2f1f941918a682382bb4f74b4c80140a0125db7"}]},"branch":"refs/heads/master"},"f2f2fcb9a3212604d564d94c9130cd47bb287800":{"kind":"REWORK","_number":3,"created":"2021-05-24 15:12:03.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/3"}}},"commit":{"parents":[{"commit":"aa318516ea3b2b85f4c98ef8fead8e00c0857a95","subject":"Import merge_configs and merge_yaml from Kolla Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa318516ea3b2b85f4c98ef8fead8e00c0857a95"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-24 09:12:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-24 15:09:53.000000000","tz":60},"subject":"WIP: Use merge_configs and merge_yaml to generate custom config","message":"WIP: Use merge_configs and merge_yaml to generate custom config\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2f2fcb9a3212604d564d94c9130cd47bb287800"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2f2fcb9a3212604d564d94c9130cd47bb287800"}]},"branch":"refs/heads/master"},"50b94fa903f604a43b1f75da43a4519aefef097a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-05-25 15:33:48.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/4"}}},"commit":{"parents":[{"commit":"8075e43e950655f343c824a14c8c71e776590994","subject":"Import merge_configs and merge_yaml from Kolla Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8075e43e950655f343c824a14c8c71e776590994"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-24 09:12:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-25 15:32:23.000000000","tz":60},"subject":"WIP: Use merge_configs and merge_yaml to generate custom config","message":"WIP: Use merge_configs and merge_yaml to generate custom config\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50b94fa903f604a43b1f75da43a4519aefef097a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/50b94fa903f604a43b1f75da43a4519aefef097a"}]},"branch":"refs/heads/master"},"901b426a88aa130b6aa013204fcd58018d4e306d":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-05-26 08:56:31.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/5"}}},"commit":{"parents":[{"commit":"c09358f3143384a47885cb987b75d7409abd58b8","subject":"Import merge_configs and merge_yaml from Kolla Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c09358f3143384a47885cb987b75d7409abd58b8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-24 09:12:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-05-26 08:55:11.000000000","tz":60},"subject":"WIP: Use merge_configs and merge_yaml to generate custom config","message":"WIP: Use merge_configs and merge_yaml to generate custom config\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/901b426a88aa130b6aa013204fcd58018d4e306d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/901b426a88aa130b6aa013204fcd58018d4e306d"}]},"branch":"refs/heads/master"},"40968f40e936d5d2fa459526714c05e697b5a7b7":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-07-08 16:56:06.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/6"}}},"commit":{"parents":[{"commit":"96191f5e7f7412e4168e132d74409084dac782d6","subject":"Import merge_configs and merge_yaml from Kolla Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96191f5e7f7412e4168e132d74409084dac782d6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-24 09:12:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-07-08 16:55:52.000000000","tz":60},"subject":"WIP: Use merge_configs and merge_yaml to generate custom config","message":"WIP: Use merge_configs and merge_yaml to generate custom config\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40968f40e936d5d2fa459526714c05e697b5a7b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/40968f40e936d5d2fa459526714c05e697b5a7b7"}]},"branch":"refs/heads/master"},"7957a3f79df6a804dec9a1511e312779aae89e37":{"kind":"REWORK","_number":7,"created":"2021-07-28 18:42:55.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/7"}}},"commit":{"parents":[{"commit":"96191f5e7f7412e4168e132d74409084dac782d6","subject":"Import merge_configs and merge_yaml from Kolla Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96191f5e7f7412e4168e132d74409084dac782d6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-24 09:12:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-07-28 18:42:27.000000000","tz":0},"subject":"Use merge_configs and merge_yaml to generate custom config","message":"Use merge_configs and merge_yaml to generate custom config\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7957a3f79df6a804dec9a1511e312779aae89e37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7957a3f79df6a804dec9a1511e312779aae89e37"}]},"branch":"refs/heads/master"},"425852a7f6a79bcd19a8b7c4d0dcb0e67c9e28cd":{"kind":"REWORK","_number":8,"created":"2021-07-29 08:43:50.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/8"}}},"commit":{"parents":[{"commit":"96191f5e7f7412e4168e132d74409084dac782d6","subject":"Import merge_configs and merge_yaml from Kolla Ansible","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/96191f5e7f7412e4168e132d74409084dac782d6"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-05-24 09:12:55.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2021-07-29 08:43:18.000000000","tz":0},"subject":"Use merge_configs and merge_yaml to generate custom config","message":"Use merge_configs and merge_yaml to generate custom config\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/425852a7f6a79bcd19a8b7c4d0dcb0e67c9e28cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/425852a7f6a79bcd19a8b7c4d0dcb0e67c9e28cd"}]},"branch":"refs/heads/master"},"2abee026c1e44e89d4b3f9370aa0fa4583c06c94":{"kind":"REWORK","_number":9,"created":"2021-08-10 11:28:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/49/782749/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/9"}}},"commit":{"parents":[{"commit":"2545155a8679025f488979d54212bef07b6b4eb5","subject":"Use merge_configs and merge_yaml to generate Kolla configs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2545155a8679025f488979d54212bef07b6b4eb5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 11:11:03.000000000","tz":60},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2abee026c1e44e89d4b3f9370aa0fa4583c06c94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2abee026c1e44e89d4b3f9370aa0fa4583c06c94"}]},"branch":"refs/heads/master"},"747290f0a4ae1030c271efa424b1a83158b881d7":{"kind":"REWORK","_number":10,"created":"2022-09-15 08:41:50.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/10"}}},"commit":{"parents":[{"commit":"6e2f5a1884745dbb596ce6f127e7eb875ab63cdc","subject":"Merge \"Fix IP allocation limit in overcloud inventory discover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2f5a1884745dbb596ce6f127e7eb875ab63cdc"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-09-15 08:40:02.000000000","tz":60},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/747290f0a4ae1030c271efa424b1a83158b881d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/747290f0a4ae1030c271efa424b1a83158b881d7"}]},"branch":"refs/heads/master"},"98bb887fb3161193e52e8e8612cde7a4fc95666f":{"kind":"REWORK","_number":11,"created":"2022-09-15 09:20:36.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/11"}}},"commit":{"parents":[{"commit":"6e2f5a1884745dbb596ce6f127e7eb875ab63cdc","subject":"Merge \"Fix IP allocation limit in overcloud inventory discover\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e2f5a1884745dbb596ce6f127e7eb875ab63cdc"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-09-15 09:20:05.000000000","tz":60},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98bb887fb3161193e52e8e8612cde7a4fc95666f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98bb887fb3161193e52e8e8612cde7a4fc95666f"}]},"branch":"refs/heads/master"},"b8f6e0fd19259c16b8dfd1511609384c4c1895c0":{"kind":"REWORK","_number":12,"created":"2022-09-15 09:22:58.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/12"}}},"commit":{"parents":[{"commit":"9cbd1e2f52b84a418ba1b3ec5c1119713774c926","subject":"CI: Drop centos8s and rocky8 jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9cbd1e2f52b84a418ba1b3ec5c1119713774c926"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-09-15 09:22:51.000000000","tz":60},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8f6e0fd19259c16b8dfd1511609384c4c1895c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b8f6e0fd19259c16b8dfd1511609384c4c1895c0"}]},"branch":"refs/heads/master"},"ad21504c0fe3709411f4b4e5c948854d67158cc1":{"kind":"REWORK","_number":13,"created":"2022-12-12 17:05:01.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/13"}}},"commit":{"parents":[{"commit":"13b005fafaf7e73a2a0187db731ff2247b9218da","subject":"Merge \"Support copying nova_compute Kolla configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/13b005fafaf7e73a2a0187db731ff2247b9218da"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2022-12-12 17:03:26.000000000","tz":0},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad21504c0fe3709411f4b4e5c948854d67158cc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ad21504c0fe3709411f4b4e5c948854d67158cc1"}]},"branch":"refs/heads/master"},"c80af21def36606cfc4f19ea9f967555d17bb2e4":{"kind":"REWORK","_number":14,"created":"2023-01-11 13:56:26.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/14"}}},"commit":{"parents":[{"commit":"ebd5799fd4b77a138cf7790a8094eb859e8898f3","subject":"Merge \"Replace references to Docker Hub by Quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebd5799fd4b77a138cf7790a8094eb859e8898f3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-01-11 13:47:39.000000000","tz":0},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c80af21def36606cfc4f19ea9f967555d17bb2e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c80af21def36606cfc4f19ea9f967555d17bb2e4"}]},"branch":"refs/heads/master"},"614b8bf0c1b0ee91185f73300d5c8b6e85ea7369":{"kind":"REWORK","_number":15,"created":"2023-01-11 17:27:51.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/15"}}},"commit":{"parents":[{"commit":"ebd5799fd4b77a138cf7790a8094eb859e8898f3","subject":"Merge \"Replace references to Docker Hub by Quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebd5799fd4b77a138cf7790a8094eb859e8898f3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"rocky Cloud User","email":"will@stackhpc.com","date":"2023-01-11 17:27:39.000000000","tz":0},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/614b8bf0c1b0ee91185f73300d5c8b6e85ea7369"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/614b8bf0c1b0ee91185f73300d5c8b6e85ea7369"}]},"branch":"refs/heads/master"},"79e62f703bda0d9056732b928ff76c944757efef":{"kind":"REWORK","_number":16,"created":"2023-01-11 17:35:19.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/16"}}},"commit":{"parents":[{"commit":"ebd5799fd4b77a138cf7790a8094eb859e8898f3","subject":"Merge \"Replace references to Docker Hub by Quay.io\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ebd5799fd4b77a138cf7790a8094eb859e8898f3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"rocky Cloud User","email":"will@stackhpc.com","date":"2023-01-11 17:35:03.000000000","tz":0},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79e62f703bda0d9056732b928ff76c944757efef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/79e62f703bda0d9056732b928ff76c944757efef"}]},"branch":"refs/heads/master"},"c0e34b73bff41e0c30e7d033a276f5308f680893":{"kind":"REWORK","_number":17,"created":"2023-03-07 10:22:48.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/17"}}},"commit":{"parents":[{"commit":"77ca0771958713804d2c65b8e33824a6573132fe","subject":"CI: Disable disk image build in seed-upgrade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/77ca0771958713804d2c65b8e33824a6573132fe"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-03-07 10:01:12.000000000","tz":0},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0e34b73bff41e0c30e7d033a276f5308f680893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c0e34b73bff41e0c30e7d033a276f5308f680893"}]},"branch":"refs/heads/master"},"3d7d5d2551496862564161670bc841e19890e867":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2023-03-07 10:29:27.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/49/782749/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/18"}}},"commit":{"parents":[{"commit":"5a4f3f356632f6489db4bc8f4b1812aaa015725a","subject":"Pass through kolla inventories as is","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a4f3f356632f6489db4bc8f4b1812aaa015725a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2023-03-07 10:29:18.000000000","tz":0},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d7d5d2551496862564161670bc841e19890e867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d7d5d2551496862564161670bc841e19890e867"}]},"branch":"refs/heads/master"},"9c1564232eebc6bdf72a2f6778076393165a2a60":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2023-04-19 14:07:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/49/782749/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/19"}}},"commit":{"parents":[{"commit":"69ecea26d714c2237a39b947ed29896389a13ea1","subject":"Pass through kolla inventories as is","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69ecea26d714c2237a39b947ed29896389a13ea1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-19 14:06:46.000000000","tz":60},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c1564232eebc6bdf72a2f6778076393165a2a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9c1564232eebc6bdf72a2f6778076393165a2a60"}]},"branch":"refs/heads/master"},"373edbecb1d2603f6e25176fc0f9b47c34cd511e":{"kind":"REWORK","_number":20,"created":"2023-04-19 14:09:33.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/49/782749/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/20"}}},"commit":{"parents":[{"commit":"69ecea26d714c2237a39b947ed29896389a13ea1","subject":"Pass through kolla inventories as is","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69ecea26d714c2237a39b947ed29896389a13ea1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-19 14:09:11.000000000","tz":60},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/373edbecb1d2603f6e25176fc0f9b47c34cd511e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/373edbecb1d2603f6e25176fc0f9b47c34cd511e"}]},"branch":"refs/heads/master"},"2ea190f2b52fbab998149d26082102fc3f99aa32":{"kind":"REWORK","_number":21,"created":"2023-04-19 15:40:15.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/49/782749/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/21"}}},"commit":{"parents":[{"commit":"69ecea26d714c2237a39b947ed29896389a13ea1","subject":"Pass through kolla inventories as is","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/69ecea26d714c2237a39b947ed29896389a13ea1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-19 15:09:23.000000000","tz":60},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ea190f2b52fbab998149d26082102fc3f99aa32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2ea190f2b52fbab998149d26082102fc3f99aa32"}]},"branch":"refs/heads/master"},"9bfd19293b2686cfd94448ee72d37ee34e348592":{"kind":"REWORK","_number":22,"created":"2023-04-20 09:19:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/49/782749/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/22"}}},"commit":{"parents":[{"commit":"ed76f1987ce0d204480e7c3b3fc3dd184000b23d","subject":"Pass through kolla inventories as is","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ed76f1987ce0d204480e7c3b3fc3dd184000b23d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2023-04-20 09:19:34.000000000","tz":60},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9bfd19293b2686cfd94448ee72d37ee34e348592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9bfd19293b2686cfd94448ee72d37ee34e348592"}]},"branch":"refs/heads/master"},"955c15f13e727a79330fe3a1afef80b39f041ad8":{"kind":"REWORK","_number":23,"created":"2023-05-30 15:52:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/782749/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/23"}}},"commit":{"parents":[{"commit":"45b968ab765d9f93f1cf879d39c4de0cb9c8e015","subject":"Pass through kolla inventories as is","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45b968ab765d9f93f1cf879d39c4de0cb9c8e015"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-05-30 15:52:23.000000000","tz":120},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/955c15f13e727a79330fe3a1afef80b39f041ad8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/955c15f13e727a79330fe3a1afef80b39f041ad8"}]},"branch":"refs/heads/master"},"526b37e99c556d9a40652902c9b3296a81eba0b8":{"kind":"REWORK","_number":24,"created":"2023-05-30 15:54:17.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/782749/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/24"}}},"commit":{"parents":[{"commit":"45b968ab765d9f93f1cf879d39c4de0cb9c8e015","subject":"Pass through kolla inventories as is","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45b968ab765d9f93f1cf879d39c4de0cb9c8e015"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-05-30 15:53:57.000000000","tz":120},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/526b37e99c556d9a40652902c9b3296a81eba0b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/526b37e99c556d9a40652902c9b3296a81eba0b8"}]},"branch":"refs/heads/master"},"428ef10fa4d4a281f0c690876fc3cbfff41c2bbb":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2023-05-30 16:47:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/49/782749/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/49/782749/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/49/782749/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/49/782749/25"}}},"commit":{"parents":[{"commit":"c75a32e72fa2d31e494b83bd406e42040e8de15d","subject":"Pass through kolla inventories as is","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c75a32e72fa2d31e494b83bd406e42040e8de15d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-08-10 10:53:16.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2023-05-30 16:47:14.000000000","tz":0},"subject":"Use merge_configs and merge_yaml to generate Kolla custom config","message":"Use merge_configs and merge_yaml to generate Kolla custom config\n\nThis patch adds new functionality - merging base \u0026 environment specific\nkolla config. This allows you to place common settings in the base\nconfiguration and only keep environment specific settings in the\nenvironment directories.\n\nChange-Id: Id4588f4529a4522e68e22ce58711cb927fa68a9d\nStory: 2002009\nTask: 42903\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/428ef10fa4d4a281f0c690876fc3cbfff41c2bbb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/428ef10fa4d4a281f0c690876fc3cbfff41c2bbb"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
