)]}'
{"id":"openstack%2Fkayobe~783316","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I33d2b8ea08903d558e2a42231de48edc6567cadf","project":"openstack/kayobe","branch":"stable/train","topic":"story/2008753","hashtags":[],"change_id":"I33d2b8ea08903d558e2a42231de48edc6567cadf","subject":"Revert \"Performance: skip LVM configuration if no groups configured\"","status":"MERGED","created":"2021-03-26 14:53:14.000000000","updated":"2021-04-14 18:37:48.000000000","submitted":"2021-04-14 18:36:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"783316-story/2008753","meta_rev_id":"eb4cd0091de0b8f14a04052dfb511d543abec3c0","_number":783316,"virtual_id_number":783316,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-14 18:36:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-04-14 18:37:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-04-14 18:37:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-14 09:27:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 16:21:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-26 17:18:05.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-04-08 09:04:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"b1c6c8f50b5758145240da5ac73da0f2e1cb17da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-26 14:53:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a053e839e4e9e30a6156bf7244c45a7c522152a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 16:21:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b1da2e6c02744419b7cc43a086c3bdb : SUCCESS in 4m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/462ea5c50bcc46808334d820cb192ec6 : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c520e96f5f34cd791e14a3b56a23059 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c02af0f74d44a138b3390799867ae45 : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c5b768b127d44d1b0080248a46eab04 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/254e8ee3ee7347bd8ff399e47c262202 : SUCCESS in 6m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7ed23a78a1ae4d2b8b5d7c8acb13c0fa : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6e1fd1eb7b5b4e74b5574435d5968cee : SUCCESS in 9m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/27b93ca13d0249d086a27e2169233258 : FAILURE in 5m 27s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/b072c09db9c441ad9528767aa5ef14f3 : SUCCESS in 1h 02m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/428a4968e2c84d8bba4ee1274ad32d93 : SUCCESS in 56m 39s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/5a06baaf1f6d4039a1a22f1ab5459f90 : SUCCESS in 13m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/06ffe97870254bfcb657aa6d7b092614 : SUCCESS in 14m 22s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6d6657bbee3b4fd7b69a6ec40bf21bd2 : SUCCESS in 1h 26m 17s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/706152bbd3cf4217a614cd30c5723229 : SUCCESS in 29m 09s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2b0f7595a5214055a3f0e1d01d8eae59 : FAILURE in 4m 41s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/4b83d3015f98440c8a34890f56876bd9 : SUCCESS in 35m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b9095100cff2ad3a537037f807a5da2aa74cb726","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-26 17:18:05.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06356f813b2de7f05b723ae246e0ff0c9fb2de21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-30 12:05:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"83e90907dd940c0257a056f058379aefda12d788","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 13:53:32.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/534adcef2d1f4615b945e5fc6d3c6e64 : SUCCESS in 4m 27s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e1fac49797ff4fa1ab3fb61fd1da30ff : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3100076add6438b93fd6ea1c9aa2be0 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94be0b6105b74c68814d5b7f2aec0a25 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2463b5831dc1468394d351ab66d7ddc7 : SUCCESS in 4m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c23221cab53439786a4a4ea2752036a : SUCCESS in 5m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b9b9527e2c9448c7a6670fdb23504a5c : SUCCESS in 4m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e6f375e1ddcd41ec8ee345284aa9f138 : SUCCESS in 8m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6a89d2b710134da093f94bfd47c8f982 : SUCCESS in 14m 51s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/5a229f5676c549bdaaab94edd136d593 : SUCCESS in 50m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5e1b9fe107464574ac9da438e147c624 : SUCCESS in 44m 16s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/75da861ca86f475bb4dff0b4c8ce1aa1 : SUCCESS in 12m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7fda503e943043fea6ee41e45c6191f4 : SUCCESS in 15m 44s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b1019323c9f14f8cb5b568df91be6760 : SUCCESS in 1h 37m 08s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/641ac4e1a3e04abb9ae843cf79d878b9 : SUCCESS in 32m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2404285eb0b94b47a9c695f1c7954e0c : SUCCESS in 26m 43s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/5644ffa21eec4bcc9d6f5c6e2310e66e : SUCCESS in 33m 28s","accounts_in_message":[],"_revision_number":2},{"id":"5768e3e2d9250ed5aeb3c194336f4274301ac880","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-30 15:00:56.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bf9ed8bfde6f19220efab1e42ffdc513d3b2ae01","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-08 09:04:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"43c4d1cedea6f9f489fd10bc65ef3a39d220f0dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 09:04:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5efa1f9284dfd54e594c0fa559ded7c09c0d19ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 10:20:46.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bcdf68511d304dcb92e1e65ab4248ea7 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1299c8950ea4403f9736c59bda255fb8 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a19081d1622d41a790e56a38de176b8f : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5dfdb5e2843c40f88e79eb8b9844bd84 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0447cf03d2d04e0e97c6a49643937de3 : SUCCESS in 9m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/43d1400591bc45b8850025f81dce8000 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a485e98ceb7d4918813e8056afa395ef : SUCCESS in 8m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3c7efdb5198a42c28b3424cbed79a6d0 : SUCCESS in 19m 35s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/e6f3a0fd0da042e0a70315cb51c73fed : FAILURE in 50m 57s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0b36c0143ffb4a57a80b7d272be241c6 : SUCCESS in 1h 04m 07s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/2231f06102da41f1b59177f6457e948c : RETRY_LIMIT in 4m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/66ca52805d8344fa8a1e15be98c93a81 : SUCCESS in 17m 21s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/1f98f5cc59294c838188482305190afb : FAILURE in 5m 11s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/a94903db550d41fd8335200e445ad857 : SUCCESS in 27m 27s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e22d25fcac534993b9327e9f3a255c22 : SUCCESS in 30m 25s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0b29d88a1b674bbc888ed11abc11b6fa : FAILURE in 4m 22s","accounts_in_message":[],"_revision_number":2},{"id":"6e2bcb77a79dd06df94487594b734009c4eb0f75","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-08 10:24:21.000000000","message":"Patch Set 2:\n\nrecheck\n\ndistutils.errors.DistutilsError: Could not find suitable distribution for Requirement.parse(\u0027pbr\u0027)","accounts_in_message":[],"_revision_number":2},{"id":"822fb3214abf6f6090215f500671389fbe9b4b9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 11:12:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4cf785403a94465ab70a67c4254b7f7d : SUCCESS in 4m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/29864f00fe70411396e217e2006cb344 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd164df52489474d86568820d6a627e7 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/636ad226def6413ab4a33d2c16ac6b75 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9795c6db6a874b2b82ef66a847aceb58 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/587a1dc0383d4c57a5c43eac71f79825 : SUCCESS in 5m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/41c6d0e5931e4273903d9ce983ce6a1e : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd1ac554c27847169cec8f2d88d1fb8e : SUCCESS in 7m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3cba79f437c04dc78c6090e21884caba : SUCCESS in 18m 10s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/d0b4de795b8641c59e1f33672ce63435 : FAILURE in 42m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/17461b78677544e398f2f9cf6682d4af : SUCCESS in 46m 21s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/d82ecf7421b5487691b5a6d3b66ae9b6 : SUCCESS in 15m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7e77ccc8b4bc424c94623860159ac682 : SUCCESS in 16m 57s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/1595bcd661044e0eada2000d24ccce4e : FAILURE in 41m 08s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/a897d00dd7664c2aba4f93a8f83ff88f : SUCCESS in 24m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d7aaaf035426439daaea01cd4ef22496 : SUCCESS in 33m 51s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/d8173c2fe16d40a883bcd44be1db0603 : SUCCESS in 42m 09s","accounts_in_message":[],"_revision_number":2},{"id":"55ee3940dd7c3040451706416e146878e9bf3b85","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-09 10:56:49.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6ffce97bcf6da5b0d4848cbf3c620d3b092eeb38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 11:57:19.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9766d284c8a497387a0b04ea23cd3a6 : SUCCESS in 4m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7a1e299917b04a8db569984877b2579a : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d78afc5552214b669fc9999bdce092de : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/046437045b7e42a799dddc8b422cf261 : SUCCESS in 4m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/acd3be42afbf436899787e208fb48413 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c3cf4fb58484e7d98949ce5235f211c : SUCCESS in 7m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c6b388990448472ba2f182405f858999 : SUCCESS in 4m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81a1b7247aa74f64aa765dcafe0d7b57 : SUCCESS in 9m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/817aa5bf9c5c43e38536560847466a3e : SUCCESS in 17m 44s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/6cf767b7ac9d47a19619cd4476524ce1 : RETRY_LIMIT in 4m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0bf3d07dfcef4ee0a8005d33a0305119 : SUCCESS in 55m 46s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/a35b2d325bab4b47a010c70d1dcd4bb5 : RETRY_LIMIT in 3m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8de0399ec5154d978be383de1f459531 : SUCCESS in 18m 56s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6e68720c69934773a55e02334874c62d : FAILURE in 4m 43s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/9951c759b37640a8828e9945980ebff3 : RETRY_LIMIT in 3m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/928acf5ecd194569b4d70fcd2e5725c6 : SUCCESS in 27m 54s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/ace83312aca74ee78af6c46e164771c8 : FAILURE in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e51f3882bae85e7a8b65cbf45f297b7548a8a9db","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-12 09:38:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a93c6e7e1b191a93d684a8646827fcfe948ec511","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 11:19:56.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa6f62a8027e42db83d73c9cbfac2dd1 : SUCCESS in 4m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c739e34f5dc24244aefb0c18c6e9eaee : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c745beaca764488aa93e1583d582204d : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8c36fe9f506460188917fa0cb25eb55 : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d295b467d1854267ade4f3196a3c2b8b : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edbcf0f5345c427aba61eb34a25a019b : SUCCESS in 5m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5a325e957bca463c85a4afb022246981 : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/79c77b8a971642569bd2b783a00ffa1a : SUCCESS in 8m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91de77dfd31d4e13bd575eeadfc7253c : SUCCESS in 19m 21s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/676fe7df1ee2468d9147c2e259686d37 : SUCCESS in 54m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b3953982e1cb4dd5a4ed59dfd5143d63 : SUCCESS in 46m 22s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/6e44f538a6ef46368e09ec58b797aee6 : SUCCESS in 15m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/42d06a863de0450ab9fa29ced44dfc40 : SUCCESS in 15m 03s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/74ad4b2b6d244c2dab100e6c9051b6b9 : SUCCESS in 1h 40m 39s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/9310598202d740fc89a2056350f3e8b9 : FAILURE in 23m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f18f1ca0e2d54785ad115d029606f92b : SUCCESS in 25m 47s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6e54521c03894e938e2122dba863b323 : SUCCESS in 52m 56s","accounts_in_message":[],"_revision_number":2},{"id":"1fcb0feab5600b3f7e902512f3201855d83decb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-14 07:59:29.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"685d2db2418b588f78da2bc43843818828a9fd4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 09:20:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/609419549c95496b87f22ca7aca4ea5c : SUCCESS in 4m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bf457b96b7d64a64a30fc9cdd534703e : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2f951d12f7b4553860a5e7ddbcfdc2b : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdccd18d9b164590be6c154a1bd43340 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8dbe0640b0894aec905ad567b3de6020 : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a18800f775024de7b06045e9cae7cfc4 : SUCCESS in 7m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6af4fda22eef455698b7a7bf6a9a2e2d : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3f32851c994947aeb897d38b5c9e3395 : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6ad6f8bd1a5140bc8fac7fbc2bf8cf47 : SUCCESS in 15m 46s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/9508867a8deb4cfdb54c0e2c25ce8475 : SUCCESS in 57m 27s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f84e5fb7224a4d79b8ec5e0b4fc2d768 : SUCCESS in 57m 02s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/5ba7c3ce89ce454dbd0b6e83fe530e3e : SUCCESS in 15m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ed5bffb32e494692a716cd14574e38ac : SUCCESS in 15m 45s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/af4152ad636340feb04a9792ee0f8151 : SUCCESS in 1h 19m 47s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/ec79402e03874f5fb8dd85f76791f82b : SUCCESS in 17m 28s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/96a89ce117cb40fa8500233f1a83b7f6 : SUCCESS in 24m 20s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/7a06acfd40a647e89143034eb4b9449a : SUCCESS in 33m 20s","accounts_in_message":[],"_revision_number":3},{"id":"570e103dd875f4b89351e11f77a062da10ebc75e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-14 09:27:28.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6dceab93d29ffb4bea85743fad94573159aab326","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 09:27:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7385779363c81eeb8de5d47d5a70ab2d9baf16af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 11:36:47.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/cfae8c5b622a46b98dd5380240b1657c : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28a4a2c66fa14b1aa7abbba07d5b43c0 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8509dc9697c745758e6fa4223b4e7e73 : SUCCESS in 5m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/91d6d91db4b449f3addfacf1882e3b69 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce7b1c1aba3e4b55adb578e3de39a193 : SUCCESS in 6m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8275078a84f74e41978caf4764e7e2ea : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5aa919ca2fb8434e892d164164084fc5 : SUCCESS in 8m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4fdf1eeb57b46d98f80b6c60c29b3d3 : SUCCESS in 17m 49s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/68eab1d8e1704f09900128e7af3bd24f : SUCCESS in 1h 08m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/47c47168fdeb47bf9e80ea11f07d0d70 : SUCCESS in 54m 29s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/66fbbdb3fa0c4271bd9f7cdb09fdf16b : SUCCESS in 14m 53s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2921d8230d2142fbbb18658d7b0c5b31 : SUCCESS in 15m 11s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f18f4c8398d54c5aa6ab7a2defd5d49b : SUCCESS in 1h 24m 25s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/21a53c0ea1cf4f479a65b301dcd14649 : SUCCESS in 17m 51s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ff581a811f2b4f6fb5bea708bf00444b : FAILURE in 17m 13s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/932032807f4642b98a17c1dc9a049cdb : SUCCESS in 29m 07s","accounts_in_message":[],"_revision_number":3},{"id":"9b0220d310942b5065dac232613bb7f2f457dfaf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-14 11:48:19.000000000","message":"Patch Set 3:\n\nrecheck\n\ngrafana repo issues","accounts_in_message":[],"_revision_number":3},{"id":"f0b289000503ac442690931e3ce5fb963f9e0318","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 12:59:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d5549531d0d492cacd4c39054cd1943 : SUCCESS in 4m 37s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a5d3e20f65cd409bb6bd299706074938 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f679ffb35a3241808197d220e1fad811 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a595749f4b1c473db2ecfd992773116f : SUCCESS in 4m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3db9de4eacca4c0aba438cac0a14640e : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d314c73b4ee94a86a8e25fca345a195a : SUCCESS in 6m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/00ffe236bc5d4ca9b05b6f6d89bffb60 : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/70281dd50794487a84bda5cb0fd2fedb : SUCCESS in 7m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1144f55377c14b618f4ccdbad117a19b : SUCCESS in 19m 15s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/0e26adfd583747f18abd66074ebe69a7 : SUCCESS in 44m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e73f013f90dd4aa9b4d95c9e367021df : SUCCESS in 56m 21s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/84d74676b19d4ba5b461ab74f6ce121b : SUCCESS in 12m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b1861a2572c84ccab1329079b7f23521 : SUCCESS in 15m 47s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b2e611349d364ba5a1f8048ecd761ed3 : SUCCESS in 1h 09m 25s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/d9fccdba51e94150b928ff88ca3c2518 : SUCCESS in 21m 36s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6317db75f63140cf9d00c88f81074489 : SUCCESS in 21m 26s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f171d3abd56143e7a7ac8fbf171ce517 : SUCCESS in 37m 58s","accounts_in_message":[],"_revision_number":3},{"id":"4074ea95006d6c0bfe2b045d70fd34d4ea07a9e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 12:59:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a0c1f1a35bc4e37e13d170a16412a655ce8ecb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 14:30:35.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/88ecf7218dc9481d92887a71171339f7 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/52600f9db7bb45c18e29feaf4fa6b19e : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ef6d2ec258548b68ec79e934ca75041 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d7d6c37bc37949e0b600d35956f741a4 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8bcfb1edc242cba2aa5c1bf309b011 : SUCCESS in 10m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d0dbc4418a3e433e92a024ccb3fe99fd : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/63bb91c672124a5f90ba8810c4f7ef8c : SUCCESS in 9m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8acc4bac022f49eabf76ded298d32e3c : SUCCESS in 19m 40s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/e6b547ceaa034f9a80f79d80fe1b2644 : SUCCESS in 1h 01m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9fa9fb36ebdb44039bacd3006d92c898 : SUCCESS in 56m 15s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/1503f72e360f4eacbc655aa32faac521 : SUCCESS in 14m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/55f340d2e9dc4d94a1504170ba9a2c63 : FAILURE in 17m 10s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/92c26d16930b44baba6a411333c7c925 : SUCCESS in 1h 29m 18s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/bda8e6b333c146f386989ac222eddb5b : SUCCESS in 23m 27s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/45a7af5079a94737898997043f593341 : SUCCESS in 28m 03s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/65a55e9277f14c66940360c34ad57620 : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":3},{"id":"62dca29d2141824626639b351b62320a5ce13e28","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-14 15:27:32.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"be44a98c473cf9f1ec83ddf59076cd82506e0182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:56:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d5f4f674dae4607a3038469e541494c : SUCCESS in 4m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/4d627889e1b94c70a1422a74d7d6e726 : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fc3cea059df3460189cf3d9ed3d1bcc6 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3404b9a9e44a400b8cea9a520e487fb0 : SUCCESS in 4m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/797e6c12deba4231afd46dd0e0cac8e2 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a5c4e5c966d4f4b85e08e578d1468f4 : SUCCESS in 6m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/690f6535b0314d328c5213138ccc7477 : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ff0505699e2749acb13cd137574741b9 : SUCCESS in 8m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e24946010a304946b63509592af80793 : SUCCESS in 18m 29s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/4603271b92ea41af863d51dcc437e157 : SUCCESS in 1h 13m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/bc8e07c148664a5a891d1b5dc4ad4e8b : SUCCESS in 1h 05m 51s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/39b089e42fad4e029a246e4cca197a5e : SUCCESS in 14m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/aa85bf9c321744989f48c6a46952727c : SUCCESS in 16m 13s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b847c03ca08f4adfa73783c85dc3ac02 : SUCCESS in 1h 25m 40s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/013662e963ad40c789d75a7735f0b3c8 : SUCCESS in 19m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4ac0a88226c642d3a3cafa66471bc6c8 : SUCCESS in 27m 03s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/9586cce279e64a3c921756a8bd8a9774 : SUCCESS in 39m 57s","accounts_in_message":[],"_revision_number":3},{"id":"cacc2a432cda9aac822e9b0c1d5d4b0d67987ee5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:56:38.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"34b11cb760fc741202e1d30467d0aa4f5aa24c10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 18:36:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6c29b168f615427480e22d5ea14898b8 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ef634a4072ff4b37be661bc3dd00b1f9 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7f4fa30be784b06b1c1717dc44d5ca8 : SUCCESS in 4m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e6dc4271e2a4406a951a076362de548f : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60ebf8fd086c42e0b1d02f04b9711649 : SUCCESS in 6m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/94c5a107fdde43a0b7e224036252855e : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/35be050cc09746ed88b915bf040246fe : SUCCESS in 11m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9cde7413b6a140769af23a751246bebf : SUCCESS in 16m 09s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/5d16dd29bf344a6098bce98bfd53c230 : SUCCESS in 1h 02m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9db3bc53f9c749cb9b5fbd0a7107a27e : SUCCESS in 50m 32s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/af8a6ad0735b4de1899d57d8cfbccd9a : SUCCESS in 16m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/308f82d0dfa34c189640df4bce059f99 : SUCCESS in 16m 05s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/1b9965c310c4478f865741bfb9d5c29e : SUCCESS in 1h 37m 48s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/92abfd0cd80349f49a8787cad2c61a09 : SUCCESS in 21m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7300f4c249d64a6eadfedc26cf2dc8d0 : SUCCESS in 28m 36s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/77f03d9804fb442dbdff4d203248e99d : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":3},{"id":"a5f3ce91366f3da9dd0a0ea5f4bab9967f8c9a97","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 18:36:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"205efebea3561364563ab23035018abf40a7e0a4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 18:37:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90506db403e44e118c1ac0fb61f47e68 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c12192c18a31c1a9e3223deff5c0862910da8258","revisions":{"7a31b3212ba8ca2caf000a59f7cb458e5852fed2":{"kind":"REWORK","_number":1,"created":"2021-03-26 14:53:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/783316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/783316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/783316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/783316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/783316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/783316/1"}}},"commit":{"parents":[{"commit":"5886f715cdfe8c2ab39787c97f9abaf4ad1179c0","subject":"Merge \"Update IPA docs and test build with extra-hardware\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5886f715cdfe8c2ab39787c97f9abaf4ad1179c0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-26 14:52:30.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-26 14:52:30.000000000","tz":60},"subject":"Revert \"Performance: skip LVM configuration if no groups configured\"","message":"Revert \"Performance: skip LVM configuration if no groups configured\"\n\nThe reverted change relies on the apply parameter for the include_role\nplaybook, which was introduced in Ansible 2.7. Requirements for the\nTrain release allow Ansible 2.6, which produces the following error:\n\n    ERROR! Invalid options for include_role: apply\n\nThis reverts commit 5fd45d67dea05e0ba3d4800b1d7b351b8b365565.\n\nChange-Id: I33d2b8ea08903d558e2a42231de48edc6567cadf\nStory: 2008753\nTask: 42114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a31b3212ba8ca2caf000a59f7cb458e5852fed2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7a31b3212ba8ca2caf000a59f7cb458e5852fed2"}]},"branch":"refs/heads/stable/train"},"424e4926e6bf4de5c539c3b01170f5fc414df8e4":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-03-30 12:05:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/783316/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/783316/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/783316/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/783316/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/783316/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/783316/2"}}},"commit":{"parents":[{"commit":"5ca6b6148acf613ed988408cb2d260dc1e2bb920","subject":"Merge \"CI: Fix kayobe-tox-molecule job\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5ca6b6148acf613ed988408cb2d260dc1e2bb920"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-26 14:52:30.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-30 12:05:53.000000000","tz":0},"subject":"Revert \"Performance: skip LVM configuration if no groups configured\"","message":"Revert \"Performance: skip LVM configuration if no groups configured\"\n\nThe reverted change relies on the apply parameter for the include_role\nplaybook, which was introduced in Ansible 2.7. Requirements for the\nTrain release allow Ansible 2.6, which produces the following error:\n\n    ERROR! Invalid options for include_role: apply\n\nThis reverts commit 5fd45d67dea05e0ba3d4800b1d7b351b8b365565.\n\nChange-Id: I33d2b8ea08903d558e2a42231de48edc6567cadf\nStory: 2008753\nTask: 42114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/424e4926e6bf4de5c539c3b01170f5fc414df8e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/424e4926e6bf4de5c539c3b01170f5fc414df8e4"}]},"branch":"refs/heads/stable/train","description":"Rebase"},"c12192c18a31c1a9e3223deff5c0862910da8258":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-14 07:59:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/16/783316/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/783316/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/783316/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/783316/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/783316/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/783316/3"}}},"commit":{"parents":[{"commit":"67103fb5d7e2aeba056c71ef20131c87d47de19e","subject":"Fix help texts of inventory options","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/67103fb5d7e2aeba056c71ef20131c87d47de19e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-26 14:52:30.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-14 07:59:29.000000000","tz":0},"subject":"Revert \"Performance: skip LVM configuration if no groups configured\"","message":"Revert \"Performance: skip LVM configuration if no groups configured\"\n\nThe reverted change relies on the apply parameter for the include_role\nplaybook, which was introduced in Ansible 2.7. Requirements for the\nTrain release allow Ansible 2.6, which produces the following error:\n\n    ERROR! Invalid options for include_role: apply\n\nThis reverts commit 5fd45d67dea05e0ba3d4800b1d7b351b8b365565.\n\nChange-Id: I33d2b8ea08903d558e2a42231de48edc6567cadf\nStory: 2008753\nTask: 42114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c12192c18a31c1a9e3223deff5c0862910da8258"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c12192c18a31c1a9e3223deff5c0862910da8258"}]},"branch":"refs/heads/stable/train","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
