)]}'
{"id":"openstack%2Fkayobe~783337","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I33d2b8ea08903d558e2a42231de48edc6567cadf","project":"openstack/kayobe","branch":"stable/stein","topic":"story/2008753","hashtags":[],"change_id":"I33d2b8ea08903d558e2a42231de48edc6567cadf","subject":"Revert \"Performance: skip LVM configuration if no groups configured\"","status":"MERGED","created":"2021-03-26 15:37:11.000000000","updated":"2021-06-10 08:25:12.000000000","submitted":"2021-06-10 08:23:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783337-story/2008753","meta_rev_id":"143f041bfa54fb217f6959bc2686bdf63a138f8a","_number":783337,"virtual_id_number":783337,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-10 08:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-06-10 08:25:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-06-10 07:17:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-06-10 07:17:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 16:40:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-26 17:17:00.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"ff3880a4df64f51194e84ac7c7b8187fecd26aed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-26 15:37:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21556d068aa98430ad53e1138673c6cd1a071715","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-26 15:39:44.000000000","message":"Topic set to story/2008753","accounts_in_message":[],"_revision_number":1},{"id":"86562834421ef9bc3f52cad1f156528daf88a634","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 16:40:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0b38e93b3224014bd1c7ac2318f80d7 : FAILURE in 3m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ab60acfaff154b33a03856b7c4af26f7 : FAILURE in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d6e65accf244014bd24e2c0f130adce : FAILURE in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3cb3607eaa145c591b8a0de45b428b4 : FAILURE in 3m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fbe3a4e076284723ba14444ff206e729 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a57e0eac86845c3b7d62c09f64b4414 : SUCCESS in 4m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ecd74e505e3f4c77b8b8f0cd9f5aae36 : FAILURE in 3m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5f4f5c76df304ab39acc3da008a8de58 : SUCCESS in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/934939da337f422aaf6d429827c97181 : SUCCESS in 17m 48s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b3a5695698ea4744b6b802a5bf0065d8 : FAILURE in 53m 44s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b98d252863e34781938f481c80c441f2 : FAILURE in 22m 39s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/dde8a8d0153749efb16fd17e4dd6ef97 : SUCCESS in 18m 25s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/edddeea7dbd040aab2daa6a909d7f4c2 : FAILURE in 35m 24s","accounts_in_message":[],"_revision_number":1},{"id":"93ad86bce36991622d7b6355aaf5a7e0d97d3215","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-26 17:17:00.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6d8a719083b2384c0cedf326cf926684585b5caa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-07 11:14:42.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"16bc0241dadc4cd5bdb45f5b2aa28fcd9a08f137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 12:05:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f414b40529c84aa39083a2be6415100e : SUCCESS in 4m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b00cf19a0ed74fcd981e6ce6bd7a44ae : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c37b0868c114163bbcbc793a050aa0b : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc4e1281ab9411cadc8856bf014fe1e : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9793f99229a4dcbb867cc9588972b14 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95718ae91c79423d8f8c4f8fa045f165 : SUCCESS in 4m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/736c87d2a2bc47098f31ed79f6d437d5 : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a5549810ac64496ba337b22e0e619db : SUCCESS in 8m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/efd9ad6bd88147379402bbaa43219fbc : SUCCESS in 18m 25s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/8cdf96686e924e6eb4afdc31d3bf8b8d : FAILURE in 47m 47s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/d34caacc45a74908b36cdc71f2f877a9 : SUCCESS in 22m 18s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/7c9e2f4787484f7e9dbe167d26a2d05d : SUCCESS in 36m 19s","accounts_in_message":[],"_revision_number":2},{"id":"461dffe7d742c1be7ed33c7a017944a081cab5f5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-07 13:07:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"4099e3a33635c53e148cb17338acb7bc5cd7b434","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 13:26:49.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54422d5602bb425eb68ff259e0dd1f0b : SUCCESS in 4m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/226c649396034807a776d8ed7797a7f3 : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e77d923912e4afda50903281f7b8837 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/013aed905ffb4ca59c4fc5fbb2bbe36c : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0d1c6a1af0b14cb7ab5e5da9636ff099 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5d34b166944c90a5c14cf78427336d : SUCCESS in 4m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0f5506177e454bb0b503a54d92d1a1e8 : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/70b17d430cbe4141a3e8ead0505113c0 : SUCCESS in 8m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0bd5ae7fa4544960a1139b1f78fb35dc : SUCCESS in 16m 26s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5e44df20920549efb3d5b65b6da70fe5 : RETRY_LIMIT in 3m 37s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/08352d629fcc4ea88ceae9258d2be670 : RETRY_LIMIT in 4m 09s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/33b1bb68c0be4d7eba184b5eee2023ac : FAILURE in 4m 27s","accounts_in_message":[],"_revision_number":2},{"id":"7333786c60678b191ec94c5623fe591ce206a968","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-09 10:57:44.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"622bd52517741d0f11246970a7f098902259401a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 11:23:03.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb52c37a2c544953a06ec03c81bfb6b4 : SUCCESS in 5m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2877a2fa212849a795e16d15711f9a2e : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59ef2ae9244a4a41b316bab058d51ef6 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f5c9b2aaceb446d8615b7a93631d736 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1944b2b8c23441df8eb6e7c49a041b74 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0333738ddc4e4533a2480fba48fa9748 : SUCCESS in 4m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e4d51af4505e4db2825f2c522e770fd2 : SUCCESS in 5m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/de4a382fb80640049aa43db44c93e602 : SUCCESS in 9m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e0e47100fa7a4076ad445bc84d1dc316 : SUCCESS in 18m 20s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/0d022531bcef4854be95272a9993dea0 : RETRY_LIMIT in 3m 07s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/895630eae5194d649615124c9fa85c24 : RETRY_LIMIT in 4m 59s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/670996b0391d4f1993881c0ba747882b : FAILURE in 4m 08s","accounts_in_message":[],"_revision_number":2},{"id":"6ab3b56e0fe4a9074176d0f9f7fb753b859337f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-14 07:59:16.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"4d58ec25680d5f0492d965d90bfa58d5b7b57189","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 08:52:16.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e3162068a276466b8074a69d277ac7cb : SUCCESS in 5m 42s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5ebeae33ce2846849ce3daa2e5bd1776 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3ecfb41ea094af2b9137de5176f8ee3 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/522e047b3d4c4e25a72c7762af064d60 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d5d6e00507f74f438af55edf723de9f8 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa3cd8e63beb4280b2af7b363d68a473 : SUCCESS in 5m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/76d7f1355ecc4530bf33fca5829c87bc : SUCCESS in 6m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a06935b2f2aa4c9d8139029090911a91 : SUCCESS in 8m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/89bfe1c531e444e58c40b1af43011743 : SUCCESS in 18m 56s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/06d9f48421654ffd9685f8a540205392 : SUCCESS in 51m 14s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/42b0b8e159894b6d8b5ec19498a8d8ab : SUCCESS in 21m 28s","accounts_in_message":[],"_revision_number":3},{"id":"21f37b9ebd16aaff4ce34a446d1f07b4ddee691b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-19 09:16:47.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ecdb7dc17613c52c05fe96e599098401ec363f67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-09 21:15:25.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"b6d8dac317d51dc52bcd8cbd5105b31737f89bbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 22:16:30.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d2bce6568a744a7b3b1217be2359093 : SUCCESS in 4m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/bb9fa17dfcbd4c32bf15f1c53d9e95ff : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5351dbf2efc2477789e6ab889a6345ee : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b65f4b473c84e49bd236a7726cdb04c : SUCCESS in 4m 07s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c6c551049b7244c399cf17453d0d6240 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cfdd204e3d34e5caa66a4cbfe11ccf1 : SUCCESS in 4m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/606ccc13311344dc8d25e759522987e3 : SUCCESS in 5m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f533fd4a2e93402e9c8b0a31a5284a74 : SUCCESS in 8m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e18e7ff4e57c498797f39583cf7d72d0 : SUCCESS in 16m 36s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9ce7556299134293a5f0ec76ad4bf733 : SUCCESS in 59m 36s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/02a4a239dc6e416b95cd7d7c4820dc61 : FAILURE in 53m 54s","accounts_in_message":[],"_revision_number":4},{"id":"9a7ff60ee6792272e6663d4453856129b847f0c8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-10 05:24:59.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"77499057e7d5d9ca2e698ddb26803dce934f70a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 06:52:10.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/98f105debe5a43069942964578f2ebe5 : SUCCESS in 4m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7646038529b74349b2b5649f22a7e0b8 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77e243b6cc5f43488d6df1405a74e900 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea14b4473d15466393a89603d4d0dc61 : SUCCESS in 3m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c675e2bd64824c3fa4acefc8c1181a2c : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b20adb006b394979bf1c2620973914b0 : SUCCESS in 4m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b09976eaadc04d1680e95ac567f018f0 : SUCCESS in 4m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bf89374f7a55408a919aa16c396ad641 : SUCCESS in 9m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9def9edb95f345549ea57d91b876fabb : SUCCESS in 18m 42s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/02965284ba8d4e6ba88e7fb5d472fee0 : SUCCESS in 1h 21m 27s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/8e1507d5a9514cf1bc696a16019f98e8 : SUCCESS in 21m 13s","accounts_in_message":[],"_revision_number":4},{"id":"282e1f76315248da151854a7558087bad4938eb8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-06-10 07:17:56.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c21f3f56cf8b70235e0f7e6d08eba281a0bad328","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 07:18:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"9232c714144fb10cb9ece88db6db32dd052154f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 08:23:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3ea1995f797d4fb5a16d116d15e532f9 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a5c3b0493ca498aa31c076fd16452da : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfaf6d2b169146209c2d216417b41aaa : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d219f501b2d644ea92d0eb79d1c2b578 : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9ac7fa03aed476c8ebb93fe2363a9e8 : SUCCESS in 4m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/de1ea508322a4af28ce6e242200a645b : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3071b44080574a369719c8de32ca24c9 : SUCCESS in 9m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91c887a63d324eaaa305e2222e0a125a : SUCCESS in 17m 59s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/af845280025443f984dd3769169339a4 : SUCCESS in 1h 02m 56s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/049abe50567c449b89c1782c26fd2b70 : SUCCESS in 19m 11s","accounts_in_message":[],"_revision_number":4},{"id":"454a7f835e9c01cfa41e9fac1e50d8eac20694e6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 08:23:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"5b23f125cdbe2ab7b3caf51506d8bbe831315f93","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 08:25:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e33552cc1e4f66a3d568f0fa71119f : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"836a714508cc494231b88f43fafd017397ece920","revisions":{"3c205b623d1af1501c094c60e44a031d161724a1":{"kind":"REWORK","_number":1,"created":"2021-03-26 15:37:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/37/783337/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/783337/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/783337/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/783337/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/783337/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/783337/1"}}},"commit":{"parents":[{"commit":"84c463e32d0c7f8234d478543e48589eb46386ba","subject":"Fix filtering of network names set to an empty string","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/84c463e32d0c7f8234d478543e48589eb46386ba"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-26 15:35:30.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-26 15:35:30.000000000","tz":60},"subject":"Revert \"Performance: skip LVM configuration if no groups configured\"","message":"Revert \"Performance: skip LVM configuration if no groups configured\"\n\nThe reverted change relies on the apply parameter for the include_role\nplaybook, which was introduced in Ansible 2.7. Requirements for the\nStein release allow Ansible 2.5 and 2.6, which produce the following\nerror:\n\n    ERROR! Invalid options for include_role: apply\n\nThis reverts commit 5eca75c31f4ffe0c875b0bcdcc6245ca174799b3.\n\nChange-Id: I33d2b8ea08903d558e2a42231de48edc6567cadf\nStory: 2008753\nTask: 42114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c205b623d1af1501c094c60e44a031d161724a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3c205b623d1af1501c094c60e44a031d161724a1"}]},"branch":"refs/heads/stable/stein"},"d02ccdce061b07d63c4c4ff5dffbb84822c3484a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-07 11:14:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/37/783337/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/783337/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/783337/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/783337/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/783337/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/783337/2"}}},"commit":{"parents":[{"commit":"9bf903666e4afad6608dca8c460fe4e833c1302a","subject":"CI: Fix stable/stein jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9bf903666e4afad6608dca8c460fe4e833c1302a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-26 15:35:30.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-07 11:14:42.000000000","tz":0},"subject":"Revert \"Performance: skip LVM configuration if no groups configured\"","message":"Revert \"Performance: skip LVM configuration if no groups configured\"\n\nThe reverted change relies on the apply parameter for the include_role\nplaybook, which was introduced in Ansible 2.7. Requirements for the\nStein release allow Ansible 2.5 and 2.6, which produce the following\nerror:\n\n    ERROR! Invalid options for include_role: apply\n\nThis reverts commit 5eca75c31f4ffe0c875b0bcdcc6245ca174799b3.\n\nChange-Id: I33d2b8ea08903d558e2a42231de48edc6567cadf\nStory: 2008753\nTask: 42114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d02ccdce061b07d63c4c4ff5dffbb84822c3484a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d02ccdce061b07d63c4c4ff5dffbb84822c3484a"}]},"branch":"refs/heads/stable/stein","description":"Rebase"},"83ff0cd843a51cf775ca1ea1b28b145e3405ca9a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-14 07:59:16.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/37/783337/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/783337/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/783337/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/783337/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/783337/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/783337/3"}}},"commit":{"parents":[{"commit":"125d399bdae572ab06fe5fb633a7a0cad9f72ce0","subject":"Fix installation errors with Python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/125d399bdae572ab06fe5fb633a7a0cad9f72ce0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-26 15:35:30.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-14 07:59:16.000000000","tz":0},"subject":"Revert \"Performance: skip LVM configuration if no groups configured\"","message":"Revert \"Performance: skip LVM configuration if no groups configured\"\n\nThe reverted change relies on the apply parameter for the include_role\nplaybook, which was introduced in Ansible 2.7. Requirements for the\nStein release allow Ansible 2.5 and 2.6, which produce the following\nerror:\n\n    ERROR! Invalid options for include_role: apply\n\nThis reverts commit 5eca75c31f4ffe0c875b0bcdcc6245ca174799b3.\n\nChange-Id: I33d2b8ea08903d558e2a42231de48edc6567cadf\nStory: 2008753\nTask: 42114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/83ff0cd843a51cf775ca1ea1b28b145e3405ca9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/83ff0cd843a51cf775ca1ea1b28b145e3405ca9a"}]},"branch":"refs/heads/stable/stein","description":"Rebase"},"836a714508cc494231b88f43fafd017397ece920":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-06-09 21:15:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/37/783337/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/783337/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/783337/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/783337/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/783337/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/783337/4"}}},"commit":{"parents":[{"commit":"b826104b0a10fc75f777b7510e397cf00cf09bf7","subject":"Use released version of mrlesmithjr.mdadm","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b826104b0a10fc75f777b7510e397cf00cf09bf7"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-26 15:35:30.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-06-09 21:15:25.000000000","tz":0},"subject":"Revert \"Performance: skip LVM configuration if no groups configured\"","message":"Revert \"Performance: skip LVM configuration if no groups configured\"\n\nThe reverted change relies on the apply parameter for the include_role\nplaybook, which was introduced in Ansible 2.7. Requirements for the\nStein release allow Ansible 2.5 and 2.6, which produce the following\nerror:\n\n    ERROR! Invalid options for include_role: apply\n\nThis reverts commit 5eca75c31f4ffe0c875b0bcdcc6245ca174799b3.\n\nChange-Id: I33d2b8ea08903d558e2a42231de48edc6567cadf\nStory: 2008753\nTask: 42114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/836a714508cc494231b88f43fafd017397ece920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/836a714508cc494231b88f43fafd017397ece920"}]},"branch":"refs/heads/stable/stein","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
