)]}'
{"id":"openstack%2Fkayobe~783395","triplet_id":"openstack%2Fkayobe~master~I0c46e5fcc6979027dc6f1bc5cc49e923a205f654","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I0c46e5fcc6979027dc6f1bc5cc49e923a205f654","subject":"CI: increase Ansible Galaxy retries \u0026 add delay","status":"MERGED","created":"2021-03-26 17:26:23.000000000","updated":"2021-03-29 10:00:13.000000000","submitted":"2021-03-29 09:58:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"783395","meta_rev_id":"cd5047e1c23b426cacb70dc814bdc65f7d21c6b4","_number":783395,"virtual_id_number":783395,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-29 09:58:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-29 07:51:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-03-29 08:09:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-26 17:28:19.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2021-03-26 17:32:07.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-03-26 19:44:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07bc2007e0830a93f386dadd46c4741e40fa48c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-26 17:26:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0cf7c3f846befa3465136c5d032b2fea67fc3bab","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-26 17:28:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"46c87f01662e4bb198bbb4f9b49aa52817d7f3a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-26 17:29:08.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"20c2ac423d0efe526f9fcc4e4811c6eed5e35900","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-26 17:29:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"da3eee04d4e2f8a50417c292e73d64beb63bdc27","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-26 17:32:07.000000000","message":"Patch Set 1: Code-Review-1\n\nCan we update control_host_bootstrap() / control_host_upgrade() at the same time?","accounts_in_message":[],"_revision_number":1},{"id":"eeb8438097aee631a3d61be41263bf2109df41f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-26 17:34:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9d3acec80bc274a40f88365ecc94dbecaf4427d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-26 19:44:26.000000000","message":"Patch Set 3: Verified+1\n\n(1 comment)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff2d01ee02144d05a8d763a8a23e39a7 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2d300a9f0d94064bdd6a4d936cbacc4 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e971f817dfe14e03a11bae4fa2d836a1 : SUCCESS in 9m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c08e56306124c4199fd11534b5aabda : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a82e31c083547b085490346227debeb : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59637d93cade45f998b20b08ba81e64f : SUCCESS in 8m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cab3d134b89a4b7dab38c43db3eebe3f : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9bb95343800e4024bbdf6496f22b9e89 : SUCCESS in 10m 20s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8aa2c195986a4d8c81c758148406aae5 : SUCCESS in 15m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3009f323e48e4a30bafa2ed714858169 : SUCCESS in 1h 01m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/314231bd85ad49b58b06ac37b46f07f5 : SUCCESS in 43m 55s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/278c3386a8b9499196b84db20c0010b7 : SUCCESS in 1h 04m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/046bea8c07ee465ba9eacb47a7058289 : SUCCESS in 17m 36s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/237156fd1ece42849b5d787ee4bc4b09 : SUCCESS in 13m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d10678e2a0074135a68478f27585d281 : SUCCESS in 2h 03m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3917757a326949d38f58b83e719c60a7 : SUCCESS in 27m 15s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9be6dc21a1f49e0b08809eb22de1f86 : SUCCESS in 24m 53s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d221590cb209425cbb698eb2f88f0390 : SUCCESS in 25m 54s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/8904417894ae4fe68fbbf2d60e4d38c8 : SUCCESS in 13m 42s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0597fa35e701447f910242de858d7b7b : SUCCESS in 14m 13s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"405a74277ce067466f7d81b31b41174a9fbed7f9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-29 07:51:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"558b468aafb7cfc4a1c44c95b9de7a7b15b0c8c8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-29 08:09:40.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"15444c180983478f0b2d09a79316ccd607d27d60","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 08:10:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cd3f653dca3b46ee532e3c40e1e33ae67921aacb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 09:58:54.000000000","message":"Patch Set 3: Verified+2\n\n(1 comment)\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcacc31d2b9f418d8b0681060fbbfbd2 : SUCCESS in 8m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/00ff95cc99d342de9125b9899ab013a0 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d87341a7e6645e2acef4be38aa959d0 : SUCCESS in 6m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/097e061753b3455388bf4df865fc3dc2 : SUCCESS in 8m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4563f8c0704243fd938c3ed0788a2891 : SUCCESS in 7m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eb7d0bd88f154bba8c99b5e597a93f04 : SUCCESS in 11m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/41a677e3abdc4a04afdfcf11a4ebd9bf : SUCCESS in 19m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/670ba25d1e2046079216fe441853692e : SUCCESS in 1h 03m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6acca2fec0444c0d883fbbb6476993a2 : SUCCESS in 46m 08s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/2d853ea34b6d472d8648eeb49b416676 : SUCCESS in 1h 01m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/35f74227a34b45859a492915a14b5a65 : SUCCESS in 16m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/860df40f624f4a66b35f1a0998950b32 : SUCCESS in 14m 36s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f330e9894ee94feaa2954123f6735319 : SUCCESS in 1h 46m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c7143263f95d4847884095d477259d15 : SUCCESS in 27m 11s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/785e03694e2c4953b533281ccd977bc9 : SUCCESS in 23m 40s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2e002007724243b685b3d689711be13c : SUCCESS in 26m 09s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f8dc5c47e5814b219426deb7a8e05129 : SUCCESS in 12m 43s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae23416b79e84027ae584e71f2df823a : SUCCESS in 10m 13s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":3},{"id":"1333931de2e02508074915f5bcbcdad0730a7314","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 09:58:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"cd5047e1c23b426cacb70dc814bdc65f7d21c6b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 10:00:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f565580dcc174ff5b83507382d1abe4c : SUCCESS in 54s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"df00ba22e7e9b7bbbd299e669d5219862f57cb08","revisions":{"b9e58cd10d2fd04a0fa166e14b9416f5f89c36aa":{"kind":"REWORK","_number":1,"created":"2021-03-26 17:26:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/95/783395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/783395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/783395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/783395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/783395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/783395/1"}}},"commit":{"parents":[{"commit":"775620733bc2d848e5eeac3e18a2b0734e750a83","subject":"CI: Fix IP address detection in baremetal compute test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/775620733bc2d848e5eeac3e18a2b0734e750a83"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-26 17:24:05.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-26 17:24:05.000000000","tz":0},"subject":"CI: increase Ansible Galaxy retries \u0026 add delay","message":"CI: increase Ansible Galaxy retries \u0026 add delay\n\nWe still see flakiness when downloading content from Ansible Galaxy,\noften HTTP 520. This change increases the retries from 3 to 10, and adds\na 5 second delay between attempts.\n\nChange-Id: I0c46e5fcc6979027dc6f1bc5cc49e923a205f654\nRelated: https://github.com/ansible/galaxy/issues/2429\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9e58cd10d2fd04a0fa166e14b9416f5f89c36aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9e58cd10d2fd04a0fa166e14b9416f5f89c36aa"}]},"branch":"refs/heads/master"},"55f42851c7b4b8a959b3c068017cb256b958f9a2":{"kind":"REWORK","_number":2,"created":"2021-03-26 17:29:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/95/783395/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/783395/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/783395/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/783395/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/783395/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/783395/2"}}},"commit":{"parents":[{"commit":"775620733bc2d848e5eeac3e18a2b0734e750a83","subject":"CI: Fix IP address detection in baremetal compute test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/775620733bc2d848e5eeac3e18a2b0734e750a83"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-26 17:24:05.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-26 17:29:01.000000000","tz":0},"subject":"CI: increase Ansible Galaxy retries \u0026 add delay","message":"CI: increase Ansible Galaxy retries \u0026 add delay\n\nWe still see flakiness when downloading content from Ansible Galaxy,\noften HTTP 520. This change increases the retries from 3 to 10, and adds\na 5 second delay between attempts.\n\nChange-Id: I0c46e5fcc6979027dc6f1bc5cc49e923a205f654\nRelated: https://github.com/ansible/galaxy/issues/2429\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55f42851c7b4b8a959b3c068017cb256b958f9a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/55f42851c7b4b8a959b3c068017cb256b958f9a2"}]},"branch":"refs/heads/master"},"df00ba22e7e9b7bbbd299e669d5219862f57cb08":{"kind":"REWORK","_number":3,"created":"2021-03-26 17:34:54.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/95/783395/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/783395/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/783395/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/783395/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/783395/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/783395/3"}}},"commit":{"parents":[{"commit":"775620733bc2d848e5eeac3e18a2b0734e750a83","subject":"CI: Fix IP address detection in baremetal compute test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/775620733bc2d848e5eeac3e18a2b0734e750a83"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-26 17:24:05.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-26 17:34:40.000000000","tz":0},"subject":"CI: increase Ansible Galaxy retries \u0026 add delay","message":"CI: increase Ansible Galaxy retries \u0026 add delay\n\nWe still see flakiness when downloading content from Ansible Galaxy,\noften HTTP 520. This change increases the retries from 3 to 10, and adds\na 5 second delay between attempts.\n\nChange-Id: I0c46e5fcc6979027dc6f1bc5cc49e923a205f654\nRelated: https://github.com/ansible/galaxy/issues/2429\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/df00ba22e7e9b7bbbd299e669d5219862f57cb08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/df00ba22e7e9b7bbbd299e669d5219862f57cb08"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
