)]}'
{"id":"openstack%2Fkayobe~783546","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~Ib8c71fb4d775f5a234a433bf61434e8e46faf047","project":"openstack/kayobe","branch":"stable/victoria","topic":"fix-extra-kernel-options-stable/victoria","hashtags":[],"change_id":"Ib8c71fb4d775f5a234a433bf61434e8e46faf047","subject":"Support extra kernel options specified as strings instead of lists","status":"MERGED","created":"2021-03-29 08:06:55.000000000","updated":"2021-03-29 12:16:56.000000000","submitted":"2021-03-29 12:15:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783546-fix-extra-kernel-options-stable/victoria","cherry_pick_of_change":780585,"cherry_pick_of_patch_set":1,"meta_rev_id":"c26a7916d35ec4c4cf603e954e865e8c8297888b","_number":783546,"virtual_id_number":783546,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-29 12:15:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-03-29 08:07:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-29 08:53:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-29 08:53:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-29 08:06:55.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-03-29 10:06:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6d559e72dae1f725ce26d89bff60937ac109276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-29 08:06:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"dd3dcdc1226ed5eddfcefd35f2b067059bc4874d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-29 08:07:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"77cd5ad0a37cab38d80a4314eb20d6039e42fdeb","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-29 08:53:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"704e1ee6968778e62b34df882868576c049cedf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 10:06:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/561e8f25402948e4a28ac232c4413cb0 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b90c3215035f4940b40f0c2247e2a755 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bf1549d1eaa485d8669e294337cd785 : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/736d08ec8b5a4d22b8563bd8b99cbace : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d6217f76249476dba005b69721d47f2 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/75e5d6fbc03a48468b798e54f62c4f6f : SUCCESS in 5m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/de5839d7fe514f3bbecb98654dcc6221 : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9364b7acd6c84e18aba55246934476e0 : SUCCESS in 9m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94627f8670864d5d9cf4c27301526061 : SUCCESS in 19m 56s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/64ec9c08ae924acfa98c101d4ca667e0 : SUCCESS in 1h 00m 28s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ae81cae07aa74bb6a6fcc5b6f28d5713 : SUCCESS in 1h 03m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9586891239694d59af91f7870b9aadfd : SUCCESS in 16m 05s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d34204b7fb58445a8881b446ec16710f : SUCCESS in 1h 56m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/34f6b8bafee34da692b76d9ece690ff0 : SUCCESS in 26m 33s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/14f9772803da43ec83320cfb219027d8 : SUCCESS in 25m 31s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/54345308e4e9416e8a26ba53539ad78c : SUCCESS in 10m 26s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"e27ab8c8e640b2a3800b93a467a2301da883b893","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 10:07:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3b913431a58d6408fe60ab8109735a13b2c8a1ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 12:15:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca4fe372d82641059b94383864c2037d : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c9b51392bde4058b48402bb58144cee : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22db1918b2e04a29901e1d2662d50306 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc192a0e939345c0a6909fa1af40fad7 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12e83b3c4be444df8c1ac04a73bfe1c4 : SUCCESS in 3m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b9ec3a5e7dd74bf6bc9d1943799129e7 : SUCCESS in 4m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/eaa5b09dfd5c489083551b2c72d83329 : SUCCESS in 7m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6a6c5d4a2008469fa7b0bfbc9383905c : SUCCESS in 22m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0d391a097627461b9b0ca899d491d38d : SUCCESS in 45m 51s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/40fafefc21174876ab28f030f48979ca : SUCCESS in 1h 03m 55s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7566ca72d89742ccadb4e2f9d7b3eabb : SUCCESS in 11m 39s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e3510a4a1fb64665bdd47e74efc84f53 : SUCCESS in 2h 06m 12s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ba2abaebc7e7465ca10b207cfbb5be0a : SUCCESS in 21m 40s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4c9653e022c143d492cb1497771f0fdc : SUCCESS in 24m 03s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/803f95f353cf4658a83865019294552d : SUCCESS in 8m 48s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"4ad5044f26f1aaf2f33f393de7f36776450462ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 12:15:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c26a7916d35ec4c4cf603e954e865e8c8297888b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 12:16:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6de1596fe282452b9a3be5ca1c53d79e : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ff7e7548858b4fa689d08a1b0126af87 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e3f49dc4d4b14db8bbcd297028c6c4cc50ebbaab","revisions":{"e3f49dc4d4b14db8bbcd297028c6c4cc50ebbaab":{"kind":"REWORK","_number":1,"created":"2021-03-29 08:06:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/46/783546/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/783546/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/783546/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/783546/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/783546/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/783546/1"}}},"commit":{"parents":[{"commit":"46cd7e1337c8053659b5056d37cb64f23defb53a","subject":"Fix formatting of extra kernel options for Bifrost","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/46cd7e1337c8053659b5056d37cb64f23defb53a"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-03-15 13:29:07.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-29 08:06:55.000000000","tz":0},"subject":"Support extra kernel options specified as strings instead of lists","message":"Support extra kernel options specified as strings instead of lists\n\nChange-Id: Ib8c71fb4d775f5a234a433bf61434e8e46faf047\n(cherry picked from commit 1c8507113a08536ff68477438f0c80e1b957c03e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3f49dc4d4b14db8bbcd297028c6c4cc50ebbaab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e3f49dc4d4b14db8bbcd297028c6c4cc50ebbaab"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
