)]}'
{"id":"openstack%2Fkayobe~783550","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I0c46e5fcc6979027dc6f1bc5cc49e923a205f654","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"I0c46e5fcc6979027dc6f1bc5cc49e923a205f654","subject":"CI: increase Ansible Galaxy retries \u0026 add delay","status":"MERGED","created":"2021-03-29 10:41:59.000000000","updated":"2021-03-29 19:05:35.000000000","submitted":"2021-03-29 19:04:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783550","cherry_pick_of_change":783395,"cherry_pick_of_patch_set":3,"meta_rev_id":"f02e95522781518defb8c1dbae0e6b51bc7d2a59","_number":783550,"virtual_id_number":783550,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-29 19:04:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-29 14:57:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-03-29 16:29:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-03-29 16:29:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-29 12:20:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-29 14:57:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-03-29 16:29:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"318c0134063829ac9c3aa855f31f65ff5e1da821","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-29 10:41:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a9d94a186d785130d3ff2d12be37e1ef3d6dc774","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 12:20:17.000000000","message":"Patch Set 1: Verified+1\n\n(1 comment)\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe5f67bed69c400f9e39771fca2eeb4b : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b5f9bcfc57a948359d2741abb828f86c : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bba4f326fae34154ac2bb4cdf3a15fb3 : SUCCESS in 5m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a7121313af8646fd81ed56e8961ee363 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc44d954c19b4bcfa659cbafe3f2be12 : SUCCESS in 7m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0a913a4ae89843c290ef2984245da39b : SUCCESS in 7m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cd007a52068f49cab88b2b82fd08e16a : SUCCESS in 9m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/93b9152309a2412198d4cb3bfd890ba5 : SUCCESS in 16m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0e5f20844a1544f797e3038f864e0b4d : SUCCESS in 56m 34s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0419eab7d4e440d9a959b41e315705b5 : SUCCESS in 15m 52s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b1f569c18f91497caa7b50626163eb97 : SUCCESS in 1h 37m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/20919af394fe49089f68aced664ab642 : SUCCESS in 24m 14s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e5dde2010016411c812d49fdf825ad37 : SUCCESS in 26m 04s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/a65e9ad3e3684c42b15d0d7269d27b3d : SUCCESS in 12m 29s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"3d28890450b1a84ad1d935be5c05ede27915623c","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-29 14:57:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e7cfc73171728ba19588fe09cc91a403f81c1cb2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-29 16:29:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a0cdc1cdc8f035de7681514063a0effd88c6d00a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 16:30:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"df20dae2b03563aa4276e5322c1cabfade82883f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 19:04:08.000000000","message":"Patch Set 1: Verified+2\n\n(1 comment)\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0c3985865e6460590816ac2f80e769f : SUCCESS in 8m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/32159955478048cb813492c54d646d6c : SUCCESS in 7m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/07924da3aaa140009734c01c28301138 : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3587115ae9064ce781e17ca20a05a9a8 : SUCCESS in 12m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d7de866133cf4c3580137563c6cd56c7 : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/43677baaa3b94de6aa78e4e4ecfe3ce5 : SUCCESS in 8m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e60c97af47e0471bbc7f3d207023ab3e : SUCCESS in 16m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/56c929f259c445a486637c0589955995 : SUCCESS in 1h 19m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/78dcb33da5bd428ba005da3511d16424 : SUCCESS in 19m 07s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bcbeb827f14145bb855f245258cc2fe0 : SUCCESS in 2h 30m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ec86d7ffdaa74b4687f1669a615bdfe6 : SUCCESS in 25m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bed38429040a481b96abb1e8d19a5cb0 : SUCCESS in 26m 33s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/36ef43ceb1204405bdc232a03b56393e : SUCCESS in 10m 45s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"b3f0be319b808a6fcbff26d38594f6e0650ebd67","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 19:04:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f02e95522781518defb8c1dbae0e6b51bc7d2a59","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 19:05:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adedccc567d243aeb3ed55a03ace8a5c : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5e8bc03746b1be704c88f7586edbb3cdd3de628","revisions":{"d5e8bc03746b1be704c88f7586edbb3cdd3de628":{"kind":"REWORK","_number":1,"created":"2021-03-29 10:41:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/50/783550/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/50/783550/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/783550/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/783550/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/50/783550/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/50/783550/1"}}},"commit":{"parents":[{"commit":"ab039ef64094d89f7049aa5a42e7ccd3c992f023","subject":"Merge \"Update seed VM image to CentOS 8.3\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab039ef64094d89f7049aa5a42e7ccd3c992f023"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-26 17:24:05.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-29 10:41:59.000000000","tz":0},"subject":"CI: increase Ansible Galaxy retries \u0026 add delay","message":"CI: increase Ansible Galaxy retries \u0026 add delay\n\nWe still see flakiness when downloading content from Ansible Galaxy,\noften HTTP 520. This change increases the retries from 3 to 10, and adds\na 5 second delay between attempts.\n\nChange-Id: I0c46e5fcc6979027dc6f1bc5cc49e923a205f654\nRelated: https://github.com/ansible/galaxy/issues/2429\n(cherry picked from commit df00ba22e7e9b7bbbd299e669d5219862f57cb08)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5e8bc03746b1be704c88f7586edbb3cdd3de628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d5e8bc03746b1be704c88f7586edbb3cdd3de628"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
