)]}'
{"id":"openstack%2Fkayobe~783692","triplet_id":"openstack%2Fkayobe~stable%2Fvictoria~I18f67eda4e48058f3f402b9d0e692d6eeb02855f","project":"openstack/kayobe","branch":"stable/victoria","hashtags":[],"change_id":"I18f67eda4e48058f3f402b9d0e692d6eeb02855f","subject":"Avoid fact gathering for group_by","status":"MERGED","created":"2021-03-30 08:33:22.000000000","updated":"2021-03-31 18:49:12.000000000","submitted":"2021-03-31 18:40:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"783692","cherry_pick_of_change":778720,"cherry_pick_of_patch_set":3,"meta_rev_id":"7f47894e9bf6591c24984cd3b0e0110c89cf0c16","_number":783692,"virtual_id_number":783692,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-31 18:40:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-03-30 11:46:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-03-31 08:19:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-03-31 08:19:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"CC":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-30 11:46:20.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-03-30 13:08:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-30 20:26:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"}],"messages":[{"id":"973f7a0d65a62fc20c6869a5dd2afd4f367364d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-30 08:33:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"811b3e12196d4ff190e9d5f1b0eb816f67e347d8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-03-30 11:46:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2ca219a21d3e835d7153412a1839aea3211a3566","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 13:08:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c306c7055fe145d7ae2d04e1432cfff7 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb51c1ac0dbb455cb72a9b650a5785d6 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b36c7562159b4437986ec81f81687596 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c4dde4274704266b96a85df55ed58ad : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39944152eb8e455090bab44ef972b033 : SUCCESS in 6m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6febbdc12dcc434faf5aa2e3998d61d1 : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3fe6a38cd0e74abaa859ffc54bcee1a8 : SUCCESS in 11m 46s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3fac336378b74456ac822cf3b4435167 : SUCCESS in 19m 33s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ff57f9e4a4f4437caab9c2170663dffe : SUCCESS in 45m 46s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/111e3136665c4b46b51ea852c3742c2d : SUCCESS in 1h 05m 01s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d1fa194df0774db8bf842ef4756c53ab : SUCCESS in 15m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/855ed3efd4514b6c9fefc615f5ee0e43 : RETRY_LIMIT in 1h 30m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/920813fed878467aa8685365e3fab433 : SUCCESS in 28m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/da37497f983a459f8928d56abf66275a : SUCCESS in 24m 59s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/334b18601cf64710b59d75ccedb78787 : SUCCESS in 11m 10s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"7929293474bcf5f1314b1f714da668c8f1ed7779","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-30 15:02:57.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"56f4162448edaf70352848e8455b125653f3a265","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 19:00:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/34ab5bfee8d24ea2991f40ea112d3a97 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e392be5fda6d4d31bdb6e6a65d8ea492 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2896d637a43143e9a2e4679503212362 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b1966ce7f804aa0b84ba2bb055e0a3b : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c979c100791447c9815811acbf005b3b : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/dbc78ccddce147abb5a7b2d5eeb53fd7 : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dfd4c1a49e8a45f6bec02de08d6f023f : SUCCESS in 10m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/623432b6464e45139ab1e9de7b588512 : SUCCESS in 23m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f71fb3d5ef774c3d97b7b4d8d631ae0d : SUCCESS in 59m 34s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/4a5d82ac0f984a35a1caed4e2a31c21b : SUCCESS in 58m 03s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7893f52efb5b417a88c4be7345501f9d : SUCCESS in 13m 47s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ca7a8a91de3142118e1c52a46bbbb64c : RETRY_LIMIT in 21m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8067b2fd75754fdaad0d3067e432fdcb : POST_FAILURE in 36m 48s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bbd94e9c59814b078ef894fc3a3a2ffc : SUCCESS in 24m 03s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/6fd7279ddf63481ca2d2d978e8b64c30 : SUCCESS in 11m 56s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"e94e251f4f7b5e0cf185bbf210e6110c49bdaccc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-30 20:26:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"8ac476911018fcc40b6ac5037a01b367347feb3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 00:06:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7486de5f91f14f6782b6ca15502c6b16 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2467373577b740019d43e77125e2430a : SUCCESS in 6m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/613a9e6aecc14938a558f5e989d2caf4 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b331ee1004a34e67b671bac0fc1abcb3 : SUCCESS in 7m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccb78a73f48545e2a3342168a8a2a848 : SUCCESS in 7m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cf5522ba8d074a2e8c2e3286ef415dbc : SUCCESS in 7m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/75773d867bc94d08852768dc7be4f271 : SUCCESS in 11m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac17c3f009494ac8973a56f4b1443f2b : SUCCESS in 22m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/38f9223c074240089afacce67d5ae527 : SUCCESS in 55m 17s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/b1d4119069e849b18ca8b8b344594ec9 : SUCCESS in 1h 04m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/335fc571a7e7446fb8b835bd0663ffa3 : SUCCESS in 13m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ebaee97000874779b7e575d08f8fe241 : SUCCESS in 1h 51m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b8e6f1585212455ea4807b3eea51f53b : SUCCESS in 24m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f6033450b0d84da38724361493403462 : SUCCESS in 25m 14s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/21bebf9794934ba28d356d9e8e1e6091 : SUCCESS in 10m 37s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"4293b3bf4575aea59a70e97af4f361534c34fa61","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-31 08:19:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b64299c23adbd5f33eced93c191cdfe576a7cf1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 08:21:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"299e5628c5ce17f65d5a91f45d5b35eff0bdabf2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 12:46:54.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7cb5f9190b74159aab7002f0125438f : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4174e1694080405494555702e305d5bf : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28df12890fbc447a8b7f0fb09f0f40f1 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/963489f5048443c78fb4e747c7f71201 : SUCCESS in 7m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/84866d34e25a484f869a791ecc73c556 : SUCCESS in 5m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/926cae8f8b044f81aba654c2c4782cad : SUCCESS in 9m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/764f83f1fe0b4b7aa2af4117bbdd6007 : SUCCESS in 19m 47s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/925ad4330aa54179970b253b4f2160e8 : SUCCESS in 1h 11m 55s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/d83519c764f64d428d606b15d73c2aaa : SUCCESS in 1h 18m 12s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/338656c00df14befb9e23a52a0a38d04 : SUCCESS in 18m 00s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/beafbba013984fc7b8e2bda6fb2a4c78 : RETRY_LIMIT in 1h 23m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/40a28c0823474c8e8e32fbc5b88748c4 : SUCCESS in 40m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d0e2952a44af424997a80f32bc2b34f9 : SUCCESS in 31m 11s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ee5c2852c1d343b8afc2be33b013d00b : SUCCESS in 14m 42s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"1a0e5b53a5e9e7977f4c0c7f84d67f23761b2dd8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-03-31 14:03:00.000000000","message":"Patch Set 1:\n\nrecheck\n\nNo space left on device","accounts_in_message":[],"_revision_number":1},{"id":"7b7b014b854b13a411f4603660a948331bb3e6c0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-03-31 14:05:54.000000000","message":"Patch Set 1:\n\nrecheck\n\nOut of disk space","accounts_in_message":[],"_revision_number":1},{"id":"0692dc9b5924aa36ce5e6d32c9a449e705206363","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 16:07:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fafce75bafb4ef382961e39679ba3ff : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edb1c1c6340046168c9d9e0524e2649d : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65a6e754b0f74eeba0fe301885e5a8d1 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ca6680993ea44adbb45870add4d24854 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/642dcff4c8234792ba304f1c396fe235 : SUCCESS in 6m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/79cf8a7e7c7a41f6ac656fc1998b33dc : SUCCESS in 4m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/23ce655bf69b45cea31e67370c42dea6 : SUCCESS in 9m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5b57bf346f464a42882e1fcb5662f030 : SUCCESS in 16m 11s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6419d6f951bd43ffb2aa1e170172dc1e : SUCCESS in 1h 04m 17s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/dfac2a70989d459da08fdeba9653a30b : SUCCESS in 1h 00m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/901865bb866f4f1faf6aa7eeb991d348 : SUCCESS in 16m 19s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/03da96998fa14aa58ac37b3db9d240c4 : SUCCESS in 1h 58m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a46cd8c32f354e1f9aef78b55645c199 : SUCCESS in 27m 47s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9b7fefe2c9fd492b9ecaf1bd4753ad05 : SUCCESS in 27m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/82ff4b764ea14bfa984fdc44fa97e35e : SUCCESS in 11m 15s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"83c6923995265861e293b6bbcf2ba7e7f408894b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 16:08:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c84ac2c2fdc2bdc4b91351ec3daa799c480e0858","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:40:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa2dd92fd6c1410bad0de2058bf3bd47 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e4f6bf8b571142aaaa129b55ba9d9fde : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b52a21673ea94a039262278c2254c4d8 : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/307b0bb6355543af847ff3e11ef8d442 : SUCCESS in 7m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb9407495e01450f9c9f02b1f9537768 : SUCCESS in 5m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6c82b7c2d6f9438984078e805fe0c531 : SUCCESS in 9m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef4110e1db9f4ee5921d3f2ebd71f485 : SUCCESS in 17m 45s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d314e13ef5a34694b31af9e2576fc36d : SUCCESS in 47m 51s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/4282b850c87e4dee991058f8ae220683 : SUCCESS in 1h 57m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f3f9c49c2d454ea3a285d64ddb5c5219 : SUCCESS in 18m 04s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/71e64f7dce9840ec9794fc35fca03fcc : SUCCESS in 1h 50m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f7a2b8a6fe934ac6ac34a071fae8485f : SUCCESS in 26m 37s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/42ad60583ee941b68915867c11d56942 : SUCCESS in 25m 26s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ba215620e4cc4d4dbcc4c25592ac02b7 : SUCCESS in 10m 44s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"02c59a5a6dd4ee655af7bdcd3e83175b61c93cf5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:40:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7f47894e9bf6591c24984cd3b0e0110c89cf0c16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-31 18:49:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ed93e8902004047bcb61ed88c2d4d45 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"71ee0ef6ce3ee39a8322a521fcccc4a49f11fb2f","revisions":{"71ee0ef6ce3ee39a8322a521fcccc4a49f11fb2f":{"kind":"REWORK","_number":1,"created":"2021-03-30 08:33:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/92/783692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/92/783692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/783692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/783692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/92/783692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/92/783692/1"}}},"commit":{"parents":[{"commit":"f9a96e5f6a60be2cc4f86dec815ff1631ec45834","subject":"Merge \"CI: increase Ansible Galaxy retries \u0026 add delay\" into stable/victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f9a96e5f6a60be2cc4f86dec815ff1631ec45834"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-04 14:54:13.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-03-30 08:33:22.000000000","tz":0},"subject":"Avoid fact gathering for group_by","message":"Avoid fact gathering for group_by\n\nThere are a few places where we have a play with one task that uses the\ngroup_by module to generate groups that are used in a subsequent play.\nThese plays do not need to gather facts for their target hosts.\n\nIn particular, the kolla-openstack.yml playbook should really be able to\noperate without access to remote hosts.\n\nThis change also adds changed_when: false to all group_by tasks, since\nthey show as changed but haven\u0027t really made any changes outside of\nAnsible.\n\nTrivialFix\n\nChange-Id: I18f67eda4e48058f3f402b9d0e692d6eeb02855f\n(cherry picked from commit 388dfa369c9608f252dc946123c1801999931f6f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71ee0ef6ce3ee39a8322a521fcccc4a49f11fb2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71ee0ef6ce3ee39a8322a521fcccc4a49f11fb2f"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
