)]}'
{"id":"openstack%2Fkayobe~784448","triplet_id":"openstack%2Fkayobe~master~I41d9169f0312108e42bb12c52c6c7ee2509cf2ab","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I41d9169f0312108e42bb12c52c6c7ee2509cf2ab","subject":"Support additional pass through host variables","status":"MERGED","created":"2021-04-01 16:43:38.000000000","updated":"2021-05-18 13:25:42.000000000","submitted":"2021-05-18 13:24:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"784448","meta_rev_id":"447eb54b715b50f67d65a63917cd155876a0e39d","_number":784448,"virtual_id_number":784448,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-18 13:23:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-05-18 11:41:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-10 09:54:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-05-18 11:41:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-01 18:36:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-10 09:54:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-05-18 11:41:03.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"3fce72f2e301e457c0ba72da39e3c4a1d63d1fc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-01 16:43:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e40374fe18e91620dd5bc7c94301a9e59db0c8f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 18:36:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e812267502f24048b36924017e09b99d : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a07357fb7f194fbab454d20fee8d26a0 : SUCCESS in 7m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8c592e0f3784444b9e4c9b022e070bd : SUCCESS in 7m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f02cfb36f714d32a0edf9d784763f62 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/169e3802a4b641c1a78d71aba0709e8f : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4896494db2be4fbb96ccce7a576d6c46 : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d1a85cc13804f449b56a3b4711bbb2a : SUCCESS in 6m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e3fe2365862f4bfdb1d67b907eefee0e : SUCCESS in 7m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/12589fee277f489ab2f023a4a44b5997 : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cc7dc0ef46974f6ca89682f0154d55fd : SUCCESS in 21m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e5c6605e714a41bbbf8da3db2f9e26b7 : SUCCESS in 1h 21m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8da35b75a7c94435aeeb8b1816ab545d : SUCCESS in 46m 28s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/4ed27f8e85d747d29b125d63a3468286 : SUCCESS in 59m 07s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0192935ced0f4843b2c65b1df64070c1 : SUCCESS in 17m 50s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d14f9e4ffa0a498c82bebd6bd442913a : SUCCESS in 13m 16s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f4c38966f6e04f1994e65b2835bd8569 : SUCCESS in 1h 50m 09s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/433156a5277a454d8be8ca259b39ab0f : SUCCESS in 25m 54s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4805d06a67d44f06a223a9c293a57d45 : SUCCESS in 25m 57s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/af439e8a403642ef8792336bb9b2f2c6 : SUCCESS in 33m 49s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/c3ac4f74f29a494ca6d6d19dcbc28686 : SUCCESS in 12m 54s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ba8afde994c4c5d937aa736918e4747 : SUCCESS in 9m 41s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"40f114e6a9333d99bf2af649c4b294003f0e2bcf","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-05-10 09:54:38.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6945801522483e42a14c51c5475162baf9b7fa63","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-05-18 11:41:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a1b3b2e1385650700cea7332563c59c607021fb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 11:41:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5ce82762a15ba78166f702172d3bf642778a2e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 13:23:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/743c07d8963d4f3aa5191e1fb6010d92 : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f0b9f090e4a4a849afe0f2434745c78 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0bea509b06a4d76b0fd6e467116d9b8 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/102c0cb5a999419683c2f08613ab9755 : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7800a6283be94cd4a6ce12bf4daee9f2 : SUCCESS in 5m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f6d5a3600b0d4f8291002af9966798f6 : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/702cde8df87b47db9c29f62f6a9da13e : SUCCESS in 13m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2f1e514674cd468d96cee0aabd4cd805 : SUCCESS in 20m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/2280f9588aaf4cdeb970ae4359e0c279 : SUCCESS in 59m 30s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0c6004a152104220bf1fc0f62064c0e7 : SUCCESS in 42m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dc329445e5f244da902b21f54aaf8656 : SUCCESS in 1h 01m 42s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4ab2a13998a546719bdcfe090ae79117 : SUCCESS in 16m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d4a9a602a344e038be18c44fbd5eaa3 : SUCCESS in 12m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6dbbe9d3892f40a39ff1ec83cc889578 : SUCCESS in 1h 41m 11s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8301ea064ece462d8f6e1e7728d31576 : SUCCESS in 29m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/caed33c4452c4b8684db423d91138c23 : SUCCESS in 25m 43s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/76a2dd11faf94230a80c62f1de3ee6ee : SUCCESS in 24m 14s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bf609ed3edf14496bfb943bbc680829f : SUCCESS in 13m 25s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dd87bea127404731b794e751b1c73346 : SUCCESS in 10m 16s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/kolla-feature-flags.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh","accounts_in_message":[],"_revision_number":1},{"id":"e66550af5f154fbd9e47b2e3ad19fe64ef75f137","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 13:24:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"447eb54b715b50f67d65a63917cd155876a0e39d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-18 13:25:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd044088a2d64653863c969249417834 : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76c4076db8324e2b800931e9e19a4cb9 : SUCCESS in 1m 19s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"45a7963810830f3fe17ce7869e0458356f31a947","revisions":{"45a7963810830f3fe17ce7869e0458356f31a947":{"kind":"REWORK","_number":1,"created":"2021-04-01 16:43:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/48/784448/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/48/784448/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/784448/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/784448/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/784448/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/48/784448/1"}}},"commit":{"parents":[{"commit":"f8575af9c07aef00f0f80cf732f33c7fbcbb28ab","subject":"Merge \"Avoid fact gathering for group_by\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8575af9c07aef00f0f80cf732f33c7fbcbb28ab"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-01 16:41:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-01 16:42:33.000000000","tz":60},"subject":"Support additional pass through host variables","message":"Support additional pass through host variables\n\nAllow passing through additional host variables from Kayobe to Kolla\nAnsible without overriding the entire list.\n\n* kolla_seed_inventory_pass_through_host_vars_extra\n* kolla_seed_inventory_pass_through_host_vars_map_extra\n* kolla_overcloud_inventory_pass_through_host_vars_extra\n* kolla_overcloud_inventory_pass_through_host_vars_map_extra\n\nStory: 2008797\nTask: 42201\n\nChange-Id: I41d9169f0312108e42bb12c52c6c7ee2509cf2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45a7963810830f3fe17ce7869e0458356f31a947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/45a7963810830f3fe17ce7869e0458356f31a947"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
