)]}'
{"id":"openstack%2Fkayobe~784564","triplet_id":"openstack%2Fkayobe~master~I215f927d8249aadd9cc7bb44893e0bdf5f26510a","project":"openstack/kayobe","branch":"master","topic":"remove-inventory-host-list","hashtags":[],"change_id":"I215f927d8249aadd9cc7bb44893e0bdf5f26510a","subject":"Fix help texts of inventory options","status":"MERGED","created":"2021-04-02 12:37:27.000000000","updated":"2021-04-12 19:59:45.000000000","submitted":"2021-04-12 19:58:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"784564-remove-inventory-host-list","meta_rev_id":"fbdd9080cca22362852768509739014bd5d92599","_number":784564,"virtual_id_number":784564,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-12 19:58:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-04-08 09:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-04-06 09:35:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-04-08 09:06:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-06 10:56:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-02 14:28:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 09:35:34.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-04-08 09:06:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"e3f42646fd91339a703c10db930c8775a5afdd91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-02 12:37:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f39b27c56ba773a8d16a9d89ac924db25d9e53e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 14:28:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b11430d67994195ab6249dd5bb92db0 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6c226f852f04136b44881cbfef7381c : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c0fae3d87d514c54911f96bd13f6647b : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bd3977d21189416388930910c24ca0d9 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c5436975bc74c6fb25011e3c06df5e0 : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/622946805ac646d6a0c681f6217f2dde : SUCCESS in 8m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0bde78d8d94d484fb90b8ae88b52fc1e : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/08928ee0ff02472090e5be142783775f : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f4bb89556b6f46a1b6ea97dd6f6d5bbe : SUCCESS in 19m 14s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f31a5f267c1740019957ad7847e7dfc5 : SUCCESS in 1h 00m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/681b01b023834ba0a343c7d7be3d2ab3 : SUCCESS in 48m 22s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ba75dc4055ba4fd493c073df40c811d0 : SUCCESS in 1h 11m 06s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/58eb7733b6f04189a04b546586923073 : SUCCESS in 17m 14s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/170f99d8014445f58dea77415a711c95 : SUCCESS in 12m 08s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e78cff82791e4734a32e075540b3c225 : SUCCESS in 1h 45m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/744a05c0723d4db2a8c111246de7b4ea : SUCCESS in 27m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c40af7e3775e49599f65f43e3175bb61 : SUCCESS in 23m 08s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/bab70c812ead4fcf8a0b398481fcceb4 : SUCCESS in 24m 37s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/ef31107a0df14c899d273880929e3bdf : SUCCESS in 12m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/12ea831cc6c747afa572db9b7b0a38de : SUCCESS in 10m 02s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"f9fa158b85c47d1e6538793aa957d1e031402afa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-06 09:35:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"08ad0ba92edfd83e416a8ea7d74fc9bcfa2a1417","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-06 10:56:46.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"76c5968cc3fbcda05e449f248194fcd4161f5ed4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-08 09:06:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fb6f3dc4e617af47a7f71102c2b561f41fdc9516","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 09:10:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"464c6dd0d657e3fce1d5a7d55b1608fd467e5f05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-08 11:06:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec07ab2c9fc04b42b183a16af4c52d73 : SUCCESS in 6m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59f8e13362a3484c84e5f0bfa0c1f39b : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfa6696a09f947c68987ae25e9c9e817 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64835183ea4447d58d01a1713322ae89 : SUCCESS in 7m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c35f0ea3a5af4551acabe7ad9756c13f : SUCCESS in 5m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/03613a81642f440a998adfc60fb52bd7 : SUCCESS in 11m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7ba767eef0bf4416ad9c1d6124154379 : SUCCESS in 22m 39s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/5636f5fc489d45e2943dbc06a5a7ffa5 : FAILURE in 51m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/622371edfd3540dbb3d0eb99b2a0e0f5 : SUCCESS in 47m 47s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/486bfe2e700e4e03ae0726b758ee71ab : SUCCESS in 1h 10m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7804ea23a3194c398aa3cbf7fb1c28be : SUCCESS in 17m 38s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c3bb7125c354f15b2cc2ef98527b7f8 : SUCCESS in 14m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/86260f8f75f84dc494d48862b99fe2c0 : SUCCESS in 1h 49m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/bfa14583f0654d7c9ad7bce75d11bf7c : SUCCESS in 28m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5915bf6d45254d618cde4316e84dfab0 : SUCCESS in 23m 28s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5825e820c92549ecb5cf4b5d28fb4555 : SUCCESS in 26m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f7a10b79c85f4112b3f37ccb870538d3 : FAILURE in 7m 27s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0df096952bc84e6ba2d05a55929c66a0 : SUCCESS in 12m 03s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"e1c46f117e57c710961ee3c8e22f1347b6a64d60","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-08 11:26:54.000000000","message":"Patch Set 1:\n\nGot this in tenks when creating a veth pair:\n\nRTNETLINK answers: Operation not permitted\n\nI\u0027ve seen that error in other places on CentOS stream.","accounts_in_message":[],"_revision_number":1},{"id":"356a6b53a9e92dac574c69d85648b439b1205cc6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-12 15:54:59.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b1002b1daf63db0ae733f3996238d8550031bbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 18:04:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d5c9b80eddf64f169ea296199c5276a6 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97191f3006454511a7adeedff43094ff : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0cf42e58eba4f27abb01504c688ed02 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/711e3a4d86a04e0590c06923ac511757 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33f9552bbaa54c3c88058fd71c458772 : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0489766e5ea0429fbc88eb07aee8019a : SUCCESS in 9m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/57f936e44353411292c4ecee1a3b79e3 : SUCCESS in 7m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/afdfa5f7bb3846b0bfc9c3e8191e2a74 : SUCCESS in 11m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/01b4295969d1447f90ac8996a1001c20 : SUCCESS in 20m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d7ccd424cdf2487c97b55a4f60ddb501 : SUCCESS in 56m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ec26e7bf4a7471ebb6953e8daa74233 : SUCCESS in 44m 41s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0ec8341256e14348bdb5a5a3dee9329c : SUCCESS in 1h 03m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/4db3c13e4a624fa587088f91730a4aa4 : SUCCESS in 19m 11s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e476ce0ffe294c809b9a297937c5122c : SUCCESS in 13m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b98c7831f44f4ef6a2b214060cad1292 : SUCCESS in 2h 02m 44s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c579870383d947a98059106287c6a8e3 : SUCCESS in 29m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7c3e17961a64ddfba34517926e2ad03 : SUCCESS in 25m 17s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7ef2334f80954b39a9103c42544ff929 : SUCCESS in 29m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/0393cca6c12748898592e6a9bc5d0ed2 : SUCCESS in 11m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b43af9637ada4fd5b0de919df2ba30da : SUCCESS in 10m 18s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"116bfbec798ab9b0cf7b0cde0da09fe8db1a4263","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 18:05:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"682ed7942d05ceffeebc152d3b706fff9a30982d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 19:58:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d755516757c343259ae49831899ca496 : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e9d95f2f1fe84538a0abceb5d3e257e7 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ac36c5fa443243f5a0e260a05515c28c : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a84a46c3c589460881cff8ce52d2fd40 : SUCCESS in 9m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f980e76d03424eb8a78d684c85d97fa6 : SUCCESS in 6m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5d883a56ae15489caecb5dfc6dd8e3e9 : SUCCESS in 12m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6d629803d0ce4de7bc71115685a2369d : SUCCESS in 21m 58s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/809a9083ff3b4758a74866643cabd07e : SUCCESS in 1h 03m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9588c92ff5140f6a8d32db55d4e0d9e : SUCCESS in 48m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f9641ec9dd36417abbc7dc3c10545f8a : SUCCESS in 1h 07m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1a18941a8d46467dad57504deedb322f : SUCCESS in 17m 55s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d36336eca81043d39aeb811b0cce4668 : SUCCESS in 13m 08s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cf1b1362f68a43ffa51c62c846db46dc : SUCCESS in 1h 51m 46s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/53a4e481b37441e0b9f01e51677136a8 : SUCCESS in 33m 24s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8f615609dbb5496ca0919c54130f33e8 : SUCCESS in 25m 30s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/19bb7bebe8c844db9f2855264f195610 : SUCCESS in 25m 33s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c4b451a55bd0470f87d57445a1d4b87b : SUCCESS in 14m 06s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d203432ea7540fdb6379012999900db : SUCCESS in 10m 43s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"3c922532029e9e6364156499c618d755d234010f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 19:58:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"fbdd9080cca22362852768509739014bd5d92599","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 19:59:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18be39428c7a4e09baf01c4587377e20 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e0b180e13543682b0196c96e39f103cbbae697dc","revisions":{"e0b180e13543682b0196c96e39f103cbbae697dc":{"kind":"REWORK","_number":1,"created":"2021-04-02 12:37:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/64/784564/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/64/784564/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/784564/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/784564/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/64/784564/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/64/784564/1"}}},"commit":{"parents":[{"commit":"ea54f822999e6bc43e195f38cfd66caf05fc10ef","subject":"Detect multiple inventories automatically","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ea54f822999e6bc43e195f38cfd66caf05fc10ef"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-02 12:31:15.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-02 12:31:15.000000000","tz":120},"subject":"Fix help texts of inventory options","message":"Fix help texts of inventory options\n\nPassing the Kayobe or Kolla Ansible inventories as comma-separated host\nlist was never supported since we check that inventory directories\nexist.\n\nAlso fix the check for the Kolla Ansible inventory directory, which was\nusing the wrong variable (parsed_args.kolla_venv).\n\nChange-Id: I215f927d8249aadd9cc7bb44893e0bdf5f26510a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0b180e13543682b0196c96e39f103cbbae697dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0b180e13543682b0196c96e39f103cbbae697dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
