)]}'
{"id":"openstack%2Fkayobe~785031","triplet_id":"openstack%2Fkayobe~master~Iaa69e5303e554e6b30a23a3e84e36d09972888c9","project":"openstack/kayobe","branch":"master","topic":"systemd-networkd","hashtags":[],"change_id":"Iaa69e5303e554e6b30a23a3e84e36d09972888c9","subject":"Ubuntu: refactor network.yml playbook","status":"MERGED","created":"2021-04-06 18:55:30.000000000","updated":"2021-04-23 11:36:51.000000000","submitted":"2021-04-23 11:35:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"785031-systemd-networkd","meta_rev_id":"70c576457658bd83a5f9a9f7e1dd2cdb288a54eb","_number":785031,"virtual_id_number":785031,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-23 11:35:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2021-04-21 09:21:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2021-04-23 09:41:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-09 13:27:11.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2021-04-23 09:41:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 20:52:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-09 13:24:19.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2021-04-09 13:27:11.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-04-21 09:21:25.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2021-04-23 09:41:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"cf6619cfbd295d3c19a3dc7e6275c71c3ffd8b66","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-06 18:55:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2dec44de7f9101a58351e63394a266af32796449","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 20:52:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff253fc33d154c5e9f0840830e56f042 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55d6b1a31a7240ea99b338d4136c92d5 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04ec247831a24106bc77dc37ec789347 : SUCCESS in 6m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2873828e23a74491bef721766e3b25c0 : SUCCESS in 7m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cfcf5c17040c45faa916dab7f96ba47f : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3d4a765a13944b5a46221e380cc3d61 : SUCCESS in 8m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/68d1804bb6934861af54439dc6674a2f : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c465750c366c4a42b1ba00ecc35ca7aa : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/afb61bef71f246f5b93e9226e5ee5bf6 : SUCCESS in 20m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e55ad6103ef84ff08be3d37365bef057 : SUCCESS in 46m 55s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bffd659aad84fcca2ea94705eee8410 : FAILURE in 39m 33s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ab5ae678d001479b9a82bdde2423576c : SUCCESS in 1h 03m 52s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/da8c8c4f92d7484a8e446cca3ef3d1c6 : SUCCESS in 16m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94df9322192d4536a323b88507862b77 : SUCCESS in 13m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d6de93474378434caed5192e106ebb6b : SUCCESS in 1h 46m 28s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0a8d3a550a2146d68eeec463f5cf7ca4 : SUCCESS in 26m 22s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d48a349c52fd4bc4878b4b8e5b273610 : SUCCESS in 24m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3e4432ffee564ea1bd8ff4356dde5711 : SUCCESS in 25m 02s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/fdc8d912c4b54093b8a41d7fd4c56488 : SUCCESS in 12m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f811f0f34e0b45d5b9a9c7a5f562581f : SUCCESS in 10m 24s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"9d79351a74d08c55d58be5a7056135d2d53a109a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 08:38:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4dfc1e9ae71ef91f15bda1c1ced704cd3d67d18c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 10:32:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26dce7f481194d3eb7353a4c3df7daab : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be1972ca3e50402893d262e29fdc2c66 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f143093fb5647f89cbbb8547186b4b0 : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9e56b246bd404e858c5731d990b7b980 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/00e605e699ab4d76a0e5295bb792af6d : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/453e5e7a04a14afaa8d39288ae32fb90 : SUCCESS in 9m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/13a09937d9614add8b6163a1f0e84cfc : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f1443c24119c49afb95d554fb0cd0322 : SUCCESS in 11m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/956864a374ae4023a893f791b5bd2b34 : SUCCESS in 19m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/62b20da90cfe44b1bebd8ab88c4e27ec : SUCCESS in 1h 03m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3068d8b29c624fa0898a49b118ce3afd : FAILURE in 45m 30s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/ea40ad18482342cfb1b12bb2af5f5452 : SUCCESS in 59m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4f5317ddd26a4050899e16d30e7273b3 : SUCCESS in 16m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/959d2f59fd7a41e6a5f98c609a547d35 : SUCCESS in 13m 54s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a86272d216d44414926e3d659d5110aa : SUCCESS in 1h 47m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d2711637d4304675926d9e68b21c9aaa : SUCCESS in 29m 25s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aec8c7c65efc4ec7b0ff690efe4ab0cf : SUCCESS in 26m 30s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9d52afeef6eb493db546901e4f4263ba : SUCCESS in 27m 16s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/3c53e94de137492a9cc58f29669e58d0 : SUCCESS in 12m 41s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/443752c98344436e91595f486cfcccbd : SUCCESS in 10m 32s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"f0e900a213583ed041f16aadb63557029bbd0db6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 10:55:00.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"bd5d849383fe916dd143201c53b2ba445151c5e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 12:43:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3344b5e45bd14b8b906fe64a61f735c0 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85177c2cfab4428bb8a8df4af7388ba1 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fec7093a840f462fa5e9373f281668cb : SUCCESS in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74fbcba186ac4ee8812e33bcbab41340 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea4139b29b57481ab6195a4fb0e81c52 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a484232fc1b403abdb3e389a3e9ce86 : SUCCESS in 8m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e8b4d434d7974a0dabafdf2e9011447d : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f94a04430e644390bd671c0cce379ceb : SUCCESS in 11m 03s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f7a7cea3d29742779615d6ca00f79f63 : SUCCESS in 21m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3dd1d11b8d234c8eb8dc59b02678b97c : SUCCESS in 1h 02m 46s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9d171f6dd4934584a873345e1f4d44ee : SUCCESS in 46m 00s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/27c2256e763c4e86a36bf8bd7f47c729 : SUCCESS in 54m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4643b297167d41edaad47002a24138c4 : SUCCESS in 16m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0489acae4ad44fabb2257e694e20e87 : SUCCESS in 14m 55s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6f4827c33c7d40b895ee43d2bd40b5f2 : SUCCESS in 1h 45m 42s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/89332179ccef41198515eb94ac1930e9 : SUCCESS in 25m 30s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2e6efaf883440b5a703125a62566d4a : SUCCESS in 24m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2e774243a7b64a64814ddf2bc75186c9 : SUCCESS in 26m 57s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/83603721f8d4458b8d55c1ac9f6cae85 : SUCCESS in 12m 04s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fdab1b949c046efb9797085cbbff70b : SUCCESS in 10m 49s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":2},{"id":"3c8be7a6830c3531f5ec1971a5f3415c0efaeff7","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 14:40:16.000000000","message":"Topic set to systemd-networkd","accounts_in_message":[],"_revision_number":2},{"id":"5ad6c7504d0e3fe3aeec2d328bcf01e745abf5ea","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-09 13:24:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9d2b9ba0cfb8364f09f30b2ed38c88d5c0f54fa3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-09 13:27:11.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"97e582cdb97deb4cee4324a0d3d3c1beb7aeba71","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-04-21 09:21:25.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"accaadec3fc6c4741b5c9c1ad57698facf7785ff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-21 10:00:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"190670d84e20abaf1d8c5466c951e516fe9b855e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-23 09:41:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5e7dac4423b83f20586504ef13efe15f6de6a964","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 09:41:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e262a7d7ffae8a7dc66adfebb9ed305ac5848514","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 11:35:14.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f0129cc73ee4d94905ced78c9a1761c : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d366fb80186b43df99cb16931b7ed1e4 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc8a742ee68c414bad081de3b538b263 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dc5e04f881c4f549d5ab3644d0a6d8c : SUCCESS in 9m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5eaf48fc8354476eaa8a2c1e3642b579 : SUCCESS in 7m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8c852fdacf164a009e7b923003e7783a : SUCCESS in 10m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/be922332a4ec44fe86e50bfee334c183 : SUCCESS in 21m 09s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f9a9666a481a4afd814c379a1de0a5af : SUCCESS in 1h 03m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/316eb2c78c774a27a525f36c9f599749 : SUCCESS in 46m 37s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/016231432aaf47d6972fe64c1b23771b : SUCCESS in 48m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1e4a563deec44fc2ad9389afdc3e9125 : SUCCESS in 17m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6230dfff1044653a4ca7062391a2d61 : SUCCESS in 12m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c6e0453351cb47a99fd34899d122f269 : SUCCESS in 1h 51m 22s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8684b09634a04e86bb9c3039a277badc : SUCCESS in 27m 55s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6bd220cb33b54843bffce267b15dc98f : SUCCESS in 24m 44s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fc02e71b4d8f419ca5466793b8393433 : SUCCESS in 27m 09s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/d5d9f91aef06451bae37c1195f9cb759 : SUCCESS in 14m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da76eecc6414453f852f83631e0efd4a : SUCCESS in 11m 58s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"46967c760864f8a29014fa82c8b43716c597924c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 11:35:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"70c576457658bd83a5f9a9f7e1dd2cdb288a54eb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 11:36:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7702baf8df3e4c47b898a81b9be112a4 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ae2ed2215aeb6daee037d7376c883ae2830bce2e","revisions":{"e690ccfee1437e1c7c623cbc278ec88f62d2a5a4":{"kind":"REWORK","_number":1,"created":"2021-04-06 18:55:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/785031/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/785031/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/785031/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/785031/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/785031/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/785031/1"}}},"commit":{"parents":[{"commit":"97f2f3b5c7159c05f33268f01a9e10c10a68c82a","subject":"Refactor veth configuration role variable generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97f2f3b5c7159c05f33268f01a9e10c10a68c82a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 16:58:15.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 18:36:41.000000000","tz":60},"subject":"Ubuntu: refactor network.yml playbook","message":"Ubuntu: refactor network.yml playbook\n\nThe network.yml playbook is reduced in complexity, now performing\nvalidation before including a role specific to the OS family of the host\n(network-redhat or network-debian). This will allow the network\nconfiguration to diverge between OS families without adding excessive\nconditionals.  Currently the content of the roles is identical.\n\nStory: 2004960\n\nChange-Id: Iaa69e5303e554e6b30a23a3e84e36d09972888c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e690ccfee1437e1c7c623cbc278ec88f62d2a5a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e690ccfee1437e1c7c623cbc278ec88f62d2a5a4"}]},"branch":"refs/heads/master"},"ae2ed2215aeb6daee037d7376c883ae2830bce2e":{"kind":"REWORK","_number":2,"created":"2021-04-07 10:55:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/31/785031/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/31/785031/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/785031/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/785031/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/31/785031/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/31/785031/2"}}},"commit":{"parents":[{"commit":"97f2f3b5c7159c05f33268f01a9e10c10a68c82a","subject":"Refactor veth configuration role variable generation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/97f2f3b5c7159c05f33268f01a9e10c10a68c82a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 16:58:15.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-07 10:54:55.000000000","tz":0},"subject":"Ubuntu: refactor network.yml playbook","message":"Ubuntu: refactor network.yml playbook\n\nThe network.yml playbook is reduced in complexity, now performing\nvalidation before including a role specific to the OS family of the host\n(network-redhat or network-debian). This will allow the network\nconfiguration to diverge between OS families without adding excessive\nconditionals.  Currently the content of the roles is identical.\n\nStory: 2004960\n\nChange-Id: Iaa69e5303e554e6b30a23a3e84e36d09972888c9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae2ed2215aeb6daee037d7376c883ae2830bce2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ae2ed2215aeb6daee037d7376c883ae2830bce2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
