)]}'
{"id":"openstack%2Fkayobe~785032","triplet_id":"openstack%2Fkayobe~master~I77aec0160eb7e4dd763326bfe6e3d9a44b248108","project":"openstack/kayobe","branch":"master","topic":"systemd-networkd","hashtags":[],"change_id":"I77aec0160eb7e4dd763326bfe6e3d9a44b248108","subject":"Ubuntu: support policy-based routing in systemd-networkd","status":"MERGED","created":"2021-04-06 18:55:30.000000000","updated":"2021-04-23 11:52:50.000000000","submitted":"2021-04-23 11:50:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785032-systemd-networkd","meta_rev_id":"39f7c9c5be67d40da26e8268493662c18183b140","_number":785032,"virtual_id_number":785032,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-23 11:50:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2021-04-21 09:11:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2021-04-23 09:42:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2021-04-23 09:42:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-06 19:43:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-21 09:11:51.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2021-04-23 09:42:12.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"c59ae38c85fbd3afe807a7bcfa420f5f55562c87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-06 18:55:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02927342f821e5f8ea98e20bc5be7d6d8b98b4c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 19:43:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d0dea3c9cf524ecf94b286c229c23091 : FAILURE in 9m 50s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/661345d8b14c47f989852a0268141461 : SUCCESS in 36m 24s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/cecbcec1b2d84ed3980029e1945a0d7b : FAILURE in 21m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/de9cfede7e164f95ab718d7f2df630b8 : FAILURE in 13m 01s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/18cd33e50dbe420ca3eec351d49be852 : SUCCESS in 9m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a5e0f169eac699d547bcc68b47c002df0883b6be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 09:24:00.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"590a71c36e32304bda3eecc312ba67b8966ab1aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 10:29:40.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2a428dd0387544f0a843d1588ad338b4 : SUCCESS in 1h 03m 08s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58f24bd0ed504184913f9cfc7af42e23 : SUCCESS in 45m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/920ef80e5fc14b19955554a393b8e05e : FAILURE in 14m 01s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/378fac126de64c179c7d7af147c0dda4 : FAILURE in 13m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/268282d2021645c6803a089fa93bc772 : SUCCESS in 11m 09s","accounts_in_message":[],"_revision_number":2},{"id":"5a9b0e1d7a5e443793873b56109d638fff20b6c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 14:35:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"b750f30ef1105637366a35fae1606fe10ed1f25a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 14:39:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"2825a6cc0f7383f4680efd306b1e937fc7738ef1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 14:40:30.000000000","message":"Topic set to systemd-networkd","accounts_in_message":[],"_revision_number":4},{"id":"5b501da2ae68fe63f256662b798168256babed53","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 14:40:37.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":4},{"id":"ba46538b35f106098a569cf451c284989002fbc1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 15:48:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5c271533c3bf4972bc05941a9b508870 : SUCCESS in 1h 06m 11s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef9b79f070844485a849a72bd6f45aeb : SUCCESS in 48m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/78624d62c83243a6a227b11ce761354c : SUCCESS in 17m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7e82c63dda36491bbb2c8cb87428480c : SUCCESS in 13m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01706ac7fbb74b05bd88df81dff895d6 : SUCCESS in 25m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/84812b602e2d43c6b9a442658758726e : SUCCESS in 10m 55s","accounts_in_message":[],"_revision_number":4},{"id":"8eb98a6470aad44579a9b839d0232c38e1847d10","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 16:35:52.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"1f231c0b35370b3426631078bd5855429d88c89f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 16:36:12.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"db1d395832265a6333da903e995fb31f7e660932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-07 16:38:41.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"185b595115d081b49278317d24da0c71f51eb363","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 18:26:05.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0be3789022c42359e7b1f178314d09c : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57cded60c353414a863bdfd5ba00f1f6 : SUCCESS in 6m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9dc5029e7ae4242bddba8d69dcf30ed : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91121d0e4a8c4cecbf91ae99fbc9c9d4 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cea9cd090e73440f8cb46a357c57ce42 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af312ce957fd4eb0bf5bf02be9450d2b : SUCCESS in 8m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d95dacf4afce42dda306eedb7eb1761a : SUCCESS in 6m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/595d2835a6ea43b5adfe09e74cf0d930 : SUCCESS in 10m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c50243ffbb4b4ab0b30f6c3f2b7494f8 : SUCCESS in 18m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c56a229d9aee46f49514b46b7c69a61e : SUCCESS in 56m 40s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/42086aaf73ff407083f743185db0a1a1 : SUCCESS in 46m 11s\n- kayobe-overcloud-tls-centos8 https://zuul.opendev.org/t/openstack/build/5464bec48f1f48e7a888e5f675d9d001 : SUCCESS in 1h 26m 09s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cc792ef3a6c447cc8f33c708d90ccfe0 : SUCCESS in 23m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e83ea463637e4da3a94bd4a7ef4ebb08 : SUCCESS in 12m 06s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/648fd791b7564fa7a35d78a7fde10e8c : SUCCESS in 1h 39m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a776cf4f2dc74faab3df496dc391cd1c : SUCCESS in 25m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0de7f288ae0b4036b7648dff9f8562f6 : SUCCESS in 23m 27s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fa3353623c8a43da944864366db5c84a : SUCCESS in 25m 15s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4ed2720ec2ef46bf9c7c6ba085252bc9 : SUCCESS in 11m 54s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/140aa565348d4648a860c9c82a8ca766 : SUCCESS in 10m 04s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":6},{"id":"9a103e55a180680cf3c352bdfe17287d12fc0d97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-21 08:16:39.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"ad14b1ebd67a1aa76d5aa9c64437671d3b33c434","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-21 08:23:57.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"8f584a89c7b9794deb7162cab3caa2c9b6bbdcb1","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-04-21 09:11:51.000000000","message":"Patch Set 8: Code-Review+2\n\nLooks good as far as I can tell. Shame about the different formats.","accounts_in_message":[],"_revision_number":8},{"id":"5bfd2246fbf850c91e7283a8b99ccad5433c1d9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 10:05:03.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2618e25ce6e442d3a6f211cffec4323b : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f876f87e6ed64dcd8f70a96f6561593a : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1d645e226f324e9db3ad42a331f09e8d : SUCCESS in 6m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10082f40d298453e87c1fba1483bec89 : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e679237dbc8c402489130bdc727f151a : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/575ccc925e334a92ac5dfaadb7e89404 : SUCCESS in 8m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7f999ffcffe24cc19550c143fa95e6af : SUCCESS in 6m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3fb585931a9e47a29eb971cb946703f7 : SUCCESS in 10m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b9d69691410c461a86cd42823c51c105 : SUCCESS in 21m 19s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/290c80df39ef44ff9ed7db18025003d3 : SUCCESS in 1h 12m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c42ca1c3924d4dbdac063df75555162a : SUCCESS in 46m 11s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6036dfbc1eda4095b728149de8f0b483 : SUCCESS in 1h 02m 18s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2e86fff84f8a44789175f6d1f572b3e0 : SUCCESS in 17m 33s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/99d651d7c99749d287708afb1dbec2be : SUCCESS in 16m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0359a63ba41942e481d8a529eb24e70e : FAILURE in 1h 39m 13s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c17820a9e8ce4ff380717009c7263aba : SUCCESS in 31m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/44eaecb951f445b9b72c0d0310a6002f : SUCCESS in 24m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d815640051fb4d779159cc7ceb270ad1 : SUCCESS in 30m 01s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/bee311ddab49406a889d4559ef5c5310 : SUCCESS in 11m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d5a8b7846eaf4a5daab0c7f8d64d2ad8 : SUCCESS in 12m 04s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":8},{"id":"c03a60e7e4365b9ca73d8ce0e0a23e453491910c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-22 15:25:52.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"ca8a39ed03e9586e1261ddf23c68ca1b445c6f4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 17:13:33.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c35e525b52f468e96f8651da4f1c7a8 : SUCCESS in 8m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9acae993fe04bf6844abbfa6ff2a85e : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd49be9735a047aca818bb03c4fd5ec3 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a32b37f8db7a465eafe0fc6b7073c70c : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/071dad7b74984e32a30bd1fea8dfc1f6 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f11d6139a0784d9ba8631be00ff7c573 : SUCCESS in 9m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cf2a6d757168456096fdc1473bb2ebd0 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f42e7392f52a46d792fcdfe645616497 : SUCCESS in 10m 14s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/721f95aa1ac44f71a9a140fd4907222c : SUCCESS in 22m 56s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/afd4a04f951b4eb19ff99ced2d93e959 : SUCCESS in 1h 06m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b1925a5ad3584ef1b18f501037ed5c7d : SUCCESS in 1h 01m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/081f7cec621d4f43854e3bdc4f342c2e : SUCCESS in 1h 06m 17s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/eae7eb72ab344cd48ea7a5bd0522712c : SUCCESS in 20m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/776284ebc94547aeb85f8dec13747204 : SUCCESS in 14m 59s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/78062923054c4f82bbaa86e534363446 : SUCCESS in 1h 41m 37s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d9b0538b3a7c4a779c8059daa4e839a4 : SUCCESS in 29m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04ec55b84a31406f93f1d45aba834078 : SUCCESS in 27m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8ef88945400e42fb90fa32a1a3de52d7 : SUCCESS in 27m 38s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/90a65930455f4fa7bf0492327f27e9aa : SUCCESS in 12m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e8ec4da6ee4e400b9dca060e54e899ac : SUCCESS in 10m 26s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":8},{"id":"ccd09ccc89be44332070dcf0a0723f53f4f57df5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-23 09:42:12.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"cad434152e048ddd68faaacbf4036759d6b9ef90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 09:42:30.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"17baa5786ccd74f22916f8611597f9e20ec341ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 11:50:28.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80582a5b738540ce9459253bf60a4e71 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d978483b516647478e483b23ffef0340 : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/106fc2acd98e4c399514bdb4b9a6f35d : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f924a51e801d466396c9036760de9971 : SUCCESS in 7m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/526c2493ab7b45728714a6494c908189 : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8e5f9af12abc4746b9ccaf24530a48a1 : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f0ede33b7dd14049a79ef4c305927afa : SUCCESS in 21m 07s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/36ec38ca634c4f218810c54d17167611 : SUCCESS in 54m 21s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/08f255d6428c487483f7f62e35e15bc6 : SUCCESS in 40m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/01f924b5196b4bf3a44e6fe7e2a520fe : SUCCESS in 54m 30s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/87a48b201d7641e59a04bbe6f7948dc7 : SUCCESS in 17m 03s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50e7ef249e7246ac98e289145414ec2e : SUCCESS in 13m 09s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/6859d359db7542fdb5931a4fdc24d931 : SUCCESS in 1h 45m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/90e5a8c14c5f484f87afcaaafbc9ff39 : SUCCESS in 31m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d671dafd7ce3421ebe6d994e8e574b41 : SUCCESS in 26m 26s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9c154e8996374142aaf74947e6549f9f : SUCCESS in 25m 26s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/a62a473dbe3643a58fa187c7ec0356f8 : SUCCESS in 11m 51s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9425517b22b84fe69836cc34f54d3d8f : SUCCESS in 10m 12s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":8},{"id":"9c2de80a53501df8385ab82f58fd184726b1542a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 11:50:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"39f7c9c5be67d40da26e8268493662c18183b140","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 11:52:50.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56164d80f95143e68b2f4f8f0e973004 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"181352f13a52ea947bf3a45e10462e4f3c3ab4c0","revisions":{"7cba24ea863d442c2e10bb6500045067526b3eb0":{"kind":"REWORK","_number":1,"created":"2021-04-06 18:55:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/785032/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/785032/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/785032/1"}}},"commit":{"parents":[{"commit":"fee81e34537e2023f1928a78e532f672b88b8fcb","subject":"WIP: Ubuntu: support systemd-networkd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fee81e34537e2023f1928a78e532f672b88b8fcb"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 16:10:39.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 18:55:00.000000000","tz":60},"subject":"WIP: Ubuntu: support policy-based routing in systemd-networkd","message":"WIP: Ubuntu: support policy-based routing in systemd-networkd\n\nTODO:\n* support string format rules in networkd\n* support dict format rules in CentOS\n* test policy-based routing\n\nStory: 2004960\nTask: 42217\n\nChange-Id: I77aec0160eb7e4dd763326bfe6e3d9a44b248108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7cba24ea863d442c2e10bb6500045067526b3eb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7cba24ea863d442c2e10bb6500045067526b3eb0"}]},"branch":"refs/heads/master"},"ab72cb6e3f1599d07efbec5bf4523d9ecf6ba01c":{"kind":"REWORK","_number":2,"created":"2021-04-07 09:24:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/785032/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/785032/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/785032/2"}}},"commit":{"parents":[{"commit":"6f6232676dd2650c929afdeaab865fdd65c41915","subject":"WIP: Ubuntu: support systemd-networkd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f6232676dd2650c929afdeaab865fdd65c41915"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 16:10:39.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-07 09:23:46.000000000","tz":60},"subject":"WIP: Ubuntu: support policy-based routing in systemd-networkd","message":"WIP: Ubuntu: support policy-based routing in systemd-networkd\n\nTODO:\n* support string format rules in networkd\n* support dict format rules in CentOS\n* test policy-based routing\n\nStory: 2004960\nTask: 42217\n\nChange-Id: I77aec0160eb7e4dd763326bfe6e3d9a44b248108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab72cb6e3f1599d07efbec5bf4523d9ecf6ba01c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ab72cb6e3f1599d07efbec5bf4523d9ecf6ba01c"}]},"branch":"refs/heads/master"},"ee48f96461bba17478cc520853ac05d4e73ff0a1":{"kind":"REWORK","_number":3,"created":"2021-04-07 14:35:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/785032/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/785032/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/785032/3"}}},"commit":{"parents":[{"commit":"dd165b7e5c23f87e1f0206cdb52ca924d1714d69","subject":"WIP: Ubuntu: support systemd-networkd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dd165b7e5c23f87e1f0206cdb52ca924d1714d69"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 16:10:39.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-07 14:32:22.000000000","tz":60},"subject":"WIP: Ubuntu: support policy-based routing in systemd-networkd","message":"WIP: Ubuntu: support policy-based routing in systemd-networkd\n\nTODO:\n* support string format rules in networkd\n* support dict format rules in CentOS\n\nStory: 2004960\nTask: 42217\n\nChange-Id: I77aec0160eb7e4dd763326bfe6e3d9a44b248108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee48f96461bba17478cc520853ac05d4e73ff0a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee48f96461bba17478cc520853ac05d4e73ff0a1"}]},"branch":"refs/heads/master"},"8fedd04b5b1b247d303e9e5996e8119bca2b5e0c":{"kind":"REWORK","_number":4,"created":"2021-04-07 14:39:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/785032/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/785032/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/785032/4"}}},"commit":{"parents":[{"commit":"f5f8468ac03d211f193807718f6e1ce75064fcc9","subject":"Ubuntu: support systemd-networkd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5f8468ac03d211f193807718f6e1ce75064fcc9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 16:10:39.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-07 14:39:04.000000000","tz":60},"subject":"WIP: Ubuntu: support policy-based routing in systemd-networkd","message":"WIP: Ubuntu: support policy-based routing in systemd-networkd\n\nTODO:\n* support string format rules in networkd\n* support dict format rules in CentOS\n\nStory: 2004960\nTask: 42217\n\nChange-Id: I77aec0160eb7e4dd763326bfe6e3d9a44b248108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8fedd04b5b1b247d303e9e5996e8119bca2b5e0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8fedd04b5b1b247d303e9e5996e8119bca2b5e0c"}]},"branch":"refs/heads/master"},"eb9cd57250f7c157a527dd2fc0330a29fb881462":{"kind":"REWORK","_number":5,"created":"2021-04-07 16:36:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/785032/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/785032/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/785032/5"}}},"commit":{"parents":[{"commit":"f5f8468ac03d211f193807718f6e1ce75064fcc9","subject":"Ubuntu: support systemd-networkd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5f8468ac03d211f193807718f6e1ce75064fcc9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 16:10:39.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-07 16:36:09.000000000","tz":0},"subject":"WIP: Ubuntu: support policy-based routing in systemd-networkd","message":"WIP: Ubuntu: support policy-based routing in systemd-networkd\n\nTODO:\n* support string format rules in networkd\n* support dict format rules in CentOS\n\nStory: 2004960\nTask: 42217\n\nChange-Id: I77aec0160eb7e4dd763326bfe6e3d9a44b248108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb9cd57250f7c157a527dd2fc0330a29fb881462"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/eb9cd57250f7c157a527dd2fc0330a29fb881462"}]},"branch":"refs/heads/master"},"89bdaaccedead2c345a4570d7c9078fe20c2dab2":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-04-07 16:38:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/785032/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/785032/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/785032/6"}}},"commit":{"parents":[{"commit":"f5f8468ac03d211f193807718f6e1ce75064fcc9","subject":"Ubuntu: support systemd-networkd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f5f8468ac03d211f193807718f6e1ce75064fcc9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 16:10:39.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-07 16:38:32.000000000","tz":0},"subject":"Ubuntu: support policy-based routing in systemd-networkd","message":"Ubuntu: support policy-based routing in systemd-networkd\n\nThis change adds support for policy-based routing via systemd-networkd.\nDue to differences in the configuration mechanism, routing policy\nrules are configured via dicts for Ubuntu, while remaining as strings\non CentOS. Ideally we would support both formats.\n\nStory: 2004960\nTask: 42217\n\nChange-Id: I77aec0160eb7e4dd763326bfe6e3d9a44b248108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89bdaaccedead2c345a4570d7c9078fe20c2dab2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/89bdaaccedead2c345a4570d7c9078fe20c2dab2"}]},"branch":"refs/heads/master","description":"Edit commit message"},"1ca665eb63440de82082de15dd5ce51d2fae0f64":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-04-21 08:16:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/785032/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/785032/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/785032/7"}}},"commit":{"parents":[{"commit":"33be25c92505f9cb906dc9c9e6d5a3d7c2b08d1e","subject":"Ubuntu: support systemd-networkd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33be25c92505f9cb906dc9c9e6d5a3d7c2b08d1e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 16:10:39.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-21 08:16:39.000000000","tz":0},"subject":"Ubuntu: support policy-based routing in systemd-networkd","message":"Ubuntu: support policy-based routing in systemd-networkd\n\nThis change adds support for policy-based routing via systemd-networkd.\nDue to differences in the configuration mechanism, routing policy\nrules are configured via dicts for Ubuntu, while remaining as strings\non CentOS. Ideally we would support both formats.\n\nStory: 2004960\nTask: 42217\n\nChange-Id: I77aec0160eb7e4dd763326bfe6e3d9a44b248108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ca665eb63440de82082de15dd5ce51d2fae0f64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ca665eb63440de82082de15dd5ce51d2fae0f64"}]},"branch":"refs/heads/master","description":"Rebase"},"181352f13a52ea947bf3a45e10462e4f3c3ab4c0":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-04-21 08:23:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/32/785032/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/32/785032/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/32/785032/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/32/785032/8"}}},"commit":{"parents":[{"commit":"3bbf736d8d4672460ab42576d1679b2465b06810","subject":"Ubuntu: support systemd-networkd","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3bbf736d8d4672460ab42576d1679b2465b06810"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-06 16:10:39.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-21 08:23:57.000000000","tz":0},"subject":"Ubuntu: support policy-based routing in systemd-networkd","message":"Ubuntu: support policy-based routing in systemd-networkd\n\nThis change adds support for policy-based routing via systemd-networkd.\nDue to differences in the configuration mechanism, routing policy\nrules are configured via dicts for Ubuntu, while remaining as strings\non CentOS. Ideally we would support both formats.\n\nStory: 2004960\nTask: 42217\n\nChange-Id: I77aec0160eb7e4dd763326bfe6e3d9a44b248108\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/181352f13a52ea947bf3a45e10462e4f3c3ab4c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/181352f13a52ea947bf3a45e10462e4f3c3ab4c0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
