)]}'
{"id":"openstack%2Fkayobe~785624","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I1b86c5f58d0c3b63269766eaa8e76af18b5fadb1","project":"openstack/kayobe","branch":"stable/train","topic":"pbr-fix-stable/train","hashtags":[],"change_id":"I1b86c5f58d0c3b63269766eaa8e76af18b5fadb1","subject":"Fix installation errors with Python 2","status":"MERGED","created":"2021-04-09 11:29:36.000000000","updated":"2021-04-13 19:00:59.000000000","submitted":"2021-04-13 18:59:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785624-pbr-fix-stable/train","meta_rev_id":"cd936aadc36d3f7e472870640ad14a7e23fe2b6b","_number":785624,"virtual_id_number":785624,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-13 18:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-04-13 08:00:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-04-13 08:13:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-04-13 08:13:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-13 19:00:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-09 12:15:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-13 05:49:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-04-13 08:00:26.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-04-13 08:13:12.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"}],"messages":[{"id":"d8ddd82b9df27ff9e25eb0c99073f9bda8359569","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-09 11:29:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d5e78de70751632bc6443af72cf046639484961d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 12:15:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/ab80dcc82701470ab797c6a2b6140a9e : FAILURE in 40m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5243765f3ccc7e2c5ed70e5fc31b6d4566132857","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-09 12:26:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"590599246b61aec628a29d3d6dc56801079083f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 13:29:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/7048469c028d4e46bc27f4786434f508 : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":2},{"id":"b2ceae04c4e003d073fbc23fddec13c4ceb7aaa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-09 13:41:27.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0bfeb74d929aec71dc1bd1db82aa3562bddc67c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 15:15:12.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bab5463b4d5e49a7ba5006207aec462b : SUCCESS in 4m 53s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/213ba8e4ad82421a8721b5a81f3d27d4 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d6adab1f74d4138a5f9209ffb477964 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c33c0db15e8454bbac078b5adc71399 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ec44831b0e0241719f29848d4f23d354 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d2d1c69c1e143368de2e78261bc1c28 : SUCCESS in 6m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4b497bd5ef824ce9bca7fb398e1d77ac : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/08d1845c499d459e9f8a6b4f21a81dc2 : SUCCESS in 8m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0bc75f33a0e545b1a9b4629a2ff631b9 : SUCCESS in 19m 49s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/1480517d08df4fb0ae4a98e268209f98 : SUCCESS in 1h 00m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a499d1a7cce64989960a7eb12f321dde : SUCCESS in 43m 40s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/ec0a375aa10444bd96c3713f571d5751 : SUCCESS in 14m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4287697215354f84b87d0ca2b6ec28fc : SUCCESS in 13m 49s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/fa9affb64f9e4a348a82a446d37cda08 : SUCCESS in 1h 30m 00s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/557fce5ff4ff43b787ee0ece0563f94d : FAILURE in 26m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2660c61f1d174b489c8c3981f34d3108 : SUCCESS in 26m 23s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a8e0fe6b3ee94a37a1a42cfd4e942bc6 : SUCCESS in 27m 36s","accounts_in_message":[],"_revision_number":3},{"id":"1c55deb4fd5cfffe5fa703a82d08bb80a2c158c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-12 07:38:35.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"9c4fe06376cdd33b3310a35de58444a7d73ac47c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 09:21:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2bc039d88a04fd3863168494331a2b6 : SUCCESS in 6m 18s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/268af99c07b24013b5df752118bf3cf0 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9428a02004aa466792335f578603c101 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aede2bd70a248d0b980c619de885ca2 : SUCCESS in 4m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b03f2f42f478463b8acd48b43ec9b863 : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1b36145a5aa4beda46affebf185b6d6 : SUCCESS in 7m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae78726a95c04be89ed6ee76e6819ed5 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cc3a99a1465a46bb95babdcfefba2651 : SUCCESS in 8m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fefd6787130b4cd48ea25ef909a58eea : SUCCESS in 18m 22s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/c0f8f03c8497476b985b9ec3ac6d15ea : SUCCESS in 54m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9f90ccbac2f44aaf92ee6a8e462c6b3d : SUCCESS in 1h 02m 12s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/ade2f3122d3c463daf3ddb48b6b85165 : SUCCESS in 12m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/32aa2323eb414edc9990f6cc415b61c3 : SUCCESS in 13m 47s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/c5674131982b413a94c6eb623ffad6d6 : SUCCESS in 1h 37m 44s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/0025b4958f1b4a4aaf88325011dd830e : SUCCESS in 21m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b3e6c4ab19524a8a92ba99fde1a389af : SUCCESS in 26m 44s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/1fa607df45704d71a424f4119ee18d8a : SUCCESS in 30m 57s","accounts_in_message":[],"_revision_number":4},{"id":"06ccce79db33ea84b1339ccabe22ee4c94d4f12f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-12 11:05:56.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"5e35652f28eb1f5070b957f434da177e16e30734","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 12:08:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da303b03a31445d28fcc4c894ec38d22 : SUCCESS in 5m 16s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/265cf4102fcb49ed8d907d5f95682f50 : SUCCESS in 9m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/77fa82d282344ac9b1e835c6725b0d90 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44a8df1d74044b88b3a4506a1c2751f6 : SUCCESS in 4m 39s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/95b69c3c79b047cca57d8fa6357d4521 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/545f87b9a36f463eaf029bbbbb9f81be : SUCCESS in 8m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/54b16d08df7445d8b4ed9e8e5f506883 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cedfeb0eb2b54ab2b44f44bf691772fb : SUCCESS in 9m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9892b6168fcc43e7aa8b60ad61ba4641 : SUCCESS in 18m 40s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/aa87fcfd6f914ea6835767af13bab270 : SUCCESS in 1h 00m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7782f0e862e042a8940f7f1cdf236896 : SUCCESS in 47m 38s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/ad759298ac1a413f93f6b8371340add0 : SUCCESS in 12m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5fc9c489b8464b0c944f3e21dcfba4e1 : SUCCESS in 15m 31s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/acf2d618816e44629b4c1eee0ed99a02 : FAILURE in 3m 23s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/ea3a86c0b0a74e26a8d63d693db9563f : SUCCESS in 24m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/289a5086eb5c4d2aae46b8290eb82b48 : SUCCESS in 25m 03s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/49e302cec1f2417eb08e9fef938f9d11 : FAILURE in 3m 13s","accounts_in_message":[],"_revision_number":4},{"id":"c4207dfbf1eb716d5fe82e8e5e01c378c21a6318","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-12 13:08:29.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3c82d5d5d2eb59af40b81b938491971afcbaa7a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-12 13:10:59.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"d8c0b004b9f0b53a595220d8965c3bc3f4339d9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 13:11:28.000000000","message":"Patch Set 6: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"51733a3bd32de73f2f6b0a4d9e2f9b9a8d693a9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-12 19:31:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"67bc7da58ac98a7ff26c07fccac24aaf6b5d1cc7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 21:21:24.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1cd38e4b1ff4404e9afe9f45de6e449a : SUCCESS in 4m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/704e687ede084748a6944cdfe62df925 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cce660b398984116a9fa1f42308fd08a : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edc791a0f79942b099039674edaf7e53 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3cfeb88e4aa74f8997be0dd312f2a45a : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3138ddf93cb24b47b6b992f1a3919295 : SUCCESS in 8m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/295ab7fe4ca84a7db332bb81993b4f09 : SUCCESS in 5m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f601a31630fe44c29b167b3b7d13ec39 : SUCCESS in 10m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d2676738705c463597d6cf5aa29b93f7 : SUCCESS in 24m 50s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/8a26e083177343de8b8623a0de79b919 : SUCCESS in 1h 04m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/299d24e6ab254780875c866a9b87316c : SUCCESS in 1h 00m 55s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/3baec510409f413e98aebeca55be7cc9 : SUCCESS in 16m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f80b109d3ce949ca8a2e8a1e0194d46b : SUCCESS in 25m 00s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8801d4fb359a4d148abaf9d09f10cc25 : SUCCESS in 1h 34m 28s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/7f838c7fa1b446fa8894df2401df2b75 : SUCCESS in 26m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/928e4cb0a0af4339bb7eb1a67cdc2309 : SUCCESS in 40m 21s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/d8deace9e3e34ecdb5ea29a5d4381777 : SUCCESS in 41m 03s","accounts_in_message":[],"_revision_number":7},{"id":"55867b7c4bf960ba23b31bb67391748a21b8ccc8","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-13 05:11:20.000000000","message":"Patch Set 7: Backport-Candidate+2","accounts_in_message":[],"_revision_number":7},{"id":"214ba09d9e642a38d4e83a896a9a33a039999271","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-13 05:12:38.000000000","message":"Removed Backport-Candidate+2 by Pierre Riteau \u003cpierre@stackhpc.com\u003e\n","accounts_in_message":[],"_revision_number":7},{"id":"e9f64dbc9667e317d6c5a12330603939fe480910","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-13 05:12:45.000000000","message":"Patch Set 7: Review-Priority+2","accounts_in_message":[],"_revision_number":7},{"id":"cb6ffd6b833dea9f0044a5a4a21c10e657062d5b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-13 05:49:42.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6dd31110c2514982d5470e33c5931bb679cf16f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-13 06:03:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"649f7275dc8f79f597a5f51bc4727031f7008669","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-13 06:04:16.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"dcae997f9d5f7464f3a641db10ae7b4152507de2","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-04-13 08:00:26.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"0c62d76856c009b9c9004bd67307ad4137a53627","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-04-13 08:13:12.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"f04caf11981282a011b0b213cdf687066ce4d26b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 08:31:12.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8e389f79c25446e9d9b63d7358792dc : SUCCESS in 5m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/47e45e2ed7f641fdbdff2aad3efd509a : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48c64954aa03487580013f77086ba5bb : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9021f14a05a245b28b69eb307c2943c8 : SUCCESS in 4m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/be3cc924a7384775bed0aaac69b9294e : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f087fe39c1b43428be70e3d375b6b40 : SUCCESS in 6m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f4a7349973624455b7b8bd1550575cab : SUCCESS in 5m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9a6e7fd3a87d464db8c77f1f361f7d29 : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c6673afecf6a434dbbd6889985a0b54d : SUCCESS in 17m 13s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/00d909c378c04d02a7eb1063566ec299 : SUCCESS in 1h 00m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4e4a3506e97140c0abf7bd9fc516a7da : SUCCESS in 48m 41s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/f725e10e52f044d2bd3f98adefa615cb : SUCCESS in 17m 46s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4cf8fa8b02d1491494c061a3d5655e27 : SUCCESS in 18m 26s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/4d9de39865e742c9930ba1b03d248cba : SUCCESS in 1h 30m 51s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/ef56a0cd2ae74d10a80bf6cc77411f3b : SUCCESS in 20m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/05d4dd2c1849457a8566296e9900e148 : SUCCESS in 27m 12s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f5641df9b86b4d489e13a241692d1d68 : SUCCESS in 23m 44s","accounts_in_message":[],"_revision_number":8},{"id":"5ad754da774acc535f06f45287fbb154b84d19b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 08:55:56.000000000","message":"Patch Set 8: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":8},{"id":"af88d980e9a7fdf8bc0f042720dd2e69d14d92be","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-13 15:29:52.000000000","message":"Patch Set 8:\n\nrecheck","accounts_in_message":[],"_revision_number":8},{"id":"22daa3937c2598ca6b817c4c46caa9978fbdf450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 17:17:53.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f1be58e95084a66b7618b3db9df2400 : SUCCESS in 4m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c349f5a6a2fd48619eb4eda5a3cc4325 : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/388227333e014ab1a67bb15d19a4f53e : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6127dce4131243178eb47f420f1e31df : SUCCESS in 5m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/14516b01f8b94890a4845ff820f9e345 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6b2738cd9164d3d9413053b81fa35be : SUCCESS in 8m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/88f2b0505b6b4b59b04020f776ee5539 : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9334f52888e645db8b65d08aaa843fa8 : SUCCESS in 10m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0878f056d4bb478595b5056f8e58afd6 : SUCCESS in 21m 16s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/c9a8bdcf09574522b4c20440ea95c479 : SUCCESS in 1h 04m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6e74cf6e7d154e31a6ec88ddc3c1d848 : SUCCESS in 59m 47s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/9585cd67dbf545138ae727ec6607db8e : SUCCESS in 12m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f319fea66bb54c72a4d18cf21bdab230 : SUCCESS in 15m 58s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/31af8892cf3748efb58cbb0c91425cf4 : SUCCESS in 1h 39m 15s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/72a9c9446af64392a9ad1a599b4c1c1a : SUCCESS in 18m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e7bed1f3bb98494891f82a10820a4263 : SUCCESS in 32m 44s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/bbd3538ab69c40809beeb22ad3afc6b8 : SUCCESS in 37m 27s","accounts_in_message":[],"_revision_number":8},{"id":"687cc1a6a04f6c10fe0dcf83764be3cb34d0d2e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 17:18:28.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"4dd2d0e24e04655b5c7fa7adcb888ff18b4d8ffd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 18:59:39.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/481a719583a4443494e81423533e36e1 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9aef9ba1c25e48258276a3042e633ed8 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48922299f47542b2812cc39f9317c942 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e4052d65c889426c8963478ec9019619 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b779b98ff4740b9b8a3fa4bca3283ac : SUCCESS in 6m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7e4b28a8c1ba49f3bb1f82e9db6345ec : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2de9c47c09234d02b7d0928c06e0f042 : SUCCESS in 8m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3b98b9b29dca4b95a2f45e30c035ad47 : SUCCESS in 18m 59s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/a5cb764f2ca5410ab2c887b1832fe143 : SUCCESS in 53m 30s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3f98ae834ccd4ffe8122e23535f97c57 : SUCCESS in 56m 56s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/d0514c484fb744ca82b5da659ca62a08 : SUCCESS in 17m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3d4a5b5ec91b416587f75381b07298ea : SUCCESS in 16m 55s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/9df67ff3aacb4cb9b8a0a4d96efe1f25 : SUCCESS in 1h 29m 26s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/4a4f053a1e57456cbb134d2125426063 : SUCCESS in 25m 51s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a7382aac1bf94de089d05cb7f8b054db : SUCCESS in 35m 44s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/51f657f696744628aa6dfc9651f37092 : SUCCESS in 42m 48s","accounts_in_message":[],"_revision_number":8},{"id":"80f3f2662691c2cda6aa3b281e0e0c1662f707bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 18:59:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"617dd0b4b2a733c135a9cd7cb77764a264b28e0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 19:00:59.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39e186dc763d411e828a33f7a20a4939 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"020e9329b5c4befb5ce6d342042f82001ffc2f24","revisions":{"d8152517838f1ea864595c3d655c5f0c45a6ad25":{"kind":"REWORK","_number":1,"created":"2021-04-09 11:29:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/24/785624/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/24/785624/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/24/785624/1"}}},"commit":{"parents":[{"commit":"1032cebdf80182be1fc7b00cce65bac2bed613f3","subject":"Merge \"Update seed VM image to CentOS 8.3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1032cebdf80182be1fc7b00cce65bac2bed613f3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-09 11:29:27.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-09 11:29:27.000000000","tz":120},"subject":"[DNM] Debug train jobs","message":"[DNM] Debug train jobs\n\nChange-Id: I1b86c5f58d0c3b63269766eaa8e76af18b5fadb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8152517838f1ea864595c3d655c5f0c45a6ad25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d8152517838f1ea864595c3d655c5f0c45a6ad25"}]},"branch":"refs/heads/stable/train"},"2bef90c3913b083acae4807ac06bbb0d561ea3ea":{"kind":"REWORK","_number":2,"created":"2021-04-09 12:26:34.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/24/785624/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/24/785624/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/24/785624/2"}}},"commit":{"parents":[{"commit":"1032cebdf80182be1fc7b00cce65bac2bed613f3","subject":"Merge \"Update seed VM image to CentOS 8.3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1032cebdf80182be1fc7b00cce65bac2bed613f3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-09 12:26:28.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-09 12:26:28.000000000","tz":120},"subject":"[DNM] Debug train jobs","message":"[DNM] Debug train jobs\n\nChange-Id: I1b86c5f58d0c3b63269766eaa8e76af18b5fadb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2bef90c3913b083acae4807ac06bbb0d561ea3ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2bef90c3913b083acae4807ac06bbb0d561ea3ea"}]},"branch":"refs/heads/stable/train"},"1d570541f9bbbd4a0e93c7da3930b83d069f81c2":{"kind":"REWORK","_number":3,"created":"2021-04-09 13:41:27.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/24/785624/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/24/785624/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/24/785624/3"}}},"commit":{"parents":[{"commit":"1032cebdf80182be1fc7b00cce65bac2bed613f3","subject":"Merge \"Update seed VM image to CentOS 8.3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1032cebdf80182be1fc7b00cce65bac2bed613f3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-09 13:41:20.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-09 13:41:20.000000000","tz":120},"subject":"Fix installation errors with Python 2","message":"Fix installation errors with Python 2\n\nPython 2 jobs recently started failing frequently with the following\nerror:\n\n    distutils.errors.DistutilsError: Could not find suitable distribution for Requirement.parse(\u0027pbr\u003e\u003d2.0\u0027)\n\nThe root cause appears to be that indirect requirements are handled by\neasy_install which doesn\u0027t like some index servers. A suggested\nworkaround is to install pbr directly with pip first [1].\n\n[1] https://github.com/googleapis/google-cloud-python/issues/3757#issuecomment-422384453\n\nChange-Id: I1b86c5f58d0c3b63269766eaa8e76af18b5fadb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d570541f9bbbd4a0e93c7da3930b83d069f81c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d570541f9bbbd4a0e93c7da3930b83d069f81c2"}]},"branch":"refs/heads/stable/train"},"1e011fe36c798cb62d206758af6e5bd7b3a527d0":{"kind":"REWORK","_number":4,"created":"2021-04-12 07:38:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/24/785624/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/24/785624/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/24/785624/4"}}},"commit":{"parents":[{"commit":"1032cebdf80182be1fc7b00cce65bac2bed613f3","subject":"Merge \"Update seed VM image to CentOS 8.3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1032cebdf80182be1fc7b00cce65bac2bed613f3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-12 07:35:59.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-12 07:35:59.000000000","tz":120},"subject":"Fix installation errors with Python 2","message":"Fix installation errors with Python 2\n\npython 2 jobs recently started failing frequently with the following\nerror:\n\n    distutils.errors.DistutilsError: Could not find suitable distribution for Requirement.parse(\u0027pbr\u003e\u003d2.0\u0027)\n\nThe root cause appears to be that indirect requirements are handled by\neasy_install which doesn\u0027t like some index servers [1]. Try updating\nsetuptools first.\n\nDisable IPA builds on CentOS 7 which are failing due to the same issue\nand would require a fix inside DIB itself.\n\n[1] https://github.com/googleapis/google-cloud-python/issues/3757#issuecomment-422384453\n\nChange-Id: I1b86c5f58d0c3b63269766eaa8e76af18b5fadb1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e011fe36c798cb62d206758af6e5bd7b3a527d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1e011fe36c798cb62d206758af6e5bd7b3a527d0"}]},"branch":"refs/heads/stable/train"},"34b14a09d05000c74c8373fd8500cd912c578b89":{"kind":"REWORK","_number":5,"created":"2021-04-12 13:08:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/24/785624/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/24/785624/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/24/785624/5"}}},"commit":{"parents":[{"commit":"1032cebdf80182be1fc7b00cce65bac2bed613f3","subject":"Merge \"Update seed VM image to CentOS 8.3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1032cebdf80182be1fc7b00cce65bac2bed613f3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-12 13:08:15.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-12 13:08:15.000000000","tz":120},"subject":"Fix installation errors with Python 2","message":"Fix installation errors with Python 2\n\npython 2 jobs recently started failing frequently with the following\nerror:\n\n    distutils.errors.DistutilsError: Could not find suitable distribution for Requirement.parse(\u0027pbr\u003e\u003d2.0\u0027)\n\nThe root cause appears to be that indirect requirements are handled by\neasy_install which doesn\u0027t like some index servers [1]. Install the\nlatest setuptools first to work around the issue.\n\nDisable IPA builds on CentOS 7 which are failing due to the same issue\nand would require a fix inside DIB itself.\n\n[1] https://github.com/googleapis/google-cloud-python/issues/3757#issuecomment-422384453\n\nChange-Id: I1b86c5f58d0c3b63269766eaa8e76af18b5fadb1\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/785869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34b14a09d05000c74c8373fd8500cd912c578b89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34b14a09d05000c74c8373fd8500cd912c578b89"}]},"branch":"refs/heads/stable/train"},"0f706d3bcb6eaef316f7a6a34ea6676c3bda3326":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2021-04-12 13:10:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/24/785624/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/24/785624/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/24/785624/6"}}},"commit":{"parents":[{"commit":"1032cebdf80182be1fc7b00cce65bac2bed613f3","subject":"Merge \"Update seed VM image to CentOS 8.3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1032cebdf80182be1fc7b00cce65bac2bed613f3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-12 13:10:17.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-12 13:10:17.000000000","tz":120},"subject":"Fix installation errors with Python 2","message":"Fix installation errors with Python 2\n\nPython 2 jobs recently started failing frequently with the following\nerror:\n\n    distutils.errors.DistutilsError: Could not find suitable distribution for Requirement.parse(\u0027pbr\u003e\u003d2.0\u0027)\n\nThe root cause appears to be that indirect requirements are handled by\neasy_install which doesn\u0027t like some index servers [1]. Install the\nlatest setuptools first to work around the issue.\n\nDisable IPA builds on CentOS 7 which are failing due to the same issue\nand would require a fix inside DIB itself.\n\n[1] https://github.com/googleapis/google-cloud-python/issues/3757#issuecomment-422384453\n\nChange-Id: I1b86c5f58d0c3b63269766eaa8e76af18b5fadb1\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/785869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f706d3bcb6eaef316f7a6a34ea6676c3bda3326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0f706d3bcb6eaef316f7a6a34ea6676c3bda3326"}]},"branch":"refs/heads/stable/train"},"1ca107c82daba734b5e0cb8e682f8341f29ae139":{"kind":"REWORK","_number":7,"created":"2021-04-12 19:31:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/24/785624/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/24/785624/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/24/785624/7"}}},"commit":{"parents":[{"commit":"1032cebdf80182be1fc7b00cce65bac2bed613f3","subject":"Merge \"Update seed VM image to CentOS 8.3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1032cebdf80182be1fc7b00cce65bac2bed613f3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-12 19:31:34.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-12 19:31:34.000000000","tz":120},"subject":"Fix installation errors with Python 2","message":"Fix installation errors with Python 2\n\nPython 2 jobs recently started failing frequently with the following\nerror:\n\n    distutils.errors.DistutilsError: Could not find suitable distribution for Requirement.parse(\u0027pbr\u003e\u003d2.0\u0027)\n\nThe root cause appears to be that indirect requirements are handled by\neasy_install which doesn\u0027t like some index servers [1]. Install the\nlatest setuptools first to work around the issue.\n\nDisable IPA builds on CentOS 7 which are failing due to the same issue\nand would require a fix inside DIB itself.\n\n[1] https://github.com/googleapis/google-cloud-python/issues/3757#issuecomment-422384453\n\nChange-Id: I1b86c5f58d0c3b63269766eaa8e76af18b5fadb1\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/785869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ca107c82daba734b5e0cb8e682f8341f29ae139"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ca107c82daba734b5e0cb8e682f8341f29ae139"}]},"branch":"refs/heads/stable/train"},"020e9329b5c4befb5ce6d342042f82001ffc2f24":{"kind":"REWORK","_number":8,"created":"2021-04-13 06:03:33.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/24/785624/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/24/785624/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/24/785624/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/24/785624/8"}}},"commit":{"parents":[{"commit":"1032cebdf80182be1fc7b00cce65bac2bed613f3","subject":"Merge \"Update seed VM image to CentOS 8.3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1032cebdf80182be1fc7b00cce65bac2bed613f3"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-13 06:03:24.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-13 06:03:24.000000000","tz":120},"subject":"Fix installation errors with Python 2","message":"Fix installation errors with Python 2\n\nPython 2 jobs recently started failing frequently with the following\nerror:\n\n    distutils.errors.DistutilsError: Could not find suitable distribution for Requirement.parse(\u0027pbr\u003e\u003d2.0\u0027)\n\nThe root cause appears to be that indirect requirements are handled by\neasy_install which doesn\u0027t like some index servers [1]. Install the\nlatest setuptools first to work around the issue.\n\nDisable IPA builds on CentOS 7 which are failing due to the same issue\nand would require a fix inside DIB itself.\n\n[1] https://github.com/googleapis/google-cloud-python/issues/3757#issuecomment-422384453\n\nChange-Id: I1b86c5f58d0c3b63269766eaa8e76af18b5fadb1\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/785869\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/020e9329b5c4befb5ce6d342042f82001ffc2f24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/020e9329b5c4befb5ce6d342042f82001ffc2f24"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
