)]}'
{"id":"openstack%2Fkayobe~785673","triplet_id":"openstack%2Fkayobe~master~Ic9d3d03192f515a055e41351a0bf25da7a7b0684","project":"openstack/kayobe","branch":"master","topic":"multiple-environments","hashtags":[],"change_id":"Ic9d3d03192f515a055e41351a0bf25da7a7b0684","subject":"Exclude shared directories when creating environments","status":"MERGED","created":"2021-04-09 15:47:25.000000000","updated":"2021-04-22 11:39:25.000000000","submitted":"2021-04-22 11:37:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"785673-multiple-environments","meta_rev_id":"6c651b3ac4a904a732b2b0344074ae4a2a63c32b","_number":785673,"virtual_id_number":785673,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-22 11:37:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-04-22 11:39:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-04-22 11:39:25.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2021-04-22 09:51:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-09 16:01:35.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-04-09 17:41:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-12 11:51:00.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-04-19 09:37:19.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"ad94a0db1520da08748eb0b07a3bc98780a8cfe2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-09 15:47:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"149137804f3d5ff2492b13b5781b18db63f279aa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-09 16:01:35.000000000","message":"Patch Set 1:\n\nDidn\u0027t you plan to skip ansible \u0026 hooks?","accounts_in_message":[],"_revision_number":1},{"id":"8e1586bc7c2c92523e15900309c53ea3b0469bb2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-09 16:42:33.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Didn\u0027t you plan to skip ansible \u0026 hooks?\n\nI did, but then I thought they don\u0027t hurt if they are copied into environments/\u003cenv-name\u003e. They can be merged manually.","accounts_in_message":[],"_revision_number":1},{"id":"2bbe4712a57b001411699a1158f86818b381809b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 17:41:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23288a987f90467e96d1133721c10b3e : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74a058aa91394034a914fb2090de401c : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1131a1316cd94b77880fb8ccc6e3e295 : SUCCESS in 7m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc25d60b485a4b2dbb5765bdf9de9d47 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10b9bcf661af498d8ab30a73ea3e2d17 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aff036fcc77c417faf243167e713e760 : SUCCESS in 9m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/80f9ad741d114f4396dcf2f9e3e0a350 : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0951038f95f54b1e99010015da38ce83 : SUCCESS in 11m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/726f2c0fb9d345779d43eae3048a0061 : SUCCESS in 20m 29s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ee8202b701ec4ceaa1bf62707ebaa61c : SUCCESS in 1h 02m 03s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09fec9063797404193227ac716a76e41 : SUCCESS in 44m 21s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/700a3450edf44571aec48ebbfb21749e : SUCCESS in 1h 11m 16s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2b2ecb21f21e44af843b8283d5404d41 : SUCCESS in 19m 28s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb1abad0e23a419f91d0dc4c9b4f90e7 : SUCCESS in 12m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/626f276276d04e589ef1281916cf977a : SUCCESS in 1h 50m 48s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/58746cdb415b4df0b8738facdaf71a20 : SUCCESS in 37m 34s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/139c9f2f0b394fbe81320cf0b9e5468d : SUCCESS in 27m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/db06507f0e6f429092d626929c3bc5a4 : SUCCESS in 23m 06s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/42511e924c254d81ade4b34b0624dd62 : SUCCESS in 15m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1eb51e051910464fa6b78a103282709c : SUCCESS in 10m 50s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":1},{"id":"08e10d580ef79731b5eb4dbf82366458d84ad776","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-04-12 11:51:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fa47267db588c8601b82c8e3136d3aee53f9a62b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-13 09:06:11.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"d72179a49f76c9f61c5f9ca9126c4c1a25920e34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-13 10:51:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/094c91beea9e4dbf9124b0ba74297dc1 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e30f8a14891f428da33e96859d05bb44 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a42295b946fe45e2b1ef81b1474d9709 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5505e64e1b464eb3bc5969fb085bb5d4 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70068a724aaa4aec88c22dd5bcba8491 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9562173221c847c1b6db0c69a4a737a7 : SUCCESS in 8m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7fba3a34a1f04d429681a8ea05d7160c : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9a1db06387e445b9b442945780ecc5f1 : SUCCESS in 10m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/764cca1f8a4c484dbf94d51dc6b9cac7 : SUCCESS in 20m 26s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ac7f40a2d4914d1c86ecb42ab777b949 : SUCCESS in 1h 09m 09s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/07d386f20f91439da9925c13fafa411a : SUCCESS in 48m 42s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/57661fb4f98347cfa04d84f109d4bf04 : SUCCESS in 1h 01m 28s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f333203e1cb54f7e95d7939c5d69e877 : SUCCESS in 13m 40s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ae4ca3eaa53a48fa9f624088b09eb269 : SUCCESS in 12m 13s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e1e54cc7c022404798e4e0dfa2967a34 : SUCCESS in 1h 43m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/67cd868f84d3494a9a8e1907481d7966 : SUCCESS in 23m 47s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d0cbf67b48784d5f8926c551354998af : SUCCESS in 25m 01s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8aa853387e5c4a8896b9cbdfb347c1bd : SUCCESS in 19m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c70ee471274d4bcd8a4ec6285145c6c6 : SUCCESS in 11m 31s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/488431d5f0294958a4d8ea6494604cf7 : SUCCESS in 10m 26s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":2},{"id":"f15c752b62fcfaa43b3f1aae697eeb6d565655c7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-19 09:37:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b414aeb3acc5b92d0b1d5685bb852baf224ce16c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 09:37:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4dd3c0523015134dd24d70e8b9013b79a95fe021","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 11:12:28.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5f1f60935004000902eea0d3593028e : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f93c4351d99428ebfe79c7e6f92472a : SUCCESS in 7m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5fc55b9af23a4aefad1edfe43fcfd7e4 : SUCCESS in 6m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e3eb44050f4b469b6408c8b65a1d30 : SUCCESS in 8m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a67f2f5da3444e1aeb889932e468e7c : SUCCESS in 6m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b52153afce5f4cc0b08ea15a309add63 : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8e827c945df04ff48259e1e4e8a5681b : SUCCESS in 18m 51s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/e53ad1b37b14452e9a8d8e1bf2669d56 : SUCCESS in 52m 27s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/914a8f738a63475fb0d27d8b55bb6bbb : SUCCESS in 46m 53s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/18dbecd8e45249188bec5707a8ccc403 : SUCCESS in 54m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/67e90d96939d478ea33650f49f776757 : SUCCESS in 16m 02s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7abbc0c2f74b419eb7a9f5ea9aaea17c : SUCCESS in 12m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9a9e2e2561154e42ac45f2558507a58f : FAILURE in 1h 34m 26s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3f437b9cb649495ab38d594d1aeb9df4 : SUCCESS in 29m 48s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d261d0ad42004134b4cf17fa7aa3dbb2 : SUCCESS in 24m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/16a5a16727794be4b6dd39454902168c : SUCCESS in 25m 09s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ef21ec20af0f44708fbe9f21a0184508 : SUCCESS in 11m 50s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6b7dbf4bcc32423b93be5df003b1e33c : SUCCESS in 10m 05s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"67f1f09df80d0884034daf48ff146445756ee7b7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-19 12:26:26.000000000","message":"Patch Set 2:\n\nnova.exception.TooOldComputeService: Current Nova version does not support computes older than Wallaby but the minimum compute service level in your system is 53 and the oldest supported service level is 54.","accounts_in_message":[],"_revision_number":2},{"id":"813a8a9f286aaac62480b75a56d523388f157d2e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-21 09:32:45.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c78c7cdcc1ebb5fd4b051062012c54df786dfcab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 11:08:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1044ee86ed584935afb4393a6a0cfa4e : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/521b5f40a9694419bebd5284190f9e5c : SUCCESS in 6m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6fc61e6204774c4a87aa7c250f115acd : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9462313c7ff542d681a31cb991acfc34 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29fc7e6e3e4a489d81e2343488394366 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f9a590543f45f4b2e25a4e503f290d : SUCCESS in 8m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/48610b294c8545308a74c3c4f313206b : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c2a88ecef4294d7db961f37c2bc9f9d4 : SUCCESS in 10m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5ef34c8436354985982a1104fefda088 : SUCCESS in 20m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d0f62a1128c447f9a0348886e1e8b8b9 : SUCCESS in 1h 03m 39s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ef70f4e64cb4cd7b9a79732b6b99c68 : SUCCESS in 46m 39s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9f9be922e3b14fa79a850a6abc828137 : SUCCESS in 1h 00m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/1ab0c610d11943b793e4c3db69c27242 : SUCCESS in 17m 08s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ef549de318194cbb8178bc6763bfdd3f : SUCCESS in 13m 38s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/0d3a752a2b8d4288b79d7ff30b35eab6 : FAILURE in 1h 32m 31s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/899c9293cc574a4a842553d1da3e7289 : SUCCESS in 27m 08s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d2e69b2499045789bd63dbb2aa1e372 : SUCCESS in 21m 52s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cdce29fbc2fc4f1c8bcd0a577f474183 : SUCCESS in 28m 37s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/13e3c1ecc8d04740a8fb6b606126cd8c : SUCCESS in 11m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0cc9c3b723954335a303b2e71f87a6fa : SUCCESS in 9m 57s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":2},{"id":"4045dede62de93b29cd89acb264d7322f68dbb14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-22 07:22:15.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"b3b3efe27c54e12e93a7a76ce091683dfb0ef419","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 09:35:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b3272b37a1f04c508d40ba13190c72d6 : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f4bacf2c8ab49c18a6c20fcabb828dc : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/606780c6be304fe09fad71344012343a : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa5772bfab424ddc9db7146a3532a018 : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8141be7b137472fa78e84a11a6db745 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1443bbd4e0304626bb94fccfd8d63937 : SUCCESS in 10m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f1be2dfad6334893b86a09422e11cdf1 : SUCCESS in 7m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ad4dd442f76745b28fd58e3a8fa08079 : SUCCESS in 11m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e8340588b6334573967e258c96f2fc13 : SUCCESS in 22m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9e674419b6dc4a2fb2885a6863648946 : SUCCESS in 1h 08m 12s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d9ab9f8300fe4306a074de77d6e442ee : SUCCESS in 50m 51s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d3c56e315a1048cfacf3a125a0e04f46 : SUCCESS in 1h 03m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b0015ca231c1430c8d97ed8f58efd688 : SUCCESS in 17m 19s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/167e4c89d2d74ee9af96eb88716ef295 : SUCCESS in 14m 40s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/cc73ddccee194723bd7eaf35b8306983 : SUCCESS in 2h 06m 21s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c13b06f1275b43a2a8d4a4151a4b7228 : SUCCESS in 28m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d11e29e4fb941b2ae0839e54fd1082a : SUCCESS in 25m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9dbdf3f14bd041b792f9a643cc7494df : SUCCESS in 28m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8b8d05cb31cb41b889cd14ccdf5da956 : SUCCESS in 12m 52s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e9082d1dd83e4bf69b8f21a7ab6a7aeb : SUCCESS in 9m 44s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":3},{"id":"c765599c38286f2fc3e321e32077a7b5b8881e98","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-22 09:51:53.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a8a7eb1c9dfecf4af99e9bed505401cd935b33f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 09:52:06.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"46e092434618697363e00c2c98c9cf0a917acb73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 11:37:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a7574a503154a96aaa2c28bd9b15f4a : SUCCESS in 7m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ecbec8bb6294f5aac6b7b03a25d349f : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5451b7b509774c1881b6043f99bdaebc : SUCCESS in 7m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a296ae001a64376b84cabd61b79e068 : SUCCESS in 9m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9f06408c681f47169f35df1ff89d5874 : SUCCESS in 6m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1d3a26c5588544ea90ed9516664028f0 : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1a0a3f2eb5eb4f199f4381f826817167 : SUCCESS in 21m 47s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/39ca452a20d747ceb52371ec0c8d1270 : SUCCESS in 59m 17s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d9969ef23984f358b969dc093883069 : SUCCESS in 47m 26s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/9a88a184608e429db48b885798eb3b5e : SUCCESS in 1h 13m 06s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0832867edc114057a3e3c6f05f614010 : SUCCESS in 17m 17s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2283482c8a224b3aab5e2d68e1ea5db5 : SUCCESS in 13m 23s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b6fcd0ae2f5e41dba96c183e4e10e835 : SUCCESS in 1h 43m 49s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/ca5c8eb16f8046a6a6ab075522485c6a : SUCCESS in 29m 27s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5129cfab4e154112843e1a72e2030b18 : SUCCESS in 24m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/2407d85d69d444abb44045511412adfe : SUCCESS in 28m 30s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/e41705a5d58646db8881cb78614de7c7 : SUCCESS in 10m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56da5671d70a4d5aa8cf2ad942739625 : SUCCESS in 12m 49s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"181521cca3e43cac126ad86517104ab984162b5f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 11:37:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"d67aa94c1aa55be76de3782e5f5321a9f03a4977","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 11:39:25.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/096a3a166b214020af2646fa538e01fa : SUCCESS in 59s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b78218adc6f9a88de50ecf9d0cab96ee6287da19","revisions":{"445cb26a41fa41b0f1fe658a510d25b2fd48100c":{"kind":"REWORK","_number":1,"created":"2021-04-09 15:47:25.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/785673/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/785673/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/785673/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/785673/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/785673/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/785673/1"}}},"commit":{"parents":[{"commit":"1a20aa071b135b4a3d7a78824c6e26360c178203","subject":"Add documentation for multiple environments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1a20aa071b135b4a3d7a78824c6e26360c178203"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-09 15:47:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-09 15:47:12.000000000","tz":120},"subject":"Exclude shared directories when creating environments","message":"Exclude shared directories when creating environments\n\nIf we try to create a new environment from a shared configuration inside\nthe same repository, we can produce an infinite loop by copying the\nenvironment into itself. Ignore any existing environments directory\nduring environment creation.\n\nChange-Id: Ic9d3d03192f515a055e41351a0bf25da7a7b0684\nStory: 2002009\nTask: 40038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/445cb26a41fa41b0f1fe658a510d25b2fd48100c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/445cb26a41fa41b0f1fe658a510d25b2fd48100c"}]},"branch":"refs/heads/master"},"cc8fc5fb285f4b9d6f3e592fe1b111956bf022d3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-04-13 09:06:11.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/785673/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/785673/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/785673/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/785673/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/785673/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/785673/2"}}},"commit":{"parents":[{"commit":"4dba1f3ee115e25afe4d982fc59e332c364fa170","subject":"Merge \"CI: use retries for control host bootstrap in seed VM jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4dba1f3ee115e25afe4d982fc59e332c364fa170"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-09 15:47:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-13 08:36:50.000000000","tz":120},"subject":"Exclude shared directories when creating environments","message":"Exclude shared directories when creating environments\n\nIf we try to create a new environment from a shared configuration inside\nthe same repository, we can produce an infinite loop by copying the\nenvironment into itself. Ignore any existing environments directory\nduring environment creation.\n\nChange-Id: Ic9d3d03192f515a055e41351a0bf25da7a7b0684\nStory: 2002009\nTask: 40038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc8fc5fb285f4b9d6f3e592fe1b111956bf022d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cc8fc5fb285f4b9d6f3e592fe1b111956bf022d3"}]},"branch":"refs/heads/master"},"b78218adc6f9a88de50ecf9d0cab96ee6287da19":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-04-22 07:22:15.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/73/785673/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/73/785673/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/785673/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/785673/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/73/785673/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/73/785673/3"}}},"commit":{"parents":[{"commit":"4ad010a5a60811135b43b9ac9d16c8cf01d781dc","subject":"CI: switch to quay.io for container images","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ad010a5a60811135b43b9ac9d16c8cf01d781dc"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-09 15:47:12.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-22 07:21:27.000000000","tz":120},"subject":"Exclude shared directories when creating environments","message":"Exclude shared directories when creating environments\n\nIf we try to create a new environment from a shared configuration inside\nthe same repository, we can produce an infinite loop by copying the\nenvironment into itself. Ignore any existing environments directory\nduring environment creation.\n\nChange-Id: Ic9d3d03192f515a055e41351a0bf25da7a7b0684\nStory: 2002009\nTask: 40038\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b78218adc6f9a88de50ecf9d0cab96ee6287da19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b78218adc6f9a88de50ecf9d0cab96ee6287da19"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
