)]}'
{"id":"openstack%2Fkayobe~786181","triplet_id":"openstack%2Fkayobe~master~I8f99adf19d192db318e8d76d611c3a15adc9ef08","project":"openstack/kayobe","branch":"master","topic":"remove-ntp-yml","hashtags":[],"change_id":"I8f99adf19d192db318e8d76d611c3a15adc9ef08","subject":"Remove ntp.yml configuration file","status":"MERGED","created":"2021-04-14 09:04:04.000000000","updated":"2021-04-27 11:04:24.000000000","submitted":"2021-04-27 11:02:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"786181-remove-ntp-yml","meta_rev_id":"f12dd43e41dba19d85db05a663dd32ea070bf2e0","_number":786181,"virtual_id_number":786181,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-27 11:02:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2021-04-26 07:10:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-04-19 09:23:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2021-04-26 07:10:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-14 11:14:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-19 09:23:40.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-04-26 07:10:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"}],"messages":[{"id":"ee26b525726bb7e457a91e97f6e6e12122d1b355","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-14 09:04:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a674d43e3ebd472062cc9bb808e45533d91a2d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 11:14:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cac09d6c61c6423da64020db48a4d01d : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d5ff07ac9504848b84c42b408ed00d9 : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1e3d0dcfe5514e6b8e0ac05afec842e0 : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e91e63ae9fcb416f98f38d7de29fd654 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbf8d10133de40e6b7f51dbcaef2863c : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40789bcf764d448db6385ddb41a745a2 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/169b913240db405fb3bed91bc5c34340 : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/64ea9e2db935443895929eaa8889c828 : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c373ad88f262416188f78be5052e021b : SUCCESS in 11m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e85a2833940f49c18b1bd6dbf5922247 : SUCCESS in 21m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/401ef4efc09f40db971974dac707e852 : SUCCESS in 58m 53s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ed7764757c6b40c0b3f87a82ca219394 : SUCCESS in 45m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c898dddd0554464da1bdd65ba8c14f15 : SUCCESS in 59m 43s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/84f3c99ca14341bbb26529b5074ebaa0 : SUCCESS in 16m 52s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69e1f91bcfce47199a598ce22c50f99d : SUCCESS in 13m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/bbcaca5c4d554fe7b2720f60612abee4 : SUCCESS in 2h 05m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d02b01d387c440ab8159e55846d26224 : SUCCESS in 31m 10s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70ce2c81eb6e46be81a744e12f0130bc : SUCCESS in 24m 12s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/f927415dbd6a46f6aa23c8973736e0c6 : SUCCESS in 26m 07s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b8e116e2a90944d4a34dc2be40fe48f7 : SUCCESS in 11m 30s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/59677f9b651843b8bc3e3c36c1905148 : SUCCESS in 10m 38s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"6112fbb4bd4fad5624f56bfa3f76a9037c5a3803","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-19 09:23:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f6e5553cac233d0d2e1530a74eb758b6e6efa016","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-26 07:10:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fbebfdaa30e46f0d6efa8a71f1bea91d2ada6638","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 07:10:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5ab2b4f21bbe3fe5e051f958ae1619bed4a7f6f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 07:44:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c8e8c181f5c4b938b6d87c8cffaf52b : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb9065e32f424523af6391e948a932f5 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf30d3dfe77c48b39f05110e66aab7eb : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7cb66a532034172bde706e65ddaaa02 : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cca65e6dde0341dd97f2501883f4ceeb : SUCCESS in 3m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a43d2ae9b8084d4a988a7601452f90b4 : FAILURE in 7m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bccbc4cc257549678e072349f2e2669d : FAILURE in 5m 56s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d27ce5c2cc3e49b3a3d8465f0c79098d : FAILURE in 7m 48s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7e991cabd6634b68b61e8fe042163f00 : FAILURE in 8m 47s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e43958e86207473aab266b3d07d4adc8 : FAILURE in 6m 22s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/1bf00adfc2324e0ab14d49757c2639b3 : FAILURE in 15m 41s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6d1d3af3f48e432ca6a1d6f704c6fcca : FAILURE in 9m 23s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c9c76c357474e4397926498ea71aea8 : FAILURE in 7m 17s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e5c493e21aff460eb7c895c655809e45 : FAILURE in 9m 30s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/c78be47607614cfea87a23a88d4f2536 : FAILURE in 7m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b60b433d921478484a3ad642904688d : FAILURE in 5m 45s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/eb5bbca8afe9492d917d55e237b65251 : FAILURE in 6m 36s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/c8b2d33bb83247939fc6f6a00ed2980d : FAILURE in 7m 22s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6a7603c8cb7c45369eb1d0b558d45463 : FAILURE in 6m 21s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"7b06474e3b04633e7e38bddea12bd9a98df54cc7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-27 06:51:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f3641028cf91e8ad089f1c30b297b061e2a7267d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 08:46:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6217fb0379ae44e08f32c8d0e55711be : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/541667c5b17d4ff5b7adf6aa91f75fb5 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a81fd0c63ac4bc0b267ad1406c7156c : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/996817cd0cc946ff8a99500f5fc953e6 : SUCCESS in 9m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55fcff9bdc8d42eda444072f2aa11667 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5df49baf8b384a95b699ace9acd4afec : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aa2ecf08177b4d01970aab47b171b537 : SUCCESS in 4m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/48327d15081c4e6d9557468a01386bbf : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0478dfcf14de4318a1ec7b62cdd37954 : SUCCESS in 10m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/456f9bdd9127490488d40f3a008a52d3 : SUCCESS in 19m 10s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3daf07fd2eaf4e0d8e459dd7042bd85a : SUCCESS in 1h 04m 15s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/893507d0bfcb49f08f7c64e04b98e33d : SUCCESS in 43m 43s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/f028c468ef264efd931952feae4835aa : SUCCESS in 1h 08m 24s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/50c0df8b7e764a1d8c25197178c84310 : SUCCESS in 17m 12s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/39cfe3acf2ba4b7fb8ad1f3f1f9ff5e1 : SUCCESS in 13m 56s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9bf4e1d87b814004854392296d04ef0d : SUCCESS in 1h 53m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/8082d0848d0e45358ea441fb8d9f8d4e : SUCCESS in 27m 39s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5bc7c498e82149ce86e6b97ca993c68a : SUCCESS in 25m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/955141ec77ab4457ae29626a20086fdf : SUCCESS in 27m 20s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ae5668d6df734135b1f1a9886c48af43 : SUCCESS in 12m 05s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ee29c85806ca43eda71331834c6e3876 : SUCCESS in 11m 00s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"451fc964452218110a22efeef303e5a079822ed1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 08:46:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f37f5b885e70fad5384df8b74fe943e7da78c20e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 11:02:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd057de38aa141519e6fb7342260b7d5 : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/951a25831c64428c8ba49b563e3f2711 : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27c2e901255e4fa6bde095148e03d809 : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65aec095ee204516b79115e1d8390b61 : SUCCESS in 10m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a46f6aa6caed4225afcd2ee00327b394 : SUCCESS in 5m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c1ee5fded2dd4037a1590a2c36d291a0 : SUCCESS in 6m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/416c0de7187445f6aa9c7de44a3ef6a6 : SUCCESS in 12m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/34d37ae6d22f4c28a7c1169332599278 : SUCCESS in 21m 11s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9eea6a4625264d588dcc08119c39e370 : SUCCESS in 1h 04m 48s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/83f71a6f956747f69c37ba98a2a180cd : SUCCESS in 48m 12s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/bbb0b1842c46478083361083b8538bfc : SUCCESS in 1h 06m 40s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f581ec1b547543479fb0d6bec01a76c1 : SUCCESS in 17m 48s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d50124e680d14b649816ca1687227fde : SUCCESS in 15m 20s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/75c5d486f81e4587b598cbb6f1911b65 : SUCCESS in 2h 13m 27s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/72e2ae1589db41bb97d3c9c73ad56ece : SUCCESS in 45m 49s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c90ab5650bbb4ad18a9c894d1506e84e : SUCCESS in 27m 56s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/fe4c1d11bb894b119387b42cd02799b6 : SUCCESS in 26m 56s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/05926cbabe044124add01956f9ff7f91 : SUCCESS in 13m 17s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a977d1883f1b4bcdb2131d24b3b0240e : SUCCESS in 10m 31s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"2766b96557c9282e03054eb241b5b9c5fdb94daa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 11:02:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f12dd43e41dba19d85db05a663dd32ea070bf2e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 11:04:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e4bb1186b20488e8afd3f94281427ae : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12c99e73491e4e6c8faafd675ae8518b : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04fffeb50c5f2792c099641222bc38d0ec29e33b","revisions":{"04fffeb50c5f2792c099641222bc38d0ec29e33b":{"kind":"REWORK","_number":1,"created":"2021-04-14 09:04:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/81/786181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/81/786181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/786181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/786181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/786181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/81/786181/1"}}},"commit":{"parents":[{"commit":"4dba1f3ee115e25afe4d982fc59e332c364fa170","subject":"Merge \"CI: use retries for control host bootstrap in seed VM jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4dba1f3ee115e25afe4d982fc59e332c364fa170"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-14 09:03:15.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-14 09:03:15.000000000","tz":120},"subject":"Remove ntp.yml configuration file","message":"Remove ntp.yml configuration file\n\nTimezone configuration is using time.yml since the Ussuri release.\n\nChange-Id: I8f99adf19d192db318e8d76d611c3a15adc9ef08\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/04fffeb50c5f2792c099641222bc38d0ec29e33b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/04fffeb50c5f2792c099641222bc38d0ec29e33b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
