)]}'
{"id":"openstack%2Fkayobe~787347","triplet_id":"openstack%2Fkayobe~master~Id6690dc9a486af0ed7d42555d84cf77d6de0b6af","project":"openstack/kayobe","branch":"master","topic":"quay-io","hashtags":[],"change_id":"Id6690dc9a486af0ed7d42555d84cf77d6de0b6af","subject":"CI: switch to quay.io for container images","status":"MERGED","created":"2021-04-21 11:01:23.000000000","updated":"2021-04-22 10:17:42.000000000","submitted":"2021-04-22 10:15:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787347-quay-io","meta_rev_id":"489f408475271f002a16cdd4517a302a60745e08","_number":787347,"virtual_id_number":787347,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-22 10:15:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-04-22 08:19:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-22 05:52:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2021-04-22 08:19:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-22 10:17:42.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-21 12:05:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-21 15:12:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"82206a550f6c951f3987b36c218a3e153341cf43","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-21 11:01:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"73bc4f7d8a07b231b115028a0da4d6f53d8e4c34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 12:05:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3982bd0c695a416cb0afe091493a891b : SUCCESS in 1h 01m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0461d7610c6e083e0d544ae005799fd8ba1619dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-21 15:12:14.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"09dd4318374ff785dfae94b6da246674ee6a9d39","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-21 15:18:54.000000000","message":"Patch Set 2: Code-Review-1\n\nNeeds config for other jobs","accounts_in_message":[],"_revision_number":2},{"id":"9a780f0914bfbb7e65f509ebb814187171dbf847","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-21 15:25:10.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"53351998c1b4c656a437c1a00b6a7fe0096382f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-21 16:39:01.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d28009a73e7ca71d78983088554ed58849be3985","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-21 16:45:43.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e4b3cc0fecbc39efff76d2232f0bb199cd0d23a4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-21 16:45:56.000000000","message":"Patch Set 4: Review-Priority+2","accounts_in_message":[],"_revision_number":4},{"id":"97c4a8a1a32c99a352ebd968c0c04fb9ee72d987","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 18:01:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42523596a0ee4d9595eff5c7bdc6c30a : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/503c16de6c404618b1b99ff167005191 : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0f7ee60523fc490195ecd865a15e6347 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a48c941915864c0081c8a6a7b8323688 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/822364b248ed43d187c8f7fa1ab8b2b0 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34ed9676580d44f28deee0d8c20f4a3d : SUCCESS in 7m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae4bdedd56c648cb8022e251a816ba66 : SUCCESS in 5m 34s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/973ee95bd82848f4a804fc828f22b70b : SUCCESS in 9m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2178ea55f7f84067bce8a93eeb59e074 : SUCCESS in 17m 15s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/3beebc278c744854b51f476b198832f1 : SUCCESS in 50m 18s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c189d12b298b4a469124cf651049ec64 : SUCCESS in 38m 31s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/0c349af37b5e409390fc802cabb4a2fe : SUCCESS in 54m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/7b92ded46f5b46f2b40b29b4780e349b : SUCCESS in 15m 22s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/04c75eee83cf4778983e8de5bbbdf654 : SUCCESS in 21m 39s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e77dd3d8d96e4b049c6812eb47de1314 : FAILURE in 1h 19m 25s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/4bd9170bb63044538868c7dc9dd06999 : SUCCESS in 55m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c7ced2eb35f04808ac5019002e978e3d : SUCCESS in 21m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b0b32c4d834d4a2bb282f8afec59387c : SUCCESS in 48m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b33f1487e1434f0f8c96f03a374bb60f : SUCCESS in 10m 47s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/281e6e157b674db0aa8afa15bc6f4578 : SUCCESS in 9m 31s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":4},{"id":"4e47b703cda7b7ed7a1d12b51f72cf0cf5148056","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-21 18:52:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"c4e0656253232e8e41b8b4231993548dcaee3135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 23:04:40.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9477fae49717467d90419b4d96b6af46 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7c284dd43c149d9a69f719846620054 : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ced05b9e0d124fa2a347393066e8caab : SUCCESS in 8m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f70fb41ca49945aeb205a1c661b70c3c : SUCCESS in 7m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/89dc21a5f83c49199d9efa79b9c1423f : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f78ad1603954ffe8e0cb9a71fb6b22c : SUCCESS in 12m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f1099bdad4314bcbb93c5d107d57590a : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/38389ec646c443628de251fba821c98b : SUCCESS in 12m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/daed75c362f943c997b371d2b4be454d : SUCCESS in 16m 34s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/21c989990afb44e08c98cf721e95a0e8 : SUCCESS in 1h 00m 45s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1e231fdf0ba54fbca8d2d074015ffd2e : SUCCESS in 44m 17s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2199b9d62b654c6ca6bc1eacba964ac9 : SUCCESS in 54m 21s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/d8436655a5254abd82a01fd5636ae347 : SUCCESS in 17m 46s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4c066f83de014e52a83d06d99a7ff69d : SUCCESS in 11m 26s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9e7565d4ff224b698db5f6c47b98c5db : SUCCESS in 1h 47m 42s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/58bc67f146544d6d9dee6cfbd92b1b2d : SUCCESS in 29m 41s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4e700a5d6d654284a9a7bcb3bde23384 : SUCCESS in 25m 51s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/50edc07bf2d844f08cc2afcdb57501fd : SUCCESS in 27m 15s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/eb5ceabbaf9649658567c30425cfa3ec : SUCCESS in 11m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb1173b0556f4b4ab601bc4df2ff28bf : SUCCESS in 10m 58s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh","accounts_in_message":[],"_revision_number":5},{"id":"6c52691a51fea2d2c4edd7b785822c2e11dcf4d7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-22 05:52:03.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"46236f1c9da9ccd03661891c62654a4ae1bb9fe5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-22 08:19:38.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ea59345fcbfd2a01594e1b0d2076e3ecee163c40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 08:19:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"87f8993775af4359eca2a207c9b8a5a8f2f31542","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 10:15:41.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30b91090d93e4474b250edf51e8c5c53 : SUCCESS in 7m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c95c7547ed5b49c5a98d3b034c955d1c : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a507cfe74fb6468a8f775f13352e9c02 : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31065b41fe6474e9961019516c586ad : SUCCESS in 11m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/acb14b3cd1b541d4a433e76d1eb8b6fe : SUCCESS in 6m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e95cc43f534b4021bda1b5a6fd9d31fc : SUCCESS in 12m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eb7b2826b89244b3a104b400c670177f : SUCCESS in 21m 01s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/013f34a4433447d7adf6748bd7f76631 : SUCCESS in 1h 11m 00s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6403b5c3cfc4434784b398c5512f465f : SUCCESS in 40m 29s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/2ceb3cbd336744ab80aecf360071b1c0 : SUCCESS in 1h 11m 33s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/6b24e844644a4deb8538b117f6c392d8 : SUCCESS in 17m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19887e413bd644a1a74a686e2436d90e : SUCCESS in 15m 54s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3e53c17698854727a55f4cf77bc543e7 : SUCCESS in 1h 54m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/fc0d1baaef874312a6493c83dd85e270 : SUCCESS in 26m 51s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/34817d145624480a83546fddc37548d3 : SUCCESS in 19m 42s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4914740998ec41e6ab522ceed8c61d5e : SUCCESS in 23m 35s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/3f128ab045e442eaac0b6f69f776ee04 : SUCCESS in 12m 36s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54c1e99443ee4066b51a2f1ef44f41e2 : SUCCESS in 10m 42s\n\nWarning:\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":5},{"id":"b25eb944ac459f5c11d18aa110dd56db4d20dcbb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 10:15:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"ff7e48652cc1e44f3a9a2ee5a27401e3a4deb858","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-22 10:17:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fa21ed849f64bf3a01e854a1c89d692 : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"4ad010a5a60811135b43b9ac9d16c8cf01d781dc","revisions":{"38933a37fffaff83a0751d58ee9d02538503f8be":{"kind":"REWORK","_number":1,"created":"2021-04-21 11:01:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/47/787347/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/787347/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/787347/1"}}},"commit":{"parents":[{"commit":"4dba1f3ee115e25afe4d982fc59e332c364fa170","subject":"Merge \"CI: use retries for control host bootstrap in seed VM jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4dba1f3ee115e25afe4d982fc59e332c364fa170"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-21 11:00:52.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-21 11:01:12.000000000","tz":60},"subject":"WIP: CI: switch to quay.io for container images","message":"WIP: CI: switch to quay.io for container images\n\nChange-Id: Id6690dc9a486af0ed7d42555d84cf77d6de0b6af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38933a37fffaff83a0751d58ee9d02538503f8be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38933a37fffaff83a0751d58ee9d02538503f8be"}]},"branch":"refs/heads/master"},"6387d614795764e30b16721435467b3027d853bd":{"kind":"REWORK","_number":2,"created":"2021-04-21 15:12:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/47/787347/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/787347/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/787347/2"}}},"commit":{"parents":[{"commit":"4dba1f3ee115e25afe4d982fc59e332c364fa170","subject":"Merge \"CI: use retries for control host bootstrap in seed VM jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4dba1f3ee115e25afe4d982fc59e332c364fa170"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-21 11:00:52.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-21 15:11:24.000000000","tz":120},"subject":"CI: switch to quay.io for container images","message":"CI: switch to quay.io for container images\n\nChange-Id: Id6690dc9a486af0ed7d42555d84cf77d6de0b6af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6387d614795764e30b16721435467b3027d853bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6387d614795764e30b16721435467b3027d853bd"}]},"branch":"refs/heads/master"},"a5cdc083fe9a661ec2bb27f3ea97b91efe078981":{"kind":"REWORK","_number":3,"created":"2021-04-21 15:25:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/47/787347/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/787347/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/787347/3"}}},"commit":{"parents":[{"commit":"4dba1f3ee115e25afe4d982fc59e332c364fa170","subject":"Merge \"CI: use retries for control host bootstrap in seed VM jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4dba1f3ee115e25afe4d982fc59e332c364fa170"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-21 11:00:52.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-21 15:24:54.000000000","tz":60},"subject":"CI: switch to quay.io for container images","message":"CI: switch to quay.io for container images\n\nBased on the discussions from recent Kolla meetings [1][2][3],\nwe established that, to avoid the DockerHub pull limit issues\nin our CI, we will try using quay.io.\n\nThe images have already been published via Kolla CI\nfor master/Wallaby [4] and Victoria [5].\n\nThis change applies only to CI.\nThe user-facing move is yet to be discussed.\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-17-15.00.log.html#l-102\n[2] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-24-15.00.log.html#l-137\n[3] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-31-15.01.log.html#l-50\n[4] https://review.opendev.org/c/openstack/kolla/+/781130\n[5] https://review.opendev.org/c/openstack/kolla/+/781899\n\nSee related Kolla Ansible change\nI22e11a994801447526ac26fc2815a1679876225e\n\nChange-Id: Id6690dc9a486af0ed7d42555d84cf77d6de0b6af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5cdc083fe9a661ec2bb27f3ea97b91efe078981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a5cdc083fe9a661ec2bb27f3ea97b91efe078981"}]},"branch":"refs/heads/master"},"d859743575eea5f9dc32d6ea72d5b7ea5e17b501":{"kind":"REWORK","_number":4,"created":"2021-04-21 16:39:01.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/47/787347/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/787347/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/787347/4"}}},"commit":{"parents":[{"commit":"4dba1f3ee115e25afe4d982fc59e332c364fa170","subject":"Merge \"CI: use retries for control host bootstrap in seed VM jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4dba1f3ee115e25afe4d982fc59e332c364fa170"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-21 11:00:52.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-21 16:38:29.000000000","tz":120},"subject":"CI: switch to quay.io for container images","message":"CI: switch to quay.io for container images\n\nBased on the discussions from recent Kolla meetings [1][2][3],\nwe established that, to avoid the DockerHub pull limit issues\nin our CI, we will try using quay.io.\n\nThe images have already been published via Kolla CI\nfor master/Wallaby [4] and Victoria [5].\n\nThis change applies only to CI.\nThe user-facing move is yet to be discussed.\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-17-15.00.log.html#l-102\n[2] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-24-15.00.log.html#l-137\n[3] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-31-15.01.log.html#l-50\n[4] https://review.opendev.org/c/openstack/kolla/+/781130\n[5] https://review.opendev.org/c/openstack/kolla/+/781899\n\nSee related Kolla Ansible change\nI22e11a994801447526ac26fc2815a1679876225e\n\nChange-Id: Id6690dc9a486af0ed7d42555d84cf77d6de0b6af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d859743575eea5f9dc32d6ea72d5b7ea5e17b501"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d859743575eea5f9dc32d6ea72d5b7ea5e17b501"}]},"branch":"refs/heads/master"},"4ad010a5a60811135b43b9ac9d16c8cf01d781dc":{"kind":"REWORK","_number":5,"created":"2021-04-21 18:52:23.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/47/787347/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/47/787347/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/47/787347/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/47/787347/5"}}},"commit":{"parents":[{"commit":"4dba1f3ee115e25afe4d982fc59e332c364fa170","subject":"Merge \"CI: use retries for control host bootstrap in seed VM jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4dba1f3ee115e25afe4d982fc59e332c364fa170"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-21 11:00:52.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-21 18:52:07.000000000","tz":120},"subject":"CI: switch to quay.io for container images","message":"CI: switch to quay.io for container images\n\nBased on the discussions from recent Kolla meetings [1][2][3],\nwe established that, to avoid the DockerHub pull limit issues\nin our CI, we will try using quay.io.\n\nThe images have already been published via Kolla CI\nfor master/Wallaby [4] and Victoria [5].\n\nThis change applies only to CI.\nThe user-facing move is yet to be discussed.\n\n[1] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-17-15.00.log.html#l-102\n[2] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-24-15.00.log.html#l-137\n[3] http://eavesdrop.openstack.org/meetings/kolla/2021/kolla.2021-03-31-15.01.log.html#l-50\n[4] https://review.opendev.org/c/openstack/kolla/+/781130\n[5] https://review.opendev.org/c/openstack/kolla/+/781899\n\nSee related Kolla Ansible change\nI22e11a994801447526ac26fc2815a1679876225e\n\nChange-Id: Id6690dc9a486af0ed7d42555d84cf77d6de0b6af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ad010a5a60811135b43b9ac9d16c8cf01d781dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4ad010a5a60811135b43b9ac9d16c8cf01d781dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
