)]}'
{"id":"openstack%2Fkayobe~787733","triplet_id":"openstack%2Fkayobe~master~I23eb5116d8d1944167a0d50cafe5b99f766ad687","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I23eb5116d8d1944167a0d50cafe5b99f766ad687","subject":"Disable Bifrost cleaning by default","status":"MERGED","created":"2021-04-23 12:56:59.000000000","updated":"2021-04-27 12:43:00.000000000","submitted":"2021-04-27 12:41:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"787733","meta_rev_id":"ac7fa772f55776f21dfd13ff30b84ecef2f578af","_number":787733,"virtual_id_number":787733,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-27 12:41:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2021-04-23 16:42:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-23 16:41:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2021-04-23 16:42:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-23 13:12:48.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-04-23 13:14:50.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2021-04-23 13:17:40.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2021-04-23 13:29:06.000000000","updated_by":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"reviewer":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"state":"REVIEWER"},{"updated":"2021-04-23 14:45:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-23 16:41:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"}],"messages":[{"id":"498554c35f45246652b6c5c63951efa3c1a7b36b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-23 12:56:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3617ae10d179d001f38a564de875cbfeeac7d2fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-23 12:57:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"a695963bd9aabcc961654176a14179dee2702ff3","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-23 13:12:48.000000000","message":"Patch Set 2: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"db1dd49d3299adc9461b65f15c03ccc9eb0b83f1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-23 13:14:50.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nAgreed, let\u0027s just fix the commit and reno.","accounts_in_message":[],"_revision_number":2},{"id":"289455da3890d8e19fb50156d3dab651996539d5","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-04-23 13:17:40.000000000","message":"Patch Set 2: Code-Review+2\n\nThe easiest thing to do..","accounts_in_message":[],"_revision_number":2},{"id":"f9ad2aaf53b3adc58e37a7c7ff91d6a1978b2cab","author":{"_account_id":782,"name":"John Garbutt","email":"john@johngarbutt.com","username":"johngarbutt"},"date":"2021-04-23 13:29:06.000000000","message":"Patch Set 2: Code-Review+2\n\nGood call.","accounts_in_message":[],"_revision_number":2},{"id":"923395838f83cb90934cfb3e3f2c4f2316a24a4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 14:45:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/546c8f99aadc4f94835a20c1995cae1c : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3a0936f87b0412dbf89c1969b358b7c : SUCCESS in 6m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1aba037089944840b6f08db302b4efba : SUCCESS in 9m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aafe30282eb447e5ae47a9d9cfa76b19 : SUCCESS in 9m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef3b459ae51147b1b293a47a9d50a894 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6736ec326bfb4c7dac7d46ca2b64d211 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f619ed2c7c5949739dfcf4c84a47a291 : SUCCESS in 4m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/54a4e6b576674e87b54edcb0431318a1 : SUCCESS in 6m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e826f3a0b6a74d46bc2e50501851f286 : SUCCESS in 11m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a6d6afca06de4fbc91468adbd02b1b6d : SUCCESS in 21m 49s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/d22675b9536240c3ab8618d1bacafcff : SUCCESS in 1h 02m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b4ad010e7824eafa5e328f3c285b94e : SUCCESS in 48m 55s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/dcdef0a287234a67875c10994769dd3d : SUCCESS in 1h 08m 58s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/b3e073a4cedb4d019f206b7d75d01f0a : FAILURE in 9m 54s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1a97b152e8294aa1989917777937d2cf : SUCCESS in 13m 52s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/9d2e6c8b9cbf4965a5a470fe6b91f620 : SUCCESS in 1h 46m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d1c0192382f34eddaa9ee4c4f679c65d : SUCCESS in 30m 18s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaaada11c4f64ae9b463a538f0ef3e26 : SUCCESS in 18m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4a86363ce8c648f78169b3f2aee51f82 : SUCCESS in 23m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/61fadf7a3a93449fa9175e8f5b68c096 : SUCCESS in 11m 32s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7ae8c19808604bc7ae7010786a2a80f8 : SUCCESS in 9m 38s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"d647b5e992806ed0e1a044b19191020d38f6de83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-23 15:03:46.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"3b1106544a430615a5c8d6fcf276954738151db6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-23 16:41:23.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"abb10f76fe703b92d95003ffd363a6a05c157e54","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2021-04-23 16:42:17.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dc6261fdd576693296416615b79f195c54d9ae3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 19:16:10.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62a1c811f3474a10a431638196e7af21 : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7095011bd60e4f5dbbfdafa45bdf455d : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/049a74ad495b42ad91c95c278266cfab : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a093fa41acb1462c9b06ab8b6a965c11 : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b8fbd66048a4a2b91b96130847af0cc : SUCCESS in 6m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55b25e3b402e4542bad9b723d96557cb : SUCCESS in 8m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb4dc2af265b4410966f6422230d97ce : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5bf8e3aa24c14103b7b54e8203f5ace7 : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9ef3bd4ecb1c49048ab9eaa93d25655b : SUCCESS in 11m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0bccbd9d088447cf80ee5e7d9b6dfbfc : SUCCESS in 15m 41s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7e370c4c2a7e444b9d314a06187eadf0 : SUCCESS in 58m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b476a58e75e4f39b28ea682d1861d32 : FAILURE in 47m 27s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/578aac5ce7e542008731be4da4335cac : SUCCESS in 58m 11s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/0589e22743f5438dbf2e564c620491db : FAILURE in 9m 01s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d118eb8cc057478d886dcccf9aeaed18 : SUCCESS in 14m 18s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5ed1e6e406eb4e4292b230c77ac3c592 : SUCCESS in 1h 53m 23s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/6cddc91848054364949a24d76320dc46 : SUCCESS in 30m 56s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3373ff84c42e43cea44135a1519435ef : SUCCESS in 24m 53s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3bc072a2abd64994a088397ecd0a8273 : SUCCESS in 29m 58s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/97e34ec3a9f84366aae578b0e410c6f3 : SUCCESS in 11m 18s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f58aece36d844bdb3f56568496140e9 : SUCCESS in 9m 12s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"9fc5aaaef82ad68f3695a4504ca296a446457044","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-23 19:32:19.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3dca7c71308963bc1bc5e3ab96794baa15883dad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 21:25:05.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a66a219f4efb48329629e5fd70726f72 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2810178331714beaa9fb2ee40ef10ffc : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/deeadfae0ed84abd9e7dd4b4df696699 : SUCCESS in 7m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e443d84560d4a0eb86916740bf912d9 : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3746423366ff4e48b77128f6ed677e3c : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfc53e20b0f44686968ff26916ac9a7f : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98d5413df2e940e4bc4f3eaac4e0c212 : SUCCESS in 4m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ec1ad0dca48b46d2aed438702f65f5a1 : SUCCESS in 6m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/23340b4576354d9696fc8740410adcdf : SUCCESS in 11m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b1611c6d098c4e1aaca9ac8499ea1c85 : SUCCESS in 23m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/b24d2c8b057440ee99833f6adafc4dc1 : SUCCESS in 58m 36s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/796f9f636ba74c6988958c68b8ca5ebe : SUCCESS in 42m 18s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d14b40d927ca44ccaced6582d644031f : SUCCESS in 1h 14m 19s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/244c00a1ba114cae99ca17975423bb6d : FAILURE in 8m 47s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1c7c6758a42c48f4a4e642c5c29dc1e5 : SUCCESS in 11m 57s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3c4151fe5573436d99a81b5a62d98438 : SUCCESS in 1h 38m 18s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/65dbd552244440aa932927c65f0a88bf : SUCCESS in 27m 44s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3ede7061b5114b7e9c201f16fa17cff2 : SUCCESS in 31m 06s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/8fba735dfd6a40f5bd534b2fe696fc85 : SUCCESS in 25m 16s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8d6510d5c3584a24abba6873ebbc7cae : SUCCESS in 13m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/522521c3e69e4184a213aae57fbbd3ea : SUCCESS in 11m 13s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"d989c8815eeef4834464feb84c6eb24a5131d946","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-27 06:52:18.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"6353e24803c310c437d5c5b645a2c76769dcbe62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 08:10:44.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd9296444be547c8939348fcd044a321 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e1d3421752d486e881461b6916183ed : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb71cb8a24d949508840ce458b243b5b : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c99332e44e44a1ebab33888bb752cb9 : SUCCESS in 7m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6038229bd5b6411690eb0cb97a386b24 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a008fde214d240f69f42430a8ce6573f : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17376f028cb94c6aa87a18fccce0a99f : SUCCESS in 4m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/30625f35894d42ca8e84a40432cf6ee3 : SUCCESS in 7m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a10ad59fd8684ff19e80188d71bc26ca : SUCCESS in 10m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/34751bce229a4b24be73ced2f3e25fbd : SUCCESS in 20m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/7e53c4af73c44b81b41111e920ed7c5d : SUCCESS in 1h 01m 42s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/804fe27f0b0147d6a82f308cb9cbf71f : SUCCESS in 42m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/6ae7f37735f8404cafcbd43354578d1f : SUCCESS in 1h 01m 59s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/11c36b98ea284ec4b343f959ae7289f3 : SUCCESS in 18m 06s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6be7f314c6a644f488e98fb72f4731b0 : SUCCESS in 12m 12s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/c4690a907aa74d99a05ae65d4be6005f : FAILURE in 1h 14m 14s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/d3416c84806f4f90af0e5beacfb6ecd1 : SUCCESS in 28m 35s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8d76c7cf8988431b96a431079f37cbe0 : SUCCESS in 25m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d93bef8c89d045b197a5e2726f767e21 : SUCCESS in 24m 43s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/f1e34d2bcab04fb68a60cd980ebcc5a0 : SUCCESS in 13m 10s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b67a217473854f77a685f308a8ea2f57 : SUCCESS in 12m 15s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"6dc82b4205b35efcf51cd4394c0a1429a8a2186c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-27 08:53:55.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"44b3fd498781b1393915adb41cfd956add7e1f32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 11:06:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56b060980b614e9da67ba5d508ae1665 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62a11012e94344e1b297356657d71a41 : SUCCESS in 8m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a0f699c2aa53421787479ff9838f1ee7 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/534e5061e65a45d0b5383eedea33e138 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90e7c1b8237f40d08bdb268e3506d31a : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/148e740328df4f8298a682b4a903d332 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b3d3fff765c461bbdf35ef47e97be65 : SUCCESS in 6m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a3dcc3a3f8434f6cae2ba9d5d64872f1 : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dd33438be6554d7abdb1c02fab535d3a : SUCCESS in 11m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/836540868d0840f8b3c5e0d89e2af578 : SUCCESS in 19m 27s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/9a4a10db81e541a2a43ed8c720af3a0e : SUCCESS in 1h 03m 31s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5c28f0f520784e22b47c8d1eac070d6e : SUCCESS in 55m 01s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/763024d91e9341eba1545948ae0a6789 : SUCCESS in 1h 09m 44s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9089be18b77a429a92491122e367383c : SUCCESS in 20m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eedfcab962cb494a8d34ad9761f9dedc : SUCCESS in 12m 36s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5501636f515d4167979e89d9b1ff5ef2 : SUCCESS in 1h 40m 00s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/0ee36773db0c4d16b5d9ce6cb1845b1e : SUCCESS in 25m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/62a3cc939fbe45d6b3b7afe3568259bb : SUCCESS in 23m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/520ce307bd2b4d4e9bdca9441db61b12 : SUCCESS in 24m 52s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/305c0a443f854fa1b8aff919fa74a171 : SUCCESS in 14m 14s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/672f6a431b594b6086403c173be55797 : SUCCESS in 10m 59s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"20291fbcb18261e1df4fdeb06af09bca544ae13e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 11:06:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c488abb1407b37cd2bc58785b0b38551725ca450","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 12:41:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86754d34973a49c199822132f8a5c83c : SUCCESS in 6m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1d4dc54d28348e2adca35f914065755 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8b1b6d7ff1144bfbb97d50b7ec541d6 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27af966645ea4f3f8f23835cd44adfad : SUCCESS in 8m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ab2048a2a99416f8a0c08fba3bfe72e : SUCCESS in 3m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ec9051e614534a4f87ac9bd33bad0e08 : SUCCESS in 7m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b08b5a8f8d3248788cf135ad240a158c : SUCCESS in 10m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3b2ec628b6254f13a64eb3949e8b1b9a : SUCCESS in 21m 36s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6dbed71967204f819ca95884f1c953eb : SUCCESS in 1h 17m 07s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/03b794361849415db1e4499961517018 : SUCCESS in 46m 38s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/421f9c68898049ad8e0d24b472198289 : SUCCESS in 58m 46s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/2d4db654451d4c6492b89041bf3d5b28 : SUCCESS in 18m 45s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb00bfb5a03b46f38c4b52ebd39b0647 : SUCCESS in 11m 34s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/b5331d14b7f24abc8a063fc5f3451ff0 : SUCCESS in 1h 26m 45s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/24a2d540822c4c9c823d038ea23ddcbc : SUCCESS in 27m 42s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/70f92b44cf09440b879470054331089a : SUCCESS in 27m 18s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/457e7f52f8784f5099f43cebd1cab7d6 : SUCCESS in 27m 40s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/622772b013e54a968028219da6496215 : SUCCESS in 12m 29s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/beb3cedd383649559a5dd6090f25e42a : SUCCESS in 11m 09s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":3},{"id":"407f69ced4e5e2365260a2f18473f57aea59f4ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 12:41:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"ac7fa772f55776f21dfd13ff30b84ecef2f578af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 12:43:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af4489070a164043a8ae79c9488ade81 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1633ba5efa3d470d84fe4a1d53a08bd1 : SUCCESS in 1m 04s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d45812d3badc04f7e99fbf4207baf5c45096dc90","revisions":{"725bd53e77c6cf5fc5f01258a504c418de81acba":{"kind":"REWORK","_number":1,"created":"2021-04-23 12:56:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/33/787733/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/787733/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/787733/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/787733/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/787733/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/787733/1"}}},"commit":{"parents":[{"commit":"1561c27fba7e62e2acc74b15f34a0f37ad1e9408","subject":"Merge \"Refactor veth configuration role variable generation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1561c27fba7e62e2acc74b15f34a0f37ad1e9408"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-23 12:40:50.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-23 12:55:08.000000000","tz":60},"subject":"Disable Bifrost cleaning by default","message":"Disable Bifrost cleaning by default\n\nCleaning was disabled by default in Bifrost until the Victoria release,\nwhen the default was changed. The \u0027kayobe overcloud provision\u0027 command\ndoes not handle cleaning states when moving nodes from manageable to\navailable. This can lead to provisioning failures.\n\nThis can This change disables cleaning to maintain backwards\ncompatibility with earlier versions of Kayobe.\n\nChange-Id: I23eb5116d8d1944167a0d50cafe5b99f766ad687\nStory: 2008850\nTask: 42363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/725bd53e77c6cf5fc5f01258a504c418de81acba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/725bd53e77c6cf5fc5f01258a504c418de81acba"}]},"branch":"refs/heads/master"},"5a4a7e3e1b255bf17946f0d8cd557359c770b980":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-04-23 12:57:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/33/787733/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/787733/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/787733/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/787733/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/787733/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/787733/2"}}},"commit":{"parents":[{"commit":"1561c27fba7e62e2acc74b15f34a0f37ad1e9408","subject":"Merge \"Refactor veth configuration role variable generation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1561c27fba7e62e2acc74b15f34a0f37ad1e9408"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-23 12:40:50.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-23 12:57:05.000000000","tz":60},"subject":"Disable Bifrost cleaning by default","message":"Disable Bifrost cleaning by default\n\nCleaning was disabled by default in Bifrost\n(Ie1198768889bd468176cd68c8ccb48791c724262) until the Victoria release,\nwhen the default was changed. The \u0027kayobe overcloud provision\u0027 command\ndoes not handle cleaning states when moving nodes from manageable to\navailable. This can lead to provisioning failures.\n\nThis can This change disables cleaning to maintain backwards\ncompatibility with earlier versions of Kayobe.\n\nChange-Id: I23eb5116d8d1944167a0d50cafe5b99f766ad687\nStory: 2008850\nTask: 42363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a4a7e3e1b255bf17946f0d8cd557359c770b980"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5a4a7e3e1b255bf17946f0d8cd557359c770b980"}]},"branch":"refs/heads/master"},"d45812d3badc04f7e99fbf4207baf5c45096dc90":{"kind":"REWORK","_number":3,"created":"2021-04-23 15:03:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/33/787733/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/787733/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/787733/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/787733/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/787733/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/787733/3"}}},"commit":{"parents":[{"commit":"1561c27fba7e62e2acc74b15f34a0f37ad1e9408","subject":"Merge \"Refactor veth configuration role variable generation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1561c27fba7e62e2acc74b15f34a0f37ad1e9408"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-23 12:40:50.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-23 15:03:40.000000000","tz":0},"subject":"Disable Bifrost cleaning by default","message":"Disable Bifrost cleaning by default\n\nCleaning was disabled by default in Bifrost\n(Ie1198768889bd468176cd68c8ccb48791c724262) until the Victoria release,\nwhen the default was changed. The \u0027kayobe overcloud provision\u0027 command\ndoes not handle cleaning states when moving nodes from manageable to\navailable. This can lead to provisioning failures.\n\nThis change disables cleaning to maintain backwards\ncompatibility with earlier versions of Kayobe.\n\nChange-Id: I23eb5116d8d1944167a0d50cafe5b99f766ad687\nStory: 2008850\nTask: 42363\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d45812d3badc04f7e99fbf4207baf5c45096dc90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d45812d3badc04f7e99fbf4207baf5c45096dc90"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
