)]}'
{"id":"openstack%2Fkayobe~787775","triplet_id":"openstack%2Fkayobe~master~Ibd20a72bd29e20de016fe41de2f1691302e7e24d","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ibd20a72bd29e20de016fe41de2f1691302e7e24d","subject":"[release] Sync Kolla Ansible feature flags and inventory for Wallaby","status":"MERGED","created":"2021-04-23 15:55:25.000000000","updated":"2021-04-28 18:28:16.000000000","submitted":"2021-04-28 18:26:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787775","meta_rev_id":"7c6837b482a10685ab0649d136eb6cc58d83d72e","_number":787775,"virtual_id_number":787775,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-28 18:26:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-04-28 16:22:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-28 08:53:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2021-04-28 16:22:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-23 19:35:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-27 06:53:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"CC"},{"updated":"2021-04-28 08:53:20.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2021-04-28 16:22:09.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"eb5f3c347a2ed79d96ea5fcf7b9b5cf6a010809f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-23 15:55:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6cd2bfa3dc87cf8afa94ab5d42baadd94f0a6435","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-23 16:16:42.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f681c653ce9406de3b49d011a0896a8072acf011","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 19:35:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93ec4b6f87de4b79990c16ae22bcde9a : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be7ea6b524c94a038372b4bfb843ee1e : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/62004ef6dfb245b98fc0119949b1e2a4 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a5421a38fe614eb7b8dc6669015fa326 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04f2f2ec63fa4e54adad2a9594c0b60f : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a8b2322ef834f549e6e4c47b0987ebe : SUCCESS in 9m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2e6003f7a5a24a3fa769cc86fe14173d : SUCCESS in 7m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ec7dda263fb841f6afb8d5e8a8aefb4d : SUCCESS in 11m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7a4d1e89875f4f29bd456b44e8a85905 : SUCCESS in 20m 33s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/6757282efe204a5ba15a592c06494118 : SUCCESS in 1h 03m 41s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a78bf8bd576846f28d5e7cb92c9b31c4 : SUCCESS in 42m 59s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/d228771fbff14b49bb5cb3313020eac5 : SUCCESS in 1h 11m 45s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/f11fd986103d42349f60f7a7aadca558 : FAILURE in 8m 43s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/faa73f6e566147fba80e85533f3fd50b : SUCCESS in 11m 28s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/123cc3e9a1a84ec3a5578d4c817d1af4 : SUCCESS in 1h 37m 34s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/578e581dc3a84ef393e7a633ebf72adf : SUCCESS in 29m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a1579105f76f487c8d6a4fa46fe87b96 : SUCCESS in 22m 29s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/df1a3f4864bf47cf97c56b0c7c1210f0 : SUCCESS in 26m 07s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/246d523a62c9424980b63002feeb8825 : SUCCESS in 13m 00s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3c8948d9f3c94e9f8ca10838d0c61b3e : SUCCESS in 10m 44s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"ecbcc3f8089b92d8060344a61c141b8abb875c88","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-23 19:49:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8eb6d9247401eed3b5ed989b0e6bedc8f8c24ce7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-23 21:30:37.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be527da5d0324f388f01f89fe991e5d5 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc1c05651e604ced8447e33f913d4eb8 : SUCCESS in 6m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3029276db5144620aae161f53451c9aa : SUCCESS in 7m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c49344d90924f1dbab8a31607d5fbfc : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01532b5631af449395462161bcbfa0f8 : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99280bb78a6a4caca32b220c61062ac2 : SUCCESS in 8m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bcc7b71112ba44539bf92e928b7f9a3b : SUCCESS in 6m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/26be74a27b644c519cb6c9763d903276 : SUCCESS in 12m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/31f5e5a02955451e8e144803171fd3f8 : SUCCESS in 17m 18s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/703196e2c7494655ad014f17bed33133 : SUCCESS in 1h 04m 59s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4f17149e1ad74f57a1f7e5f4b7e5fb01 : SUCCESS in 45m 44s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/02a5c980b8c14228937afc0908ab658c : SUCCESS in 1h 07m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/631c0b7a7fd940c89515c3d3970497d8 : FAILURE in 8m 57s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7cb5494d44e647e6b727451f58600d9a : SUCCESS in 12m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/4c1928ccf99d4d60a63ba112c5d4d6dc : SUCCESS in 1h 37m 28s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/2d2747840c97438ba7ef7d05127cce4b : SUCCESS in 29m 33s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/86c82755529e4df5be288602b52c3e62 : SUCCESS in 22m 04s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/86e36d0599ad4c8a8465da1517032830 : SUCCESS in 24m 44s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/ea34d1354394464a813f55479dec6c95 : SUCCESS in 11m 34s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b38bfd9cad45495bb088cf9f45d2bb70 : SUCCESS in 8m 30s\n\nWarning:\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"b5c7f97fd914ff55f590e3baf9520fc12ea4bf43","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-27 06:53:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"bc80f130389aed7425252bb993ac4144108ea5c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 08:41:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbe98f7f73f04a2bbe3ef5f1083504e2 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14ce477277ca4dc28849188fa568d88e : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28505928b3594641899a34b4cd897cf9 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/383c5ff58fc14725996e20fa80a93ac1 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7ff67dd783c4240bef11756c1bde18d : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d4563a5b44c485ba9468206905434f8 : SUCCESS in 8m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/de9a3c159fef4501b15024f78b7da673 : SUCCESS in 9m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/419b9d1ce811485bae4343b1b404341d : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0851eeaabb54485483942caf04a6ef14 : SUCCESS in 19m 14s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/453eb882dd134d039b97a5c263566da1 : SUCCESS in 1h 04m 49s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/47a1bd9955ac444aa9abafc38e630361 : SUCCESS in 44m 54s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c674f12000f744c6996557e0b2193b28 : SUCCESS in 1h 07m 26s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/a646ffb4c48c46cbaa29e010b7d40a8e : SUCCESS in 17m 09s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dcee9a058f33483f9abb3036b6728663 : SUCCESS in 13m 33s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/3f235d9dcfd54fb09eaa5b99f87968df : SUCCESS in 1h 45m 29s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/17674d32bf614013ad7e46f3597e0329 : SUCCESS in 25m 59s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1c4ffa110814e1b82e1e6a1a583feb3 : SUCCESS in 22m 36s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/31e044c8cf7c44ff98899235850b4287 : SUCCESS in 24m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/8071569f61834d60914f40d18edc295c : SUCCESS in 11m 53s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3f46bd16ec8446548119100851e61a8f : SUCCESS in 9m 55s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"1dff6ceb4652a166cbbd4379ff0742d03ed9ae4f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-28 08:53:20.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a2944bc62632cfa729d06f5aa44b6b2b0ac79b74","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-04-28 16:22:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1f2b60c72330d0584a50dc817721735b3d58ffa8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-04-28 16:22:22.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3ea6fddf0e15cc6a16d2b6c7eddc1f5f8f5ba208","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 16:24:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3fc100bdbfd2409572a0ee041c3ccd8b18fb205d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:26:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9004b51b82a54231a1b85acc4b500d6e : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f9063d699daf4c439546472a4f885b01 : SUCCESS in 9m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad6c3ec2c34c456a9de56c87f8504278 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af18378ab5594f56a4f00a940930e5cc : SUCCESS in 9m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/04fc686a64a64da7a7a843f2bdaaf60a : SUCCESS in 9m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9c6501f7ee64e649f0695fd3391c332 : SUCCESS in 11m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6186fb887b7d4e4486131a5152e2e716 : SUCCESS in 16m 57s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ece8c4cd82cf46d8bc95a9455cad781d : SUCCESS in 1h 02m 51s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9fc93fe1d2c74bbca3886587d54bd4f2 : SUCCESS in 47m 03s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/676f792c85b4403cb3a17d4593df4dc2 : SUCCESS in 1h 05m 10s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/e2595c17db4e407ca736f3351dc42f61 : SUCCESS in 20m 58s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aed5f66bb5d0403c97fcbcb4e4790dd4 : SUCCESS in 14m 14s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d090f4532aed47219f764dabcfb730bf : SUCCESS in 1h 54m 36s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/b0770a8b117b479cae1f582b53a05e5e : SUCCESS in 30m 57s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9af51c17c59648e59671489a7d67e377 : SUCCESS in 25m 54s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5e94f744f2a24ac480349d90f3336eb2 : SUCCESS in 29m 17s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/87b3fa94c0924bbfa42828042d02afdc : SUCCESS in 15m 20s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/af038d10b338411684fcbccc12190156 : SUCCESS in 10m 03s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":2},{"id":"df329fc5904f505527706da007714e25f1cab8de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:26:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"7c6837b482a10685ab0649d136eb6cc58d83d72e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-28 18:28:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1db30c2ba0d34d73b16c532e2c17346f : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b279474e00d1588899441896157c1c767ff199b","revisions":{"98935c0d8384e570e0e8f8d64055b72ae6120c37":{"kind":"REWORK","_number":1,"created":"2021-04-23 15:55:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/787775/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/787775/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/787775/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/787775/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/787775/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/787775/1"}}},"commit":{"parents":[{"commit":"1561c27fba7e62e2acc74b15f34a0f37ad1e9408","subject":"Merge \"Refactor veth configuration role variable generation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1561c27fba7e62e2acc74b15f34a0f37ad1e9408"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-23 15:46:40.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-23 15:55:15.000000000","tz":60},"subject":"Sync Kolla Ansible feature flags for Wallaby","message":"Sync Kolla Ansible feature flags for Wallaby\n\nSyncs with Kolla Ansible feature flags for Wallaby release, at commit\n058dd6828d788b60b6803f24f53f1dfca0876e6b.\n\nAlso adds a script to automate the process, and updates the release\ndocumentation to include these steps.\n\nChange-Id: Ibd20a72bd29e20de016fe41de2f1691302e7e24d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98935c0d8384e570e0e8f8d64055b72ae6120c37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/98935c0d8384e570e0e8f8d64055b72ae6120c37"}]},"branch":"refs/heads/master"},"0b279474e00d1588899441896157c1c767ff199b":{"kind":"REWORK","_number":2,"created":"2021-04-23 16:16:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/75/787775/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/75/787775/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/787775/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/787775/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/75/787775/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/75/787775/2"}}},"commit":{"parents":[{"commit":"1561c27fba7e62e2acc74b15f34a0f37ad1e9408","subject":"Merge \"Refactor veth configuration role variable generation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1561c27fba7e62e2acc74b15f34a0f37ad1e9408"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-23 15:46:40.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2021-04-23 16:16:20.000000000","tz":60},"subject":"[release] Sync Kolla Ansible feature flags and inventory for Wallaby","message":"[release] Sync Kolla Ansible feature flags and inventory for Wallaby\n\nSyncs with Kolla Ansible feature flags and inventory for Wallaby\nrelease, at commit 058dd6828d788b60b6803f24f53f1dfca0876e6b.\n\nChange-Id: Ibd20a72bd29e20de016fe41de2f1691302e7e24d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b279474e00d1588899441896157c1c767ff199b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0b279474e00d1588899441896157c1c767ff199b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
