)]}'
{"id":"openstack%2Fkayobe~787934","triplet_id":"openstack%2Fkayobe~master~I3d85b341cf541cfb9b486021e66b316833fe9130","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I3d85b341cf541cfb9b486021e66b316833fe9130","subject":"Use released version of mrlesmithjr.mdadm","status":"MERGED","created":"2021-04-26 08:10:35.000000000","updated":"2021-04-27 00:33:51.000000000","submitted":"2021-04-27 00:32:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787934","meta_rev_id":"a4d72ae0a8daaeac8662f760e141984bf98e6fdc","_number":787934,"virtual_id_number":787934,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-27 00:32:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-04-26 11:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2021-04-26 08:44:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":2,"date":"2021-04-26 08:13:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-04-26 21:05:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-26 08:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"}],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}],"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 08:13:36.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2021-04-26 08:14:51.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2021-04-26 08:44:25.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"REVIEWER"},{"updated":"2021-04-26 08:54:34.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-04-26 09:18:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 11:23:37.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"}],"messages":[{"id":"80eb3638b5f92d845508bd2af200fc5eeb31ad94","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 08:10:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d5e21018f693c5ac898801d4979583f146867a6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 08:11:42.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"2447b224be5f81103efdf1180c5105592b8c8d28","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2021-04-26 08:13:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fe8502b69eb2f55b60cab477bcd3195936874223","author":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"date":"2021-04-26 08:44:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"461eb500e12ae28be0653dfd476c3970a16b1c57","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-26 08:54:34.000000000","message":"Patch Set 1:\n\nHere is the diff from the commit to the tag. Note that the config file location is changed from /etc/mdadm/mdadm.conf to /etc/mdadm.conf for RH distros.\n\ndiff --git a/tasks/arrays.yml b/tasks/arrays.yml\nindex 4832179..a19da92 100644\n--- a/tasks/arrays.yml\n+++ b/tasks/arrays.yml\n@@ -23,7 +23,7 @@\n\n   command: \"{{ update_initramfs }}\"\n   when: array_created.changed\n \n-# Capture the raid array details to append to /etc/mdadm/mdadm.conf\n+# Capture the raid array details to append to mdadm.conf\n # in order to persist between reboots\n - name: arrays | Capturing Array Details\n   command: \"mdadm --detail --scan\"\n@@ -106,31 +106,32 @@\n         array_check.results[0].rc \u003d\u003d 0 and\n         array_removed.changed\n \n-- name: arrays | Ensure /etc/mdadm/ directory exists\n+- name: arrays | Ensure {{ mdadm_conf }}\u0027s directory exists\n   file:\npath: /etc/mdadm/\n+    path: \"{{ mdadm_conf_path }}\"\n\n     state: directory\n+  when: mdadm_conf_path is defined\n \n-- name: arrays | Ensure /etc/mdadm/mdadm.conf file exists\n+- name: arrays | Ensure {{ mdadm_conf }} file exists\n   copy:\n     content: \"\"\ndest: /etc/mdadm/mdadm.conf\n+    dest: \"{{ mdadm_conf }}\"\n\n     force: no\n \n-# Updating /etc/mdadm/mdadm.conf in order to persist between reboots\n-- name: arrays | Updating /etc/mdadm/mdadm.conf\n+# Updating mdadm.conf in order to persist between reboots\n+- name: arrays | Updating {{ mdadm_conf }}\n   lineinfile:\ndest: \"/etc/mdadm/mdadm.conf\"\n+    dest: \"{{ mdadm_conf }}\"\n\n     regexp: \"^{{ item }}\"\n     line: \"{{ item }}\"\n     state: \"present\"\n   with_items: \u0027{{ array_details.stdout_lines }}\u0027\n   when: array_created.changed\n \n-# Updating /etc/mdadm/mdadm.conf in order to not persist between reboots\n-- name: arrays | Updating /etc/mdadm/mdadm.conf\n+# Updating mdadm.conf in order to not persist between reboots\n+- name: arrays | Updating {{ mdadm_conf }}\n   lineinfile:\ndest: \"/etc/mdadm/mdadm.conf\"\n+    dest: \"{{ mdadm_conf }}\"\n\n     regexp: \"^ARRAY /dev/{{ item.name }}\"\n     line: \"ARRAY /dev/{{ item.name }}\"\n     state: \"absent\"\ndiff --git a/tasks/debian.yml b/tasks/debian.yml\nindex ba51213..ed2dfaa 100644\n--- a/tasks/debian.yml\n+++ b/tasks/debian.yml\n@@ -4,6 +4,8 @@\n\n     name: \"mdadm\"\n     state: \"present\"\n \n-- name: arrays | Set update_initramfs cmd to update-initramfs -u\n+- name: arrays | Setting distribution-specific facts\n   set_fact:\n     update_initramfs: \"update-initramfs -u\"\n+    mdadm_conf: \"/etc/mdadm/mdadm.conf\"\n+    mdadm_conf_path: \"/etc/mdadm\"\ndiff --git a/tasks/el.yml b/tasks/el.yml\nindex 5bfdd5d..a0d8f5a 100644\n--- a/tasks/el.yml\n+++ b/tasks/el.yml\n@@ -4,6 +4,7 @@\n     name: \"mdadm\"\n     state: \"present\"\n \n-- name: arrays | Set update_initramfs cmd to dracut -f \n+- name: arrays | Setting distribution-specific facts\n   set_fact:\n     update_initramfs: \"dracut -f\"\n+    mdadm_conf: \"/etc/mdadm.conf","accounts_in_message":[],"_revision_number":1},{"id":"01c7381907cfbdfb3eb7c9d4f4b3eb96a9658629","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 09:18:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/568a6372ce334a76a67e7f3bc2d6c461 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ad4d81d7b6b4bde9de370122d9f5933 : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2dae2b6886084907ac1ed8e0504c6f1d : SUCCESS in 7m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/161544a9688e4bc0ba4663b1cbabc27e : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3ecbb89f5b6442989cafb39d1773a16 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdfc1037756640a6bc9e329652fb1529 : SUCCESS in 8m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/463e0405a2fb48b7b3d114ce4d3bea8c : SUCCESS in 5m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/957d6c116c2a4038bed00e2f3f1b3fbe : SUCCESS in 11m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b82a6022c2274971a0ac29f2ed40dc87 : SUCCESS in 17m 12s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/a08444f959f748caa1e4435c6526c0b1 : SUCCESS in 1h 02m 41s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d1cb702fc55e412888c1ae292b85457c : SUCCESS in 36m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/c28a5c3fedc94c3781093768b97818db : SUCCESS in 1h 05m 49s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/de061b6506cf420fa52b42279bb056a4 : SUCCESS in 17m 51s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/89644aab59eb495e95aad1fe9d4df1a1 : SUCCESS in 15m 07s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/7f2c23e6d224460ca08acc046c9333c1 : FAILURE in 6m 43s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/7d96a658c9fb4dd38bdff09b77a481eb : SUCCESS in 29m 37s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/61be7a754da740d1bd9ca1e6420ae684 : SUCCESS in 24m 10s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/08268501aaed4e088e90ff640861b405 : FAILURE in 5m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/b52b004f75704d7588250c3da7810a60 : SUCCESS in 10m 33s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e1f9df63c054491a7f7c2c96e9d8e35 : SUCCESS in 8m 38s\n\nWarning:\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/release.sh","accounts_in_message":[],"_revision_number":1},{"id":"5171d87dd5c29d0046d273eb0b6187acea21bd20","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-04-26 11:23:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"329a9dd57005c53172098b10ac50998825762371","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 12:25:58.000000000","message":"Patch Set 1:\n\nThis will need to be rechecked once the victoria fix has landed. Not using Depends-On to keep the commit SHA the same.","accounts_in_message":[],"_revision_number":1},{"id":"a707ff144eaef8e2add8596b8cdd792c08988143","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 21:05:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"71b56fd896fd038989606cbfc069e00208bd22de","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 21:05:52.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"83f4202dc3d31610563d15a77c8547f85fb6ddc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 22:59:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/509243ab6457429baa68a91ab8dd7fc6 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92102a72f3df4089aacfea629ab5de1d : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f51128ec1514ec7adac5225168040c0 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02ad976dc20a4337912d4ddfce0926ad : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7227ba858d244eeabbdff8d4b4f65f22 : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a3f3b0af7394b6c9a627f3658321e3c : SUCCESS in 9m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1775a20caf7e413ca1cc3754a6e21c77 : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ec77726258dd4629b0a2fa868b951563 : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a94ee08f67014db682a51907bfe5f4d6 : SUCCESS in 20m 54s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/ab5e7b488b1c4cac95235c121999bcf0 : SUCCESS in 44m 56s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/50fd9d7824734998a82a027fb8525dae : SUCCESS in 41m 20s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e6c08cc896c142f6a1edef2da0cdcc61 : SUCCESS in 49m 54s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/9e50553315844f7ea43574cc7597a736 : SUCCESS in 14m 16s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eeb6dadccd0546bb80361d9ad2cb35e9 : SUCCESS in 9m 55s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/5a4839da4a774ab8980a5377edbb9db6 : SUCCESS in 1h 53m 16s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/e14b5abac12644d9a096eebd7f9b6ee5 : SUCCESS in 30m 19s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0ca05708785e4f19aca7ad03142bcf4c : SUCCESS in 22m 50s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/e61702efaf0f4bce85c48184c5bfd156 : SUCCESS in 28m 32s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/7c610a4d7c1c4382922277dd94f86316 : SUCCESS in 12m 48s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1b5fefee834c48bda3d9d7d936eb7d1f : SUCCESS in 7m 58s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"5f5dda37c317c7edd06aac165da6cc58a583ae8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 22:59:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39421111203631e7b22a6373922b471e736afa65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 00:32:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa22f098e68a4e83a1c557a004d91273 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d21e38c790814dfbb61db8bfc0111bc9 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beb49aed9e674df6834f8e23dc3bfb0a : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21f3da09afd74c0e8362c118cf7af268 : SUCCESS in 8m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e2faa38438494d10bc49507289073a18 : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/16c57892bd574beea0c470f371bb0b5b : SUCCESS in 9m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d020db9e3be1448d8793f4673eeba87a : SUCCESS in 16m 13s\n- kayobe-overcloud-centos8s https://zuul.opendev.org/t/openstack/build/f6a47ae998da4f84b761bdcdb8785e60 : SUCCESS in 44m 52s\n- kayobe-overcloud-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bb24dd6c6eb445d9aa97f4f99c10da60 : SUCCESS in 34m 40s\n- kayobe-overcloud-tls-centos8s https://zuul.opendev.org/t/openstack/build/e710575e4f634dd288d5d3fe005e0db3 : SUCCESS in 54m 31s\n- kayobe-overcloud-host-configure-centos8s https://zuul.opendev.org/t/openstack/build/70e89dd6c49f43a18ef967ff1c7253f8 : SUCCESS in 18m 39s\n- kayobe-overcloud-host-configure-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09ca3baae782480c90d22a973a9e1d19 : SUCCESS in 13m 00s\n- kayobe-overcloud-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/94a77830c54b4fa2affb3a561ec21d2d : SUCCESS in 1h 31m 20s\n- kayobe-seed-centos8s https://zuul.opendev.org/t/openstack/build/3b0d839c43474e0e8623726578656041 : SUCCESS in 28m 02s\n- kayobe-seed-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b486b8ae90514c72bae70b7f5143423f : SUCCESS in 24m 21s\n- kayobe-seed-upgrade-centos8s https://zuul.opendev.org/t/openstack/build/d61e5fc2ecaf4b0d963564f7fc66abab : SUCCESS in 18m 55s\n- kayobe-seed-vm-centos8s https://zuul.opendev.org/t/openstack/build/5a31d201ddbd4476ab1b9d44d792fc0f : SUCCESS in 12m 08s\n- kayobe-seed-vm-ubuntu-focal https://zuul.opendev.org/t/openstack/build/317a39f5943c49ceb5800a529d9d9423 : SUCCESS in 11m 00s\n\nWarning:\n  Comments left for invalid file tools/release.sh\n  Comments left for invalid file tools/ansible-galaxy-retried.sh\n  Comments left for invalid file tools/test-ansible.sh\n  Comments left for invalid file roles/kayobe-diagnostics/files/get_logs.sh","accounts_in_message":[],"_revision_number":1},{"id":"e7cdf1348bdddcfbc5c0275e34bda9807299aa45","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 00:32:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"a4d72ae0a8daaeac8662f760e141984bf98e6fdc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 00:33:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/129d3cf9e0db48acb2b006b827fbfd75 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7411e624dea77a1b94d91bd37c9a388aa5011bb4","revisions":{"7411e624dea77a1b94d91bd37c9a388aa5011bb4":{"kind":"REWORK","_number":1,"created":"2021-04-26 08:10:35.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/34/787934/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/787934/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/787934/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/787934/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/787934/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/787934/1"}}},"commit":{"parents":[{"commit":"83951b01acb4f775e9d4e61c3986dcba9135eaf4","subject":"Merge \"zuul: add more irrelevant files\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/83951b01acb4f775e9d4e61c3986dcba9135eaf4"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-26 08:04:45.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-26 08:04:45.000000000","tz":120},"subject":"Use released version of mrlesmithjr.mdadm","message":"Use released version of mrlesmithjr.mdadm\n\nAn initial version of the mrlesmithjr.mdadm role was released 2 days\nago. There are only minimal changes compared to the commit we were\nusing. This should avoid the following issue seen in CI:\n\n[WARNING]: - mrlesmithjr.mdadm was NOT installed successfully: - the specified\nversion (5be3ee7d330aa17317897bd104dc87ff0df11915) of mrlesmithjr.mdadm was not\nfound in the list of available versions ([{\u0027id\u0027: 152374, \u0027url\u0027: \u0027\u0027, \u0027related\u0027:\n{}, \u0027summary_fields\u0027: {}, \u0027created\u0027: \u00272021-04-24T02:51:30.453954Z\u0027, \u0027modified\u0027:\n\u00272021-04-24T02:51:30.453980Z\u0027, \u0027name\u0027: \u0027v0.1.0\u0027, \u0027version\u0027: \u00270.1.0\u0027,\n\u0027commit_date\u0027: \u00272021-04-23T22:50:00-04:00\u0027, \u0027commit_sha\u0027:\n\u0027b8a9853b7eafa9afc62bbb7728f64fd5dc9f54de\u0027, \u0027download_url\u0027:\n\u0027https://github.com/mrlesmithjr/ansible-mdadm/archive/v0.1.0.tar.gz\u0027, \u0027active\u0027:\nNone}]).\nERROR! - you can use --ignore-errors to skip failed roles and finish processing the list.\n\nChange-Id: I3d85b341cf541cfb9b486021e66b316833fe9130\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7411e624dea77a1b94d91bd37c9a388aa5011bb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7411e624dea77a1b94d91bd37c9a388aa5011bb4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
