)]}'
{"id":"openstack%2Fkayobe~787936","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I3d85b341cf541cfb9b486021e66b316833fe9130","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"I3d85b341cf541cfb9b486021e66b316833fe9130","subject":"Use released version of mrlesmithjr.mdadm","status":"MERGED","created":"2021-04-26 08:22:58.000000000","updated":"2021-04-26 16:47:39.000000000","submitted":"2021-04-26 16:46:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787936","meta_rev_id":"ab3fc8649d3c826cc920f78d01d31e9035fc478d","_number":787936,"virtual_id_number":787936,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 16:46:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-04-26 11:22:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-04-26 12:23:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2021-04-26 12:23:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-26 12:23:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 08:54:08.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-04-26 09:25:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 11:22:41.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-04-26 12:23:15.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"2f58122169ce2e3b969a3aeee643c352b1a9e74b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 08:22:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"873b38a45b406f21430cb6d9664128676eb58682","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-26 08:54:08.000000000","message":"Patch Set 1:\n\nHere is the diff from the commit to the tag. Note that the config file location is changed from /etc/mdadm/mdadm.conf to /etc/mdadm.conf for RH distros.\n\ndiff --git a/tasks/arrays.yml b/tasks/arrays.yml\nindex 4832179..a19da92 100644\n--- a/tasks/arrays.yml\n+++ b/tasks/arrays.yml\n@@ -23,7 +23,7 @@\n   command: \"{{ update_initramfs }}\"\n   when: array_created.changed\n \n-# Capture the raid array details to append to /etc/mdadm/mdadm.conf\n+# Capture the raid array details to append to mdadm.conf\n # in order to persist between reboots\n - name: arrays | Capturing Array Details\n   command: \"mdadm --detail --scan\"\n@@ -106,31 +106,32 @@\n         array_check.results[0].rc \u003d\u003d 0 and\n         array_removed.changed\n \n-- name: arrays | Ensure /etc/mdadm/ directory exists\n+- name: arrays | Ensure {{ mdadm_conf }}\u0027s directory exists\n   file:\n-    path: /etc/mdadm/\n+    path: \"{{ mdadm_conf_path }}\"\n     state: directory\n+  when: mdadm_conf_path is defined\n \n-- name: arrays | Ensure /etc/mdadm/mdadm.conf file exists\n+- name: arrays | Ensure {{ mdadm_conf }} file exists\n   copy:\n     content: \"\"\n-    dest: /etc/mdadm/mdadm.conf\n+    dest: \"{{ mdadm_conf }}\"\n     force: no\n \n-# Updating /etc/mdadm/mdadm.conf in order to persist between reboots\n-- name: arrays | Updating /etc/mdadm/mdadm.conf\n+# Updating mdadm.conf in order to persist between reboots\n+- name: arrays | Updating {{ mdadm_conf }}\n   lineinfile:\n-    dest: \"/etc/mdadm/mdadm.conf\"\n+    dest: \"{{ mdadm_conf }}\"\n     regexp: \"^{{ item }}\"\n     line: \"{{ item }}\"\n     state: \"present\"\n   with_items: \u0027{{ array_details.stdout_lines }}\u0027\n   when: array_created.changed\n \n-# Updating /etc/mdadm/mdadm.conf in order to not persist between reboots\n-- name: arrays | Updating /etc/mdadm/mdadm.conf\n+# Updating mdadm.conf in order to not persist between reboots\n+- name: arrays | Updating {{ mdadm_conf }}\n   lineinfile:\n-    dest: \"/etc/mdadm/mdadm.conf\"\n+    dest: \"{{ mdadm_conf }}\"\n     regexp: \"^ARRAY /dev/{{ item.name }}\"\n     line: \"ARRAY /dev/{{ item.name }}\"\n     state: \"absent\"\n\ndiff --git a/tasks/debian.yml b/tasks/debian.yml\nindex ba51213..ed2dfaa 100644\n--- a/tasks/debian.yml\n+++ b/tasks/debian.yml\n@@ -4,6 +4,8 @@\n     name: \"mdadm\"\n     state: \"present\"\n \n-- name: arrays | Set update_initramfs cmd to update-initramfs -u\n+- name: arrays | Setting distribution-specific facts\n   set_fact:\n     update_initramfs: \"update-initramfs -u\"\n+    mdadm_conf: \"/etc/mdadm/mdadm.conf\"\n+    mdadm_conf_path: \"/etc/mdadm\"\ndiff --git a/tasks/el.yml b/tasks/el.yml\nindex 5bfdd5d..a0d8f5a 100644\n--- a/tasks/el.yml\n+++ b/tasks/el.yml\n@@ -4,6 +4,7 @@\n     name: \"mdadm\"\n     state: \"present\"\n \n-- name: arrays | Set update_initramfs cmd to dracut -f \n+- name: arrays | Setting distribution-specific facts\n   set_fact:\n     update_initramfs: \"dracut -f\"\n+    mdadm_conf: \"/etc/mdadm.conf","accounts_in_message":[],"_revision_number":1},{"id":"6dd1905d894e2d672d62209e44a11c029ed4d5ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 09:25:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0cce3d39e524472b9211a6e1963545b : SUCCESS in 5m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f6d5063760b1452db130ccc1dc9a80a8 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/179ba07f50804c598b2e25948c78535b : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6baaec94835741ae9c4697b28a747f90 : SUCCESS in 5m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/46c6930d6d624047aeaacade646f00c1 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50dfd3c7dabf46abb96a8eedbefa11be : SUCCESS in 4m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c02a526e52f34c8e9d01f6abe69708c6 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/31450e975d37442bb1ab1e8301078fdd : SUCCESS in 7m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cd3104d5512844fe906383c5593061c1 : SUCCESS in 19m 18s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/7b3cdbc29a7a4f50a1887ba83fb58cf5 : SUCCESS in 43m 59s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/08f4f2fb50e6445ab322731489e7da43 : FAILURE in 56m 10s","accounts_in_message":[],"_revision_number":1},{"id":"4155019cb35d2ae268fbb1f3082c4a04c082df04","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 09:30:53.000000000","message":"Patch Set 1:\n\nrecheck\n\nmirror timeout","accounts_in_message":[],"_revision_number":1},{"id":"51b534ea1f76c2afc399b900993b76e746f5a232","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 10:30:21.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/791b12398427465fa72960de5b963478 : SUCCESS in 5m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ff19b412cf0d42ec8e5c2e6831cc3dc6 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eeb5bda9a8174be2806b7c4a56fc31d1 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a80d581023c403d82858aa82b89df51 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/00f775e5c3254c9f9a2edafeccd528df : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b241ce98e174e2185a9a10b06c1fab7 : SUCCESS in 4m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a67a07846d6741589028629e9c7eee8b : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8ae990b3ccc94935ae4d2d547afbd9d5 : SUCCESS in 8m 34s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8ec677116c14404a895633e61ae70ebb : SUCCESS in 18m 21s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/22b360c36dc04c91a6fe5362eeef2cee : FAILURE in 53m 52s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/4551e3ddb5cb49d882fefd6238a7e52a : SUCCESS in 19m 35s","accounts_in_message":[],"_revision_number":1},{"id":"221a8fc7056141fe4f69c4f0dbec95af1436467c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 10:37:54.000000000","message":"Patch Set 1:\n\nrecheck\n\nmirror timeout","accounts_in_message":[],"_revision_number":1},{"id":"433fdfd0ebd7dea85dbcec4d373afd1e4ec007c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 11:02:42.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"bd4db39aae39e25c4f52ae12ff4fb66c53ca5023","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 11:04:58.000000000","message":"Patch Set 2:\n\nI pushed a new version without change because py35 tests had failed. Infra issue?\n\nFileNotFoundError: [Errno 2] No such file or directory: \u0027/home/zuul/.local/share/virtualenv/wheel/3.5/embed/1\u0027","accounts_in_message":[],"_revision_number":2},{"id":"ff9b01773bbf822314d1dd97b093eb50243690dd","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-04-26 11:22:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"f3dab81e9a66683e2eb25a4bfa772a4ba4db33b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 12:06:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c5075b2b4ca4d60865d033620e3836b : SUCCESS in 4m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/99ca91f27cdc4982b3ad60f33e869f62 : SUCCESS in 6m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/063c69c7877148e08bb982ebc3b2f1c2 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c09e19f15ea4ce1a45a3e41450e7d83 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/efc31139e0d0492691d995cd6f9706c5 : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73331979c5dd4e11a0ddfdc86ce3f1c8 : SUCCESS in 4m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5a70089eec4046a49c36abbfad8a13cf : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5a3488e072be4eb7a6c24b075727b5ce : SUCCESS in 8m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/813ea324c3d84c608d433f9cde94c20a : SUCCESS in 18m 59s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/721bac52d8d44f42ad809966cb3dcc6d : SUCCESS in 1h 02m 47s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/1065cb8f34644ca5aa09c14a472c1650 : SUCCESS in 22m 38s","accounts_in_message":[],"_revision_number":2},{"id":"e7c7add5ffdf7d32ff6c4c97e5a7063ed9128ee2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-26 12:23:15.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9ee6cc0311297b04a85b2d593c4a7be97012ed06","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 12:23:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0ee623e5b66f1f7d001abe079d213723674de02b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 12:23:51.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"8ff3b7be03a16907905be9151b067f48ef0f628c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 13:23:07.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/17f41d1d862446e08ae4655cfd52390f : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa5e6c49cf5e4d8abe1e2baf5cfafb2f : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c491f103e63e4382a78cc31883a6dfdc : SUCCESS in 3m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8fa0d3298024f40a68e68b990b2b4b5 : SUCCESS in 3m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fab10fb21d1342d2b0b80944095e7916 : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/10db5f5cd7434bf08fedec36fc5a5386 : SUCCESS in 4m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3a4ffab5cd9b47a289db3b15ba3b4923 : SUCCESS in 8m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9a52710a15ab4787906688fdbd21db04 : SUCCESS in 17m 18s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/0d72aaae2e604fb9836112fb4e308072 : FAILURE in 54m 07s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/5aa7e5e959814e459b558fd9214cbe43 : FAILURE in 53m 50s","accounts_in_message":[],"_revision_number":2},{"id":"a6b8fa2ea9e2dd21dab2592e74ba7c11d2fdaf9e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 13:28:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"82f9e66fc8115ed7c6facb724b6fdc63213d00fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 14:55:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cd53569a55f41d48a51649d9e78e106 : SUCCESS in 4m 39s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d58b32feec3f493c9901c3e58e5e976f : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0b8b1721ec14aec8f6a4491b06e4aa9 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8de94b4401c40d685ab08f4c7027594 : SUCCESS in 3m 58s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/d1b2adf32f214ef0a4b8bd74db166bcb : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9eed3f1938c4f73a6d79feb893ba003 : SUCCESS in 3m 52s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d5bb95d792ec40cd949422598ccc83fd : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/84c6016af26847ebb38a55caa5a26364 : SUCCESS in 8m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6aaba55859b5437f850840638ce0b57d : SUCCESS in 16m 40s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/34d44fe895a244fb9eb269331317868c : SUCCESS in 1h 02m 18s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b82ab4190912413b85219943c395e115 : FAILURE in 53m 27s","accounts_in_message":[],"_revision_number":2},{"id":"e34225cec94cfea0758e5390fdb05f691ab2fcf5","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 14:59:14.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c1967f3d239a20301c97fb398ba7b412e26d5c5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 15:47:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2abf7d4d251849b3b30a3a5f8e14eac3 : SUCCESS in 6m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/70fc05255839422db750ba09de596681 : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a96c47f1b4664ba7840f3a71ace69f9d : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0fb4b8068e942768dc3f808aa1c1154 : SUCCESS in 6m 34s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/adc5e945ac3c400186e7f4e1042bf1e3 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e727e5e92747a2999a184d7af0e07e : SUCCESS in 4m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5d3fcff4f7c94b02a857ff12e48971c1 : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4d3aec2424bd47a9937fdc96d64b5791 : SUCCESS in 10m 07s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f2be2a35c31477e90bbb16b7335785e : SUCCESS in 15m 06s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/0cfb15be458c4be1858dc1ed6119ea63 : SUCCESS in 45m 13s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/13b7f69eb5174a6baa6639e45f3f8710 : SUCCESS in 20m 50s","accounts_in_message":[],"_revision_number":2},{"id":"a98cfd13189891c219217a1c7f365fd8dc0decd9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 15:47:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"dda2c1cee4680317dac6f02cb939847ba5a0d97d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:46:23.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/88a69a41fb564ef0a2540e5af03229d9 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c32e7ff804b74d72950b579c6e83f5ea : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b5f98211d7433699df35e9deb9a4c3 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5c6b715965834dd789c69318513dfbf7 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cce3dbc458e4fe9b7a20a6d89db547e : SUCCESS in 4m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2e7f62e51097499282b83258f68aaa3d : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1a5e21a7042c4609bc44ea9e3ec1b32b : SUCCESS in 8m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ef23c18394854f3494fb6d7b16467794 : SUCCESS in 15m 15s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/2285f13708f145fb82359f06a6405956 : SUCCESS in 57m 08s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/5f66f3e393094320891173bf8e009f04 : SUCCESS in 16m 08s","accounts_in_message":[],"_revision_number":2},{"id":"c2e12054727ebfcd77f134c5d49b47f25ccbc310","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:46:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ab3fc8649d3c826cc920f78d01d31e9035fc478d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:47:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fba32bec4c447e39d38f30e3ebacb26 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b826104b0a10fc75f777b7510e397cf00cf09bf7","revisions":{"d76838ee60b86fd40d3e17974f44483f173beb5e":{"kind":"REWORK","_number":1,"created":"2021-04-26 08:22:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/36/787936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/787936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/787936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/787936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/787936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/787936/1"}}},"commit":{"parents":[{"commit":"125d399bdae572ab06fe5fb633a7a0cad9f72ce0","subject":"Fix installation errors with Python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/125d399bdae572ab06fe5fb633a7a0cad9f72ce0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-26 08:04:45.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-26 08:22:50.000000000","tz":120},"subject":"Use released version of mrlesmithjr.mdadm","message":"Use released version of mrlesmithjr.mdadm\n\nAn initial version of the mrlesmithjr.mdadm role was released 2 days\nago. There are only minimal changes compared to the commit we were\nusing. This should avoid the following issue seen in CI:\n\n[WARNING]: - mrlesmithjr.mdadm was NOT installed successfully: - the specified\nversion (5be3ee7d330aa17317897bd104dc87ff0df11915) of mrlesmithjr.mdadm was not\nfound in the list of available versions ([{\u0027id\u0027: 152374, \u0027url\u0027: \u0027\u0027, \u0027related\u0027:\n{}, \u0027summary_fields\u0027: {}, \u0027created\u0027: \u00272021-04-24T02:51:30.453954Z\u0027, \u0027modified\u0027:\n\u00272021-04-24T02:51:30.453980Z\u0027, \u0027name\u0027: \u0027v0.1.0\u0027, \u0027version\u0027: \u00270.1.0\u0027,\n\u0027commit_date\u0027: \u00272021-04-23T22:50:00-04:00\u0027, \u0027commit_sha\u0027:\n\u0027b8a9853b7eafa9afc62bbb7728f64fd5dc9f54de\u0027, \u0027download_url\u0027:\n\u0027https://github.com/mrlesmithjr/ansible-mdadm/archive/v0.1.0.tar.gz\u0027, \u0027active\u0027:\nNone}]).\nERROR! - you can use --ignore-errors to skip failed roles and finish processing the list.\n\nChange-Id: I3d85b341cf541cfb9b486021e66b316833fe9130\n(cherry picked from commit 7411e624dea77a1b94d91bd37c9a388aa5011bb4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d76838ee60b86fd40d3e17974f44483f173beb5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d76838ee60b86fd40d3e17974f44483f173beb5e"}]},"branch":"refs/heads/stable/stein"},"b826104b0a10fc75f777b7510e397cf00cf09bf7":{"kind":"NO_CHANGE","_number":2,"created":"2021-04-26 11:02:42.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/36/787936/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/787936/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/787936/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/787936/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/787936/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/787936/2"}}},"commit":{"parents":[{"commit":"125d399bdae572ab06fe5fb633a7a0cad9f72ce0","subject":"Fix installation errors with Python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/125d399bdae572ab06fe5fb633a7a0cad9f72ce0"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-26 08:04:45.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-26 11:02:31.000000000","tz":120},"subject":"Use released version of mrlesmithjr.mdadm","message":"Use released version of mrlesmithjr.mdadm\n\nAn initial version of the mrlesmithjr.mdadm role was released 2 days\nago. There are only minimal changes compared to the commit we were\nusing. This should avoid the following issue seen in CI:\n\n[WARNING]: - mrlesmithjr.mdadm was NOT installed successfully: - the specified\nversion (5be3ee7d330aa17317897bd104dc87ff0df11915) of mrlesmithjr.mdadm was not\nfound in the list of available versions ([{\u0027id\u0027: 152374, \u0027url\u0027: \u0027\u0027, \u0027related\u0027:\n{}, \u0027summary_fields\u0027: {}, \u0027created\u0027: \u00272021-04-24T02:51:30.453954Z\u0027, \u0027modified\u0027:\n\u00272021-04-24T02:51:30.453980Z\u0027, \u0027name\u0027: \u0027v0.1.0\u0027, \u0027version\u0027: \u00270.1.0\u0027,\n\u0027commit_date\u0027: \u00272021-04-23T22:50:00-04:00\u0027, \u0027commit_sha\u0027:\n\u0027b8a9853b7eafa9afc62bbb7728f64fd5dc9f54de\u0027, \u0027download_url\u0027:\n\u0027https://github.com/mrlesmithjr/ansible-mdadm/archive/v0.1.0.tar.gz\u0027, \u0027active\u0027:\nNone}]).\nERROR! - you can use --ignore-errors to skip failed roles and finish processing the list.\n\nChange-Id: I3d85b341cf541cfb9b486021e66b316833fe9130\n(cherry picked from commit 7411e624dea77a1b94d91bd37c9a388aa5011bb4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b826104b0a10fc75f777b7510e397cf00cf09bf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b826104b0a10fc75f777b7510e397cf00cf09bf7"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
