)]}'
{"id":"openstack%2Fkayobe~787937","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I3d85b341cf541cfb9b486021e66b316833fe9130","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I3d85b341cf541cfb9b486021e66b316833fe9130","subject":"Use released version of mrlesmithjr.mdadm","status":"MERGED","created":"2021-04-26 08:25:04.000000000","updated":"2021-04-26 17:33:08.000000000","submitted":"2021-04-26 17:31:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"787937","meta_rev_id":"1f1e2b14e796cd9c8519e2141ac5048e3261707e","_number":787937,"virtual_id_number":787937,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-26 17:31:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2021-04-26 11:22:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2021-04-26 12:23:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2021-04-26 12:23:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-26 09:36:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-26 11:22:48.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2021-04-26 12:23:28.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"45c56a2bca714df7a3c208f58962a8e787571d6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2021-04-26 08:25:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cfed30c92ce483b1f5972776c223d5022d83fcd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 09:36:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ed1cb8800ebf4f59b313280fd2ef3986 : SUCCESS in 6m 33s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6179e6c7210f4eb2a63f43436b677561 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33fb0800b4d14d6e92aeb76d3aff7e46 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96261ee26c6d4918936efd27a70e27f2 : SUCCESS in 4m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bc3a8739cded490d98e0cb466d5193a9 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c09e4e37f53d4a0e8c4f67afabd8c086 : SUCCESS in 6m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0a62645485664a42a7ea402cb0626e36 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d39e485f1101458bbf3ec2c936c010ca : SUCCESS in 9m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3e0818e5483348198251be1fa38af7bb : SUCCESS in 16m 47s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/69dbee2daae74d28883b9e7ac69af378 : SUCCESS in 1h 05m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f2d0f974b2f04997971d4ec3d9053d1e : SUCCESS in 42m 01s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/b67dadb62099405f822e4f47d1ec1004 : SUCCESS in 15m 54s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d38e6264cc3b469c83e7a6e050f2db5c : SUCCESS in 13m 44s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/dd0b65a3f5dd439f96d2affdee8329c3 : SUCCESS in 1h 07m 03s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/69203cbe64d54e208b976f8f5b05f7e0 : SUCCESS in 19m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5582015600db48279c54333d3f4b4425 : SUCCESS in 24m 56s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/c44d6e3bc95c458a98518ae3a5eccf8d : SUCCESS in 41m 27s","accounts_in_message":[],"_revision_number":1},{"id":"243ddc28c0bc3ed601920d10099bcbae8be76b7a","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2021-04-26 11:22:48.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0b24661918319e45f8fe2e902e435cc544db8664","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-26 12:23:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c41778ae6ec8e885df7f80e7d0b94da7d4ea665","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 12:23:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2a1fe92221eaefb7eb15a0249b87fb3a608a2d1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 13:23:08.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"8d7bd42ddd97ae8a7df688b05af5d721ecd78106","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-04-26 14:14:06.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"910d881ddd8a3f8d48cddba9580d853c0f6d1587","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:04:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76e26ece4e5342eb8bec6b308ea750fa : SUCCESS in 5m 14s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2c78be14d74f4a499c99b659a88d1466 : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd4839c654bf429381603167691b9e08 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2de31282b4a4fbda10260ab1fac0094 : SUCCESS in 4m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ed0938abdb6f46b49fa96df285494918 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3af99af1d3c24b8eb91eaefdd438b598 : SUCCESS in 6m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/eb4b9e46b06b4bf6811c319c411a5a9d : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/03b77c8910bf4148be6c33b64494c848 : SUCCESS in 8m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9b7f4e4e56d64367ab13ec7ea681700a : SUCCESS in 18m 46s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/1c209fa8ad6b40ea97c18393dc766967 : SUCCESS in 57m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/575f4f7418ca403b995690effaca119d : SUCCESS in 57m 50s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/87433a1a95b14c34b912217e2215f98f : SUCCESS in 12m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7b68c5ed99ce4498b9a97d89ed45d87d : SUCCESS in 15m 39s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/9f712803ea5d40cc8e9b6da5482510a7 : SUCCESS in 1h 45m 05s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/a8c5ea438f84448698462f7b8ecdca09 : SUCCESS in 21m 20s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f1f95e9645284a0cb9032e7a445f226f : SUCCESS in 23m 04s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/c1b624be1a8947599709490302231374 : SUCCESS in 26m 29s","accounts_in_message":[],"_revision_number":1},{"id":"933a9082b00f78b446390f94af93480165d0cbfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 16:04:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a00d5c19aa98af887ceb6a1365439cb5604cfab4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 17:31:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/24af4c7f039c42c59c9ad9723685b3d4 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4219b4bdcf6c4138b32779d9add32cb1 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8a780b45a724aadab09c900855919fb : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c97defd2f07941a487bffe6428d1e4bc : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3ef51bb2c7c4022bd893bcb84339619 : SUCCESS in 10m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/40a0ed9033f54a389ad6f53ebba4cbd3 : SUCCESS in 5m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8d92dd45c3bf449a8c575db4247168fe : SUCCESS in 8m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0755495cb31d4801b7720290a4032b08 : SUCCESS in 18m 30s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/7c5a2a8b6a43479dbe4b2552a290c6ca : SUCCESS in 50m 20s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fc3a4dd89c354611bb2d395dfa53a621 : SUCCESS in 55m 49s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/d856078e13974e858c5530ebc765c1fa : SUCCESS in 12m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/83024fac3a294f51b06a515171a0660b : SUCCESS in 16m 07s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/53786cd353104ba89af71d984f4c7cb5 : SUCCESS in 1h 22m 18s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/5d6d2c932bb544fa8a48fdf24cd4d41e : SUCCESS in 21m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/36e246df2de64cc1bb83b5432981c661 : SUCCESS in 23m 10s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/e8f27e81de0740429971f3857e43c309 : SUCCESS in 33m 00s","accounts_in_message":[],"_revision_number":1},{"id":"80479843c5767267a455aef7c8b554e7ebfcac41","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 17:31:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1f1e2b14e796cd9c8519e2141ac5048e3261707e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-26 17:33:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/173aacfd53784c7aa6fa961b2022e5ce : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8b4d2b1eccc2f6d2bd2a2df276333ece127ad3e","revisions":{"f8b4d2b1eccc2f6d2bd2a2df276333ece127ad3e":{"kind":"REWORK","_number":1,"created":"2021-04-26 08:25:04.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/37/787937/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/37/787937/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/787937/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/787937/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/37/787937/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/37/787937/1"}}},"commit":{"parents":[{"commit":"b57f536e62f2e876b2835cf872bd4fcc1b054a8b","subject":"Update setuptools before installing kayobe","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b57f536e62f2e876b2835cf872bd4fcc1b054a8b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-26 08:04:45.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2021-04-26 08:24:48.000000000","tz":120},"subject":"Use released version of mrlesmithjr.mdadm","message":"Use released version of mrlesmithjr.mdadm\n\nAn initial version of the mrlesmithjr.mdadm role was released 2 days\nago. There are only minimal changes compared to the commit we were\nusing. This should avoid the following issue seen in CI:\n\n[WARNING]: - mrlesmithjr.mdadm was NOT installed successfully: - the specified\nversion (5be3ee7d330aa17317897bd104dc87ff0df11915) of mrlesmithjr.mdadm was not\nfound in the list of available versions ([{\u0027id\u0027: 152374, \u0027url\u0027: \u0027\u0027, \u0027related\u0027:\n{}, \u0027summary_fields\u0027: {}, \u0027created\u0027: \u00272021-04-24T02:51:30.453954Z\u0027, \u0027modified\u0027:\n\u00272021-04-24T02:51:30.453980Z\u0027, \u0027name\u0027: \u0027v0.1.0\u0027, \u0027version\u0027: \u00270.1.0\u0027,\n\u0027commit_date\u0027: \u00272021-04-23T22:50:00-04:00\u0027, \u0027commit_sha\u0027:\n\u0027b8a9853b7eafa9afc62bbb7728f64fd5dc9f54de\u0027, \u0027download_url\u0027:\n\u0027https://github.com/mrlesmithjr/ansible-mdadm/archive/v0.1.0.tar.gz\u0027, \u0027active\u0027:\nNone}]).\nERROR! - you can use --ignore-errors to skip failed roles and finish processing the list.\n\nChange-Id: I3d85b341cf541cfb9b486021e66b316833fe9130\nDepends-On: https://review.opendev.org/c/openstack/kayobe/+/787936\n(cherry picked from commit 7411e624dea77a1b94d91bd37c9a388aa5011bb4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8b4d2b1eccc2f6d2bd2a2df276333ece127ad3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8b4d2b1eccc2f6d2bd2a2df276333ece127ad3e"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
